)]}'
{"id":"openstack%2Fneutron~817529","triplet_id":"openstack%2Fneutron~stable%2Fxena~I110772330c0e7298a09b94212f7d56600b5a9bde","project":"openstack/neutron","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-01-06 09:22:43.000000000","reason":"Removed by Lajos Katona using the hovercard menu"},"33652":{"account":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"last_update":"2022-01-25 10:25:36.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-01-25 10:25:36.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2022-01-25 08:37:50.000000000","reason":"Bernard Cafarelli replied on the change"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-01-25 07:44:34.000000000","reason":"Slawek Kaplonski replied on the change"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-01-25 10:25:36.000000000","reason":"Change was submitted"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2022-01-25 10:25:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I110772330c0e7298a09b94212f7d56600b5a9bde","subject":"[stable-only] Update Placement allocation when min bw rule direction is changed","status":"MERGED","created":"2021-11-11 10:18:46.000000000","updated":"2022-01-25 10:27:28.000000000","submitted":"2022-01-25 10:25:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"817529","meta_rev_id":"5125ce3845602464d828c30feafb6ab5243cb3b2","_number":817529,"virtual_id_number":817529,"owner":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-25 10:25:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-01-25 08:42:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},{"value":1,"date":"2022-01-25 07:29:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2022-01-25 08:37:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-01-25 07:44:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-01-25 08:42:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-11 11:04:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-11 11:51:14.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-11-11 12:24:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-15 15:05:44.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-11-15 15:05:44.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-11-15 15:05:44.000000000","updated_by":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-01-25 05:49:07.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2022-01-25 07:29:26.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-01-25 07:29:26.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-01-25 07:29:26.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"2e07826a64a1d0304a2b470eeb87f9ff33632246","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-11-11 10:18:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c4844406933fef1efac38e0607d2fd3b4e6738c","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-11-11 10:22:38.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"54295071dd07d1f34dc72fd1c33fe2bb25f4f7ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 11:04:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9b182829094b45de8eaf6574c7e97d35 : SUCCESS in 43m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b16a8e5f80be4145a976659f04e3aee0 : SUCCESS in 43m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc24ef152fe4c6f0b2f94d319ebeed09c99b15b1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-11 11:51:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/817529/1 : SUCCESS in 1h 31m 35s","accounts_in_message":[],"_revision_number":1},{"id":"89e54de0118695d9cf128d2aa0a74915d6f889d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 12:24:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e0d4c061e7ad43a4ad15daa238a17954 : SUCCESS in 33m 35s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d74759fc5218403a8e44174c90982efb : SUCCESS in 1h 53m 22s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/562f415a0cdc44cf97e619bd08170e64 : SUCCESS in 1h 38m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fd4784e8c0684dacbd7f21c4a6570ff7 : SUCCESS in 1h 25m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2098493337504bf4a22868b69b91f8f4 : SUCCESS in 2h 03m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffccd236d43a49d2ae729280001912dd : SUCCESS in 34m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/125cd91f0c30450a9255e21d82893606 : SUCCESS in 16m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fe95f19c4bf4b6a9e19999dc06102c7 : SUCCESS in 26m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa9ffdbb11784ce590ada58dcf3fa82d : SUCCESS in 27m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6c8af44cd4143c499d546432370aad6 : SUCCESS in 22m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d7edea440a64c2989ec9a5d9dcf1a04 : SUCCESS in 10m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a6f8793214274551923a1ca3ad401831 : SUCCESS in 45m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b0b6532b61c45048f7b5f0f7eff42d2 : SUCCESS in 1h 48m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e3cafd99af3e41039a0cd2d3b1df12d7 : SUCCESS in 1h 06m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e3bb0d0e1c5349038aa55b030730ea10 : SUCCESS in 57m 52s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e6aeaee4026d46238e108e812a53dbfa : SUCCESS in 1h 02m 42s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/297463f00d234a448db74aac3b1e9e77 : SUCCESS in 1h 10m 23s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ec13f29b0dcc4960b5e8550d8ac384d3 : SUCCESS in 1h 09m 27s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/9a38872c6bc74ff38ef8b8a3bf8d3770 : SUCCESS in 1h 15m 19s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/502fb6f52e1f49288b348e6e9b24ed41 : SUCCESS in 50m 17s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4ac15650b7d8445cb3ea365d55ba81e6 : SUCCESS in 55m 29s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/975e602f01734cada3eefa00fc940b46 : SUCCESS in 33m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8fb2c64f108d42d3b058c28b22b54e6a : SUCCESS in 1h 04m 33s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e17f0b82447b49ce8f09826c7b34162a : SUCCESS in 1h 14m 41s","accounts_in_message":[],"_revision_number":1},{"id":"c9e946fe8b9fc3a996ffc518876ac7c89e466a1b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-11-15 13:46:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8fe2ad2db31165978be0bbac491aafc9f3a269d5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-11-15 13:47:33.000000000","message":"Topic bp/qos-minimum-guaranteed-packet-rate removed","accounts_in_message":[],"_revision_number":2},{"id":"347a00780d1c52fbde5f9512336bc276d9b0f1a0","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-11-15 13:47:46.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"deb6479fddd622be57cc84546e0035d2faa52450","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-11-15 13:48:21.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4cee9fd0f4a54b34790ff499e02670e8da58bf48","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 14:44:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d670c20dcc9403b8e0b5cfc6c2661cc : SUCCESS in 55m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9507146007854153a6919e9f8ab27eed : SUCCESS in 45m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8461140e89aa3fa580186ce2dc6e8b25b27f561d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-15 15:43:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/817529/2 : SUCCESS in 1h 55m 42s","accounts_in_message":[],"_revision_number":2},{"id":"ce3a527d02da8f9b13a8683ba8dd830c9e33c917","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 16:05:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cbea6d5322434b169ba7d4fa7ad6db72 : SUCCESS in 44m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0ea6cad2e63444258e2e7f4ee6f7d6a9 : SUCCESS in 2h 15m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/64b722792c2e4122a1c2a7f3cbab192e : SUCCESS in 2h 01m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/50783625591a46b5ac7c5b15a24e30fc : SUCCESS in 1h 52m 26s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/30b641af762a48f9b4ea17ccefc0f87b : SUCCESS in 2h 02m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f83721f8a8684d13ac30c1ac8b6c419e : SUCCESS in 50m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04c83f1cb8c14e1d839197d6240904cc : SUCCESS in 15m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ea46e9b64244e3d92fbb339d0cf129c : SUCCESS in 41m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dcf106d7e414104bd7c69fb9f7d78f9 : SUCCESS in 41m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e939b1f423f4862bc90f1ca5de6d8ca : SUCCESS in 35m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/065b2a155b2544b2b1a2348d9aa3b7fa : SUCCESS in 9m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b6e39aea6ed447393d8be6dd88e0fe4 : SUCCESS in 1h 05m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c386c11a779c4cc2a21be69374ec7991 : SUCCESS in 2h 08m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bb358f5b1dd049fa96eb68b6622aa7e1 : SUCCESS in 1h 05m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3cbd291c560148fbaf1da4f168ffba73 : SUCCESS in 47m 07s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8f37494f577441cab3b1cea5524f6880 : SUCCESS in 1h 16m 31s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/96f555408d404096863d06bc630b5072 : SUCCESS in 1h 10m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c3f9126bc7da463db60f7095fd1a9dce : SUCCESS in 1h 20m 15s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/28a416d4425a4ed19795963fb767dbd0 : SUCCESS in 1h 42m 57s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e8785d731b7425d96474ff4636d0dea : SUCCESS in 55m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/49fc1a3df62845ae80f82be401d1c759 : SUCCESS in 1h 04m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ea9b1d913c3d47d3a2f639f1eefe4aae : SUCCESS in 39m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6ededd6148bb45699cb75299b904e0e1 : SUCCESS in 1h 03m 31s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6351ffa5c9184494bdf168f30f0d07bc : SUCCESS in 1h 16m 21s","accounts_in_message":[],"_revision_number":2},{"id":"b9b419c1f6fe3f22ad439ee141fae2c0f4fdc319","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-11-19 14:51:13.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"090a4e4d2574fc58927fa1734461884a7e97f091","author":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"date":"2021-11-19 14:52:32.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d2ed8bafb07df9cdb00ae611f804786a4268e52a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 15:47:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4c50f29a025420882557c99c5c0ea78 : SUCCESS in 45m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/22d77f141ccc4202a32573f2c148eb8f : SUCCESS in 49m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8bf1da27862a035aca900e5aad7cafaa1ede70bb","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-19 16:28:51.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/817529/3 : SUCCESS in 1h 36m 43s","accounts_in_message":[],"_revision_number":3},{"id":"2dd8b73fd4d0eaba71d0aab184e59f0ee53ea053","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 17:39:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6f12cc8fd93c408b9286bd2540a7f8d8 : SUCCESS in 47m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c595a8a1df6146acba693b5449d88a2f : SUCCESS in 2h 41m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/183607c6fddd4cf29ccf23ffe818dc9e : SUCCESS in 1h 54m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a82d99a7a720471186431c0273c24c4b : SUCCESS in 1h 53m 49s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/df1264b72d4e4a81ad95a0825e36d28b : SUCCESS in 2h 01m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6d6385b5bbb404287cec847ccfee21a : SUCCESS in 51m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9c02270a248435aa984c1cd4bf440dd : SUCCESS in 12m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dcfb77b822d4f199c9bb20e58dcbab8 : SUCCESS in 42m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9705a2a4ff0646fcb2089932a848d313 : SUCCESS in 46m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe5321256ca144cfa94d41e3a631af01 : SUCCESS in 38m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/037231fadc594d718e93c174e377ccac : SUCCESS in 11m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8c42fef1cef84ca1a36ef2c30e548e70 : SUCCESS in 53m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3f483fd7ee564810b260f9426c359662 : SUCCESS in 2h 01m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4e3260209342485687b888bf99bc3bd9 : SUCCESS in 56m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/90470ade46e146a09e2b772cfd97732e : SUCCESS in 58m 05s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/52f3a54c478f4fda86a9cd21ee112ec4 : SUCCESS in 1h 04m 04s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5ebf3e268d804ad49a207415e8fb9e3d : SUCCESS in 1h 04m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/436139a575804ab49c7f4eecf5530c33 : SUCCESS in 1h 05m 31s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/508a130257fe4fd282a10a5e74fd7f9c : SUCCESS in 1h 31m 05s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/985af5f894934830abd559d045d1e40a : SUCCESS in 53m 36s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/53b4331c3abe481e95a208c106807981 : SUCCESS in 51m 28s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/12b4cd62959c4d6f89a3456e493c95f2 : SUCCESS in 32m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6e96d9a22c3f4da99c71903ea2c8f92d : SUCCESS in 58m 50s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/322034563d8f4546a4669c59793ea2f7 : SUCCESS in 1h 26m 25s","accounts_in_message":[],"_revision_number":3},{"id":"3ecef9f7983c4937dc30e2aeab37041cde081bc5","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-01-25 05:49:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ebcc3358f742f6bc1ba79d2428fab8a0fc234c29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-01-25 05:49:13.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"81832d3a14ba926bac6b19a61b91fd23c5ca31e5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-01-25 06:30:08.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/817529/4 : FAILURE in 40m 13s","accounts_in_message":[],"_revision_number":4},{"id":"143a3abe87726b8de06e17477f2e35d2b19d3707","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 06:39:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/851f0a64eaaa4a28a803328953a662fd : SUCCESS in 38m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4468e1ae4031493b8d14d49ca3e9f445 : SUCCESS in 48m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8dd3e236b67f409c968d4ee1c8cb25c2d4ffddaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 07:22:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f7e29eb4e1f04ae99ff8f69abb59b4d4 : SUCCESS in 31m 11s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f9b0049d739242e3aad98afaba056f62 : SUCCESS in 1h 15m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1f252d9a8bf24a38a302313c85e713f9 : SUCCESS in 51m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3422079cf7144dfb9cbe600d039e486b : SUCCESS in 1h 05m 43s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6da92212470845f294b81da369368cf4 : SUCCESS in 51m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea27da10ed2141419e7dfa2bf451c670 : SUCCESS in 32m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d92bb952b0d34d08b4049b4599c9bcae : SUCCESS in 13m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91e1dd771d2e42519503418d896cc4c0 : SUCCESS in 22m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4092509adb5a44e29673ac98ee7ce54d : SUCCESS in 19m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a440dd19d79742ab83c37c5c1e54d58e : SUCCESS in 24m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/569eadb7539f4b70a13ccf521f2c03da : SUCCESS in 8m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8cd7c4d30504481f9a80cc4d051c8d79 : SUCCESS in 34m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/84f8957be9ab48059f2f10f699e6ba47 : SUCCESS in 1h 31m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/18aaacebd651433d8dfa8c05b0912c5b : SUCCESS in 1h 08m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cc84b68f015b493685c34965db0e59a8 : SUCCESS in 53m 24s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fd4e77531a574a598cd3746964c1ff5f : SUCCESS in 1h 04m 05s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/57e692db11fc4573912960d7adec05aa : SUCCESS in 59m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7abd01410b95436a951cade7d24553d4 : SUCCESS in 1h 02m 22s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d64640b235d04506a880ab506f95553e : SUCCESS in 45m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ea3a3e8e224049339cfe8e982616a268 : SUCCESS in 57m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"eaaa0ef1f4ff6b9fdd316251cb89e980387f9cea","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-01-25 07:29:26.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"07dd150709549495d446a6c158f08d25e5526f1f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-25 07:44:34.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9d88fbd6477db3a04ab2b490b2ae0e95c97ef578","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-01-25 08:37:50.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fed3da096e72ee979a6515a3f64543b4981a36f5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-25 08:42:05.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c07be22cf022ef8083e78fa4fe791e3ae38106ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 08:42:42.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c13d44fba41a099c40d29ef1936c35a62269b4be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 10:25:36.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a2d033eb72594d18b2241628ab7057b7 : SUCCESS in 34m 23s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6bd98422651b429592917e3bea379202 : SUCCESS in 55m 39s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b80539c3228c4a64a93d5fd1b440e41b : SUCCESS in 53m 03s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/28fc40d77ed3457981b60f15bf362d1f : SUCCESS in 42m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0b89f89177634177a41499e14561a4e4 : SUCCESS in 44m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/497b7891d1084df08324f3e95a08a14a : SUCCESS in 12m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73ab84848c7349d599b78441cee01aed : SUCCESS in 26m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9dc85492e4a345a583f60074220b7342 : SUCCESS in 22m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/029130ada3564ff4b3c0f8616870b436 : SUCCESS in 9m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/54a8aaf1f10d46da82bfa68a5fc148c7 : SUCCESS in 57m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/616b9409017847909e6fbd8cc9c9f2b6 : SUCCESS in 1h 42m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/80a78070c83a4ae7b8e09c29691387d2 : SUCCESS in 1h 16m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b4c1ae4baa2e4c8ab5ecf9257cb5150a : SUCCESS in 1h 00m 40s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7e341f4151142f9a3f03dd4885de80b : SUCCESS in 55m 16s","accounts_in_message":[],"_revision_number":4},{"id":"913ae1d53a3211675f8961cb0961283eb182d3a7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 10:25:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"5125ce3845602464d828c30feafb6ab5243cb3b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 10:27:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1812787baf5446cd8ff5e0d9cbb79ebf : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"fb50dc2a10f3711781e65dd175946456cfe0d1ed","revisions":{"87e592b0da48bcc7ce1865bc21c4d95c4ba07c13":{"kind":"REWORK","_number":1,"created":"2021-11-11 10:18:46.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/29/817529/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/817529/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/817529/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/817529/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/817529/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/817529/1"}}},"commit":{"parents":[{"commit":"5e917c40e3f6baf4c46ad4c70902bac169bd6169","subject":"Merge \"[DVR] Fix update of the MTU in the DVR HA routers\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e917c40e3f6baf4c46ad4c70902bac169bd6169"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-11-03 10:03:43.000000000","tz":60},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-11-11 10:17:28.000000000","tz":60},"subject":"Update Placement allocation when min bw rule direction is changed","message":"Update Placement allocation when min bw rule direction is changed\n\nThis patch fixes logic responsible for detecting if Placement\nallocation needs to be updated. This commit is based on\n8db15cb2f3dec37441103df15ff28c7e72de682c.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/817528\nCloses-Bug: #1943724\nChange-Id: I110772330c0e7298a09b94212f7d56600b5a9bde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/87e592b0da48bcc7ce1865bc21c4d95c4ba07c13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/87e592b0da48bcc7ce1865bc21c4d95c4ba07c13"}]},"branch":"refs/heads/stable/xena"},"23b1b20b8e563ef03428f4740dd7ff8cab462b91":{"kind":"REWORK","_number":2,"created":"2021-11-15 13:46:25.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/29/817529/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/817529/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/817529/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/817529/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/817529/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/817529/2"}}},"commit":{"parents":[{"commit":"5e917c40e3f6baf4c46ad4c70902bac169bd6169","subject":"Merge \"[DVR] Fix update of the MTU in the DVR HA routers\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e917c40e3f6baf4c46ad4c70902bac169bd6169"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-11-03 10:03:43.000000000","tz":60},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-11-15 12:55:10.000000000","tz":60},"subject":"Update Placement allocation when min bw rule direction is changed","message":"Update Placement allocation when min bw rule direction is changed\n\nThis patch fixes logic responsible for detecting if Placement\nallocation needs to be updated. This commit is based on\n8db15cb2f3dec37441103df15ff28c7e72de682c.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/817528\nCloses-Bug: #1943724\nChange-Id: I110772330c0e7298a09b94212f7d56600b5a9bde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/23b1b20b8e563ef03428f4740dd7ff8cab462b91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/23b1b20b8e563ef03428f4740dd7ff8cab462b91"}]},"branch":"refs/heads/stable/xena"},"3b4ad581f797015200806db948fc2a6a9ddd8ec4":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-11-19 14:51:13.000000000","uploader":{"_account_id":33652,"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@gmail.com","username":"pszczerbik"},"ref":"refs/changes/29/817529/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/817529/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/817529/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/817529/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/817529/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/817529/3"}}},"commit":{"parents":[{"commit":"5e917c40e3f6baf4c46ad4c70902bac169bd6169","subject":"Merge \"[DVR] Fix update of the MTU in the DVR HA routers\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e917c40e3f6baf4c46ad4c70902bac169bd6169"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-11-03 10:03:43.000000000","tz":60},"committer":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-11-19 14:51:13.000000000","tz":0},"subject":"[stable-only] Update Placement allocation when min bw rule direction is changed","message":"[stable-only] Update Placement allocation when min bw rule direction is changed\n\nThis patch fixes logic responsible for detecting if Placement\nallocation needs to be updated. This commit is based on\n8db15cb2f3dec37441103df15ff28c7e72de682c.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/817528\nCloses-Bug: #1943724\nChange-Id: I110772330c0e7298a09b94212f7d56600b5a9bde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3b4ad581f797015200806db948fc2a6a9ddd8ec4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3b4ad581f797015200806db948fc2a6a9ddd8ec4"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"},"fb50dc2a10f3711781e65dd175946456cfe0d1ed":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-01-25 05:49:13.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/29/817529/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/817529/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/817529/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/817529/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/817529/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/817529/4"}}},"commit":{"parents":[{"commit":"e878b33d0a8bf8a34096bd27577eeacaf773ef26","subject":"Merge \"Make the dead vlan actually dead\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e878b33d0a8bf8a34096bd27577eeacaf773ef26"}]}],"author":{"name":"Przemyslaw Szczerbik","email":"przemyslaw.szczerbik@est.tech","date":"2021-11-03 10:03:43.000000000","tz":60},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2022-01-25 05:49:13.000000000","tz":0},"subject":"[stable-only] Update Placement allocation when min bw rule direction is changed","message":"[stable-only] Update Placement allocation when min bw rule direction is changed\n\nThis patch fixes logic responsible for detecting if Placement\nallocation needs to be updated. This commit is based on\n8db15cb2f3dec37441103df15ff28c7e72de682c.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/817528\nCloses-Bug: #1943724\nChange-Id: I110772330c0e7298a09b94212f7d56600b5a9bde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fb50dc2a10f3711781e65dd175946456cfe0d1ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fb50dc2a10f3711781e65dd175946456cfe0d1ed"}]},"branch":"refs/heads/stable/xena","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
