)]}'
{"id":"openstack%2Fneutron~817675","triplet_id":"openstack%2Fneutron~stable%2Fvictoria~Ic0bc3d951d32efadc116708bfe518a711730429d","project":"openstack/neutron","branch":"stable/victoria","topic":"bug/1947993-stable/wallaby-stable/victoria","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-11-15 09:19:03.000000000","reason":"Rodolfo Alonso replied on the change"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2021-12-06 10:12:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic0bc3d951d32efadc116708bfe518a711730429d","subject":"Cleanup router for which processing added router failed","status":"MERGED","created":"2021-11-12 09:00:56.000000000","updated":"2021-12-06 10:16:01.000000000","submitted":"2021-12-06 10:12:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"817675-bug/1947993-stable/wallaby-stable/victoria","cherry_pick_of_change":817716,"cherry_pick_of_patch_set":1,"meta_rev_id":"4d051d1d67cc24e70f6acb88b6aaae811ba3e1e7","_number":817675,"virtual_id_number":817675,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-06 10:12:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-11-15 09:19:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-11-15 11:02:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-12-06 08:35:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-12 09:17:33.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-11-12 11:46:03.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-11-12 12:36:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-15 11:02:15.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"ad198ee5c4c440199d5c51b54af40580b42665c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-12 09:00:56.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"16b00660866eb4bb602d0bef6a0fb6bcf844c0fa","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-12 11:46:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/817675/1 : SUCCESS in 1h 50m 52s","accounts_in_message":[],"_revision_number":1},{"id":"b2285bc3e67153502745548a9a159926ffb26ed3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 12:36:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge-victoria https://zuul.opendev.org/t/openstack/build/7a4bf39cefbf40618abb1d0564543a37 : SUCCESS in 2h 13m 52s\n- neutron-tempest-plugin-scenario-openvswitch-victoria https://zuul.opendev.org/t/openstack/build/a3e2d1e0dba249dea5a6d90e7fd12fb2 : SUCCESS in 2h 46m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-victoria https://zuul.opendev.org/t/openstack/build/85323800540e43139661180992f2dab4 : SUCCESS in 1h 46m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8706a87d27cc4d0cbe2ae71044d3d5e0 : SUCCESS in 32m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6acd41ddb4f54868b2afa336207140d1 : SUCCESS in 14m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6323df3ce7284635a2e756586d666db8 : SUCCESS in 24m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1f1e845c8834347b2e1f7cc817d8bcd : SUCCESS in 32m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c674afc02ae04961be217da6f537bd6f : SUCCESS in 11m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/20d52520bd4c4350934d6a18261f145d : SUCCESS in 1h 13m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2da04d2dcf2d460db9f61399d0bde630 : SUCCESS in 1h 43m 34s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5f65193fe80e41c6815a422bcf474eaf : SUCCESS in 1h 36m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/df9699bc08c541a0bedf0a7ffd4dc524 : SUCCESS in 1h 22m 56s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3bd1df9f36494fbdb95e9de3b4caed73 : SUCCESS in 1h 14m 24s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/7041dcf1ae9a46c9a2d48e6f97bb800f : FAILURE in 1h 07m 37s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a92dc9caa64b4cb9a88f028becee0a5b : SUCCESS in 57m 25s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/2c8b1544815c4b598572e3ba96bc959b : SUCCESS in 1h 06m 39s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/862ebb9bf0cf4368bf4c9f6221af2477 : SUCCESS in 1h 30m 17s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a7e9d4e6dea24f7780569680c5727158 : SUCCESS in 1h 31m 21s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a6166a56785c48b7970d4f7271ef0cef : SUCCESS in 1h 02m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b19f823bf6b84f8cab18989c47ace069 : SUCCESS in 1h 11m 27s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/32df07c1644f490080699215c0ebe093 : SUCCESS in 1h 49m 05s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b92bb5e8811a4b848fcbf942951ebe0a : SUCCESS in 1h 04m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1e354c98de054369b53729b90bd2e9cf : SUCCESS in 1h 19m 36s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1f817d25b6c441b4a107679ed6e0cdff : FAILURE in 31m 32s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a21c8bb21c7947ff94244302dfe1cc01 : SUCCESS in 1h 16m 01s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/221761991a5c492480150329a5a60a86 : SUCCESS in 3h 30m 29s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3942d272723e46a19c29ea1405ef24a6 : SUCCESS in 1h 13m 25s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/61ffa6aadafc4a159633e19647fdfbd0 : SUCCESS in 1h 24m 59s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8cbd23a7b930410eb8bd1130954a1e3b : SUCCESS in 1h 56m 42s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2394f7b46a8147779b37d38267a1909c : FAILURE in 35m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bef829dad1147bff4afb6db4d2ee0af8c8b1d25d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-15 09:19:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e4c214cbcb77e6b455e8ffd3a7008fb3b71e991d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-11-15 11:02:15.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"46a4837d80165baf2351ec04e6ef0e0c42e0a081","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-12-06 08:35:31.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cd029e90fd20c04d7a561a4bea49508ff7730071","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 08:36:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6025bf78b26609cbfeeeaa14ed22b43a383f195d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 10:12:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/492f7a33573c4f8c85826ffc55c7c77e : SUCCESS in 10m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b4531efaf434a69aa5f533dac7ccf13 : SUCCESS in 21m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bbff088a8b24d949e63e1247e39e83c : SUCCESS in 20m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c542244c6214d9491675e4687926bf1 : SUCCESS in 12m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/814fc992fe5646b3a6d23b82f08149a2 : SUCCESS in 1h 16m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1d093e38ac4449bda9e4d003896e0880 : SUCCESS in 1h 30m 10s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/eb83a4bf73e841b7b28d7065876738b1 : SUCCESS in 57m 17s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/aeb4d5ac9a3a4e0da763a0ae08214b15 : SUCCESS in 1h 03m 17s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5f66735ed9404ab2ab89a6de88541240 : SUCCESS in 1h 05m 53s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e140280ccb73479586e4412051f0b6cb : SUCCESS in 1h 25m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/93eb89de68a84148a6371ca0ba0a1b14 : SUCCESS in 1h 07m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3c24f6e523684c14ae0e9d95f240a343 : SUCCESS in 1h 15m 41s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/842e88bf123e44fd87406d036e040b06 : SUCCESS in 1h 26m 14s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/390f9380052b4856ae7aa4ec35ad31fc : SUCCESS in 57m 25s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1a88f9a6ad9c4651971b4dee375e1255 : SUCCESS in 1h 24m 38s","accounts_in_message":[],"_revision_number":1},{"id":"e1a0b75ce8b0645485e02028d33a3bdcc8f153fb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 10:12:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4d051d1d67cc24e70f6acb88b6aaae811ba3e1e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 10:16:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d62f34fc65f84d549e2434881d76d365 : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7ae03f1eb62c68dff07874e28ee39af4274d3a3f","revisions":{"7ae03f1eb62c68dff07874e28ee39af4274d3a3f":{"kind":"REWORK","_number":1,"created":"2021-11-12 09:00:56.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/75/817675/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/817675/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/817675/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/817675/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/817675/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/817675/1"}}},"commit":{"parents":[{"commit":"40f2ec340c94988b8fc7c2e5e4128cfe5ae9cbdb","subject":"[DVR] Fix update of the MTU in the DVR HA routers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/40f2ec340c94988b8fc7c2e5e4128cfe5ae9cbdb"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-10-21 10:06:25.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-11-12 09:00:56.000000000","tz":0},"subject":"Cleanup router for which processing added router failed","message":"Cleanup router for which processing added router failed\n\nIn the _process_added_router() method of the L3 agent, if processing\nrouter will fail, router_info should be cleaned to e.g. be removed from\nthe router cache so it will not be treated as updated router in next\niteration of the agent.\n\nConflicts:\n    neutron/agent/l3/agent.py\n\nCloses-Bug: #1947993\nChange-Id: Ic0bc3d951d32efadc116708bfe518a711730429d\n(cherry picked from commit 41159bd9a4178a5f8a4fdc90b034c48fa9880cac)\n(cherry picked from commit 1c17019f6c7957c95b1d12942c5898fcc111e4ab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7ae03f1eb62c68dff07874e28ee39af4274d3a3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7ae03f1eb62c68dff07874e28ee39af4274d3a3f"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
