)]}'
{"id":"openstack%2Fneutron~818059","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~I754e2f9ad5b896fcc14ff30bd2901ccec2471907","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/1939704","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-11-22 22:31:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I754e2f9ad5b896fcc14ff30bd2901ccec2471907","subject":"Execute \"migrate_neutron_database_to_ovn\" inside the same DB ctx","status":"MERGED","created":"2021-11-16 09:58:17.000000000","updated":"2021-11-22 22:33:37.000000000","submitted":"2021-11-22 22:31:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818059-bug/1939704","meta_rev_id":"2b7267004dcda1428a5213bc37765b95293a68e1","_number":818059,"virtual_id_number":818059,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-22 22:31:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-11-17 13:56:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-11-22 19:59:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-11-22 19:59:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-16 11:02:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-16 11:44:51.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-11-16 12:13:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-17 13:56:47.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-11-22 19:59:28.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"f33d356a178f3d7b95c40ba92f8379d1f22ee12f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-16 09:58:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9cdc73a1345d4af91d5118db3604a647aa785b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 11:02:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e39489198eff476e8a9f7a32d87af12b : TIMED_OUT in 1h 01m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/990b315224b74226a842561d3cb3419d : SUCCESS in 51m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"648561314e178d9b065467b44301bb73ca45791a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-16 11:44:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/818059/1 : SUCCESS in 1h 45m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9cce7c3d02b1b4780743a86157c602d13146f24f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 12:13:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/64fd92d0ceae4475ab54c0eea1f1b212 : SUCCESS in 58m 12s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/00b4f3247090481d86208d2bc7039ea9 : SUCCESS in 2h 13m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c82cb0f32c54737978ee0a938c7c069 : SUCCESS in 54m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91e84968a3534bf890407dbc99c3ffed : SUCCESS in 17m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c1240fc796449b4827035eda09c033e : SUCCESS in 48m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/997fd50c3da84b10a76f4b51aff647de : SUCCESS in 43m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd9243b6f10744069e70d0f628016dec : SUCCESS in 37m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d5c20fed7864936acfedc4aa886eb57 : SUCCESS in 9m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/55824bb5d07146939b544fac0b9abf5b : SUCCESS in 52m 08s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/243da4884b644cfbae63d67c0db7b732 : SUCCESS in 57m 07s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f04863908f264d51b84ab903e607dcfc : SUCCESS in 32m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a460a8c4fa634e50b372763547fa859b : SUCCESS in 1h 36m 47s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8da3c0eb70b2458b80bb1d8e13806380 : SUCCESS in 1h 28m 15s","accounts_in_message":[],"_revision_number":1},{"id":"41bad577deddd1226762f0eb31b25090684903d4","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-11-17 13:56:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ee7d90147b35e3e06d674ab8f0d2deb4d9fcb630","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-22 19:59:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d30dec05acc29004045354aad7ce32c070b72254","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 20:02:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"396fe60605ef31e6e57509bda19bd90ccb8a933e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 22:31:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/d3079422dd7846c1abdcedab902142a6 : SUCCESS in 1h 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dee94019327a4a4d847877faea83857c : SUCCESS in 14m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41c6e541b6b549ada8cef413d6506d87 : SUCCESS in 53m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc92d3a2f8f84974a050103f2f53a6e1 : SUCCESS in 38m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a543b8a380e442fb4f77da381a7ae1c : SUCCESS in 10m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/78c920b471e14b8f94a11e02d03d52db : SUCCESS in 50m 56s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/dc8173f0815e4c0193c1ee813e9fd345 : SUCCESS in 1h 07m 55s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/167ca511cbe340dd9f172e6d0c4d7eef : SUCCESS in 43m 49s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/cf7dca6343af46f1b01b70f23cd2cfb9 : SUCCESS in 1h 12m 10s","accounts_in_message":[],"_revision_number":1},{"id":"b5ecbf0baab2b8338c0f0c570f19707076e95112","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 22:31:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2b7267004dcda1428a5213bc37765b95293a68e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 22:33:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7df5c3d4beaa419aa8954f5f06773fff : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"98e98f995c53a9e4426a96686d884ec2333ade8a","revisions":{"98e98f995c53a9e4426a96686d884ec2333ade8a":{"kind":"REWORK","_number":1,"created":"2021-11-16 09:58:17.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/59/818059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/59/818059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/818059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/818059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/818059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/59/818059/1"}}},"commit":{"parents":[{"commit":"0b389b163bd50f04f842addf87f7a4ba664b8b5d","subject":"Merge \"Set RPC timeout in PluginReportStateAPI to report_interval\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b389b163bd50f04f842addf87f7a4ba664b8b5d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-10-05 11:12:20.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-11-16 09:57:41.000000000","tz":0},"subject":"Execute \"migrate_neutron_database_to_ovn\" inside the same DB ctx","message":"Execute \"migrate_neutron_database_to_ovn\" inside the same DB ctx\n\nAll DB commands executed in \"migrate_neutron_database_to_ovn\"\nshould be executed inside the same write context. That will prevent\nthe DB transaction errors reported in the BZ. When a OVO object is\nretrieved and updated, both DB commands should be executed in the\nsame DB transaction to ensure the DB object stored in the OVO\n(that is a stateful representation of the requested columns) is\nfresh and not modified by other transaction.\n\nRelated-Bug: #1939704\nChange-Id: I754e2f9ad5b896fcc14ff30bd2901ccec2471907\n(cherry picked from commit dfcbb4cce0f206bbf824a1062de028223a27388f)\n(cherry picked from commit 785ce06b3d06b4477730641ad431c99072f35bac)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98e98f995c53a9e4426a96686d884ec2333ade8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98e98f995c53a9e4426a96686d884ec2333ade8a"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
