)]}'
{"id":"openstack%2Fneutron~818152","triplet_id":"openstack%2Fneutron~stable%2Fxena~Iebce0929e3e68ac5be0acaf5cdac4f5833cb9f2f","project":"openstack/neutron","branch":"stable/xena","topic":"bug/1939723-stable/xena","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-11-24 13:27:26.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2021-11-23 20:41:47.000000000","reason":"Slawek Kaplonski replied on the change"},"31662":{"account":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"last_update":"2021-11-24 13:27:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iebce0929e3e68ac5be0acaf5cdac4f5833cb9f2f","subject":"[OVN] Fix port disable security dead when run neutron-ovn-db-sync-util","status":"MERGED","created":"2021-11-17 04:07:38.000000000","updated":"2021-11-24 13:34:34.000000000","submitted":"2021-11-24 13:27:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818152-bug/1939723-stable/xena","cherry_pick_of_change":816328,"cherry_pick_of_patch_set":10,"meta_rev_id":"9e934634346dbd82e46638d0686674b6809642e3","_number":818152,"virtual_id_number":818152,"owner":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-24 13:27:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-11-23 13:03:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-11-23 20:41:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-11-23 20:41:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-17 04:41:06.000000000","updated_by":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-11-17 04:41:06.000000000","updated_by":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-11-17 05:05:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-17 05:55:13.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-11-17 06:15:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-23 13:03:19.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"cbb70a23ef31fd220f944b0e2f3e569acb60da15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"date":"2021-11-17 04:07:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"77251de9369e108fedd7226c6709f0ddfc7aa95a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 05:05:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0326679db0654ff89a6e7e870a633960 : SUCCESS in 55m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/41dbd48ce6674f63893d72128d732be7 : SUCCESS in 44m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c2fe5623bd3238d9e92b2764be3ef58725b75344","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-17 05:55:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/818152/1 : SUCCESS in 1h 46m 53s","accounts_in_message":[],"_revision_number":1},{"id":"7609ac3a59f7d3377d0a8810e5a0273142cb2851","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 06:15:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8b3c342bef8b4ec081bd85a02105d120 : SUCCESS in 52m 40s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/eb2e9a65eb4846b2bab293293cbbdb4d : SUCCESS in 2h 01m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0204fa840b63442eab6f860214147cb4 : SUCCESS in 55m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d7ab689e3b9494eb0922a5c5deccd63 : SUCCESS in 16m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78ef9ea6bcc94403bd73bb4d3b89370c : SUCCESS in 46m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1bacffa601a94209a85db5f7f772bc31 : SUCCESS in 38m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d319845d16604b9ca86d08acd91d2544 : TIMED_OUT in 41m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b257a6afbadc4f5ab0a72164b551ccf5 : SUCCESS in 9m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f9f095af75774a05b976bc9803cb45b8 : SUCCESS in 49m 19s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8b5a4ad3faf7488fae86debf6b8f1f52 : SUCCESS in 56m 12s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e8353ffcec4d4f34be0f3b2590d9171a : SUCCESS in 33m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fac52b1cf2124f888f86a43d64c18924 : SUCCESS in 1h 12m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8d37d886ff154f5b95fe141690f47000 : SUCCESS in 1h 15m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a391e7fcd42dcbb23c2bd97a0b6fe5e7cb1c1797","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-11-23 13:03:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b416ebfff38e8869ab758886e03f1e20d7fea019","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-23 20:41:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a837f3773527b9b610034ea222a63244113df6b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 20:52:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2bd9a0e968020e09a8dcdce0a33f85fd289bd313","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 07:51:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/61fcc6757b3d455484ab4e370a6c7a1f : SUCCESS in 52m 02s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/98c872abd1ee489097bf5ef048b32acd : SUCCESS in 1h 42m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dfa4078d5cd416abaac2e6886699d88 : SUCCESS in 16m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fc9592695304c6997a314c2a3819491 : SUCCESS in 44m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dcb3b2e15674ffda2f4248e2dd46165 : SUCCESS in 41m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cd6bddf55f34bc6b9f40a1d2cb560aa : SUCCESS in 9m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c370f2d10a4d426aa31ba41fe6d5a094 : FAILURE in 56m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7e5560966ede45648a17832c935eb898 : SUCCESS in 59m 21s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2f7af4c22a654c7fb106295912a0c01b : SUCCESS in 30m 34s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/87932ed63f9d4883b211ef7ba2d6e4ea : SUCCESS in 1h 14m 31s","accounts_in_message":[],"_revision_number":1},{"id":"797063d2078df588a4b2c710c9dd3e01bfc0bb80","author":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"date":"2021-11-24 08:07:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dfd20c50d37449416b419e470badf41ef7474bf6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 08:59:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9bcd5c94557c4449a72c7091129fc373 : SUCCESS in 46m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/64c28ac26a8d43bea3e8b65c964ee375 : SUCCESS in 45m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f6040f88e042a31f80f1d27ed17d73ea2e62031","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-24 09:40:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/818152/1 : SUCCESS in 1h 32m 55s","accounts_in_message":[],"_revision_number":1},{"id":"8d0d5c8070c021cd1f00b38c73f8ba48347f581f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 09:50:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b07b9c0336a54c3c9e6eaf32a1233bd9 : SUCCESS in 38m 18s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/654160caad0a4b6083bd51a4cad2b409 : SUCCESS in 1h 40m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91efaa5575a8429d8871ea8efd21fa19 : SUCCESS in 51m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6532ca6ccad14f8c8d73fdd93d22a33f : SUCCESS in 11m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e767c8a7944548c1ac87bd041f953ee9 : SUCCESS in 37m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c8782d6b868451eb4a07ebe54ead507 : SUCCESS in 35m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8635674b1774cc59823c48593224b3c : SUCCESS in 33m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34d2d42874144530afc888caa022e32c : SUCCESS in 9m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0b92a451f94b43b98c0c0f7edc43aea6 : SUCCESS in 46m 38s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/08dfa6ea8de54b0b876da76889b69874 : SUCCESS in 50m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d845912acb754c6db1e57e4ea0abdeb1 : SUCCESS in 30m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4ad5c0a7cd9d416e818d9fafab26a864 : SUCCESS in 1h 08m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bcf6ef866708487783dc806957b70335 : SUCCESS in 1h 10m 52s","accounts_in_message":[],"_revision_number":1},{"id":"f4c40f486ef0bc5b9f26c9e427c8408e1a058a55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 09:52:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fc68b7cbb9886baeef46df0096b0acd00c04287d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 13:27:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/264bd142e8334e96a19b6ae5b54b1b2c : SUCCESS in 41m 51s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ad4416e099674432a7e38986e5c1e5fe : SUCCESS in 1h 39m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76a0889583c54ed1a6231c848492f89b : SUCCESS in 16m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3fda640ff324b689380c87d0619f1aa : SUCCESS in 50m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a54a9ba9a6df4522918abf890f4cb60f : SUCCESS in 39m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a073033e80124103846a0e67c0eecbe1 : SUCCESS in 14m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3bd35e6f9e2b45b7a75f07aa80d079ee : SUCCESS in 47m 39s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f97ad2a7961046fda1c4ed3cab7cccca : SUCCESS in 54m 36s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2e83272b640b41f2ae8e10bc5596127d : SUCCESS in 32m 36s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5e8a93cbf6864c37bdcb65a49d8cff61 : SUCCESS in 1h 10m 52s","accounts_in_message":[],"_revision_number":1},{"id":"94abf479249b8b4f9f80a30aaeeb0c54112a4947","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 13:27:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9e934634346dbd82e46638d0686674b6809642e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 13:34:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78bfbd2d5e7242e9bc8fdac2d378c27c : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78bc1f6fec8e216aca7cf6fbc6770631dacca0ee","revisions":{"78bc1f6fec8e216aca7cf6fbc6770631dacca0ee":{"kind":"REWORK","_number":1,"created":"2021-11-17 04:07:38.000000000","uploader":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"ref":"refs/changes/52/818152/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/818152/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/818152/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/818152/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/818152/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/818152/1"}}},"commit":{"parents":[{"commit":"e952ccec6a58d926b539814a53e019455d582c4f","subject":"Merge \"Set RPC timeout in PluginReportStateAPI to report_interval\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e952ccec6a58d926b539814a53e019455d582c4f"}]}],"author":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2021-11-11 14:00:23.000000000","tz":420},"committer":{"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","date":"2021-11-17 04:07:38.000000000","tz":0},"subject":"[OVN] Fix port disable security dead when run neutron-ovn-db-sync-util","message":"[OVN] Fix port disable security dead when run neutron-ovn-db-sync-util\n\nPort disable security should not have been in acl neutron_pg_drop, but\nwhen run neutron-ovn-db-sync-util, port disable security still added\nto ACL neutron_pg_drop. It because port disable security is not\ntrusted port.\n\nCo-authored-by: archiephan \u003cchungphan7819@gmail.com\u003e\n\nCloses-Bug: #1939723\nChange-Id: Iebce0929e3e68ac5be0acaf5cdac4f5833cb9f2f\n(cherry picked from commit 4511290b726f605384285228a28ad7b32a4b8c43)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78bc1f6fec8e216aca7cf6fbc6770631dacca0ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78bc1f6fec8e216aca7cf6fbc6770631dacca0ee"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
