)]}'
{"id":"openstack%2Fneutron~818255","triplet_id":"openstack%2Fneutron~master~I39e3b987b8546fd970a933b846ed23c8a2588258","project":"openstack/neutron","branch":"master","topic":"bug/1951296","attention_set":{},"removed_from_attention_set":{"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2021-11-17 20:45:17.000000000","reason":"Terry Wilson replied on the change"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2021-11-18 17:10:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I39e3b987b8546fd970a933b846ed23c8a2588258","subject":"ovn: Filter ACL columns when syncing the DB","status":"MERGED","created":"2021-11-17 17:59:41.000000000","updated":"2021-11-18 17:13:09.000000000","submitted":"2021-11-18 17:10:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818255-bug/1951296","meta_rev_id":"849e30d757123e497241cfea68182b320f508256","_number":818255,"virtual_id_number":818255,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-11-17 21:50:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-18 17:10:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-18 07:18:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-11-18 14:47:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-11-18 14:47:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-17 19:04:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-17 19:33:57.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-11-17 20:32:58.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2021-11-17 20:49:44.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-17 20:54:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-17 20:56:54.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-18 07:18:59.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"033b249fe91d0a51d70da42fac837d30712e5a6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-11-17 17:59:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"379fbac9fd7d29f77463cab07650eb275c5b2872","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-11-17 18:32:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d85745bb842f83d5e83359e3c4b38f428aec9c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 19:04:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9165608849cf4d5e98bcff3859e219de : TIMED_OUT in 1h 01m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8cc0eb93614f4fb4b7de55f644a309a0 : SUCCESS in 49m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"93efd178aea9d4a75886164dbdca369df9d5d273","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-17 19:33:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/818255/1 : SUCCESS in 1h 33m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d543b8e4e472c4ccdaeb28ff52b480f050cfa374","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-11-17 20:32:58.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a56355822c8579262d9f3b2d59544c1d4603053e","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-11-17 20:38:27.000000000","message":"Patch Set 1: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"af5e1609c0dfb04db2503de9d7422e47307deb72","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-11-17 20:45:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"66d6f4db5979d1c6aad244696f60bddf47edeb07","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-11-17 20:49:44.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2021/Arista_Tempest_Testrefs/changes/55/818255/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"1d5ed033214d0f97b24c549d9869d1a3f2abb0cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 20:54:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/51a784242f894516aebe5033a0de6e30 : SUCCESS in 50m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c2eabb6fd3ec43b8bdd57def3a9bb00e : FAILURE in 2h 45m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6f801376c7a473a8764929d002db67a : SUCCESS in 51m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e1cc6b55a2a4cf6b866167f85549dcc : SUCCESS in 57m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/accf9917cd99487ca1a9c4a2c8529d5f : SUCCESS in 14m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e995ff74fec45ef943f251b2de4cb9c : SUCCESS in 40m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59c197dce41e4423aeec33ff0ab31e1e : SUCCESS in 40m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f11e87d739524b889da041aa9fed348a : SUCCESS in 41m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44870eb8c44e419f9e7f0034b7dd8cac : SUCCESS in 9m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/11e1b9321b184bef8320cd464c880cff : SUCCESS in 53m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e53d262b9d654130a86611ca81325b30 : SUCCESS in 59m 33s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/79dfda2967134a28b7fd20a982b1a0ab : SUCCESS in 34m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a60a2b4885b34b2e9f0ee979160c06f6 : SUCCESS in 1h 24m 33s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/76312ea31fd34364b13efaf0fea4a9a8 : SUCCESS in 1h 19m 42s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5e2884f270d942e1bbcb0414a3f5c6e7 : SUCCESS in 53m 16s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/83bad601905049de85d025b34497f34d : SUCCESS in 54m 55s","accounts_in_message":[],"_revision_number":1},{"id":"c28341dbd5e999a0a79a85c8e57eb32ead3232a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-11-17 20:56:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"66c2518614443372b0961a8cc86cac0082f301fe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-17 20:56:54.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/818255/1/SRIOV-neutron-macvtap : ABORTED\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/818255/1/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"fef30cd1cf4b4c6133b2dfe0b93b76e9fdc261cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 21:48:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f034b45af584fd1920ac96a41b3433d : SUCCESS in 45m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/43e8228e8b9742a6afb5d16acd25c436 : SUCCESS in 44m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2b6bbb304cc22c7d5121bd1aad5cccfc3e1f0195","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-11-17 21:50:51.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2021/Arista_Tempest_Testrefs/changes/55/818255/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"52432657bdeec0d1a5a526d305f5cb1a22c7125f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-17 22:05:09.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/818255/2/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/818255/2/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"1ed53dc9838c87bfe235739a30e46132a145a5d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 23:29:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b30411d61d544a4a8c2ca6a7527eac5f : SUCCESS in 38m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d88cac1243764fc5b0fd59897508bde8 : SUCCESS in 2h 29m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd6d6c266d564c7dbab6e36def01e277 : SUCCESS in 40m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/718a57075f7840a0aea0e95a836ab994 : SUCCESS in 49m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5aa93d041a25496784601e2e654e1a22 : SUCCESS in 13m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ee31966617e4cc3ab071eebf5ea1cec : SUCCESS in 38m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/135baa6419be4494b1463ebad2dc3bb3 : SUCCESS in 38m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9aacf847c6a14a97b275c63c4db1f35b : SUCCESS in 38m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64b79ae605ea4495a34a99e9d4424d80 : SUCCESS in 13m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/628aa772e3c14982a4a0fa222ab3403e : SUCCESS in 45m 51s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4d5e63efd2fe4e2dab9ae17118277e2c : SUCCESS in 45m 08s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/66810f52938e4c55abaf1319c31ba7bc : SUCCESS in 29m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/953459f133f144fb9fcac83ebcc57fb3 : FAILURE in 16m 01s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a9217533a76046d09d229e4596c4fb62 : SUCCESS in 1h 14m 37s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c12f3eae6a124d84be447d070fd17650 : SUCCESS in 55m 54s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e433c214d46544c89f3efdd36fd11f23 : SUCCESS in 55m 11s","accounts_in_message":[],"_revision_number":2},{"id":"9b1592cf625b3d6f046118041cdd2f4c10dd74ae","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-17 23:50:25.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/818255/2 : FAILURE in 2h 53m 17s","accounts_in_message":[],"_revision_number":2},{"id":"e52ae23bbc2d9e7b3731ffd365e78392526b9fea","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-17 23:57:07.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/55/818255/2/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/55/818255/2/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"61aa58df1355b4ee31aa0e9da9513da1d851ee79","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-18 07:18:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"207b711bc56e709430bff62dca2e5ccb9877c187","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-11-18 14:47:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9f1852ee63e099e90264fc8371b11d82bcb2987c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 14:48:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"72597fe9113e7f197b72abe5a9f5c1921cca8826","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 17:10:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bf716a2162fc4bf28b1352218f94328a : SUCCESS in 43m 46s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bb12e62ddf694cabae3d3ef25eb25e50 : SUCCESS in 2h 19m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82a8eba9e3834ff7a1f6e9f17b8f9078 : SUCCESS in 53m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/324376492c614aa7bd1fbe728d785121 : SUCCESS in 13m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c722be021584d0ab877f4fac9bbbc88 : SUCCESS in 40m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/247d7fa2acec4421865f9ad561accc79 : SUCCESS in 46m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/760c550514434cbb903275085287df66 : SUCCESS in 11m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/091d32123ebc4f9f80c9e4de61b2d86f : SUCCESS in 51m 11s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e23782606cce4503bd713dcd4b40871e : SUCCESS in 1h 00m 57s","accounts_in_message":[],"_revision_number":2},{"id":"b5dc4382774dd7e7305ebd8dc38f0a66f8e13042","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 17:10:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"849e30d757123e497241cfea68182b320f508256","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 17:13:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df0e879e9aa04a1b91e3b14c7a8eaf30 : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"23b99e2f127731c85f63c88c7144aa0a111c4abf","revisions":{"0b7ddb71a72732c881e7d023fbbd57ba75cb8586":{"kind":"REWORK","_number":1,"created":"2021-11-17 17:59:41.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/55/818255/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/818255/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/818255/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/818255/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/818255/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/818255/1"}}},"commit":{"parents":[{"commit":"63f8a39d75feb4d2bfb38e82a29593375d156153","subject":"Merge \"Don\u0027t enforce scopes in the API policies UT temporary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/63f8a39d75feb4d2bfb38e82a29593375d156153"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2021-11-17 17:29:13.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2021-11-17 17:59:39.000000000","tz":0},"subject":"ovn: Filter ACL columns when syncing the DB","message":"ovn: Filter ACL columns when syncing the DB\n\nThe patch filters columns from OVN DB only to those that are used by the\novn mechanism driver. It means generated ACLs from Neutron DB and ACLs\nobtained from the OVN DB will always have the same columns. This is\nuseful for db sync script when comparing if given security group rule\nhas corresponding ACL in the OVN DB.\n\nCloses-Bug: #1951296\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\nChange-Id: I39e3b987b8546fd970a933b846ed23c8a2588258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b7ddb71a72732c881e7d023fbbd57ba75cb8586"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b7ddb71a72732c881e7d023fbbd57ba75cb8586"}]},"branch":"refs/heads/master"},"23b99e2f127731c85f63c88c7144aa0a111c4abf":{"kind":"REWORK","_number":2,"created":"2021-11-17 20:56:26.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/55/818255/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/818255/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/818255/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/818255/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/818255/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/818255/2"}}},"commit":{"parents":[{"commit":"63f8a39d75feb4d2bfb38e82a29593375d156153","subject":"Merge \"Don\u0027t enforce scopes in the API policies UT temporary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/63f8a39d75feb4d2bfb38e82a29593375d156153"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2021-11-17 17:29:13.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2021-11-17 20:49:25.000000000","tz":0},"subject":"ovn: Filter ACL columns when syncing the DB","message":"ovn: Filter ACL columns when syncing the DB\n\nThe patch filters columns from OVN DB only to those that are used by the\novn mechanism driver. It means generated ACLs from Neutron DB and ACLs\nobtained from the OVN DB will always have the same columns. This is\nuseful for db sync script when comparing if given security group rule\nhas corresponding ACL in the OVN DB.\n\nCloses-Bug: #1951296\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\nChange-Id: I39e3b987b8546fd970a933b846ed23c8a2588258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/23b99e2f127731c85f63c88c7144aa0a111c4abf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/23b99e2f127731c85f63c88c7144aa0a111c4abf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
