)]}'
{"id":"openstack%2Fneutron~818399","triplet_id":"openstack%2Fneutron~master~I751b8baeda00ce890722d96199b19a91818b2ca0","project":"openstack/neutron","branch":"master","topic":"bug/1950454","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-12-16 10:08:15.000000000","reason":"Change was submitted"},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2021-12-16 10:08:15.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2021-12-14 10:20:27.000000000","reason":"Slawek Kaplonski replied on the change"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2021-12-06 10:01:46.000000000","reason":"Lajos Katona replied on the change"},"9531":{"account":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"last_update":"2021-12-16 10:08:15.000000000","reason":"Change was submitted"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2021-12-03 07:40:18.000000000","reason":"Oleg Bondarev replied on the change"}},"hashtags":[],"change_id":"I751b8baeda00ce890722d96199b19a91818b2ca0","subject":"Add upgrade check for floating IP QoS inheritance from network","status":"MERGED","created":"2021-11-18 15:05:31.000000000","updated":"2021-12-16 10:10:18.000000000","submitted":"2021-12-16 10:08:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":38,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"818399-bug/1950454","meta_rev_id":"774672d7ee7c5467223bf188ec7c702aaac53f77","_number":818399,"virtual_id_number":818399,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-11-24 14:20:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-16 10:08:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-14 10:20:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-12-14 10:30:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":1,"date":"2021-12-07 06:15:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},{"value":2,"date":"2021-12-03 07:40:18.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-12-14 10:30:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-18 16:06:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-18 16:41:38.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-11-18 17:36:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-18 17:39:57.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-19 02:00:03.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2021-11-19 07:45:43.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"CC"},{"updated":"2021-11-19 09:47:13.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-11-22 10:01:52.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2021-11-22 10:01:52.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-11-22 10:01:52.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-12-07 06:15:21.000000000","updated_by":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"reviewer":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"state":"REVIEWER"}],"messages":[{"id":"bca8c5a35fb3a59855b3d3949c3aa6a098bafbe3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-18 15:05:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89c042fec7e44e968aff8abb48fabc85dbe669f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 16:06:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b79a458dda1e4953b5ae2c3303ea5b24 : SUCCESS in 57m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/23797f339c1f45deb7cfa46028c57142 : SUCCESS in 57m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"041df26ced2d6c7b287a7942479c02febb78472f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-18 16:41:38.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/818399/1 : SUCCESS in 1h 34m 11s","accounts_in_message":[],"_revision_number":1},{"id":"6e1c8d1d83aeffc11b718382fba25ed30349567c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 17:36:56.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b4b5d9b9e5b34ef5a9f5de90de051d9a : SUCCESS in 40m 27s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/054e7071708c427087fd849ff9abb5ed : SUCCESS in 1h 58m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/75d378ebc90c4c84adfc1b0b9be69dcf : SUCCESS in 2h 08m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a28a4287bafc41b69a097a78d6d53a65 : SUCCESS in 2h 18m 39s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/965c0aca61d745fd91f6cdf80a0426fb : SUCCESS in 2h 20m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f53575d7ee047eb85316c124f57e184 : SUCCESS in 51m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d022ee6f41a544ce8d1c00e5f65fd019 : SUCCESS in 55m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68bf46a8fbc54e62ae0ced67fa960fb6 : FAILURE in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3f96edd3c4546f89794b9fe15df785d : SUCCESS in 41m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eca0e799533f4967963f5502a79413b8 : SUCCESS in 40m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8ee9f4fede64162a178f83e27108541 : SUCCESS in 43m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e2f903220f9401e8c26cb8a4a55e56f : SUCCESS in 9m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b147bb73adf4b34861f4c7d0bfb819b : FAILURE in 45m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d6bb95712de14bb5a1b2a3aff7c15c72 : SUCCESS in 1h 20m 29s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0c6777d5058549f298b69149d50d48f9 : FAILURE in 23m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e94cbb72ea674bfbb0f0ccc5305d6bbe : SUCCESS in 58m 41s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9c59034c08434fd7a1fdc3eb08bcd98e : SUCCESS in 55m 57s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/361080ac93344421a21a63a3c280554a : SUCCESS in 1h 12m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/43917c69fec242d7896885d35586df2e : SUCCESS in 1h 20m 10s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00b4bcc57d9f4b6a909705056f07e5b5 : SUCCESS in 59m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e8d87bb07db94dd19899273eb4505636 : FAILURE in 19m 38s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4cb86924efc04ffcb2597126d3086171 : SUCCESS in 55m 56s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/34f1575ad23f4ed9bea8f1cd21ac478e : SUCCESS in 58m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b569321f34472cb82f4e82afb74bc09424c4e2be","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-11-18 17:39:57.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2021/Arista_Tempest_Testrefs/changes/99/818399/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"3d95a8d2875ec9701879921f82a1adffe98dbba5","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-11-19 02:00:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"243e9bab234946110486bda612994eb861eb7dfe","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-11-19 07:45:43.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f4b58e9ac3043c4a7c39c07ac914081f9da67a7f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-19 08:34:46.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"698f9e4335b910f7b398ad5bee4b599a8d3b9527","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-19 09:36:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"99252bd37db75e088de05247b58b3dd6362649ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-19 09:39:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"edc33f43d05280d7d83bff49a357815ae5b74627","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-11-19 09:47:13.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3d7c5a0676bd34128a8aab2aab21f4a625347235","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 10:47:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c6409d6f3bb24c2cbfbb0f15f0795ae7 : FAILURE in 52m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/128ad2628f614e129ae8d38e8e954167 : FAILURE in 57m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b6b7c26ad7b3302936805ba09d69dcfc8023b88c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-19 11:14:06.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/818399/3 : SUCCESS in 1h 33m 59s","accounts_in_message":[],"_revision_number":3},{"id":"c6249f51c92547076fca678908811893a817b999","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-11-19 11:15:21.000000000","message":"Patch Set 3: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f79d8d443517e0b2be165f6877972bcbf77e08af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 12:01:00.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ae83158ca5bf4f76b311c9592e494ec0 : SUCCESS in 49m 12s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/89ff0b3db40c4859806e0bba5c8404dd : SUCCESS in 2h 01m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7ed179c43c30457985227c45bda08001 : SUCCESS in 1h 54m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0f053d925a044da8bf74f506ef99fc5a : SUCCESS in 1h 55m 26s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c538ecec6fdb4d3c92f2f6696a4c3457 : SUCCESS in 2h 19m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fba300b440647c48b5686d0ae6d2dab : FAILURE in 45m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9858d3432bb64f26a8231da25dbaf494 : FAILURE in 46m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/489b04b794ff44ae9c8156cc90baea04 : FAILURE in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82b6bc87e4004fce8cc6a41ae0ad5bc2 : FAILURE in 37m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3554f1fa2ea4963a094675c7f3cdafa : FAILURE in 36m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19028e967855403e849f2c1dcb44dc60 : FAILURE in 36m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db7a4561ee5c4bf78775ff0762559956 : SUCCESS in 10m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/672e86e9139e4a22a49e9efeace955e8 : FAILURE in 49m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/245bbce24b3a4b27a2d81786ca3d436a : SUCCESS in 1h 24m 28s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bf9b1e1e7e5a48809ec94c1703d5ccfd : FAILURE in 27m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/978837a1438648c48a2225782c82a61f : SUCCESS in 55m 59s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3d37b60f0d7c4de9a58e920554a1630f : SUCCESS in 1h 05m 56s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5601dfdfd7654243bf14f65b33a20895 : SUCCESS in 1h 03m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b1f7625c96574dca94442d0a3946850e : SUCCESS in 1h 15m 49s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/daac3234a3194f5898df7fd31e541e39 : SUCCESS in 1h 00m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0d113972ae3646c0abb5cfb1c33651df : FAILURE in 28m 06s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/02c7920cacc742bb89646fb8577665f7 : FAILURE in 46m 35s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1466c6e8a6af4cc4bc4932eefe6b436a : FAILURE in 49m 21s","accounts_in_message":[],"_revision_number":3},{"id":"48cb2adee224ed3c0219ab7e615e4a1eeb777627","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-11-19 12:08:42.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2021/Arista_Tempest_Testrefs/changes/99/818399/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"de99b69c7f2336689d9aee0e00c1af998d25b681","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-19 12:14:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"00358cab25aa7877f90f35f6a2e6c0d97605db44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-19 12:14:54.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"96c28f91fca29be96ec2f52035b2ebb0c2834e6e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-11-19 12:56:22.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2021/Arista_Tempest_Testrefs/changes/99/818399/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"fd0f8aaac36443cdfc3db839a4ad3cac65f5934a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 13:21:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c5a48f4ba2bc4858b15660f43fdd9cba : FAILURE in 1h 00m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/04b8c893eef440d1a4ba20e5cd10d844 : TIMED_OUT in 1h 01m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9c0f5df3b615020ed54081278a6e0c7f0b79fdaa","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-19 13:53:24.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/818399/4 : SUCCESS in 1h 37m 25s","accounts_in_message":[],"_revision_number":4},{"id":"939d69876f920655c9b120d09514766947de16bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 14:51:46.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5506d393b48e4b4d88b9f49bb5243b4a : SUCCESS in 58m 46s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2ac6a19e927e495c9a511824f78eaeea : SUCCESS in 2h 33m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c2b522ea12614e06918c3e8b8d14baaa : SUCCESS in 2h 18m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/89c989ac34c64157a51332ab85753b20 : SUCCESS in 2h 29m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4d2abc36e5da4ef78502313a9de2995c : SUCCESS in 2h 25m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/788a44d38c9d4e7197494b4d5bc5875a : FAILURE in 59m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5748523bad7a4d0f994500ca6563d64b : FAILURE in 56m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88fe0a3c6e194e42987c1d151c474fd1 : FAILURE in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96618acb764b409e890314b4c478116e : FAILURE in 44m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d09bb70e8e04de2bf22fce77f7c3cad : FAILURE in 36m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1db6ff463784abea5006b1135b16ece : FAILURE in 39m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce7112a2fad1418e89413d00cc7cbe5f : SUCCESS in 10m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/75d6df298c5d4100a3b40f359934bb37 : FAILURE in 52m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2c411400d146404880e5d8fa29368b36 : SUCCESS in 1h 27m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1822f4d2d895486e82c0c267bb8d4d68 : SUCCESS in 1h 15m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/497f51bba80d4765b333f7be57232048 : SUCCESS in 56m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c6968239a0264874814a94c9194bd60b : SUCCESS in 1h 02m 15s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c05feb2087434c1d9cd392c6126b05cf : SUCCESS in 1h 07m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2f983686d3e744d7b2aeb0af301d840b : SUCCESS in 1h 21m 03s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de3d44c35de74d6d96402ad52c3d1d51 : SUCCESS in 1h 09m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/766ab0e4b986478c8f1087e6fa026f4e : SUCCESS in 1h 08m 59s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/29a139665c754400b031aef9be9c31e9 : FAILURE in 1h 00m 34s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/26a9f9e67bfe4413bbced0f720ce8e5d : FAILURE in 1h 02m 11s","accounts_in_message":[],"_revision_number":4},{"id":"f5b738c97c1f1d2c1c2e569de6d90ecec3262460","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-11-19 15:31:15.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2021/Arista_Tempest_Testrefs/changes/99/818399/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"c9d0634d60b97ebd3630f341a1ce4b3bff9dea2b","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-11-20 01:53:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a238c019d34f89c9e145b8f6eddee8801cfbe70c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-21 17:12:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"04dbfa79e45cb937996d69801f301177bf751289","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-11-22 06:34:54.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3070715566e18a5add8ab8b47bf9f0929d4e74a9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-22 10:00:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ab46cedcd06c185738b562f3365fdd9a0ba3719b","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-11-22 10:46:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"199183e51c67ebaf93b2791493c838edaf0acd71","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-22 13:58:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"da0f980f692db240aeaa66baadd874dc39290bea","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-22 14:30:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7463bc0fb2912e7995da40912df105dba2ab87dd","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-11-23 01:45:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2945886e9f9b63c1bb08a2dda1ea260382d03bf1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-24 08:33:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b66d24f0963039ff83969503972220246e537a53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-24 09:26:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f841088de638e4da54a7b51937c89d1c47986536","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 10:32:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/56521cdcccbf42d9b8ded1940636e8e6 : TIMED_OUT in 1h 01m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4b9d669bc8154f039a61af6c2aac4bd7 : SUCCESS in 51m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"676736c4a87bc5e4d91692ea9f18c752e843278e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-11-24 10:51:44.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2021/Arista_Tempest_Testrefs/changes/99/818399/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"da1d9221deded92e1898d88bc553c6640121a655","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-24 11:13:54.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/818399/5 : SUCCESS in 1h 33m 08s","accounts_in_message":[],"_revision_number":5},{"id":"bad6e44a3f0fc72741ce28deffa510537a7b0492","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-24 12:06:51.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1efd0f7f9199d8e3fe14c82a1a3b6ac4dbe19dee","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-11-24 12:14:25.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"ac0c9aaf2c4ced68dd924034e4788562a264192b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 12:16:44.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/19d0e3115d474d94987e1ca42e8b348d : FAILURE in 28m 02s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6b9a05e1d5de4105855eab6d132fe118 : SUCCESS in 2h 24m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3ad5742a0242493fbbaed84803ed476d : SUCCESS in 2h 26m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/88b2fa68f1a94648a925ca6c64720785 : SUCCESS in 1h 31m 11s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7aca64d68e8944beaf3ef524f23cf8ba : SUCCESS in 2h 43m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14ee2420193d43a1aa73d35164397d05 : SUCCESS in 49m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fbbf8cb65cc84993bd7b8f857ca46e79 : SUCCESS in 49m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e12303be5db1427c98a6d8eea7b0a2f8 : SUCCESS in 15m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b510ca7009a1457cbcd52b016ce9adef : SUCCESS in 46m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d331b1d0a5fb43e7a86ba6369c2a1b7f : SUCCESS in 43m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2325d75483ea48f5937a3070c59e4810 : SUCCESS in 38m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc0fc6042c5e45da98e5b4a51e11e798 : SUCCESS in 9m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/58892fa1ecb247f595da77e8a7d40293 : FAILURE in 56m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3458877ba09b445bbfb74c6a2cc76d67 : SUCCESS in 1h 27m 29s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/30c74e47ebca496699337b0e5c3cf8b7 : SUCCESS in 1h 05m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/abafc1fb0dff472b98e55deac88805df : SUCCESS in 55m 32s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6ff6430f2e2044069c78f75f796e7d2c : SUCCESS in 1h 02m 36s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9c9a0de9291d48cf9cdfc7914422efd5 : SUCCESS in 1h 25m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9f82ba4dd253403d924543a7b90fe8a8 : SUCCESS in 1h 17m 28s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02ab84a56bab4db1aa4004433145a742 : SUCCESS in 54m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9928befa6f434e63a765239e68d44264 : SUCCESS in 1h 26m 20s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f2cd5e8244149d0a30fd06d6d748b11 : SUCCESS in 1h 02m 42s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e78792ab64d14e0a8798547e4655bbe5 : SUCCESS in 1h 00m 27s","accounts_in_message":[],"_revision_number":5},{"id":"eed0e901b80d98a63ebac3e4375496c7b12d4b23","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-24 12:39:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"78c6a088f3e177d8869ca1010075307387779d80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-24 12:44:48.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"68829c97da75755a17d7e9b81aeee91c62c64de8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 13:46:57.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fca8b767a97043dfa3e28d49b29c831e : SUCCESS in 48m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0dd4f9781e4d404eaebfabc39d17ebcd : SUCCESS in 59m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"56a313aaf52e671476ef43b371fcce488f5b3f89","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-24 14:18:49.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/818399/6 : SUCCESS in 1h 33m 11s","accounts_in_message":[],"_revision_number":6},{"id":"b96eac06aa762d25673cce22ec20f2518279b488","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-11-24 14:20:17.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2021/Arista_Tempest_Testrefs/changes/99/818399/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"d47964c47bad36d993df55796ff6a805efa0069c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 14:34:52.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"808275ff84a61e500964e644852c2c77dd683dbd","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-24 15:09:23.000000000","message":"Patch Set 6: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ed28911e40374a457e2c5b5a7760e05b59531bfe","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-24 15:09:28.000000000","message":"Patch Set 6: -Code-Review","accounts_in_message":[],"_revision_number":6},{"id":"e6780ced7146684d10dc037c87162fc066068e7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 16:20:45.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/365a198f08ee4fabbf516d513477308a : SUCCESS in 50m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f33ef40deca6426699dcbfb20cfa07d6 : SUCCESS in 51m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"254fee2e7d9d5dc65eddbb2f53a9f53c87c29f4a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-24 16:43:51.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/818399/6 : SUCCESS in 1h 33m 46s","accounts_in_message":[],"_revision_number":6},{"id":"d1546adada289695418d87e0fba4cec86f48f70b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 18:12:04.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/99a0d69d49554637ba16a17fec564fca : SUCCESS in 44m 27s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e25d45a09214496b98cd784a33b775e5 : SUCCESS in 2h 19m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a98eccf5d5b64569b4e85817a4cbac2d : SUCCESS in 2h 07m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bd98644cb5bd4c6f923af6844f7f82c0 : SUCCESS in 2h 05m 21s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9bb6ddc9947b499486f6d21baa527ec5 : FAILURE in 2h 54m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f800adcf4694e39943e6af591a3ad3d : SUCCESS in 46m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cc086f4911842138625412a0786be3f : SUCCESS in 58m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e52f09180cc4d168fee98be8405f987 : SUCCESS in 13m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5f910cd733743469caeb1e1544d4c22 : SUCCESS in 49m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e0f2fd704a4428491ef143b5918c40a : SUCCESS in 46m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d151f6fe8ff34fafb15d15fba9a0047b : SUCCESS in 45m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a64d2c7ef08b4337a9b11d6c390918dc : SUCCESS in 11m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3e512df47ee649ecb375828c42c5660b : FAILURE in 56m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/50a7978721384dd099bac537c931a211 : FAILURE in 1h 59m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5a7683dcce3e44edb81b28b0b9d7dfa8 : SUCCESS in 1h 36m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a16ee0182d334d3996238398801a5964 : SUCCESS in 57m 10s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4071b7a2367c44e28f64b686235a7d47 : SUCCESS in 57m 14s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/aaa03ecf5e7c44f8bfbb8720ed104000 : SUCCESS in 1h 15m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7efe665ce46a4983bdfec5d01c9419bf : SUCCESS in 1h 27m 25s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ad79aa884684160b241119fcf72aa3e : SUCCESS in 55m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/70acf780f10f4df0a88ede036ed6e60f : SUCCESS in 1h 36m 26s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc5a838b382f49bda0ef8b45f0f9c069 : SUCCESS in 55m 06s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e13c44aa340e4c66bc6deea158575454 : SUCCESS in 52m 15s","accounts_in_message":[],"_revision_number":6},{"id":"52e06d31fb820dd0bc5c4979299e48ae0e0e4a73","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-11-25 01:52:44.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"268ae4b69ae93daedda36a29274049ec5074ecd1","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-11-25 01:53:37.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"da6ff8f3ba1aec577036f4d6a76e842e2dffffaa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-25 09:09:21.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"52ddf0be986fe2554bd97be7629abf0cd18b3c0e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-02 13:19:31.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d9b8c1b78d6b85bc907f4db19efe688ee5044157","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-02 13:33:53.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/818399/6 : FAILURE in 13m 42s","accounts_in_message":[],"_revision_number":6},{"id":"3c3a5d2eb0b2ec8903cb819a33dfa050179608fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 14:00:02.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/17d4e18ff8d5418cb4f1714a15446e9d : SUCCESS in 34m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ffe2bcba932a4f04bafd2435912f9535 : SUCCESS in 33m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d5eae2561a2a23a2d131a8027e3a4e572cd0dd84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 16:24:33.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8c1eccacc6f140e89f4eff452c4c2f62 : SUCCESS in 48m 32s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a5fdd72a5eb64bfaaf39847fca7d56b7 : TIMED_OUT in 2h 51m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9b44025213d547d88756c5f5c492bb78 : SUCCESS in 2h 08m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6347978b2007473daf1ce37e58c48105 : SUCCESS in 1h 37m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e1ce27d24a144320b03d7f53017dcf51 : SUCCESS in 2h 59m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9909a3d2fdb549d4960e904ccc935b3f : SUCCESS in 51m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62474a38367447bbadb3ae39b4466ccf : SUCCESS in 34m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9aac67b8add4df3a8622ce0fbee0cdc : SUCCESS in 14m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b36368f55a6e4d7eb4dbf70128c295fb : SUCCESS in 27m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f9e8250de68461d9900334ed9f22956 : SUCCESS in 25m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcaf476211594f81b4c4f73516740be9 : SUCCESS in 11m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5129371b5b0c42c4bffe3d88134a6bcd : SUCCESS in 40m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b531e65d080c4512b1c04527f59f5173 : SUCCESS in 1h 56m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/82ca19ffe10a47338937f30e4ce4865f : SUCCESS in 1h 20m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1f21491544574f88b291841abd62ae6c : SUCCESS in 47m 29s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/17661237bc6a4f5fab76c3cbff4df7ec : SUCCESS in 1h 14m 41s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/32bde02abd0d4f89a16376f9fa17e952 : SUCCESS in 1h 05m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4efbc9727aa2465c9c0a019f7dbdc1c3 : SUCCESS in 50m 14s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c79a1a1bf0b04f4a9247ed39c17437e9 : SUCCESS in 55m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dace9f52e2c34ca58ae9866c088978c7 : SUCCESS in 1h 23m 09s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e8fc727260642e78aa66ffdbfcda828 : SUCCESS in 35m 10s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c7564f5181e1430189b59be21b4d68a0 : SUCCESS in 48m 10s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5109003162094dc3854c6879c9cd7ecc : SUCCESS in 47m 56s","accounts_in_message":[],"_revision_number":6},{"id":"4cdec170a3631b09b882082eda58b2bf82189777","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-12-03 07:40:18.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ba035b2f7f520696e2ceb7769b5c4e6155a1c00a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-03 07:54:48.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/818399/6 : FAILURE in 13m 49s","accounts_in_message":[],"_revision_number":6},{"id":"86e8b3c88c7ad9d7f40def7b019706310d140540","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 08:22:28.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6db8fabc7eb740b4b0c87cb26aa6d980 : SUCCESS in 40m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ea870a918d9042509de492279987f347 : SUCCESS in 29m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6e883b718599da386136946c5904b4eb90c4d24b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 10:49:28.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/caa86bfcba25494d9bb9b38f512ce3b8 : SUCCESS in 44m 47s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a70067e1c415479bbe86583e727fd131 : SUCCESS in 2h 33m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/264ec1a7e9194031a1a8dd6fd1c47711 : TIMED_OUT in 2h 51m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bcfa030de9ab49d79c022aafaa524937 : SUCCESS in 2h 12m 00s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/cc5a6da764ab4906ad36abe899eb9432 : SUCCESS in 3h 03m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6038cb6547974a4e8baa04f5307a33c2 : SUCCESS in 43m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/989b1b0ded7a48afab9fff659e9fa0aa : SUCCESS in 34m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9592b73fb374606ae75a5a19a5cf63a : SUCCESS in 13m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b04d1a719bdd46a89f16214f6437c056 : SUCCESS in 23m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53cdc5e8bd214232ab57ecd191228cda : SUCCESS in 26m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b80acda8c65848dca2d4b8133d40b255 : SUCCESS in 9m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b02b77a47e8a4de1ae1ada37787e0327 : SUCCESS in 37m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b49293dc16cc42df98ebc2339afc754c : SUCCESS in 1h 28m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/01800667a3cb45a1b2752fb995feb9fa : SUCCESS in 1h 08m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dee3255d38c44232981c4a35094906da : SUCCESS in 58m 36s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dd7757edb59e4a1e9b49c0675ae622a2 : SUCCESS in 1h 02m 28s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/08369c4cf52444faad2108c195449a51 : FAILURE in 56m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d94e994bdcc8461189c8c609717e9137 : SUCCESS in 1h 19m 42s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/084a1564295745a0809a9c10b273442f : SUCCESS in 59m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e103d3c9acb343918a09331ed674eb66 : SUCCESS in 1h 17m 30s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e9254f6f36c40f58e1c5941101150f6 : SUCCESS in 29m 54s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bea28b6860a848aaa8d20402fad61217 : SUCCESS in 41m 04s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/26ba9fb374254b31917593ca7c91c4d2 : SUCCESS in 40m 59s","accounts_in_message":[],"_revision_number":6},{"id":"61871a358bc2459aa6be93f326af89fdc20f5db8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-12-06 10:01:46.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f3afdb6baf898285f6889fa0ca4e7e15c40b3855","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-06 10:16:13.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/818399/6 : FAILURE in 13m 47s","accounts_in_message":[],"_revision_number":6},{"id":"24d5d954f7092d4101039c9690eada3c5c03ee64","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 11:01:40.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7139d09d93bb4a00815a1b8e29f1f5d4 : SUCCESS in 34m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f1c2e8c3e60941f89a0d6f9e6e16b86e : SUCCESS in 44m 17s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bf2ae37cff0a056b761194de6a747dff1ac1e830","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 12:15:47.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1ee111bf26eb44cbaa27a27e8cfade04 : SUCCESS in 39m 38s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/199392f5bfd94e7aadb66b211eee8495 : SUCCESS in 2h 10m 19s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/136c41403add4595b5067220a3202f09 : SUCCESS in 1h 38m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4c1613ad874440699ab79d0ceefb7ab5 : SUCCESS in 1h 37m 01s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e6dbed94bc3e4873800156f13d9a0d37 : SUCCESS in 1h 41m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef881a3e8f414f44a8bdeba89c4eb647 : SUCCESS in 38m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/286eb364dc9c47a0a6648c2cbfde7668 : SUCCESS in 35m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a60a233c07b148b38fc6b69fb7ad6142 : SUCCESS in 13m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18c97ddc23874a319e255b6503bbc0ad : SUCCESS in 27m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c004d57559c4d00a97cdd31857ea798 : SUCCESS in 24m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b21b64347ea3488aa755eb0258c69d94 : SUCCESS in 9m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b42a6cf25ab7490ea575db2e5ca48f06 : FAILURE in 54m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f52951af5d254639a490c60269df84a7 : SUCCESS in 1h 14m 46s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e4ec1d1acd53447cab5b8576fa6fa4c8 : SUCCESS in 1h 07m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e260ec48119e49c2a2890afb4ff3c655 : SUCCESS in 41m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/962ebe66893041efa32730e7ea430521 : SUCCESS in 1h 04m 34s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c41b8b87fc7d4013994a3d962e69ce18 : SUCCESS in 1h 16m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/95504818e9b3426c84844de262b8d0a8 : SUCCESS in 1h 18m 23s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc92f0d82fea42b69cb9f7d62035f535 : SUCCESS in 55m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/42ef064c77d64fd6bd2cbd245f445afa : SUCCESS in 1h 11m 24s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2789e4dbdfa74edf93fbd24e42fb2b73 : SUCCESS in 29m 03s","accounts_in_message":[],"_revision_number":6},{"id":"16eb5dc82c9822148bcb5adfefb20abfd3199e21","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-12-07 06:15:21.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f4d63d59bdf771757386b571ad9c6648174dccab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 07:21:07.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/570ae6e778774efa83bb56399b8fadd4 : SUCCESS in 58m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/71e3be22c81a40b68175ab03737082d2 : SUCCESS in 1h 00m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"28d7c16eae3b5fa684457a59a495d92408c96264","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 09:16:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/82445c33267745f6be8fb0555f370626 : SUCCESS in 42m 41s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6a91d180cf0848628acf6ef82c3a6351 : SUCCESS in 2h 08m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/85809ad083874c10b0fb4467425af4d7 : SUCCESS in 2h 25m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/129cf4e9c8964039bbae4965b45dee5b : SUCCESS in 1h 31m 07s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/267510b313fe467aa2c572e3a41b1daf : SUCCESS in 2h 04m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/042b6d6bbf994c10b2639988c94bee0f : SUCCESS in 27m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1ba94f6d66647a5815ba88ade2d6d47 : SUCCESS in 38m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6461df152cd4caf81e3147a1be9f06c : SUCCESS in 11m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/977dfaac5a9f4b698383bc7f5b036e79 : SUCCESS in 34m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ea06e6c39b24df58f8ed96c53a0cbc8 : SUCCESS in 21m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e50007e2f5d04776977fc2e808f0a71a : SUCCESS in 8m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/67f81d53e513476fb35376b1fbfde88d : SUCCESS in 42m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/296beb4b7a584daba6db2044f8e8e3bd : SUCCESS in 1h 41m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b2cf0ad333904734b36854cecdd19cd5 : SUCCESS in 1h 06m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e98f6914fa014ef7a3b74fd7b3655f5f : SUCCESS in 41m 12s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/91a82aa85a184a54b861e66f29d1339e : SUCCESS in 53m 20s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d1db5546849c4b60bb7b7af5290a5f14 : SUCCESS in 1h 08m 12s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/608641a7c8064257ae9c18756458241a : SUCCESS in 1h 14m 53s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73762a8f1fc24d39ab30793468097458 : SUCCESS in 1h 19m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/61da8e5652f84f81a8a69ad809d64513 : SUCCESS in 1h 01m 33s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af0e2779ca994b8399ae573c5004b50a : SUCCESS in 37m 40s","accounts_in_message":[],"_revision_number":6},{"id":"8f307f3a13f383603923db7b76dfc10153572840","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-12-07 09:37:43.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"056936e5a8b261e014ec527b3771c5c05c557dee","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-13 09:53:40.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9f79c7f6bc9a0ed9216781ca32098aef6697dcef","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-12-14 10:20:27.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"7d2065fc7127623d2eb85282596cc97e0f6657a4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-12-14 10:30:32.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"f1ec80092a1a34ac383d4413cdd32b7a68ac68b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 10:31:32.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"75711c10d5f61beb7de0c58d5ae76776dccc8ba7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 11:10:17.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa0c8ef49bbb4c38b7fd3b4703f53ecb : SUCCESS in 46m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/899e66e98d594af4842d749abd29dd98 : SUCCESS in 47m 08s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ce325491f61f428be63bd4e639438fbfc789a6b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 11:49:03.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/01054dbccf1d452ba1c0cdf704e37865 : FAILURE in 19m 12s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5478f147d622422d8727b9d3e5a2a1a7 : FAILURE in 19m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c58163fdbd834490987aff96ad1e31af : FAILURE in 17m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/faf0f4f0a7b9437f84d773ac7f0922dd : FAILURE in 19m 12s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3c407a76de0b448c8c89766af5ab719b : FAILURE in 17m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7733208976fc4750ab3361b4dc8ab96d : SUCCESS in 35m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/028dbc19826747e4a9be2f23405c7702 : SUCCESS in 35m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fe79c53259b4f36af231183d6f3282b : SUCCESS in 13m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dcb31fcf2bb498ca104cbcd6cc34462 : SUCCESS in 32m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdf47248995842a0a7876f00fe0903be : SUCCESS in 24m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/445d574a9dec43a9868dc3bc999f7c80 : SUCCESS in 9m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e0aeff9cf2ed43bd8938852d0fcef39d : SUCCESS in 46m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d2a5cf2845f045efa06d98e0579138e9 : SUCCESS in 1h 22m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9ccb38a8f2694b7e9c59f93ec232a7da : FAILURE in 17m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2cf0b331a65640b48d9eb0f5ff8c4270 : FAILURE in 45m 57s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/103f71c4c6dd402eb048cae4bc4f9cbd : FAILURE in 49m 29s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8fea83f6d9384005886d917db5d49652 : FAILURE in 22m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4072b36ac6ca41808966340872405eaf : FAILURE in 19m 33s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6f68383845640c685e2b334f380e2d1 : FAILURE in 27m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8831fa34d9754a8eb6df812ffb5285dd : SUCCESS in 1h 12m 36s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0933f1e17fdc4021b35711860ac4a74b : FAILURE in 4m 14s","accounts_in_message":[],"_revision_number":6},{"id":"8c6d30680067afc5830f5469848efadf40b26161","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 12:23:47.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7282a9e79f6644d5a1cfd245a464ef33 : FAILURE in 23m 03s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f2e3ce14895a442baf769052469813f6 : FAILURE in 19m 21s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ba39b958cbbc4b99b82f0748aef15f23 : FAILURE in 18m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2e65f384d79b436e949c02edf39d0517 : FAILURE in 18m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4608da70d5894cd79b2a497be3f01edf : FAILURE in 23m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddccd92b352949d3bfeae1241c52e07e : SUCCESS in 26m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/088f819ad21e42f5b80cd1884544b428 : SUCCESS in 13m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/951b68e634c542ce804da8adc5d0c46f : FAILURE in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af3d3d09d003424e80fcf1005d85102d : SUCCESS in 27m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ae446de669b4e1b9fd7494efba88758 : FAILURE in 5m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d08b16598adb430e99d34a7f8af5c07e : FAILURE in 19m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/662984dbcbc64b439ec5d3c475a9fc0c : FAILURE in 18m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/21b6eb7d1b524f31bfc46b609be5d368 : FAILURE in 22m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6effbe9f85c545bb8886e60fed1b1c88 : FAILURE in 47m 17s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8866170071184302a847a918a1ec09e8 : FAILURE in 18m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bac1ed860194889b0cc3d07ee6cf02f : FAILURE in 3m 50s","accounts_in_message":[],"_revision_number":6},{"id":"258e753c29ef8c50d10f5d2b3a8eda9c1bbd1854","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-14 12:25:26.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/818399/6 : SUCCESS in 2h 04m 25s","accounts_in_message":[],"_revision_number":6},{"id":"77497a4338bdc825a41ef5cbfbd24b32b1c9447b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-15 08:05:20.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f65767725a8e144edb1e14559bad543e9190e029","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 08:57:38.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b133d9e623d487fbda654cb114cb444 : SUCCESS in 49m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/da1ead78a6444cd58ad9cdf287dcdb57 : SUCCESS in 49m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"aaddd338135b21b54c3a58d84a64c520d9bc9580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 09:30:47.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0234b2feda03491790565b034bffa1a9 : SUCCESS in 40m 07s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f30d223b56e4479aa83d768c29708081 : SUCCESS in 1h 01m 39s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fce10c6b73404e29857d780410d1bf59 : SUCCESS in 54m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/11630198e56a447f8b25e8e390a5d7c5 : SUCCESS in 51m 10s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5a37265aa7a445c292451acd996127b5 : SUCCESS in 1h 08m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2344e5f5d12248fcb4847fcc98524c34 : SUCCESS in 29m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8c3e0d8b0714c04a2fb17c56bb92aab : SUCCESS in 29m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f2aacbaad9a481289b01a9bd23978f6 : SUCCESS in 12m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8966c49733d42bebd4d1def938514e7 : SUCCESS in 21m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d12a8ed1e59e4661a0239ab7ce6f4aa9 : SUCCESS in 22m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4b62c853d944c23ac4424612efd5fb0 : SUCCESS in 9m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1080af4f3cbf42bf8227558803592e09 : SUCCESS in 43m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e7cfc2fdcce1450da1fced4706b16275 : SUCCESS in 1h 22m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/664a3cf2c2284a1c9274343a7551a828 : SUCCESS in 1h 00m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/023b22d9324745498c190fcb3e66c77c : SUCCESS in 57m 52s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e155aa5519054e7ca1ef39b2de86218a : SUCCESS in 1h 03m 39s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/64c5a23d499044b7a727e12651ddf6ba : SUCCESS in 1h 00m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9bfd8873bc744f72a0760d8a9a1f3097 : SUCCESS in 1h 17m 07s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/593dcb03110b4da3bc13c5ef37493887 : FAILURE in 14m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8f2a35d3c04447fda023feb589055563 : SUCCESS in 1h 01m 31s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2bb0ce8269f49ad9960141f234f39b7 : SUCCESS in 25m 49s","accounts_in_message":[],"_revision_number":6},{"id":"bc4fabe237e4fcbc737c63f3a9da77150518f756","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-15 09:43:12.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/818399/6 : SUCCESS in 1h 35m 00s","accounts_in_message":[],"_revision_number":6},{"id":"d9d42f83f8f36a867ba62512eb58fb5e69741cd8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-15 18:05:02.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"373253abf7d5f3fd9aef78ba074c6a43887a4d17","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 19:09:27.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f3229a8d09e4891b08e58de06ecdbf2 : SUCCESS in 53m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f70f45d9b8184968bccca22cc6ebb090 : SUCCESS in 53m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"10c7b92e21d131d3dbbfe199ecca576bb51dc63c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 19:34:22.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b571e7cb78514d9e8a5c9203800b6f8c : SUCCESS in 51m 04s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6d9b9bf35ef74a03af75415317acbd82 : SUCCESS in 1h 12m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d2d7e52e265042f5b60c4daa16dccb7f : SUCCESS in 1h 10m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4ee5bc79479245b8902be8a8ce313f6e : SUCCESS in 1h 16m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/55b5ff6219ed497a83ecc6f3588f4da8 : SUCCESS in 57m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1836e8032d241f88bbce0cc3c7e8138 : SUCCESS in 45m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ef923ef94b04013a92f792262ffbc0f : SUCCESS in 31m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/378680a334e144679ee0b51f6bb5390e : SUCCESS in 15m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed677ec6ce0248f1921ae006297648b9 : SUCCESS in 26m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8a06b140c76460c92853d15212a13c6 : SUCCESS in 26m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa11d42bc42c474a97fa4a4585979ed1 : SUCCESS in 10m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/844ac1bbef1a483698ccca1fed9ab787 : SUCCESS in 51m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/55e1ee43eabe45cbac0445cca0b24371 : SUCCESS in 1h 23m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fab00db4dcee486aad6fbcbedd451525 : SUCCESS in 52m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/376d126485d04193a358a00eddd77a26 : SUCCESS in 47m 46s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8b2bbe5856a64a2889f6ae14f389a7e1 : SUCCESS in 1h 02m 20s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0993be131a8e4e2d8be72bf17516a3a6 : FAILURE in 1h 03m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fd4f98728bc0418ea66fb6147aabb7be : SUCCESS in 1h 05m 04s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1701dd16b6e94d8c9eca30f9111b5a03 : SUCCESS in 41m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/076320df28b7414f8148e977e7d2a62b : SUCCESS in 1h 13m 22s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c07382365df64f6b8c08d92e9c04a081 : SUCCESS in 24m 30s","accounts_in_message":[],"_revision_number":6},{"id":"57420b58925e5b4a2b5f8676f6dfb1cb2ecdf9b0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-15 19:45:22.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/818399/6 : SUCCESS in 1h 39m 35s","accounts_in_message":[],"_revision_number":6},{"id":"12fe80fd0db2a7efb0c3a2607f162a42ec2d926c","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-12-16 07:10:32.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"92ca5798b8a6913412b4d781dcd37b7d296c4ada","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 07:56:11.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/37658cebbadb4b4992950bd7f3e30232 : SUCCESS in 42m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8d69912414d245b4a94319831d101878 : SUCCESS in 43m 20s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b349326e82b9d5f48d7afdfb6c0e40a12c689e51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 08:38:32.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8ebb1798128142969bfa2d1a7c67efdf : SUCCESS in 34m 01s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/866f36d381474bb8b7b3446304e73f4b : SUCCESS in 55m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c29eaf594d144d25a7d76126884f67d8 : SUCCESS in 1h 02m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d7a5cb21ec7f446a9753e53074c2ad16 : SUCCESS in 50m 14s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c868e6e1426e4098a145287ad9d312fa : SUCCESS in 56m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07826477516d4e86ac3bf9bb8cb650fb : SUCCESS in 24m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db05b940d78e4d7b8ce01ab71b0e5e6c : SUCCESS in 22m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37f91e91a3c04205b4e2d1c5009462ab : SUCCESS in 13m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e809400731bf46b6a53b5d19db80ab73 : SUCCESS in 26m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/419054ceb74246c4affcc921c4b888d1 : SUCCESS in 24m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a08771d165624bf9b7103e8f13762d28 : SUCCESS in 8m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8b05ec5f43064a7bbd20630ed02e51ef : SUCCESS in 45m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cc8b61c1640e43f295a3d6522e17cefb : SUCCESS in 1h 25m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/52a973320b7b4740ac86e4027e76f09d : SUCCESS in 1h 07m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/21954fb043d6477ba8b8677325f44161 : SUCCESS in 43m 15s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f51d5cdfa60c49429a4079fe0fdcd912 : SUCCESS in 55m 08s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b4d7d45f33ad456ea7f5f317a0e1f0c7 : SUCCESS in 1h 13m 53s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c954ff49332446769c37fda556124eda : SUCCESS in 53m 33s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4199bbc4978144d5a6a11ea6312f240f : SUCCESS in 52m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7cabd2354591490bb1269f3c6d407dc0 : SUCCESS in 1h 11m 12s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9588932d30f54d28b0eaca80baa58a09 : SUCCESS in 20m 42s","accounts_in_message":[],"_revision_number":6},{"id":"87cf624009b8dad6e4a5a269a7f92afad2be4c60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 08:39:34.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"73bb641bdc83e382c5a8c262750312330f4b64ec","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-16 09:01:03.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/818399/6 : SUCCESS in 1h 50m 00s","accounts_in_message":[],"_revision_number":6},{"id":"4ec117e53b65bbf6093b705189f6a5cfd8382b9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 10:08:15.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/529e6706b5aa4948b7ca903865205579 : SUCCESS in 54m 01s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6edf22b7170040619bd4293df878f7ee : SUCCESS in 52m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f0a488c46e904c23b85f5d3b62f80e34 : SUCCESS in 1h 02m 38s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a5fdfbacf53a4076a94969ae3ecc384b : SUCCESS in 58m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c566998d8b26424b98995e0c2d69064f : SUCCESS in 48m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42b7ff6ae2cf4a07be4803fe92666296 : SUCCESS in 32m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f309b858d9e44daab0e736e4c7baca63 : SUCCESS in 13m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/423e9a1cba554bdda9a3145ea4f2c71d : SUCCESS in 27m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/943b31ab8a2f42e0a23f7de97164b8a4 : SUCCESS in 24m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/966921218c334a7d886d481ed3ee3356 : SUCCESS in 10m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a46523ff6ae64c819dbfad57290c8f04 : SUCCESS in 44m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/903dcae2d0d94c86984267a490279af2 : SUCCESS in 1h 27m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b8eb35fa95374277b196e31a0d50e8d1 : SUCCESS in 1h 04m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/af6fc9937cb0481798632d822cd938d8 : SUCCESS in 1h 00m 00s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddb81b8682874fd1abfa555391c10a2c : SUCCESS in 55m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/725524255f9e41eba1fc2c2c44cd2215 : SUCCESS in 23m 23s","accounts_in_message":[],"_revision_number":6},{"id":"6f155c7c3b684f6a4d1e8873ccf15fa53a8b0ef5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 10:08:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"774672d7ee7c5467223bf188ec7c702aaac53f77","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 10:10:18.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a0d4104ca914628a1a5e444c7c867be : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"d7178927fdea70c2200ec409b4c3a94a8819a029","revisions":{"20bd9d2ae468679b9af84b9ac388ad480878caf6":{"kind":"REWORK","_number":1,"created":"2021-11-18 15:05:31.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/99/818399/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/818399/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/818399/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/818399/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/818399/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/818399/1"}}},"commit":{"parents":[{"commit":"8b2cbd38c734e2a1811b64e4b86640385c3bc84b","subject":"Add network QoS inheritance to floating IP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b2cbd38c734e2a1811b64e4b86640385c3bc84b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-11-18 14:27:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-11-18 15:05:00.000000000","tz":0},"subject":"Add upgrade check for floating IP QoS inheritance from network","message":"Add upgrade check for floating IP QoS inheritance from network\n\nThis new upgrade check will inform to the user what external networks\nhave at least one floating IP address and a bound QoS policy. That\ncould affect to the L3 QoS applied on the floating IP.\n\nPartial-Bug: #1950454\nChange-Id: I751b8baeda00ce890722d96199b19a91818b2ca0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/20bd9d2ae468679b9af84b9ac388ad480878caf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/20bd9d2ae468679b9af84b9ac388ad480878caf6"}]},"branch":"refs/heads/master"},"8cec0b37ab31ea62f8850cb7f4e698361c8e75f9":{"kind":"REWORK","_number":2,"created":"2021-11-19 09:36:40.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/99/818399/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/818399/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/818399/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/818399/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/818399/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/818399/2"}}},"commit":{"parents":[{"commit":"8b2cbd38c734e2a1811b64e4b86640385c3bc84b","subject":"Add network QoS inheritance to floating IP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b2cbd38c734e2a1811b64e4b86640385c3bc84b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-11-18 14:27:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-11-19 09:36:20.000000000","tz":0},"subject":"Add upgrade check for floating IP QoS inheritance from network","message":"Add upgrade check for floating IP QoS inheritance from network\n\nThis new upgrade check will inform to the user what external networks\nhave at least one floating IP address and a bound QoS policy. That\ncould affect to the L3 QoS applied on the floating IP.\n\nPartial-Bug: #1950454\nChange-Id: I751b8baeda00ce890722d96199b19a91818b2ca0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cec0b37ab31ea62f8850cb7f4e698361c8e75f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cec0b37ab31ea62f8850cb7f4e698361c8e75f9"}]},"branch":"refs/heads/master"},"fd0d9d977ee7b4414502e30eb66ba0ad8a3b81d9":{"kind":"REWORK","_number":3,"created":"2021-11-19 09:39:17.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/99/818399/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/818399/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/818399/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/818399/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/818399/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/818399/3"}}},"commit":{"parents":[{"commit":"8b2cbd38c734e2a1811b64e4b86640385c3bc84b","subject":"Add network QoS inheritance to floating IP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b2cbd38c734e2a1811b64e4b86640385c3bc84b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-11-18 14:27:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-11-19 09:39:01.000000000","tz":0},"subject":"Add upgrade check for floating IP QoS inheritance from network","message":"Add upgrade check for floating IP QoS inheritance from network\n\nThis new upgrade check will inform to the user what external networks\nhave at least one floating IP address and a bound QoS policy. That\ncould affect to the L3 QoS applied on the floating IP.\n\nPartial-Bug: #1950454\nChange-Id: I751b8baeda00ce890722d96199b19a91818b2ca0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fd0d9d977ee7b4414502e30eb66ba0ad8a3b81d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fd0d9d977ee7b4414502e30eb66ba0ad8a3b81d9"}]},"branch":"refs/heads/master"},"25f251dfb9527dd31e5306ea47a726d74ab9f11d":{"kind":"REWORK","_number":4,"created":"2021-11-19 12:14:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/99/818399/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/818399/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/818399/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/818399/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/818399/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/818399/4"}}},"commit":{"parents":[{"commit":"8b2cbd38c734e2a1811b64e4b86640385c3bc84b","subject":"Add network QoS inheritance to floating IP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b2cbd38c734e2a1811b64e4b86640385c3bc84b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-11-18 14:27:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-11-19 12:14:54.000000000","tz":0},"subject":"Add upgrade check for floating IP QoS inheritance from network","message":"Add upgrade check for floating IP QoS inheritance from network\n\nThis new upgrade check will inform to the user what external networks\nhave at least one floating IP address and a bound QoS policy. That\ncould affect to the L3 QoS applied on the floating IP.\n\nPartial-Bug: #1950454\nChange-Id: I751b8baeda00ce890722d96199b19a91818b2ca0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/25f251dfb9527dd31e5306ea47a726d74ab9f11d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/25f251dfb9527dd31e5306ea47a726d74ab9f11d"}]},"branch":"refs/heads/master"},"e1d1b30d5cdded5409ee707a0a4a4967fc48c9a9":{"kind":"REWORK","_number":5,"created":"2021-11-24 09:26:57.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/99/818399/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/818399/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/818399/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/818399/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/818399/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/818399/5"}}},"commit":{"parents":[{"commit":"42cfa055c278b1daa2c486a277e71c674836fb51","subject":"Add network QoS inheritance to floating IP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/42cfa055c278b1daa2c486a277e71c674836fb51"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-11-18 14:27:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-11-24 09:24:16.000000000","tz":0},"subject":"Add upgrade check for floating IP QoS inheritance from network","message":"Add upgrade check for floating IP QoS inheritance from network\n\nThis new upgrade check will inform to the user what external networks\nhave at least one floating IP address and a bound QoS policy. That\ncould affect to the L3 QoS applied on the floating IP.\n\nPartial-Bug: #1950454\nChange-Id: I751b8baeda00ce890722d96199b19a91818b2ca0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e1d1b30d5cdded5409ee707a0a4a4967fc48c9a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e1d1b30d5cdded5409ee707a0a4a4967fc48c9a9"}]},"branch":"refs/heads/master"},"d7178927fdea70c2200ec409b4c3a94a8819a029":{"kind":"REWORK","_number":6,"created":"2021-11-24 12:44:48.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/99/818399/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/818399/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/818399/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/818399/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/818399/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/818399/6"}}},"commit":{"parents":[{"commit":"42cfa055c278b1daa2c486a277e71c674836fb51","subject":"Add network QoS inheritance to floating IP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/42cfa055c278b1daa2c486a277e71c674836fb51"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-11-18 14:27:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-11-24 12:44:23.000000000","tz":0},"subject":"Add upgrade check for floating IP QoS inheritance from network","message":"Add upgrade check for floating IP QoS inheritance from network\n\nThis new upgrade check will inform to the user what external networks\nhave at least one floating IP address and a bound QoS policy. That\ncould affect to the L3 QoS applied on the floating IP.\n\nPartial-Bug: #1950454\nChange-Id: I751b8baeda00ce890722d96199b19a91818b2ca0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d7178927fdea70c2200ec409b4c3a94a8819a029"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d7178927fdea70c2200ec409b4c3a94a8819a029"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
