)]}'
{"id":"openstack%2Fneutron~818439","triplet_id":"openstack%2Fneutron~stable%2Fussuri~I57b274387505f42778f1d9d3ec2aa9feb5ae929d","project":"openstack/neutron","branch":"stable/ussuri","topic":"bug/1949607","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-11-24 15:53:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I57b274387505f42778f1d9d3ec2aa9feb5ae929d","subject":"[OVS][QOS] Dataplane enforcement is limited to min-bw egress direction","status":"MERGED","created":"2021-11-18 16:37:49.000000000","updated":"2021-11-24 15:56:10.000000000","submitted":"2021-11-24 15:53:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818439-bug/1949607","meta_rev_id":"f42f24c48b325973d8dddc8b44bce14e9c6c7383","_number":818439,"virtual_id_number":818439,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-24 15:53:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-11-23 13:14:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-11-23 20:35:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-11-23 20:35:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-18 19:21:02.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-11-18 20:43:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-23 13:14:20.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-11-23 20:35:25.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"5735d683ceefc25bc08769dd08988de7bfe7142e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-18 16:37:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b78b1aa37d56c146a455c696923ace4e902d2d9e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-18 19:21:02.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/818439/1 : FAILURE in 2h 03m 27s","accounts_in_message":[],"_revision_number":1},{"id":"68c0c9b7e2ca46438b1b89f7976a468d87708407","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 20:43:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/4f54a84f51a94f47881f3928bd396dd2 : SUCCESS in 43m 58s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/bbc1f22be94041a98928ada60acaf4f5 : SUCCESS in 1h 42m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/6a192f8525ce415584e69d358cae95a7 : SUCCESS in 1h 43m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05000053810747898234ba20dd346cfa : SUCCESS in 39m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56aec07d9bbc46f3860a6684c682be0c : SUCCESS in 15m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9735c2da035948b0a7ba4ab0b3344d8f : SUCCESS in 25m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c633746473094d288b9f66485de4ba62 : SUCCESS in 41m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f784c0a009644268e86761a7a354a1c : SUCCESS in 28m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e063e4d1fec14e2ea5491e0b760031d9 : SUCCESS in 8m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/361bd177b80a4c2fae24ecabbe7c5aad : SUCCESS in 1h 06m 30s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/551b3cf05aad4cdeb2e6470da435810d : SUCCESS in 1h 17m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/af13385220644576b2ab6aa56273856b : SUCCESS in 1h 17m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5529bfc5e480407e95b5af61e475e401 : SUCCESS in 1h 33m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f6363bd15bed486c88f7815da29771bb : SUCCESS in 59m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d2d7d8e424e3463899aa1bb127afb639 : SUCCESS in 1h 46m 55s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/fa870cd2534f4aa1a8e93c872dd588c2 : SUCCESS in 54m 22s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/41847ca71fc747c89df4241e63f87d5e : SUCCESS in 1h 00m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a4470f7677aa4daeaaae16c0bb5ff812 : SUCCESS in 1h 02m 26s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/68e95fe0798a443fb5ad7d390e5de2ba : FAILURE in 1h 36m 12s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/13a330a602924b85bcc1709a2759b2fa : SUCCESS in 1h 16m 48s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8d731e11863b442980cd9cafd4b62dd5 : SUCCESS in 1h 14m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d70ac1f1e7b74012936ada1bf0c7f29c : SUCCESS in 1h 03m 37s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/29752ca5fad7450fb52abbe4b5e1d8b6 : SUCCESS in 1h 34m 48s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c99a39f07fb4480d9325e1860db40621 : SUCCESS in 1h 05m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/333690e2e2c34d829dc3ca9ba6c3d41e : SUCCESS in 1h 13m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a5316017c49f488b82801e937df0394a : SUCCESS in 38m 29s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e018b7eb7fb34616ae55479bfd5ef62e : SUCCESS in 1h 19m 03s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b3e8476e474e4145b9f213a6d91066a5 : SUCCESS in 48m 22s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/1c3e834692e6477cbc93f7d553f8d13b : SUCCESS in 3h 31m 10s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c8d8373c0ddd4d5ca94dd8a9133d4005 : SUCCESS in 1h 18m 58s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c60dfaef636d41108c0ab33842edcbac : SUCCESS in 1h 32m 01s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/df968efc709a4f808f7bd1d092d48565 : SUCCESS in 2h 06m 50s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c530b3fbc36a430ca41ad2f6738a718b : FAILURE in 1h 52m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6cbcebeee9366b1403b28f0b0c220053d7a861f8","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-11-23 13:14:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5538c15fbdbea9d74c7570b747043983d393be84","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-23 20:35:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fd96f02706e64d1ed117403192006aba77fa0313","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 20:42:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7e8d86792998246428154db5065bb9014913dc81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 02:53:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/60f1914779b74d31a44ef1ec2ed374dc : SUCCESS in 46m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/928776398f924e8ca9158b28ccc06765 : SUCCESS in 11m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1d131e2c26147e0872a7af923eff6c4 : SUCCESS in 30m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c11e6cf0a9a4e2aa73f495c4803529e : SUCCESS in 32m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6c7a35cc5244da19aa6a4a1015f2cbb : SUCCESS in 9m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/edefab94968f4243a0cc4414488d9b31 : SUCCESS in 50m 38s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e2ed427358c8458a96f73a1103e7028b : SUCCESS in 1h 21m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8a7725f2642f4ddbb11193c7f4417b5e : SUCCESS in 1h 08m 54s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/71812185d0594a1f98beb9d0ae3fe32f : FAILURE in 59m 29s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/59f53c11c61547a79211507338112a0e : SUCCESS in 1h 00m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/73a9d7ee8f4a4575aec2145f21102dd0 : SUCCESS in 1h 11m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ccf8f0641f7f42b3ab6b8e2e8c3e5cb6 : SUCCESS in 1h 28m 29s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/662ee20577704c7c9b2a53c50989eeac : SUCCESS in 1h 39m 19s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f6d2cf351c047529e77d033e9e091b3 : SUCCESS in 56m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2649c34982814c96adf0e531310843e8 : SUCCESS in 1h 16m 06s","accounts_in_message":[],"_revision_number":1},{"id":"1bfa794ac416ade47454d2090666fdf4198ad267","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-24 08:47:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a575c8ad64ef646acf65675414daeb60b489faa6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-24 10:43:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/818439/1 : SUCCESS in 1h 55m 06s","accounts_in_message":[],"_revision_number":1},{"id":"855c679596774cd28d375907e2bf142c57c130cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 12:43:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/08984afb73d14c6caa03ee1c325df6bb : SUCCESS in 43m 42s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/74a11c42f8c94563a37ee010b1a3cb71 : SUCCESS in 1h 53m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/230ac6c31864434f87cc35bd20f7abf6 : SUCCESS in 1h 26m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3e9a59fd7c5447681762414ccc53333 : SUCCESS in 27m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a5fd48585654e3ca917b9a5f8041dd1 : SUCCESS in 13m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e42d2be8b2b4939805c0850f875757d : SUCCESS in 21m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/48a01945ef8648099fe5de4bcfb0d132 : SUCCESS in 21m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/946ee99cf53a4a989efc7ad44f9138b2 : SUCCESS in 30m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c3148a74c5747c68544d533e383b6c5 : SUCCESS in 9m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/04bc9ffdc4cc42ac819af00bc3dd02c8 : SUCCESS in 50m 43s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/85416a1fcbab457496930517377299c7 : SUCCESS in 1h 28m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7ee8c50949294fbf94e6b2346ff807bc : SUCCESS in 1h 25m 22s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/41c4f3cd3c3a4c38bc8efb12822e70d0 : SUCCESS in 1h 27m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/563cf9d9fbf54a68987e39ab0a5ce365 : SUCCESS in 1h 06m 40s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9c4190dd0a044e8c84e29b2664548c7a : SUCCESS in 1h 22m 06s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c33eaf0a6c874fce88b58abd2cf28b1a : SUCCESS in 1h 14m 49s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e89cc86d78514e18bbc7c226e78902ea : SUCCESS in 1h 13m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2a0f32e2ab4143baab8f5d819d18c145 : SUCCESS in 1h 08m 50s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/472fe0113c1541c68ffaef14851d6b25 : SUCCESS in 1h 49m 47s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/566c5ff7d044413bb7a350b947bce8d5 : SUCCESS in 1h 40m 45s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b0cf3570a6dc44bb98339ccda7ce5546 : SUCCESS in 1h 05m 27s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/300388dd4f8e4b4a885ee55c9e2bc9e6 : SUCCESS in 1h 04m 36s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/90f10f22014f44d585b909956627ae7e : SUCCESS in 1h 34m 43s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40656758bd7d4d8a877e8c5ed438b07d : SUCCESS in 1h 06m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/35e84c1bda054da48c949f131239aa9d : SUCCESS in 1h 23m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/58d502cebcda47e389ce7afd3d194f6c : SUCCESS in 42m 00s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/fece516c23354579936bdfb5de14c4e1 : SUCCESS in 1h 25m 44s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4be0cc19ad8e43d093950eb2f9950026 : SUCCESS in 1h 01m 51s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/204ef4e095be4cb59b421312d8af17b9 : SUCCESS in 3h 48m 33s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/57d1fbb38d6e43bcb7b35297c0a5f193 : SUCCESS in 1h 37m 45s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d22ee8ea7b254edcaa4a4670d934812c : SUCCESS in 1h 51m 22s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/dcb299be7c944d4ca09f6e67e93d2939 : SUCCESS in 2h 10m 25s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/999f15121b644408adf3987520b81371 : FAILURE in 1h 34m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4bcf55c140a393042a721fa5063133b2dc8c4330","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 12:44:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eaa8d751556c26c91d2768f95684fe963b58aeb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 15:53:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/a544d13be0774549a0da8dd413af1efd : SUCCESS in 42m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f1b6ba52a474d81a56bcbae76bf7c0b : SUCCESS in 12m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/773f142dbbfc4adcbc8ff96b8fc815bc : SUCCESS in 20m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0900eefd9d584926aea3b9cacbf202f3 : SUCCESS in 22m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08598824648248a886dcc1e29d404614 : SUCCESS in 9m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fcb4aeb489594637b8395cb9198493a5 : SUCCESS in 49m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/782a350eb07741a88deab20d7db74972 : SUCCESS in 1h 17m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b1d32019fc9b43a7aa78cdb555a29629 : SUCCESS in 57m 39s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/69a09c44e6174f6f856077d4da789735 : SUCCESS in 59m 05s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cacdadce69a44861b989147fc4817af2 : SUCCESS in 1h 03m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9ddf234240e345ab93e3fd4ceddc4194 : SUCCESS in 1h 04m 07s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1d7fa19543fd45058a9ab6f0de530371 : SUCCESS in 1h 07m 17s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9169ef2c37814e0bb48ed4b0e6c02c33 : SUCCESS in 1h 24m 51s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c77eb7fda2645c09e56c25d938e1b1c : SUCCESS in 1h 01m 36s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/171380c4c9024223a49bf4d5b8b756d7 : SUCCESS in 1h 29m 40s","accounts_in_message":[],"_revision_number":1},{"id":"3d040003db153a093393e0a1e995a47d9e989151","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 15:53:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f42f24c48b325973d8dddc8b44bce14e9c6c7383","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 15:56:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c2a78da70cb4caabb21c8f1a511b109 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a81360ff3f85cc778b1079d791079b8ac61e18c","revisions":{"7a81360ff3f85cc778b1079d791079b8ac61e18c":{"kind":"REWORK","_number":1,"created":"2021-11-18 16:37:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/818439/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/818439/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/818439/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/818439/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/818439/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/818439/1"}}},"commit":{"parents":[{"commit":"ee67324c17c7d46bf110da977d999796951a6a2d","subject":"Merge \"[DVR] Fix update of the MTU in the DVR HA routers\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ee67324c17c7d46bf110da977d999796951a6a2d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-11-03 16:01:31.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-11-18 16:37:18.000000000","tz":0},"subject":"[OVS][QOS] Dataplane enforcement is limited to min-bw egress direction","message":"[OVS][QOS] Dataplane enforcement is limited to min-bw egress direction\n\nOVS QoS extension only enforces dataplane traffic shapping for minimum\nbandwidth rules with egress direction.\n\nConflicts:\n    doc/source/admin/config-qos.rst\n\nChange-Id: I57b274387505f42778f1d9d3ec2aa9feb5ae929d\nCloses-Bug: #1949607\n(cherry picked from commit 5627c8713723c6f25ac836f4a03a00741eb809fd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a81360ff3f85cc778b1079d791079b8ac61e18c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a81360ff3f85cc778b1079d791079b8ac61e18c"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
