)]}'
{"id":"openstack%2Fneutron~819049","triplet_id":"openstack%2Fneutron~master~I00fba2dc4395c0a8cd4631d4c2c71d9c3dc429e9","project":"openstack/neutron","branch":"master","topic":"neutron_db_for_ovn_agents","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-11-24 17:30:42.000000000","reason":"Rodolfo Alonso replied on the change"},"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2021-11-24 13:16:23.000000000","reason":"Lucas Alvares Gomes replied on the change"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2021-11-25 10:24:22.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2021-11-25 10:24:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I00fba2dc4395c0a8cd4631d4c2c71d9c3dc429e9","subject":"Properly clean up ovn-northd in functional tests","status":"MERGED","created":"2021-11-24 03:29:24.000000000","updated":"2021-11-25 10:26:46.000000000","submitted":"2021-11-25 10:24:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819049-neutron_db_for_ovn_agents","meta_rev_id":"d660ceee0d80285d30bfd0388266bf986416f43e","_number":819049,"virtual_id_number":819049,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2021-11-25 10:26:46.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-25 10:24:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-11-24 17:30:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-11-24 13:16:23.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2021-11-25 04:49:35.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-11-24 17:30:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-24 03:52:20.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2021-11-24 03:52:20.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2021-11-24 03:52:20.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-11-24 05:25:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-24 05:56:05.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-11-24 06:38:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-24 07:32:17.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-25 04:49:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"b6c4dbb6c80197de6f1a5f1a2e37750e66515883","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-11-24 03:29:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec8d773017010c43c9eae2008cf715258f0262e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-11-24 04:21:26.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"6a12398d270f2ee7996786e223f88572c230280c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 05:25:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f9b853eadaf248b18a4a6b14ebdd6b91 : SUCCESS in 53m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e842fe81685546aa9bd2139517a0ede1 : SUCCESS in 56m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"912a8d44c3de8277f31133dbf629b7c2c7f5bca2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-24 05:56:05.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/819049/2 : SUCCESS in 1h 33m 57s","accounts_in_message":[],"_revision_number":2},{"id":"2b8f82938587d4067065c0f08624bf1ad90b2422","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 06:38:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7404417b1436410aa6508d73567668f3 : SUCCESS in 43m 25s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/499e97c2232d4819b497ec4917404dd3 : FAILURE in 27m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6a020b7b106a4478a964c8b179ad1763 : SUCCESS in 2h 15m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c44a8f8f235c4238a7985bd369f4185a : SUCCESS in 1h 57m 39s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/16353f1a893e43f1ae426a6bd3ad2d92 : SUCCESS in 2h 05m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1ed8ced89eae43e1bcccb160ea132481 : SUCCESS in 40m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/645a13ec20684b62b9af14844b15cf11 : SUCCESS in 51m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/441134a628a045419074369942dee97b : SUCCESS in 58m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20b7283d938d4be18b1b60757f501012 : SUCCESS in 12m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cc03165aa924fe79b3f7f9680fb1527 : SUCCESS in 46m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5df967812424417ba5abf9264ce3d2bd : SUCCESS in 41m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fbc5050230fb4af1a99fd14a4a338bce : SUCCESS in 45m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26919f60739444d5add9b9f4f0409149 : SUCCESS in 8m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2cdd774571c34f02ad4efda57d1a927d : SUCCESS in 53m 20s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84a6c8bbd1c049e2bdf695ac12fa674a : SUCCESS in 53m 43s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/755633604a4f44cc8b25d49267b2f1a2 : SUCCESS in 1h 01m 24s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/60060bda450a4654a039cbc31ffa02a6 : SUCCESS in 58m 39s","accounts_in_message":[],"_revision_number":2},{"id":"9984f97abbc0f7e52b0b1777d76a58d13b22a101","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-11-24 07:32:17.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2021/Arista_Tempest_Testrefs/changes/49/819049/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"c3e031ee2a75961615ea25e39eae4e9a0e2cc022","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-11-24 08:24:16.000000000","message":"Patch Set 2:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2021/Arista_Tempest_Testrefs/changes/49/819049/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"8eb71eaf018558cbb8b7b435b1d7148fc12d28a5","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-11-24 13:16:23.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bf684325d6dd73ad330f3b2d2f7ab912581ab790","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 14:30:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc10a8501d8d4a85aa54cec5638d0d87 : SUCCESS in 58m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0c88cad58b846cc911db89196533e19 : SUCCESS in 54m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"715b322fc1f667f5b8bb8357f3fc434e79a6e6df","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-24 14:52:17.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/819049/2 : SUCCESS in 1h 35m 21s","accounts_in_message":[],"_revision_number":2},{"id":"bc551c4e5979df645176a2d2951c265d281d1728","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 15:52:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/58ed8a2e93dc4d6ab544e3645d94f059 : SUCCESS in 50m 30s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/db02877c99524dc19d21b6a177b9a540 : SUCCESS in 1h 59m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/acc98114e2624b98921bba8a045eb35b : SUCCESS in 1h 51m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4baa4448c5e44155b01a3c06af5ff925 : SUCCESS in 1h 29m 22s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4ebbbddba3694c3ab9d4bf947cf6d15a : SUCCESS in 2h 32m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c9da2b2a24f2446a90f1472baa53af50 : SUCCESS in 41m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0082b8b90d9445b8837709c46b0eb37 : SUCCESS in 47m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ebae2522fe14baab5e98a6d80cf939a : SUCCESS in 57m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5af1c30a62242bcacd5f84c874f650c : SUCCESS in 14m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9af3391d3fc84f58bded3febd8b37c70 : SUCCESS in 44m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff807d47dec0459f899225ca9c17af84 : SUCCESS in 37m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e54ce645c5f44aea40b22bdfec43bfd : SUCCESS in 37m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19a06f12b0914be4b9e1421da0d0374f : SUCCESS in 12m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e0846182e54f43548f950a08898c72e7 : SUCCESS in 54m 26s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1113a66700e4e1facbb96f2806e12b6 : SUCCESS in 1h 11m 51s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/009d241b14f34c9eb43c89cca4850e0e : SUCCESS in 49m 59s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9a5b87ff2f6d4d8f8976288d23f98c19 : SUCCESS in 51m 24s","accounts_in_message":[],"_revision_number":2},{"id":"706daba3291a153e4b5f3237845c3777f2993960","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-24 17:30:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"231136cb7c36a17808d815d5c97b76498fb46924","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 17:31:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"320ffd8025773acc3a739488b5a70e196f8c4c1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 22:20:49.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3da60d592c8048ea868b81d875bb8279 : SUCCESS in 52m 15s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/423db50b1c1745c285d3dc8d0f8ee498 : SUCCESS in 2h 01m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f89372c68cb44121b7ce95fc80d8e94a : SUCCESS in 2h 32m 29s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/11cdbea08bf140ffac1da2d80dac8be0 : SUCCESS in 2h 37m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d9f207d5f06947bbb20d2e454a4ae6fa : SUCCESS in 1h 38m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9adf042aaf1e41dca355974aa017b673 : SUCCESS in 46m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c62aa5e66dc74779965139fbcf6f1b9d : SUCCESS in 12m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81973f1a96fe462dafe06ab688828fc7 : SUCCESS in 41m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2078fd0f003d42f39a2ea830b79b0e5e : SUCCESS in 40m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46825e60c202492085df2cbdbd926175 : SUCCESS in 9m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e327a549e86a456f974b4136677fd00c : FAILURE in 50m 27s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1f14b216e5149c388bf534098e050a1 : SUCCESS in 1h 15m 21s","accounts_in_message":[],"_revision_number":2},{"id":"f2c2d345bc2417aa49eebace4e70117a909b2a7e","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-11-25 00:15:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0d2e65f14c323ad9cea4c7f46e6b13fe2bafc7e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 01:17:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b410c519da314bab886d05abad327c62 : SUCCESS in 45m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5655e96d613f4209800de0ae0b772371 : SUCCESS in 1h 01m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4c5b29617085a6de4ea0edd0c57ec002e74e48db","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-25 01:53:45.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/819049/2 : SUCCESS in 1h 37m 26s","accounts_in_message":[],"_revision_number":2},{"id":"71a0209b1cc2625c61391cf4e314f7d62f9890ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 02:24:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9845c328c1f8421aa6e64fe8d27227db : SUCCESS in 39m 18s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ec109240d9f9429fa3f5ac43849ef161 : SUCCESS in 2h 07m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/51a91f9bc7654194b6030f47bad8137b : SUCCESS in 1h 35m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f837f8b4cefb4f6099895590786db69a : SUCCESS in 1h 30m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/08cecd6b7846492495496a664f33e161 : SUCCESS in 1h 57m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/51617bc7d8a640928389b42ae9796b2f : SUCCESS in 41m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5a1944a8b4a47c8b7619f33674a0008 : SUCCESS in 43m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8addda2dd5994386899924e0f53cbd30 : SUCCESS in 55m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eaf8a2f760546b6a6966f17a4c488ff : SUCCESS in 10m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f0201a1d26a481aae45586e165b6d39 : SUCCESS in 38m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c55fea6dede7419aa91463044e97f1ee : SUCCESS in 37m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a40948a90724d27bca9d47cd0b4bf66 : SUCCESS in 15m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/10a145254ece4dcdbc12f2ec5f6c7d06 : SUCCESS in 47m 22s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad89dd6ddcb44e6da94f1d61f5dff025 : SUCCESS in 51m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9de278931e5b4ff6996ef1c49d522ee4 : SUCCESS in 38m 58s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d61a4e6b5c348a7a9ee22d61cb39a58 : SUCCESS in 42m 53s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c125209db4ac46988ffe71d7882ee279 : SUCCESS in 1h 00m 20s","accounts_in_message":[],"_revision_number":2},{"id":"1c8ec554d8a8812e9e1f04ab6c4eb255b75aac1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 02:36:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0edacf14d1d58f9d2020d644cca24352065c9697","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 04:41:56.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/515e404307f348949a4288ab6ee0cd22 : SUCCESS in 51m 41s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d11214f18b0943fc85c11b57bb9f2589 : SUCCESS in 2h 02m 47s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d253fda29c5a49299f0c5e12045f0ff8 : SUCCESS in 1h 56m 04s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e39635cb960b442899c8b252effb64a3 : SUCCESS in 2h 02m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/87a7df6a98e248eeaef748e55dae7ab8 : SUCCESS in 1h 47m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ccc8a3581bf4b74bb68d1479cc5b6e9 : TIMED_OUT in 1h 01m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbad2e991d5b4919bc0850177447aa83 : SUCCESS in 13m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2201653cef514f488d12ba821897b6ca : SUCCESS in 39m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/297e3b86e3ee4c96b4452edf3cbe1ab7 : TIMED_OUT in 41m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/739f5bbb39144386a1996464a40d3119 : SUCCESS in 11m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/88062f2e2dec40a997749ce5fd1639d1 : SUCCESS in 52m 07s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cad50afae574742aab36ed8a16c88d7 : SUCCESS in 1h 04m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef92d26a960a439192ae4d8e0ffde0ec : SUCCESS in 40m 57s","accounts_in_message":[],"_revision_number":2},{"id":"7e1c56cfc4129159a170abb2e2f14e348b71f4f1","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-11-25 04:49:35.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"caf19214f800e5bedd5892833711f10802d2b2fa","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-11-25 05:09:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a83f220219fc3bd54e6d170c1515797074d408b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 06:13:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/71180ce290474506b004c476d8269160 : SUCCESS in 56m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6ceef9395f9245649238342a9d7a4694 : SUCCESS in 1h 01m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ff81bc6f2589f4e3ac16690cf9a69b7fc435b573","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-25 06:43:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/819049/2 : SUCCESS in 1h 32m 53s","accounts_in_message":[],"_revision_number":2},{"id":"5e3ecf7fe0c77d5dce6f34f3e145e6a1095689f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 07:33:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8ec20c50a4274fed9f052ac0f133aa9b : SUCCESS in 37m 40s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d6c2b90d1a09486dbb1de813ce97abcf : SUCCESS in 2h 18m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/139c233357f44590b86b8f6929dfbc1a : SUCCESS in 2h 21m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c16dfab54a114cce834f5e335ea5b266 : SUCCESS in 1h 38m 38s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/089b07a4e49249cdafba55de591a1ea8 : SUCCESS in 1h 45m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/68547b52b4f34eee8be48f8ccb37408d : SUCCESS in 42m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d476ab46fc264feb8f2cc0db476848c3 : SUCCESS in 55m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00bdc7f9a37f4da483655e52d56f1785 : SUCCESS in 57m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea01149f485b4b1a8e290f21ceda98be : SUCCESS in 11m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e6bc34ae678462493a62270168f1bc4 : SUCCESS in 37m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/504b6b3fe46a45bdbadbbea7f10d8eb9 : SUCCESS in 39m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f69bce31f8146009cd5c491adb5f8f4 : SUCCESS in 9m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/319f3022e864496785dc745d46e92eb3 : SUCCESS in 55m 02s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a0dbb6a6b004047839228e37c5812e4 : SUCCESS in 56m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9923788e88984f08999e626f531a9fea : SUCCESS in 44m 46s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5cff6a5796144578d10896aca564046 : SUCCESS in 55m 19s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/87ebc162fc6b4c388d98936d236b7e26 : SUCCESS in 56m 49s","accounts_in_message":[],"_revision_number":2},{"id":"8dca146b9842f8c5cbb5e9f786062e7a5bcb69aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 07:34:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a67965bafc72c642c0d8547bd28d7d2f07d21138","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 10:24:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f7334e6d50f44a36949be4f15fb5036e : SUCCESS in 44m 10s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fffa657104b943df9c1b10ec42a7ae87 : SUCCESS in 2h 44m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0751ea2c6e364ae5882d31772987d42d : SUCCESS in 2h 05m 53s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/42ed8d1b1559458397167802397e0054 : SUCCESS in 2h 46m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/936071de1b424f44b20db8d7082a9e32 : SUCCESS in 1h 45m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8275e417c7474522b202934f37bee1df : SUCCESS in 55m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eedbf8c232514bfbbd71e3b5cf660a56 : SUCCESS in 15m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2406a1314282461994b739944f555226 : SUCCESS in 40m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9767f039cf4746c9b5491169bcecb3d8 : SUCCESS in 39m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d7024baaf9547cebbc975bd20785989 : SUCCESS in 9m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b15ab3f5aed466a8ce0a5b10c589084 : SUCCESS in 48m 30s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f45bd0edc27e47bab45a6c6ced91f8e8 : SUCCESS in 1h 12m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/010c6a3880ca49efb0a85bb039f9208e : SUCCESS in 39m 11s","accounts_in_message":[],"_revision_number":2},{"id":"d15cb921389c8121d4182543fb3011fb63e8f87a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 10:24:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"406bba484d1892289126f29c27758105fc664859","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 10:26:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4a1ce3b02c749359d665338730d7b66 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"74aa86a976c5bcf42b2779e5c557d7f7f4fdac9b","revisions":{"c0b0d48122bfbe33f249dbcf12d07ba53e11430a":{"kind":"REWORK","_number":1,"created":"2021-11-24 03:29:24.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/49/819049/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/819049/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/819049/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/819049/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/819049/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/819049/1"}}},"commit":{"parents":[{"commit":"4b1bf15f7bfae842669b61400c5469a6d7269eb6","subject":"Merge \"[OVN] Update check_for_mcast_flood_reports() to check for mcast_flood\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b1bf15f7bfae842669b61400c5469a6d7269eb6"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-11-24 03:21:27.000000000","tz":-360},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-11-24 03:21:27.000000000","tz":-360},"subject":"Properly clean up ovn-northd in functional tests","message":"Properly clean up ovn-northd in functional tests\n\novn-northd.ctl is a socket, so os.path.isfile() returns false and\novn-northd is not properly killed. Use os.path.exists() instead.\n\nCloses-Bug: #1952023\nChange-Id: I00fba2dc4395c0a8cd4631d4c2c71d9c3dc429e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c0b0d48122bfbe33f249dbcf12d07ba53e11430a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c0b0d48122bfbe33f249dbcf12d07ba53e11430a"}]},"branch":"refs/heads/master"},"74aa86a976c5bcf42b2779e5c557d7f7f4fdac9b":{"kind":"NO_CHANGE","_number":2,"created":"2021-11-24 04:21:26.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/49/819049/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/819049/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/819049/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/819049/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/819049/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/819049/2"}}},"commit":{"parents":[{"commit":"4b1bf15f7bfae842669b61400c5469a6d7269eb6","subject":"Merge \"[OVN] Update check_for_mcast_flood_reports() to check for mcast_flood\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b1bf15f7bfae842669b61400c5469a6d7269eb6"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-11-24 03:21:27.000000000","tz":-360},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2021-11-24 03:54:03.000000000","tz":0},"subject":"Properly clean up ovn-northd in functional tests","message":"Properly clean up ovn-northd in functional tests\n\novn-northd.ctl is a socket, so os.path.isfile() returns false and\novn-northd is not properly killed. Use os.path.exists() instead.\n\nCloses-Bug: #1952023\nChange-Id: I00fba2dc4395c0a8cd4631d4c2c71d9c3dc429e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/74aa86a976c5bcf42b2779e5c557d7f7f4fdac9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/74aa86a976c5bcf42b2779e5c557d7f7f4fdac9b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
