)]}'
{"id":"openstack%2Fneutron~819201","triplet_id":"openstack%2Fneutron~master~Ia66475b377c683b115c8bd9a77c72d955db60fdc","project":"openstack/neutron","branch":"master","topic":"yoga-python-testing","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-01-27 17:16:07.000000000","reason":"Rodolfo Alonso replied on the change"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2021-11-26 05:40:14.000000000","reason":"yatin replied on the change"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-02-15 13:54:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia66475b377c683b115c8bd9a77c72d955db60fdc","subject":"Updating python testing as per Yoga testing runtime","status":"MERGED","created":"2021-11-24 23:43:35.000000000","updated":"2022-02-15 14:05:12.000000000","submitted":"2022-02-15 13:54:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"819201-yoga-python-testing","meta_rev_id":"786a351f4e678e7f1fe001e532288bfa2fca2034","_number":819201,"virtual_id_number":819201,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-15 13:54:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-01-27 17:16:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-12-24 08:12:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2021-12-14 09:31:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-12-16 11:52:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-02-04 07:24:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2021-12-24 08:15:12.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-01-27 17:16:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-25 00:48:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-25 01:49:05.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-11-25 02:09:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-25 05:46:17.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2021-11-26 05:40:14.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"CC"},{"updated":"2021-11-26 05:40:14.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2021-11-26 05:40:14.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2021-11-26 05:40:14.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2021-11-26 05:40:14.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2021-11-26 07:57:22.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-11-26 08:45:45.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-12-14 09:31:41.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-12-16 11:52:09.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-02-04 07:24:51.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"53fd44eeb429b62de1284c75698c70a2fda1b534","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-24 23:43:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9b8568cd910b623812baac2bc070f952964e4a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 00:48:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d7987654a0842bcb3fd4a784bc07ba5 : TIMED_OUT in 1h 01m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/37657c6630424eda829b9bebcd3bab93 : TIMED_OUT in 1h 01m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c89b5743329f10208a9a51b70a21f597b839235b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-11-25 01:49:05.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/819201/1 : SUCCESS in 2h 04m 33s","accounts_in_message":[],"_revision_number":1},{"id":"7b377837ce892e923c66235b71fa12a368fc0f5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 02:09:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9c463bad34344243b3d186324529ca91 : SUCCESS in 2m 44s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f8546487669248949022aab0be0f6396 : SUCCESS in 40m 17s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6a1a27a8cac149ea865e1e493452f7c2 : SUCCESS in 1h 50m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/64f2a38f9d2f40438371e42555f31373 : SUCCESS in 2h 22m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f5728d806ac44500b5ab54b1235ab733 : SUCCESS in 1h 39m 26s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6cd9871c9dfe477297c94b51d2f54302 : SUCCESS in 2h 19m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3ff4783b1b8347b28601a5f3db85d2ed : SUCCESS in 33m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/934d78deb8af434c86106a5bf87724f0 : SUCCESS in 51m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c86956bb1e74a348efeefcf75502e76 : SUCCESS in 47m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d776fd7145c4690800a3cec0801e73a : SUCCESS in 11m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/149810b977544290b0a86344d689e1a2 : SUCCESS in 36m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f708539e77441cfb8ba6084db669855 : SUCCESS in 35m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dae3565b82ed44f481f4dec607972106 : SUCCESS in 8m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ae76c0e647ad42048f7459eb4bf13654 : SUCCESS in 50m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/31e97914ca2845689f980149084218f4 : SUCCESS in 1h 26m 50s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5f3f0fc8d3d5450e8e51e1c14ccc968d : SUCCESS in 1h 00m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/41a9f10d962444ffa1487962b30dcd8c : SUCCESS in 51m 31s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b697218787a54bad9b35eb07a5303ea2 : SUCCESS in 52m 02s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b5ea129dbe074b6a941e3d73e45b4f46 : SUCCESS in 1h 22m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b131ec88a9a74acab35f3a5d66666b54 : SUCCESS in 1h 12m 57s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cae59841e9c4d09a5e9ed85374fa06c : SUCCESS in 51m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fcd83d64adac4854a8c2134b79c017b1 : SUCCESS in 58m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f8fcc19ea454f9c91ad06719dc06b63 : SUCCESS in 54m 09s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fbadce48be0449808d39a4356a4f5d90 : SUCCESS in 55m 52s","accounts_in_message":[],"_revision_number":1},{"id":"799d4f6498eace4cf981116109d0d1cd8b9669ff","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-11-25 05:46:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df9f7a75d20b49102a1173bc9f851323445c87f3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-25 16:11:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f25255242a8fcee945c976745b2316cc52abb262","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-11-26 05:40:14.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ef0224cf553a64d238b55c2592038df0b297bb55","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-11-26 07:57:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"39601d9650bd030906f8f7827c676a57e2654b09","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-26 08:45:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"14dd217c800ca7da2660082e4626895e4a65b755","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-11-26 08:59:59.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"7d56c4203d867476cc258e24211700dcaabe1d11","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-26 12:50:17.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"292f54130c9513a585dd450af201298bbd6fd82c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-26 15:39:39.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7b24f8a69920d084b4801a909d768b08f4523111","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-12-14 02:13:45.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"083db4d08cb924d751d0fc6108d076760ce7acb2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 03:33:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d77d00711b741a492909da24d5d8aa0 : SUCCESS in 55m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3f59ecd1cb9e4a3092e4900926b1ea7b : SUCCESS in 55m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"439761c4ceddaac29d6f75d6436e70ca20e26cea","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-14 04:05:57.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/819201/2 : SUCCESS in 1h 51m 30s","accounts_in_message":[],"_revision_number":2},{"id":"9ab4a9e3b6f0dc849dbcdb352230f0cb0181bf04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 05:44:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2523044245b74e38a06003a46dff7fc7 : SUCCESS in 4m 43s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ba8e7a1bc5f34274ba53ea7c04691c3a : SUCCESS in 40m 50s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/857b2c7a03214ce5b8fd7f190739271f : SUCCESS in 2h 13m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/638cb90f163a43f0b2bd9044b177ff5f : SUCCESS in 2h 37m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/95244d0b8d7248339ff236ac8aa32f08 : SUCCESS in 2h 01m 48s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/21ce5fed35114595a4bf6c424c0152c6 : SUCCESS in 2h 19m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c023de871fc247b5b9989b61a2f9dc9c : SUCCESS in 39m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c671f9d365f4cf996b3047cdb9095b9 : SUCCESS in 39m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2692f9c83dc3430291a61c4060dfac35 : SUCCESS in 33m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f9a6aabf2654a8cb3f0ae4077a01fe4 : SUCCESS in 15m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2514e9e7fc0a4804826854f8ab1e9331 : SUCCESS in 27m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/531ca15b3c724c349e0675c356634749 : SUCCESS in 25m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc90344ba5e84c81b173ee29b32a850c : SUCCESS in 14m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa9ededa1e5e4fc293e8d88d9e007d33 : SUCCESS in 44m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7e1905c9b43b496f87f63521459bdcd4 : SUCCESS in 1h 27m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f4abe7b27d4b4b958711a12ad7f95d6b : SUCCESS in 56m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/baca62ebf48246ac98ca4fb37aa5e6ea : SUCCESS in 1h 02m 44s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7ffa2631a3334a4fb8677681b6a52302 : SUCCESS in 1h 00m 00s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/aa0d9b72c51b4e92841ebbd5c1c58491 : SUCCESS in 1h 15m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5c99cd9471374898b9d59a090712373a : SUCCESS in 1h 19m 19s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/505b81ac833c433c866261fd8193dcf5 : SUCCESS in 1h 03m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/37f3fc35d6f4430ab9b70c1608b0a79e : SUCCESS in 1h 44m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e6ea9649e724550c96b2a5039ea2282300a9832f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-12-14 09:31:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"636e14af55012ac151467286895a530f348321cc","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2021-12-16 11:52:09.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d589d67a0526dbb160b344b6f399cb39e155e1c9","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-12-24 08:12:47.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"242bbf0321be1904e8f733cccbf4a2bd2bba8e9d","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-12-24 08:15:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ad1257801910749165c7d2183e865b0bed8c1b07","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-27 17:16:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1ae835d82484718d4b58c851146c04c3b9a22523","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 17:18:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2b0f2c25facd2624b58561d19d71db896dccf1cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 19:45:43.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f7f891305b834752b9e08a74f4f65a1d : SUCCESS in 4m 14s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9bd64114eb79479aa66c3f9fb747bf71 : SUCCESS in 45m 33s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4cca00f18cdd4c6c8c13ed4368ae53f1 : SUCCESS in 56m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d24f858650c942328812617efd260f4c : SUCCESS in 58m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3c5c72c0bb7c4118858de232d33fc14b : SUCCESS in 47m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b379f233b1ec4f44857f08e34728fd2a : SUCCESS in 54m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f53312fa8dbf402a9c7e15af0efa7123 : SUCCESS in 41m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b903cbd504e94369a5ceb14a08971eb3 : SUCCESS in 15m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d07a52f6c6b74a209ea710290f0b96e3 : SUCCESS in 28m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c54ca5ac44444c09aeb96ec4bfbc5840 : SUCCESS in 28m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6ce6c58fbef488a833d66e4b25dbd4f : SUCCESS in 12m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a042d0bb3b054e429956b8e3c5ee6b5a : SUCCESS in 55m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/45c0c55272be48318d9f8245975a9fd2 : SUCCESS in 1h 37m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/43e7711ef92c4f9ab2c0a2bf73b935bc : FAILURE in 1h 56m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/854a63d62281493bb7aa23cc8d19180a : SUCCESS in 1h 01m 06s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73fae49c9c3646a48e50d44001240dbe : SUCCESS in 1h 09m 54s","accounts_in_message":[],"_revision_number":2},{"id":"df843e4ed6d95b90a963ad48cade4de4c7d9dd2c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-01-28 01:11:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"170d7cf94105448b7587423de3503cd18f9cbb42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 01:53:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f65d69bc28f41b499e658a9176acfc0 : SUCCESS in 32m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db7950ed8085445e9eb1fffbe639ebb2 : SUCCESS in 34m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3bda9981f136e443ed0eb451c17099282cb901b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 02:56:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b0c80129a012439085bab2de34052605 : SUCCESS in 3m 09s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8a0c61460cf140edabadd0203c981349 : SUCCESS in 44m 23s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d62c3eb262f0443fb7c46c62a381b62a : SUCCESS in 1h 19m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3275b0d0f1b343ab8e2571b016711228 : SUCCESS in 1h 15m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/36018cf80e6d48f9907c69eac2bf1827 : SUCCESS in 57m 00s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/57305dc89f3145e7a62985a07fd8054c : SUCCESS in 1h 11m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1b7d04d878914542a7e30bcc44d66cc7 : SUCCESS in 31m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3367395c9a047e29330e1e8430e6871 : SUCCESS in 27m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94564328f91e44f9a0de0be53e39c011 : SUCCESS in 23m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c259bc4cb7d343d1898adb2161dc5cb1 : SUCCESS in 17m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc5cc741fc054a8d987a9cdc12276043 : SUCCESS in 34m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b9e79d900b4450e8d63de51cfa99a86 : SUCCESS in 26m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/600a797a7bf141bf9446380ab93b07b2 : SUCCESS in 10m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1439a93dd7c3405d8999ae0dce041984 : SUCCESS in 39m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d853396b535d4e5aa41e839c1c01d9c8 : SUCCESS in 1h 26m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d43d85f6b5d04dfe8520d2c4ec91cbf7 : SUCCESS in 1h 11m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e3eb5ae08ce47a2a14b4b6f327f8b4c : SUCCESS in 53m 23s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b03d8949f88e4e91a38ad688546dfa63 : SUCCESS in 1h 21m 29s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ccf24707366440539a34dc13f3f38bd6 : FAILURE in 1h 42m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6cd277c66f2f42e7b1d1946a3f15de97 : SUCCESS in 1h 01m 45s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/208875a8e5b1486d812defa285608494 : SUCCESS in 58m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5b10212b9aeb4b00a92d2f51ac3412f7 : SUCCESS in 1h 07m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"457960ba359d64bd77b74ae7cb1465ba6998f021","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-01-28 03:17:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/819201/2 : SUCCESS in 1h 53m 10s","accounts_in_message":[],"_revision_number":2},{"id":"4fb563a7a7e0c5c6787386adb1db5808f5a53e82","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-02-01 11:32:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6360d087f3a7def0a53b1eb5c180d7858ef0f3e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 12:57:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/11f01029807c43d3ba08537a0f416a67 : TIMED_OUT in 1h 21m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d267625b9e53479093b49364ad237fe9 : TIMED_OUT in 1h 21m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"80727a5c4e3a558e2c182aaefdb0f40966bda37c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 13:12:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4a0bfd7e351642c585ea428638ebd923 : SUCCESS in 2m 29s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e1f272dfe40546ffad89b9af7336a8b2 : SUCCESS in 50m 42s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2f1350f4ae41447e912d077e52181e3b : SUCCESS in 1h 01m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5ae3916c0e034bdfb979c0cdef43d7c4 : SUCCESS in 1h 01m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4df242140a964dd698004b88eeff840a : SUCCESS in 53m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3468c70c14d348bf9c9d9fab7f2ab12e : SUCCESS in 46m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6f1b81811d4b4e8eab30500df85e8874 : SUCCESS in 29m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8297e6d7d6d426da841e59d59d8f70a : SUCCESS in 33m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c77e79ea9c54866ae40a935915afec4 : SUCCESS in 25m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c20a47a9ec5148959024c00926319ea7 : SUCCESS in 14m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae734579c8424387b8148f1610059915 : SUCCESS in 23m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19ed6957797d412cb4f4365bc428239b : SUCCESS in 22m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9ae5d72110a434aa72e8a1c64df9ca6 : SUCCESS in 9m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7504d2288e6b4bc391e7fb797262fdbd : SUCCESS in 47m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c0693922fb424d6895779071ad90784d : SUCCESS in 1h 36m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e206c89938a945e9ba814e4d17a1da3e : SUCCESS in 1h 19m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e0042bb4f7f448e8c3221589d7e3c2b : SUCCESS in 39m 27s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5a111cc3b4874190b6ba5152abae3fb0 : SUCCESS in 54m 58s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8a62f1170ca344e49e22dfbb29361cad : FAILURE in 1h 07m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6556d7c054bf4dafa591926757478efe : SUCCESS in 1h 07m 56s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bd6681b70484238b112d6cb2cb60499 : SUCCESS in 45m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4f945b488a154e3dba7b373b325770d8 : SUCCESS in 1h 04m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"22ca5ccaf836f027b03b9dd1b8fdadc556a02bcf","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-01 13:39:29.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/819201/2 : SUCCESS in 2h 05m 49s","accounts_in_message":[],"_revision_number":2},{"id":"29e1fb1064d9e0370e3e6775baf8a3359cb50005","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-04 07:24:51.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fa9eafdc714b7e245d28039224d7e7105898413a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-04 08:57:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/819201/2 : SUCCESS in 1h 32m 18s","accounts_in_message":[],"_revision_number":2},{"id":"c09963e112a08ed8c45d7925d38652c98933529c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 09:06:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/04af3cd82aa74be8aae08f390000eb43 : SUCCESS in 3m 09s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/08f56c5e0fb146c69d2ef4f79a133835 : SUCCESS in 43m 17s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e83ac46309b84fdfbc41be36e7ea1e8d : SUCCESS in 1h 17m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1a30a255e54d4f34ac5f579f22142250 : SUCCESS in 49m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6b7e74d68c1f43b594d5656cfa647a65 : SUCCESS in 1h 11m 12s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2c7e110f7a284655901ea478bdd2d7aa : SUCCESS in 1h 04m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9a91c858cbd94048801f4d78eef7f230 : FAILURE in 22m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fce4fc9d543449c09c58bc69135ad4cf : SUCCESS in 24m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c7f5dfe81ef46bfb9b127e7171c8f58 : SUCCESS in 22m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e5932c689934fe582fcc5c3ed86fe84 : SUCCESS in 14m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9f4dc41331c47fd9909fc83a81a4ce2 : SUCCESS in 50m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56799e44183f403c991301667e49e4f7 : SUCCESS in 29m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df39bf6f36184731a579eaae4b409ac4 : SUCCESS in 10m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b95c8940dccb4197aadcf51da1fa8155 : SUCCESS in 59m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/550f6c39ff7540808827039c77dde8e1 : SUCCESS in 1h 27m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/60e5984ea10f494f9273d9f708188ea3 : SUCCESS in 1h 14m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/97f30d7bab984c5c99292deec8e4c300 : SUCCESS in 1h 03m 16s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6407f901193f4aac82e40a3715e900b1 : SUCCESS in 56m 18s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7b188a46372a4fe4a98d71d46e11412c : SUCCESS in 1h 21m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/705e6588447e469d829a6e473d582c1d : SUCCESS in 1h 07m 47s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c7cfb9cc2b342f5bc8dcb87993282ce : SUCCESS in 1h 18m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/af2121a1fb424d7bb7017026d46b6088 : SUCCESS in 1h 17m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"43507eb294d8f47e202b97baf819318de4ebc198","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 09:19:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d86570875264aa58d30a75cf9f428b4 : SUCCESS in 55m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ee5a0ecbdc5e4b949c48a17eb3a95651 : SUCCESS in 53m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0dfe968196d6d60d1a3cc40d1949352b6b1813c5","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-15 06:43:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ad23ab22653b0636d1cd39db5fd3b456fed03024","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 08:05:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f111e2000bd641c2a27195475a57caa2 : SUCCESS in 33m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/66e1e8f124e74c8bb0553507874b3ed1 : SUCCESS in 58m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c3f40bbfcedd597c32f68118d58ae6d27f922678","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-15 08:15:30.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/819201/2 : SUCCESS in 1h 30m 52s","accounts_in_message":[],"_revision_number":2},{"id":"72aaa9f0dd5c3a9d4ffa349236f785f88b919e8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 08:53:55.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/624bcffb87d644f6b4a9141d6875ab96 : SUCCESS in 2m 37s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c8b223fe2a4946bd874df3ec1697edce : SUCCESS in 52m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e3c6ac9e79f84905bbbd145b0a0fa142 : SUCCESS in 1h 22m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2e526a2abb94415490039e743b1df82c : SUCCESS in 1h 11m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/16bbb4a1da974d73ba215d28ba783ee1 : SUCCESS in 47m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4e7c05493ca446c89c3ed6b7d63a7a6e : SUCCESS in 1h 08m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5d85ec188e204283a51f441b4d4a365c : SUCCESS in 25m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04336e6c87d44a5cb3df32f4bf8f1736 : SUCCESS in 31m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c6d79c26eff4fdd99f3a19e860fbdb9 : SUCCESS in 40m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79dcf1cd8b9145cfa6d9ba519c194a5d : SUCCESS in 12m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a3326d4b3df42e08ae018682e8066d7 : SUCCESS in 22m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ac9a757887b46199d3147d091f2c06d : SUCCESS in 20m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cfec3797db54af981b483c331e94036 : SUCCESS in 10m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/422f07ac8fa54a8ca19a653d07f92960 : SUCCESS in 47m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8bc29c0d2b9445f8a5ec889af0d4069e : FAILURE in 2h 08m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4f72aeba1bb945d2a311454b7b860acd : SUCCESS in 54m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/032d0001027f4b559b4e86a64d6e0adb : SUCCESS in 50m 10s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5bd66c2f98a849159ae96ee65b64015c : SUCCESS in 1h 03m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e44623e2549d4b549471d97d8f03c859 : SUCCESS in 1h 37m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/21d7e0a4c22149279b1bd1ef8b09b494 : SUCCESS in 1h 30m 35s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9422786c5334132b256a10fee3bc2c4 : SUCCESS in 47m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0c3bf906fb464fcc923d614ae93be0a4 : SUCCESS in 52m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"12015d8b82c8b60aa4b3db9469a0cbbb13f1c6fb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-02-15 10:42:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"68ffc739b266f2e80a02d3326d5a2fb1b5208647","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 11:34:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/506e62bb412b4c6c877f3f4febfaf28e : SUCCESS in 47m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6b4a154404b6404a847e3f5a113eb486 : SUCCESS in 35m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cdadaa570b793b97d4ff52a90b2071b0aea6bfec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 12:08:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b69b4ba609144132a38ae72a9095d152 : SUCCESS in 3m 43s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e281eb09c732470da7e2794ef5d356dd : SUCCESS in 42m 32s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/df5819c3d0d8467a9769e5fd1acf1b12 : SUCCESS in 56m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1e8d867bbfa548f281e07e93e8cfa8f1 : SUCCESS in 57m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/92a642ba540f4ab795468c860b7065c5 : SUCCESS in 47m 51s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/610565877f064941932a229917d96c2f : SUCCESS in 56m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7c556b4cd01f4e69b32c88a9635d1077 : SUCCESS in 30m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df063b5132a146c797ab3e78a96681cf : SUCCESS in 29m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ab3758cdda345f18a87140269085310 : SUCCESS in 31m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c7ff7d884a84ca08ffc666b540c6134 : SUCCESS in 12m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea26fbda663549ca988af2835728e27c : SUCCESS in 23m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62d14f9938d94daaa5a83535895080e1 : SUCCESS in 23m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/862accf69e6d44d3abcb722bbf5c58e4 : SUCCESS in 9m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e64a183aaf8c449b925855e9a0dbe074 : SUCCESS in 44m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ccb5ff23ae4d4c11ab49fbde7d4bdd48 : SUCCESS in 1h 20m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0e91f301b1074f5baed191c31851c1a2 : SUCCESS in 1h 04m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/858f47b9f8624ade8dc627c1cf08a248 : SUCCESS in 42m 37s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ee5d19738aed46d3ba7b025f14569d9b : SUCCESS in 1h 00m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/238bbffce588495388f1944eb4ce2f4e : SUCCESS in 1h 19m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d7326b4e95d94b999245cba7440a0218 : SUCCESS in 58m 18s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d5b801fd4d4466c8b4f919cf3d43fd0 : SUCCESS in 53m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e197f95711d34de09d45e6d84863bb66 : SUCCESS in 1h 04m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"49f1f10cbaac88838f5b551c1483d413ffc40738","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 12:10:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"51fa83791dfb72820278cd89f0b4186c6b9bb2ba","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-15 12:29:54.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/819201/2 : SUCCESS in 1h 46m 31s","accounts_in_message":[],"_revision_number":2},{"id":"02dfba2f8b23842900974436bb314c8905c6e5ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 13:54:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/92d5678a226a439db6383df8180b47f7 : SUCCESS in 2m 53s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7b037a79fd0943a993894eed441a1ca9 : SUCCESS in 36m 02s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ce27738e3c7c485696251b85b737c93e : SUCCESS in 1h 25m 21s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a2b77238e6e24d65ab52c68df0f4c724 : SUCCESS in 1h 04m 00s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/332d6abfa8e94c30aa366b35640047d9 : SUCCESS in 1h 21m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/29a13b0d3c3948ff8ec04b8fe9da699e : SUCCESS in 1h 06m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa8421c550c04c3b96fa2f4db18faff8 : SUCCESS in 34m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b14712ab9ea245268464589d1baccd0c : SUCCESS in 15m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7347f7d3080a4e8288f2f269d8bc7275 : SUCCESS in 28m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17803e66e34a48fc956821be72f31374 : SUCCESS in 28m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec6bf3dd4324427abfb8a6996e4be11f : SUCCESS in 12m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/053b4c1a95b545c49c764542be4581ca : SUCCESS in 41m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a9d90f1b2e6847d58b1a5686a0565bb9 : SUCCESS in 1h 39m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8d895a4ec7a240cd929a472242d42e10 : SUCCESS in 1h 07m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4454f04644734010873916b407af3d68 : SUCCESS in 50m 19s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/594351e2e0b242ad93669736e85cfa2b : SUCCESS in 1h 08m 03s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9fb6b41b5c043229d1770c64b5b4eb5 : SUCCESS in 1h 11m 56s","accounts_in_message":[],"_revision_number":2},{"id":"f10a7c9ce18a9a5bcbd97f216eb089c821609e5b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 13:54:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"786a351f4e678e7f1fe001e532288bfa2fca2034","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 14:05:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8453639ac1124926a31ef20e10ec93f1 : SUCCESS in 2m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f9b188238092f841da6bcf13f3163d38d0df923d","revisions":{"bdbe4175877e1f1e04f8f76118d75083c3878885":{"kind":"REWORK","_number":1,"created":"2021-11-24 23:43:35.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/01/819201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/819201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/819201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/819201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/819201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/819201/1"}}},"commit":{"parents":[{"commit":"770b64b90e0ada90a40538830002fc447a8dee9c","subject":"Fix tunnel_types in ml2 ovs sample config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/770b64b90e0ada90a40538830002fc447a8dee9c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-24 23:24:17.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-24 23:24:17.000000000","tz":-360},"subject":"Updating python testing as per Yoga testing runtime","message":"Updating python testing as per Yoga testing runtime\n\nYoga testing runtime has been updated with py38 and py39\nas voting and removed the py36 testing. Unit tests update are\nhandled by the job template change in openstack-zuul-job\n\n- https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/818609\n\nthis commit makes other required changes in fucntional jobs and\nupdate the metadata in setup.cfg file.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: Ia66475b377c683b115c8bd9a77c72d955db60fdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdbe4175877e1f1e04f8f76118d75083c3878885"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdbe4175877e1f1e04f8f76118d75083c3878885"}]},"branch":"refs/heads/master"},"f9b188238092f841da6bcf13f3163d38d0df923d":{"kind":"REWORK","_number":2,"created":"2021-12-14 02:13:45.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/01/819201/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/819201/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/819201/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/819201/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/819201/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/819201/2"}}},"commit":{"parents":[{"commit":"770b64b90e0ada90a40538830002fc447a8dee9c","subject":"Fix tunnel_types in ml2 ovs sample config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/770b64b90e0ada90a40538830002fc447a8dee9c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-24 23:24:17.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2021-12-14 02:13:45.000000000","tz":0},"subject":"Updating python testing as per Yoga testing runtime","message":"Updating python testing as per Yoga testing runtime\n\nYoga testing runtime[1] has been updated to add py39\ntesting as voting and as we are testing py3.6 and py3.9\nwe do not need to test py3.7|8 explicitly. Unit tests update are\nhandled by the job template change in openstack-zuul-job\n\n- https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/820286\n\nthis commit makes other required changes in zuul.yaml and\nupdate the classifier in setup.cfg file.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: Ia66475b377c683b115c8bd9a77c72d955db60fdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9b188238092f841da6bcf13f3163d38d0df923d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9b188238092f841da6bcf13f3163d38d0df923d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
