)]}'
{"id":"openstack%2Fneutron~820190","triplet_id":"openstack%2Fneutron~master~I4ea7b86e3f544d5dddcdac562208bb8afd1fc36a","project":"openstack/neutron","branch":"master","topic":"bug/1950662","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-01-20 09:10:47.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-01-20 07:32:37.000000000","reason":"Slawek Kaplonski replied on the change"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2022-01-19 11:27:43.000000000","reason":"Oleg Bondarev replied on the change"}},"hashtags":[],"change_id":"I4ea7b86e3f544d5dddcdac562208bb8afd1fc36a","subject":"Improve DHCP RPC handler","status":"MERGED","created":"2021-12-02 15:59:26.000000000","updated":"2022-01-20 09:17:44.000000000","submitted":"2022-01-20 09:10:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":31,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"820190-bug/1950662","meta_rev_id":"119ab7c9017cf396c68e78fad7ace5440bbc349a","_number":820190,"virtual_id_number":820190,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-01-19 13:25:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-20 09:10:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-20 07:32:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-19 11:27:43.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-20 07:32:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-02 16:13:57.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-12-02 16:14:18.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-02 17:09:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-02 17:56:56.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-02 19:27:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-22 15:22:21.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-12-22 16:15:38.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"57974a3780f791529fd91c2dd877e3f5fac6863a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-02 15:59:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"adee5087cabce8ce7353b5a722784a0ae51da7ab","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-02 16:13:57.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/820190/1 : FAILURE in 13m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5afa547b4863f47ca94558cb65c24b790fbc15be","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-02 16:14:18.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/90/820190/1/SRIOV-neutron-direct : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"b076bc4f69375c505e896ca0929b533b5b3a0aa3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-02 16:26:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ad75d19d168af9d932c84285a41e895df0be5638","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-02 16:27:05.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/90/820190/1/SRIOV-neutron-macvtap : ABORTED\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/90/820190/1/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"cd2661a99bb8a1b4b842a62fcabd3234d7c2eb7a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-02 16:41:00.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/820190/2 : FAILURE in 13m 40s","accounts_in_message":[],"_revision_number":2},{"id":"19f52404915a683a348bd631f81a8d88722b01c7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-02 16:41:32.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/90/820190/2/SRIOV-neutron-direct : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"bf31aaa65ba014c5d9f72dc04af14e05bface82f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 17:09:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0297e91f2af8449ba8db133f1627e9b6 : FAILURE in 34m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6bf63bac0d464af8b0a570ea14247f48 : FAILURE in 37m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"78adccf6a8eb4ed18e90111a74e699d3a2ae5011","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-12-02 17:56:56.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2021/Arista_Tempest_Testrefs/changes/90/820190/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"d443cb006c58e1a588a2f529504c6230be8af1b6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-12-02 18:53:11.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2021/Arista_Tempest_Testrefs/changes/90/820190/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"2261b4e027f0d6da07ce8c9661cd73d575f2448e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 19:27:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4b0b1f1691334280adb13a5bddaeb7b0 : SUCCESS in 44m 41s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/eb0d399fe11a467ab175f35bb260ae57 : TIMED_OUT in 2h 50m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3464f66ca0ea4a4dac01dd1c6a55f422 : TIMED_OUT in 2h 50m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b4e43aec86e449ec815d0b0df7ca1dc5 : TIMED_OUT in 2h 50m 57s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f036ce4111b54ba594be07658cae4b78 : SUCCESS in 2h 40m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a443c804c119403394fb0d88547d3f65 : SUCCESS in 27m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5811352ee65f44249c58c3279c67a53a : FAILURE in 42m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cec72b2cdf984c61ab1bcbd2377d9a37 : FAILURE in 40m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/609e8ab40498478192a9c35749ef3e07 : SUCCESS in 12m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e885377a254642668a50e61ec0f228c1 : FAILURE in 32m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c414527eeedf4c9fae333f6293073d48 : FAILURE in 28m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6af7e1bb37fe42f4bd06bea1a5a51f12 : SUCCESS in 11m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9fd9e0f22492483b93b9da20e0710844 : FAILURE in 47m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0daee119e84d484da5d0f59d31aa870d : FAILURE in 1h 18m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/08a9eca821c64545920f7e90a4a6f3ee : SUCCESS in 56m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb564007f14f46ea9a675bdb9a66cb6f : FAILURE in 54m 24s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b75da8cbb38a4da5b9b68a9240a61b7b : FAILURE in 1h 00m 35s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4aa51cce7467447c831f81dcd6802ded : FAILURE in 2h 56m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/edc13234c4104da7885d9ddfdd3d4b25 : FAILURE in 1h 39m 07s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8dee53d67d64edfb1fbcdbaabd03ca8 : SUCCESS in 47m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e8c868f5d8864fa68a5b83b83f7afcc7 : SUCCESS in 1h 11m 58s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cebbd8bee7242a292fe88250d9b1fa1 : FAILURE in 22m 10s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/92955dd670614a89b3cb000ee18196ed : FAILURE in 31m 52s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0ee40be0cc804850a233d5d916c439da : FAILURE in 41m 39s","accounts_in_message":[],"_revision_number":2},{"id":"e3a5023d6622c9928fe413d76ebd6699dcacd54a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-02 19:28:19.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/90/820190/2/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/90/820190/2/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"c05ff074bfc6fa2dafb7b1beec3e9b512c704717","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-03 12:15:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"279371bcf74b4157903cbd557950194cccf9cad6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-03 12:30:37.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/820190/3 : FAILURE in 14m 04s","accounts_in_message":[],"_revision_number":3},{"id":"17600be6f8a2441016e25e814ee877fe01038174","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 13:27:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/69b0ccd950034033b3535b10e51b589f : FAILURE in 54m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f7708513099c4fd0b6c78b097577fb77 : FAILURE in 51m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"42cbf6ce00f16b743e5c798c1a4e7ce4c9575565","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-12-03 13:41:51.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2021/Arista_Tempest_Testrefs/changes/90/820190/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"118c3593a04806a2b93e7486bf89dbcf64281cd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 15:18:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b2bb62b64de947fabda1a10ae01efd6e : SUCCESS in 50m 25s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d8aa0abdccf2469e9e5c1fb80863ebaa : TIMED_OUT in 2h 53m 20s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a7182cca3238450289d8f9bc2ac9071c : SUCCESS in 2h 41m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/246a60ad6e0f44d5a3287fc1b177308c : SUCCESS in 1h 40m 46s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c2eae3f2872b49aea191e3c5d9f19255 : SUCCESS in 2h 25m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/279507d505284c0ca5ce52321ceb3281 : SUCCESS in 47m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/899af3db2b734e39adf730930ac010ea : FAILURE in 34m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05bad8a19f58467b884e473dd68baa2d : FAILURE in 36m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6911aa546598431cad958d3afd271867 : SUCCESS in 14m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c6a2e2c6ac24c918465fdaff09594f2 : FAILURE in 28m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/933107a98a0743278bbb9974d499e455 : FAILURE in 23m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5092affbfa3b43a5b3827812f879a359 : SUCCESS in 10m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3e9efac32c8a4f4e8c43d18fa82f1f75 : SUCCESS in 42m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5310a2fc1f8d42249514599ea8dead4b : SUCCESS in 1h 50m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/492f658a01da4bc6ac1684cb8bae04a3 : SUCCESS in 1h 11m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3b229422d7c249108ea892e4dea22589 : SUCCESS in 59m 02s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/29c2a949c26947cbb66b7631fd23a34d : SUCCESS in 1h 03m 03s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b379b643a9b84065bf8f68895f8ade5f : SUCCESS in 1h 13m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/893f325fd2e64da0a8517f2033732569 : SUCCESS in 1h 13m 55s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/925524520ba440fbb5f15fcefe5eb26f : SUCCESS in 1h 05m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c469a6ac68b848f980469fbb7b92342d : SUCCESS in 1h 20m 44s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ebda9eabe6846a3a4a52cf9f9f48d37 : FAILURE in 32m 02s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/481bd5afb7a545f0aeeab7cc9bb298bc : FAILURE in 42m 40s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8691456d4a39464cb25f194d570364bb : FAILURE in 36m 16s","accounts_in_message":[],"_revision_number":3},{"id":"f9f668e8d0a4f89cf187391d4a8f12622600b692","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-10 16:42:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5206f79995347112025438457dba53eccb2769ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 17:20:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/216eb66937b741fbbeb67d8985bf9d1a : SUCCESS in 33m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bfffdedbaf78466cac209b3db7de0485 : SUCCESS in 32m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3dfd24eb803f41c29e5c6b1d4b83eb1d5c420a3b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-10 18:35:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/820190/4 : SUCCESS in 1h 52m 16s","accounts_in_message":[],"_revision_number":4},{"id":"5ef6d8015dc38ed98f7bf0941e178cc8d6ddc97b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-12-10 19:49:21.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2021/Arista_Tempest_Testrefs/changes/90/820190/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"66c166927cbc81e83523214ae1f33e7d0793ca78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 19:53:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/36919dff48764b9caf9ff246e1783c98 : SUCCESS in 47m 47s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a2c3b25ec0a44d50b0db9c7e760ef545 : SUCCESS in 2h 26m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1b4fc1235c84443aa85d0fe83550b5ff : TIMED_OUT in 2h 50m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2d2a6b721d274c52ae7cf3930b380bdc : SUCCESS in 1h 36m 22s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4d2b89d88c86427c8ee3be34369da631 : TIMED_OUT in 3h 04m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c76c424d2e344af3abc17927f27cfc38 : SUCCESS in 39m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc216c4550ff4557a390b5866567cb8c : SUCCESS in 40m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5710c32e4284cd1bc3db81b061c0c83 : SUCCESS in 31m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef1fafa7034748c9bd9e4b2cc0ef73c3 : SUCCESS in 11m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9b3210616c3473ebbb25ae87c73239c : SUCCESS in 26m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a7b1bcb66934a9995607cfb8c117aaf : SUCCESS in 27m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b3ff80998b2427db5d6dce1af87f417 : SUCCESS in 9m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e9029ac2d2fc4124b97958d375df2bfc : SUCCESS in 46m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/35c23a1233be4c2ba34211ccd945fca9 : SUCCESS in 1h 12m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ef580be7cd114be185aebf8ce0b5087d : SUCCESS in 1h 07m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1f06417a34f6490ba3337027084c8296 : SUCCESS in 49m 24s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/247d236febce4a9090eeb85ff4e9d1f7 : SUCCESS in 49m 27s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5f01a9f21baf4358930302616c3173ff : SUCCESS in 1h 05m 46s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1e2b00d9323f451eba73758e9f3fdad7 : SUCCESS in 1h 15m 28s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d71f126649b2418681897b53470f7700 : SUCCESS in 1h 08m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9c4d76f0bc754de4af908ae22110e330 : SUCCESS in 1h 32m 15s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c614ce2c1e8342c6928f777e117d0377 : SUCCESS in 24m 54s","accounts_in_message":[],"_revision_number":4},{"id":"ba884f2ee463b9ffc3a8c44a3085afd422eef696","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-11 08:10:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7d950d20059bd9fd745da64f6243c6c69068d40f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-11 08:49:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5cd6b97242304cfa8af7fd3f32e61f6d : SUCCESS in 33m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/35c414a2a34c4c2e9d1069b5685da5e9 : SUCCESS in 32m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c179d68824342960a1215c3482795bf070d0f541","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-11 09:45:42.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/820190/4 : SUCCESS in 1h 34m 55s","accounts_in_message":[],"_revision_number":4},{"id":"182d99441e2a9d89e08fdcd4333f16eab0b512cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-11 11:03:41.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c41e2a9bc73d4e27a377c46b2bd6f8f6 : FAILURE in 43m 47s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a19c459ebd1f4ba7b5a644cbfa3cdb47 : SUCCESS in 2h 45m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/cda64832eb5b4d63aac78f2fd0bcddd0 : SUCCESS in 2h 27m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dd786c39dae64958855a395835cc618d : SUCCESS in 1h 26m 34s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/87fa1102e2694ef885e176e228012ef1 : SUCCESS in 2h 46m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cc757d4579884675b789d92c74032af3 : SUCCESS in 40m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5ac0cd2d811426f8b8993f89f498168 : SUCCESS in 51m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2919e5d50a6648c7b3919144cc4ea63c : SUCCESS in 36m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3c5a676bf6648ff9c682c0e2f36143b : SUCCESS in 13m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a69b32342a64c0a9f267d720f7d0605 : SUCCESS in 37m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3acacb865ec4b9c9a2106ba434a3ec3 : SUCCESS in 25m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6f1321162bf49ba953e8d27eeae639d : SUCCESS in 10m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f2c06928d7dd4243a1909933b90a5653 : SUCCESS in 51m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5ce08200f06149b89bc8c941723ed77a : SUCCESS in 1h 23m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7fe52585dac2477cb5cea5b51edfee07 : SUCCESS in 1h 21m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0692d128db9a4af89d2ab1d8e8651b1a : SUCCESS in 49m 45s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1046dbda5d6a4e1999ebffead0b780b1 : SUCCESS in 51m 48s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/40db117783e947a8bbdabb6c28580574 : SUCCESS in 1h 22m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d3267ab5a03f4032986dc5ac55c49a20 : SUCCESS in 1h 13m 46s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f8c5f593a5c43d4afbc0a33747a1681 : SUCCESS in 1h 16m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7a82a34037bb4fe199eb022565a2ca11 : SUCCESS in 1h 31m 22s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69df4ccf164b472e91e4611f22f491a7 : SUCCESS in 40m 48s","accounts_in_message":[],"_revision_number":4},{"id":"7bd9e2148f5f0194cf6a2e3ae30080edc4ce592d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-13 08:28:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"560c23252a19417869f8fa7e6be5a207efc65f5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 09:27:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dcdc585d88eb43789fc1ca616150fb57 : SUCCESS in 41m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8a5e74d0333a453784569f60c60a5dfb : SUCCESS in 42m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c923d9436aee10401e6b5c0b18eb3da257dc92c1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-13 10:10:22.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/820190/4 : SUCCESS in 1h 41m 00s","accounts_in_message":[],"_revision_number":4},{"id":"b20257ff8588111c82719a191fc3a4fae70ff1f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 10:52:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7c78180359984297a6d7135c7354d9d8 : SUCCESS in 44m 59s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4884c5c62f0e4cbaba5cd337036b70d9 : SUCCESS in 2h 21m 19s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/393ab956cc31479ab92821bda85c5a1d : SUCCESS in 2h 09m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0cd8bc04ed424886845ac189b8a3ec3a : SUCCESS in 1h 39m 49s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/90509475a7674f18b84fde45e76c7383 : SUCCESS in 2h 01m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fb0733672278415a842a7532569bb868 : SUCCESS in 41m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c6cb0f4a5f0491ab43d32faf7c0ce8f : SUCCESS in 35m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a00c94d713f14ab39044f8d6a50b8ded : SUCCESS in 30m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd76ca39be9f4e5cab83cbbfe28c187c : SUCCESS in 12m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a31028a1e67844439391dc285c1241db : SUCCESS in 25m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3ac729cee3647cca1e8622f19ea99fc : SUCCESS in 26m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e430ead59a2f4e32880dd3d5d9f5348e : SUCCESS in 11m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2d5390bdd0d24ba4b8541ec71c266bf3 : SUCCESS in 51m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/868a826764ec47d6871d42921897c988 : SUCCESS in 1h 55m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c9b2779fefc6446fabddff9f1a10a385 : SUCCESS in 1h 15m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c99d7ea626254010bf8683ea46fef6c3 : SUCCESS in 44m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e14a64dd500e4608854f82b33eafeea8 : SUCCESS in 1h 06m 56s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cd2388b465b4400189ad217d14c839ad : SUCCESS in 1h 12m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/251e5451516c441bbeb6565b7e37d42f : SUCCESS in 1h 19m 03s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8ed1081e24b44c0ac1c545d7a64f79c : SUCCESS in 1h 06m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3ec61c8f02ec4f3591ca74529780fbec : SUCCESS in 1h 09m 03s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18aa6ba59f0b4d35a0a3474b00e58ad8 : SUCCESS in 28m 25s","accounts_in_message":[],"_revision_number":4},{"id":"c22749475da053ac458d1f1147211ab9813ad12c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-12-22 15:22:21.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"22e60d4e52e06853d2dc4ec1dc0123f55a35d4ef","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-12-23 08:59:46.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"de7fb97e7c57720dd41318ee99f61ebf92445472","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-12-23 09:01:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"45a8a9abfeba763b38bb719b865d90893321d1f6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-24 14:52:44.000000000","message":"Patch Set 4: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6ab22cdda8f0628c3c4a870a778e757cc41ef1dd","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-12-24 15:18:32.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6848642a8baaea31dfb5c2f7f9189f0523d7ebf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-24 15:45:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4a7f2b43a8fdf83bfbaf2d1e2a8b8dcf67c94a48","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-24 16:07:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1afb0587001708810c1aa1f8b5bd86db9f0edaf1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-24 16:22:23.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"56dccea2a17ba780647e59d17e2215f97732f77e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 16:28:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b7bf3683c0494edcb2fbecd01438b99d : SUCCESS in 39m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/56d902530b3d42fba9a3b9f4c71a3625 : SUCCESS in 40m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"548e2b145e6e55b49b4861453e884f50b538057d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-12-24 16:42:01.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2021/Arista_Tempest_Testrefs/changes/90/820190/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"bd646afa164042de073c4553e6d170e236858375","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-24 17:16:18.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/820190/5 : SUCCESS in 1h 30m 08s","accounts_in_message":[],"_revision_number":5},{"id":"27e4f206c5d76edfe238c87115afb0c4afb1ed8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 17:33:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2f1fe64c85894cbe96b6c69b8d070a23 : SUCCESS in 40m 00s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/701d0327f4154135922fde414eca8389 : SUCCESS in 1h 28m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1ca824efb20e417082b53af1f7432bdc : SUCCESS in 1h 09m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/91804c08891f438db0983bef927c9efd : FAILURE in 1h 45m 33s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/cfc4ffa7f5a545448d407d24d976b21b : SUCCESS in 1h 15m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/10f024690682452bb7021ecb5d39a186 : SUCCESS in 36m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c1b6bd6b4c2472080649be4f53541e7 : SUCCESS in 25m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29ac9b27654c411b9c6fb74615ae4137 : SUCCESS in 29m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3680df7bf55b478b86683d3830dfacab : SUCCESS in 12m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9924cd405eec4129b71485e19442ffaf : SUCCESS in 24m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3271d22d241a4ec6a4a5708a8cc69d38 : SUCCESS in 22m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ad4c3fb34fb44af869d0db9770db368 : SUCCESS in 10m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f3a6307db88a48c3ba6ef0952a5da29f : SUCCESS in 42m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8a3c6965c14545b8acfaf9581fd41ba0 : SUCCESS in 1h 16m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/81840b7e363e4321b44e9f6d0171c36a : SUCCESS in 1h 01m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/828b886ecc974b6cab0334ef7f80a76e : SUCCESS in 40m 50s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f9effd47099345aba7e2dfd2cbdcfa41 : SUCCESS in 54m 49s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/61f0a70958cc4c4ebcc794aa76078d72 : SUCCESS in 1h 16m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/acd70c3a11c14f278f9ce9839a9f0a3f : SUCCESS in 56m 26s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dac95597e54040c882b227ad9411ace0 : SUCCESS in 57m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/09abac0bb63843d38c34d951b5c8a7d3 : SUCCESS in 58m 30s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/119e989b64fa4775be4ea57d5197fa88 : SUCCESS in 26m 02s","accounts_in_message":[],"_revision_number":5},{"id":"a72a3836eea80b1d890df80af1c7636b7055e4ac","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-12-27 07:06:33.000000000","message":"Patch Set 5: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"684cd0f39e096a347db36439cdfa5f461ac732e5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-04 11:34:00.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"0694944bd59e7134c7a4a0de77ed61259319b721","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-04 11:34:40.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9f2e31547de1670c005c9859df56fde5fb646536","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-04 11:39:17.000000000","message":"Patch Set 6:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/90/820190/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"12d07b585ec7abc77000323dd00f46855866c537","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 12:23:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/13842b77e505425c83023e633347a467 : SUCCESS in 44m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8533a69a17724795b1e76e899ed16662 : SUCCESS in 46m 49s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"81bcaec1d23a8106e8e3003bca2be2f43d78f966","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 13:07:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0134a2950940490cbe6906a1c7093e97 : SUCCESS in 45m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/70b052a51a044446b1e1dcab0f23795e : SUCCESS in 1h 11m 22s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/67ce1305b2cb4099b5fcd4227d37b6a0 : SUCCESS in 1h 18m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c1e77ab933484dd597d7cad42476d128 : SUCCESS in 56m 02s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b79a173e9cb347689f806c0796782e3f : SUCCESS in 46m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/389ebabfb316485e811db3b0f168acaf : SUCCESS in 28m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73ac6023bdaa431b891124b794a646b9 : SUCCESS in 35m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/398828fba1ea40eda130ba0f7ff01a17 : SUCCESS in 34m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1317a67f1e7a49aebcf94675d0f78414 : SUCCESS in 13m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61e64ea0f82440168a85587cf86d4982 : SUCCESS in 27m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23623c92a62f4bcea3c61b9505136fae : SUCCESS in 28m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aeab05625f6462288fec43293004aea : SUCCESS in 10m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5dea03fd244e412386c93ea28656c6e8 : SUCCESS in 49m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/12fb893255184491a05f8495ade335d8 : SUCCESS in 1h 25m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d51eea5332584402b2d8e1868761a3d3 : SUCCESS in 1h 16m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7e44b3e73a5f4c1e9bf8ff63242cb4a4 : SUCCESS in 58m 06s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bf059348d8c5400baf64e1e908b75acb : SUCCESS in 58m 14s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e62730611ea34d939b65431e8fa542a3 : SUCCESS in 1h 10m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a38c6a1f0a044bedadde869ed397fe52 : SUCCESS in 1h 22m 55s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/014cd154ff604af28d68fa6373ae5889 : SUCCESS in 56m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/abfa4238a0614490b8a102573ebdf0d7 : SUCCESS in 1h 08m 16s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6d5851b9d9a4bfb80698dadd8bac0d8 : SUCCESS in 22m 32s","accounts_in_message":[],"_revision_number":6},{"id":"59462949158a747a98049acc9552a8559b93e0cd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-05 15:24:30.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"115bd98d0e7f826c87e5193571306f9174d99312","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-05 17:25:57.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a6489dc23cbaaad577b631f870166a385f5d07fa","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-01-10 11:46:02.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"798c852a480eddd3b05438368c9f2b6ec60d6991","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-14 11:59:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"bcbee33727bee0180ddedd228c5340c22e1c95ba","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-01-14 12:37:30.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"f3e246c185652dac96ffed5312a33e584d449009","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-19 08:53:11.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b1a981b7bae1d69456031c571bc53cc69bfc7cfa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-19 08:55:38.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6a53c1d808743ee41996a6a085bd86bcbb7afeab","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-19 09:00:35.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"3a2ab8b04ff1a316cca84270145941459ce90c8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-19 09:02:10.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"2671409465dccbdcc2e452e71532e339df98604a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 09:45:22.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1548580e86cc468089c4cad995091394 : SUCCESS in 35m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/28715a7ac8f0475dacbbdb5567bf156a : SUCCESS in 35m 52s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"38669166607096cd9316160a4f6e3c0da68ec524","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 10:37:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5ac10db2513449a7a26390fde5f3ee69 : SUCCESS in 57m 10s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/423df07d417f401e9c47642d7fb99aca : SUCCESS in 1h 20m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2295aa96bc80473789f6251ecb658949 : SUCCESS in 1h 19m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/967f8eccbdde45919c5183fde829d240 : SUCCESS in 1h 10m 08s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7fb43c53989c4d22bbf72f3fee038ec6 : SUCCESS in 1h 10m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/57061e3c42024af1a63d731ea8c6c3a6 : SUCCESS in 32m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47cf05a1d2974747ad4a8911a487a2e1 : SUCCESS in 39m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a6fbe14577547bf8a063803ab4751f9 : SUCCESS in 35m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4a9c901615d411cafb5d54df221d874 : SUCCESS in 13m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0411691fcb3d496390753e49ec212af8 : SUCCESS in 30m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8797f4a89d04ddda2870a304a524ac5 : SUCCESS in 28m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a09a7d858ed43a0bcefad545131b535 : SUCCESS in 10m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b7718bf93086434081fa024190744224 : SUCCESS in 47m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d29bbf629e204d4bb1bb16036f78ef27 : SUCCESS in 1h 32m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2c215c501d1e4e4f81c6378c99953f35 : SUCCESS in 1h 10m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ea81b2d3d0404d9db9ae44246b2038d1 : SUCCESS in 57m 35s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bf1ed4723ab348ed84881faaefb96e2b : SUCCESS in 1h 08m 16s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/327859ee9e4a4e778e73e42bbe7c00e0 : SUCCESS in 1h 17m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d816dbdca56a435fa7a733e8874ddbce : SUCCESS in 1h 10m 40s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24322ab2bc9543f4938685fc9ddc81cc : SUCCESS in 55m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1e60486250fa48ac8edcbb2407d24a52 : SUCCESS in 1h 07m 44s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79903ce26f764c72850d7f6c451fca97 : SUCCESS in 23m 06s","accounts_in_message":[],"_revision_number":7},{"id":"0dbe1b69fb7aa19ad18e49d2566a4667642b1485","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-01-19 11:27:43.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"01c16573032726f10c6fcc8b3b9ebb61670461ed","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-19 13:25:42.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/90/820190/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"4b5c43a9672aa959516f94346ee4c70005f8239b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-20 07:32:37.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"148b260741291fe783d66c0448695d949bff34e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 07:33:21.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"d52b8bd5567a034504f3700706831814b5c10bd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 09:10:47.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/09c1b14477b544ef88cf2a11eab0ef63 : SUCCESS in 45m 16s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fec8e974c8bf4797ace8b2b16f010d85 : SUCCESS in 56m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/661a7c7bec6f4549a424bc651b6b9bec : SUCCESS in 49m 47s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/17d36aa3ef80400cb056a1c6fe698434 : SUCCESS in 52m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/212eb6442e81404c94e4acda8c39a7d8 : SUCCESS in 46m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5eab5d323104405c9b6fd1f3ea5f45c8 : SUCCESS in 32m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74aa6085a4be41cbb75055b4b9a110a5 : SUCCESS in 14m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/131151e72549475e8079ce64e7075118 : SUCCESS in 28m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29b8d2fdde7f4ca49c28311c0f7d0dbb : SUCCESS in 29m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9889e45ca8f4929ab3874803a7b95c1 : SUCCESS in 8m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/680d005a66d443d784f718029e9bc7cf : SUCCESS in 45m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4799c6a85e1d46358aa9cb66bd81c889 : SUCCESS in 1h 30m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ded1c9e2e3a14fbdb13ebb5a0bb41da2 : SUCCESS in 1h 15m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7a9d4581f2ea4594abad7eacac29c81c : SUCCESS in 47m 47s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea0837bf882e48c0a1330fe615d0c655 : SUCCESS in 54m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c5ba11eca464fadb0f7127588079c25 : SUCCESS in 24m 33s","accounts_in_message":[],"_revision_number":7},{"id":"1ba1d6e6170873a794a92f1ae2abaf86f3ad2166","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 09:10:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"119ab7c9017cf396c68e78fad7ace5440bbc349a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 09:17:44.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81d27f2b49264148a30d63cf4adf83b1 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"5710d3407bbd59932c1c9c6a3716b34c8e2ab3e8","revisions":{"309ec2b694c84d60a36d214e55d58e0fd78c5305":{"kind":"REWORK","_number":1,"created":"2021-12-02 15:59:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/90/820190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/90/820190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/90/820190/1"}}},"commit":{"parents":[{"commit":"c686a2b555a15de6f8c1759dfa6d7ff589792975","subject":"Improve DHCP RPC handler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c686a2b555a15de6f8c1759dfa6d7ff589792975"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-12-02 15:58:07.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-12-02 15:58:07.000000000","tz":0},"subject":"[WIP][DNM] Improve DHCP RPC handler","message":"[WIP][DNM] Improve DHCP RPC handler\n\nRemove unnecessary DB retrieval operations from\n\"get_active_networks_info\" method.\n\nPartial-Bug: #1950662\nChange-Id: I4ea7b86e3f544d5dddcdac562208bb8afd1fc36a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/309ec2b694c84d60a36d214e55d58e0fd78c5305"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/309ec2b694c84d60a36d214e55d58e0fd78c5305"}]},"branch":"refs/heads/master"},"91ecf379d6f9a958a48897ebd1d8f5f0e897bb00":{"kind":"REWORK","_number":2,"created":"2021-12-02 16:26:32.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/90/820190/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/90/820190/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820190/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820190/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820190/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/90/820190/2"}}},"commit":{"parents":[{"commit":"c686a2b555a15de6f8c1759dfa6d7ff589792975","subject":"Improve DHCP RPC handler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c686a2b555a15de6f8c1759dfa6d7ff589792975"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-12-02 15:58:07.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-12-02 16:26:12.000000000","tz":0},"subject":"[WIP][DNM] Improve DHCP RPC handler","message":"[WIP][DNM] Improve DHCP RPC handler\n\nRemove unnecessary DB retrieval operations from\n\"get_active_networks_info\" method.\n\nPartial-Bug: #1950662\nChange-Id: I4ea7b86e3f544d5dddcdac562208bb8afd1fc36a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91ecf379d6f9a958a48897ebd1d8f5f0e897bb00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91ecf379d6f9a958a48897ebd1d8f5f0e897bb00"}]},"branch":"refs/heads/master"},"008e86676174b9af5ee2d79ab1d7fee9f312f9ff":{"kind":"REWORK","_number":3,"created":"2021-12-03 12:15:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/90/820190/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/90/820190/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820190/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820190/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820190/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/90/820190/3"}}},"commit":{"parents":[{"commit":"c686a2b555a15de6f8c1759dfa6d7ff589792975","subject":"Improve DHCP RPC handler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c686a2b555a15de6f8c1759dfa6d7ff589792975"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-12-02 15:58:07.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-12-03 12:15:11.000000000","tz":0},"subject":"[WIP][DNM] Improve DHCP RPC handler","message":"[WIP][DNM] Improve DHCP RPC handler\n\nRemove unnecessary DB retrieval operations from\n\"get_active_networks_info\" method.\n\nPartial-Bug: #1950662\nChange-Id: I4ea7b86e3f544d5dddcdac562208bb8afd1fc36a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/008e86676174b9af5ee2d79ab1d7fee9f312f9ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/008e86676174b9af5ee2d79ab1d7fee9f312f9ff"}]},"branch":"refs/heads/master"},"00f54295a771e79eaae43dcb7b429f8ea6ceb2d8":{"kind":"REWORK","_number":4,"created":"2021-12-10 16:42:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/90/820190/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/90/820190/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820190/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820190/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820190/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/90/820190/4"}}},"commit":{"parents":[{"commit":"01debb20bf12267d16f932a678c77450cf761419","subject":"Merge \"Provide a \"priority\" value to the OVN events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/01debb20bf12267d16f932a678c77450cf761419"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-12-02 15:58:07.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-12-10 16:42:07.000000000","tz":0},"subject":"Improve DHCP RPC handler","message":"Improve DHCP RPC handler\n\nRemove unnecessary DB retrieval operations from\n\"get_active_networks_info\" method.\n\nPartial-Bug: #1950662\nChange-Id: I4ea7b86e3f544d5dddcdac562208bb8afd1fc36a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00f54295a771e79eaae43dcb7b429f8ea6ceb2d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00f54295a771e79eaae43dcb7b429f8ea6ceb2d8"}]},"branch":"refs/heads/master"},"932380d86c0703f860a9c6fb35b5607048d583c6":{"kind":"REWORK","_number":5,"created":"2021-12-24 15:45:29.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/90/820190/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/90/820190/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820190/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820190/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820190/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/90/820190/5"}}},"commit":{"parents":[{"commit":"01debb20bf12267d16f932a678c77450cf761419","subject":"Merge \"Provide a \"priority\" value to the OVN events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/01debb20bf12267d16f932a678c77450cf761419"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-12-02 15:58:07.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-12-22 00:11:06.000000000","tz":0},"subject":"Improve DHCP RPC handler","message":"Improve DHCP RPC handler\n\nRemove unnecessary DB retrieval operations from\n\"get_active_networks_info\" method.\n\nPartial-Bug: #1950662\nChange-Id: I4ea7b86e3f544d5dddcdac562208bb8afd1fc36a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/932380d86c0703f860a9c6fb35b5607048d583c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/932380d86c0703f860a9c6fb35b5607048d583c6"}]},"branch":"refs/heads/master"},"e0f387f808dec2f31ad4dec256b132df0beeaaad":{"kind":"REWORK","_number":6,"created":"2022-01-04 11:34:40.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/90/820190/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/90/820190/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820190/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820190/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820190/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/90/820190/6"}}},"commit":{"parents":[{"commit":"01debb20bf12267d16f932a678c77450cf761419","subject":"Merge \"Provide a \"priority\" value to the OVN events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/01debb20bf12267d16f932a678c77450cf761419"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-12-02 15:58:07.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-12-22 17:01:46.000000000","tz":0},"subject":"Improve DHCP RPC handler","message":"Improve DHCP RPC handler\n\nRemove unnecessary DB retrieval operations from\n\"get_active_networks_info\" method.\n\nPartial-Bug: #1950662\nChange-Id: I4ea7b86e3f544d5dddcdac562208bb8afd1fc36a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0f387f808dec2f31ad4dec256b132df0beeaaad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0f387f808dec2f31ad4dec256b132df0beeaaad"}]},"branch":"refs/heads/master"},"5710d3407bbd59932c1c9c6a3716b34c8e2ab3e8":{"kind":"REWORK","_number":7,"created":"2022-01-19 09:02:10.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/90/820190/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/90/820190/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820190/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820190/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820190/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/90/820190/7"}}},"commit":{"parents":[{"commit":"01debb20bf12267d16f932a678c77450cf761419","subject":"Merge \"Provide a \"priority\" value to the OVN events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/01debb20bf12267d16f932a678c77450cf761419"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-12-02 15:58:07.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-01-15 07:58:25.000000000","tz":0},"subject":"Improve DHCP RPC handler","message":"Improve DHCP RPC handler\n\nRemove unnecessary DB retrieval operations from\n\"get_active_networks_info\" method.\n\nPartial-Bug: #1950662\nChange-Id: I4ea7b86e3f544d5dddcdac562208bb8afd1fc36a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5710d3407bbd59932c1c9c6a3716b34c8e2ab3e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5710d3407bbd59932c1c9c6a3716b34c8e2ab3e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
