)]}'
{"id":"openstack%2Fneutron~820490","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~I9f5bd8a1404dde3a0aa163ce72aef2961f537676","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/1952770-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-12-15 02:40:48.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2021-12-15 02:40:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9f5bd8a1404dde3a0aa163ce72aef2961f537676","subject":"Do no use \"--strict\" for OF deletion in TRANSIENT_TABLE","status":"MERGED","created":"2021-12-06 09:03:07.000000000","updated":"2021-12-15 02:44:27.000000000","submitted":"2021-12-15 02:40:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820490-bug/1952770-stable/xena-stable/wallaby","cherry_pick_of_change":820349,"cherry_pick_of_patch_set":1,"meta_rev_id":"e1f1677e420e3af0f18cbaa0b47d15b537d5b7af","_number":820490,"virtual_id_number":820490,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-15 02:40:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-12-06 12:53:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-12-14 22:56:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-12-14 22:56:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-06 09:03:07.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-12-06 09:17:56.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-12-06 09:50:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-06 10:57:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-06 12:53:12.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-12-14 22:56:47.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"f6fa3abe7fa8f9c312838779b4e423e9be920e6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-12-06 09:03:07.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"eaca0919505bec936ae4bb588102bf3688f0b8e9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-06 09:17:56.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/820490/1 : FAILURE in 14m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0a65c2daa7e5840b9a5eab8a3f52c9e3bf67d510","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 09:50:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee607a07aa194f299c7bb6369b9ab03a : SUCCESS in 45m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3c697b3657f244b88cffd89c45374f92 : SUCCESS in 43m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dad5200df309e611b99d45499d18ff9d6a4e07c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 10:57:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/9819cda9e8f14188a8707a446c570313 : SUCCESS in 1h 51m 56s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/30412c028147412797b985a011636458 : SUCCESS in 1h 47m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e16d47f15884ca8beee54d284ff62b3 : SUCCESS in 44m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3384dfbdc4e849d4ba3fb4575c54d634 : SUCCESS in 11m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eafcca8a2e5c4529b7b161df62a4bc3b : SUCCESS in 27m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c1896e6e73e4f709be1a453725051f2 : SUCCESS in 29m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1ba150295154f19b82ed065e60016b2 : SUCCESS in 23m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eb8c6d904af43eb9b2a064bfe93f0fd : SUCCESS in 9m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f76d03bdec5e45fab7c06779fc30559f : SUCCESS in 1h 04m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ddcbb53dfa1c4e17b44126112953a332 : SUCCESS in 1h 23m 17s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/aaaff3b9bfea41a1b9e6102ad62a2908 : SUCCESS in 1h 20m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/025750a9da94492e864cfbab9de7cf84 : SUCCESS in 1h 05m 38s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/eeb03c7101e74f53928f68dec384fa84 : SUCCESS in 1h 03m 30s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/58f4531c058146ceb6e8aeecfb0acd1a : SUCCESS in 1h 13m 44s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/688f8b7b45c84db09708881b544df15c : SUCCESS in 1h 26m 30s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/65b160290f874e7f82ecea66249d6e06 : SUCCESS in 1h 36m 07s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddb52398ac024a5bb55ff780636f33d2 : SUCCESS in 57m 58s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d388f726b84e455fa83fe81b07098a56 : SUCCESS in 1h 22m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6e129ad391d3458cae8c9b9daa6426e4 : SUCCESS in 29m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a6b46bc2c2394c44869ae1690e3efa3e : SUCCESS in 1h 26m 13s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9ca49a42b2f64727b3afe533536b3191 : SUCCESS in 1h 13m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9e9c756b2afc435d13325a174cbd3cf46a5e3517","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-12-06 12:53:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8ca855e37b7e4ae9d8ccc5e050041e8e17b1b399","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2021-12-14 22:56:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0f138e0a3738a82869fe474a14b55c86388cb174","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 22:57:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c7a3226a1cfda253e8d3a7a67faef8e07f433051","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 02:40:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/813e2aecbd4d4ff895035249710f5f55 : SUCCESS in 13m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb52168c344d43ad9cb48167525ed898 : SUCCESS in 22m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a344796be794bfa8b61ad0d28bc620f : SUCCESS in 28m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39ebe756b9c7424cb83827973a4b4c31 : SUCCESS in 9m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/98a1d0461cdb4e5c9728e4482b3183f2 : SUCCESS in 1h 22m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/40730e1592af45fbaed7117b4b14f1f4 : SUCCESS in 1h 39m 34s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9d1e485b67c64d54829e496c85b0943a : SUCCESS in 1h 03m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5af6d25d5a8f496f978cc422c37e24ab : SUCCESS in 56m 30s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b66c494755624d4aabcdc6cb4e5baa16 : SUCCESS in 1h 30m 07s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bbea65ac42404ab5950e79f94cc2c8fc : SUCCESS in 1h 06m 08s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5309c55fb09c4fc69c04849107a40a48 : SUCCESS in 1h 13m 23s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e281d021dd95427689649c196e60dbba : SUCCESS in 38m 18s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/93a7ed3da8d14887815fcad373e5618d : SUCCESS in 1h 30m 30s","accounts_in_message":[],"_revision_number":1},{"id":"3d12b6a02a009459de29b5e6110a21783fb89a54","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 02:40:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e1f1677e420e3af0f18cbaa0b47d15b537d5b7af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 02:44:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/578ad7c2a52c406ca22c0be59c84ff58 : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0db5334ccf63226383f9fa85017c194bf01fcc38","revisions":{"0db5334ccf63226383f9fa85017c194bf01fcc38":{"kind":"REWORK","_number":1,"created":"2021-12-06 09:03:07.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/90/820490/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/90/820490/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820490/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820490/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/820490/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/90/820490/1"}}},"commit":{"parents":[{"commit":"a321c6787fe20b48d47650aa3ac2a9b7d249db61","subject":"Merge \"Set report_interval to 0 for ovs agent unit tests\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a321c6787fe20b48d47650aa3ac2a9b7d249db61"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-11-30 16:01:27.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-12-06 09:03:07.000000000","tz":0},"subject":"Do no use \"--strict\" for OF deletion in TRANSIENT_TABLE","message":"Do no use \"--strict\" for OF deletion in TRANSIENT_TABLE\n\nThere are two types of OF rules in TRANSIENT_TABLE:\n- With priority 100: these rules match by \"in_port\", that is a\n  unique identifier.\n- With priority 90: these rules match by MAC address and VLAN ID.\n  This combination (MAC, VLAN) is unique.\n\nThat means when a deleting an OF rule in TRANSIENT_TABLE, it is\nenough to specify the \"in_port\" or the (MAC, VLAN) tuple. The\n\"--strict\" parameter, added to also define the priority, is not\nneeded.\n\nBy removing the \"--strict\" parameter, these deletion commands can\nbe executed synchronously at the end of the OVS deferred context,\nwhen all the OF rule commands (addition or deletion), are executed\nat the same time. That removes the small window, detected in the\nrelated bug, when the OF rule set for a port is not complete.\n\nCloses-Bug: #1952770\nChange-Id: I9f5bd8a1404dde3a0aa163ce72aef2961f537676\n(cherry picked from commit ef7f673098c2a4574365f6f4ed20734f29309f08)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0db5334ccf63226383f9fa85017c194bf01fcc38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0db5334ccf63226383f9fa85017c194bf01fcc38"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
