)]}'
{"id":"openstack%2Fneutron~820897","triplet_id":"openstack%2Fneutron~master~Ib6b70114efb140cf1393b57ebc350fea4b0a2443","project":"openstack/neutron","branch":"master","topic":"bug/1930414","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-01-19 23:13:22.000000000","reason":"Change was submitted"},"15554":{"account":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"last_update":"2022-01-19 23:13:22.000000000","reason":"Change was submitted"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2022-01-19 23:13:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib6b70114efb140cf1393b57ebc350fea4b0a2443","subject":"Make the dead vlan actually dead","status":"MERGED","created":"2021-12-07 15:02:22.000000000","updated":"2022-01-19 23:15:58.000000000","submitted":"2022-01-19 23:13:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820897-bug/1930414","meta_rev_id":"63beb6285a18a249938935d87b961dc70a897955","_number":820897,"virtual_id_number":820897,"owner":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-01-12 16:15:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-19 23:13:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-19 18:45:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2022-01-13 14:05:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-01-19 12:49:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-14 01:45:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-19 18:45:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-07 16:41:20.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-12-07 16:47:21.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-07 20:33:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-08 07:43:39.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-01-07 12:04:50.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-01-13 14:05:44.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-01-14 01:45:37.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-01-19 12:49:24.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-01-19 18:45:15.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"9ad38f0824ed311667f422ebfd58ca59d589f01c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2021-12-07 15:02:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32886d734467b63329b98ceecae22b8790eb3228","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2021-12-07 15:03:51.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"b3cd9ffbeb732da2ad430ea6bb2f69b32e55eb92","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-07 16:41:20.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/820897/2 : SUCCESS in 1h 36m 49s","accounts_in_message":[],"_revision_number":2},{"id":"0c820a0ca90b98e219cd94537da6d0560eb0ec72","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-12-07 16:47:21.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2021/Arista_Tempest_Testrefs/changes/97/820897/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"fac0d23c8de0424dbf4a4032be04e837b455dc8f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-12-07 17:43:42.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2021/Arista_Tempest_Testrefs/changes/97/820897/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"befdc362ef23209d542f9364ada4964c8ee4ba70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 20:33:42.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c46f807b368b423aacf0dd6d5601e8a9 : SUCCESS in 43m 21s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ea7402552cc94e6f9cabe8bc6699d8c2 : SUCCESS in 2h 08m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/485ba0f35739458fa1ba1425e522ee9c : SUCCESS in 1h 39m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c80c9436824849f6bde4f346fcc7ef3d : SUCCESS in 43m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e594db4ad1a479d81a26b17ed790610 : SUCCESS in 37m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7692fbb51cca42deaaf65c0d58ba0224 : FAILURE in 9m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38bc6f8e70b44262b77610b9427995e8 : SUCCESS in 35m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fabe9e8e61245c7a9bc989143b80538 : SUCCESS in 34m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3dcea6975f74e14962ee7432d59c907 : SUCCESS in 10m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/912b056058af48eda3053b1c0e5d8c39 : FAILURE in 48m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5817101f4f6d4606a7f0c557a7aa3d1f : SUCCESS in 1h 31m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/69545ffa604448eb81edd852daaafc65 : SUCCESS in 1h 08m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/853c91e47848426baf1e8c97dd64ab33 : SUCCESS in 53m 45s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9aaf8c08a94e45298bd4706c661cdc4e : SUCCESS in 1h 06m 49s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dcd7a03b179c46f3b041b34c436defb3 : SUCCESS in 1h 19m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/19fd506ea80b4f929c65574f008945e0 : SUCCESS in 1h 01m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f20eec642024957b583d2bdcbca4db1 : SUCCESS in 23m 29s","accounts_in_message":[],"_revision_number":2},{"id":"ca825fb57e91582bae477f11b86b08cbcb24dff0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 21:34:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/89d0d0f2bd2243f3b8982fbd4241c290 : SUCCESS in 47m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1fa19aad29cf472a93a95e9304b7ec18 : SUCCESS in 41m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ca6cb20ef5acd4442f705ed9b1458006d464aa55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2021-12-08 07:32:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a03bf1e9cc0640ff7282879e2ac437be363816e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 08:20:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f23cb3b6828945a9b5c4f166729926f4 : SUCCESS in 45m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fda0a865e0ca4f648eed590aa51b6aa8 : SUCCESS in 40m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0f7cf68d34a79cc740e69f48157549baa81dc999","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-12-08 08:35:03.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2021/Arista_Tempest_Testrefs/changes/97/820897/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"2aca49f78cdd254ef0f1e799be7e8a8de589a73f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-08 09:04:54.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/820897/3 : SUCCESS in 1h 31m 04s","accounts_in_message":[],"_revision_number":3},{"id":"1979a0b7ae99460d94a97950290f00e6cf8d0b99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 09:54:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9bf3bdd5da0245309dc380ba336f52cc : SUCCESS in 1h 01m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/db0bc60f04c543a780aa64bbb3caff0f : SUCCESS in 2h 19m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e8c72cb766f2447bb55b6e2d606fecbd : SUCCESS in 1h 35m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d76062aa35e4f5ab2f4ed4d5e0f8941 : SUCCESS in 28m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/484fdc8148ea413a94568b080453867d : SUCCESS in 34m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/413dd632ff7744d088d5e3e5ecbf2f77 : SUCCESS in 11m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58b0cee4fd6042d99e1491d73b461511 : SUCCESS in 19m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9ccb2dfb77a497898119de106e5d32c : SUCCESS in 21m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfa3ebda75584b8c9b9cd7d644ad9773 : SUCCESS in 10m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a9fe736eae6448798e905abaff2b9cca : FAILURE in 59m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/61ae845f0fa143e7bde464b7e7555c34 : SUCCESS in 1h 48m 50s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0a7231f6046b447291ccf729f9a68068 : SUCCESS in 1h 17m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/24fa8c1684054403b18a9fa50a7bc8db : SUCCESS in 1h 02m 46s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/49dc9b2b79a44ccd958c6deb7a63d3a8 : SUCCESS in 59m 20s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/335a963110694986a2a2c74024ea5405 : SUCCESS in 1h 10m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9f9c9560846f4508abbeffe234222fbb : FAILURE in 1h 06m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/411207ca1d2a407c8708bfceeba0f1e6 : SUCCESS in 31m 07s","accounts_in_message":[],"_revision_number":3},{"id":"9647f5b4873ceacc62fe1f944315409051e67947","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2021-12-09 15:38:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"80ffd9e6a915dcbd66ff1d9dc89bc7b1d6bef4c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 16:32:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9cfd6daf0a24f6dacd038f34e0d5277 : SUCCESS in 47m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e4af7e17c9664358ae7d17c9ba710fcd : SUCCESS in 43m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"05d85d484e47d21b485f2603f6c41981f6af0837","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-12-09 16:54:44.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2021/Arista_Tempest_Testrefs/changes/97/820897/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"be1d134ac0c3faaa7ff9c7e922613a8c0157755e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-09 18:05:54.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/820897/4 : FAILURE in 2h 26m 22s","accounts_in_message":[],"_revision_number":4},{"id":"72a8b012000da2803c0440522be336039ea905e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 18:35:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0abe10987eae4b5fa9ac6fd36e10f079 : SUCCESS in 40m 04s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4b4e5a853e874556a57cb6922a5ee09e : SUCCESS in 2h 13m 30s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/72fe7175b1e6472f8114e943f72c780a : TIMED_OUT in 2h 50m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b73bd0769916460ca4c83080a632841b : TIMED_OUT in 2h 50m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/63482ef8078b485b8120d166283bb740 : SUCCESS in 2h 11m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/498a65c81f7b479e8f2ca7ec197b2ed0 : SUCCESS in 30m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca4bc367db0a4ec3a2590d69fc45c75d : SUCCESS in 34m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13ca38b459c74c1db986c35a60c6ebdd : SUCCESS in 13m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e91d4b3851154f6390afb55cfadc4b32 : SUCCESS in 21m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b60e921ef4624592a96e5bb4f96a5a24 : SUCCESS in 25m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2de0983733f34fa4950660f7087d678e : SUCCESS in 10m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cd9a354d603d4c239d3e8ae4786f1a71 : FAILURE in 46m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ceff9ae2c4484d47ad675f2c183e3444 : TIMED_OUT in 2h 16m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1ff92112f6844d58a8408aafb775120a : SUCCESS in 1h 09m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a0a189ee1f3d4e4fb5437970220051cc : FAILURE in 57m 57s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/449de378f62949278e2de65cf951372a : FAILURE in 1h 13m 15s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e51f817498e7413baf199aad2553a27b : FAILURE in 1h 28m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/29f007f3463740438c86c7e1af58ddd5 : FAILURE in 1h 22m 42s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67831e623f7747078869830367ef756e : SUCCESS in 59m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/43eb30130c20435191b03fd7f3cf72c0 : SUCCESS in 1h 09m 03s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d26f49f628cf4760a6491b46bec76d0d : SUCCESS in 21m 44s","accounts_in_message":[],"_revision_number":4},{"id":"e4766e7e9b9000f2b758dc698d4994c54c4400cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2021-12-14 14:00:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9345e7ad22ecb9e3495717d5257952b774480303","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2021-12-14 14:04:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4ae69c467f3d9d244937b58f10a6e822d5a5bd36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 14:47:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3fcbcbe5601c44fa82961d4ad336258f : SUCCESS in 44m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b212e8ff14f14446b916192a613f47d3 : SUCCESS in 43m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f07852ba68d317f4a5d7756a16c8512cbd71f4d1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-12-14 16:01:35.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2021/Arista_Tempest_Testrefs/changes/97/820897/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"a06a4f5b80c9fd33cb5eb8c870b3cebeda00875d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-14 16:07:58.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/820897/5 : FAILURE in 2h 06m 59s","accounts_in_message":[],"_revision_number":5},{"id":"09b4c3577fec71c2a352ef503f001113628bfa3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 16:59:23.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9e1a804ce641429b8ef1f2f53453aeb9 : SUCCESS in 46m 38s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/898999e1cde44d4abce74e5c195da1dc : FAILURE in 17m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/348e114a3e894617994610189a6c6346 : TIMED_OUT in 2h 52m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8758b90dd4dd418aabf867766ba9021f : TIMED_OUT in 2h 51m 46s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0c41261e781b46a09d46325cde4390c4 : FAILURE in 17m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41930a64ad15442881b30fe197e8fe85 : SUCCESS in 35m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16a10fc45d484649b5f6cb44611e37b2 : SUCCESS in 33m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2409e7eca4b64ffebb2ab51f12ac4d81 : SUCCESS in 14m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a80d3f0f28142a3b27af6ec32165d98 : SUCCESS in 26m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24440055cfa348d3abfa2845a87d5cb9 : SUCCESS in 27m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8333e615c5d046db971d22409b53e707 : SUCCESS in 8m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4af0879f05db4a4ba2a7a403250ace62 : FAILURE in 48m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9a1c095da109476ab9ae3f2950e41fa2 : FAILURE in 2h 14m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/dfdd51df7d42488e9b3c6e81ce33e4ac : SUCCESS in 1h 09m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/621b03e773ad462895fffe7c9154b5e3 : FAILURE in 1h 23m 20s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/710ff4e84d1642f991217f73bdc0cf2d : FAILURE in 49m 01s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/73c5c0820f85446a8243ecdec949aea2 : FAILURE in 1h 35m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dac5a6435e4542199768fa599c77335c : FAILURE in 1h 53m 31s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ff75f5d90de4d8fa582123bb86c971d : FAILURE in 17m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/199dde5c716a4b2f8ff1c7ed88350520 : SUCCESS in 1h 14m 24s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a80dd1b1b4b4cc38d2c8b3f5e3886a5 : SUCCESS in 26m 18s","accounts_in_message":[],"_revision_number":5},{"id":"1fc0b43c1da58948904c7509bed3623edc8db71c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2021-12-15 15:19:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e9b218902040f95cd322ceecc69cf53e3135bded","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2021-12-15 15:32:23.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"05a3611958309a073136a852e1c2e49bba1bd4ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2021-12-15 15:33:48.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a75ac4e97c56a835828e0d7095232d38d645e5a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 16:18:30.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/964341a11fe847369c372c3a0d771f34 : FAILURE in 42m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/812a85e4786d4fab8b0b219e1a946a9c : FAILURE in 42m 01s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6d02eb5195ef39b6002ee0a940a744a2eaff7a2f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-12-15 16:40:40.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2021/Arista_Tempest_Testrefs/changes/97/820897/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"921ab54c3a6aabe00cc6cf75e220d5a1fb70ac54","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-12-15 17:33:05.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2021/Arista_Tempest_Testrefs/changes/97/820897/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"c34c7d39b3bff6a05caf6cd207a3d49c1e4da92d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-15 17:37:31.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/820897/8 : FAILURE in 2h 03m 00s","accounts_in_message":[],"_revision_number":8},{"id":"b185b6d5d288e6394f67ab87196a4cbad44b9af3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 18:31:01.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4abb6e9d48cd4d6b937c1c88ce687c2c : SUCCESS in 35m 47s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/008acc4d632f485f89c7999d96b0e454 : SUCCESS in 1h 26m 39s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fb752f92ac5844d1a8e3d6635d4fcbac : TIMED_OUT in 2h 52m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2f0a61d631c04fc991edf88eb1a88032 : TIMED_OUT in 2h 50m 50s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/21b2205fe6804f1b8e72ceb9d67a1abd : SUCCESS in 1h 08m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2025007c27bf4b36ba5344b18caa8da5 : FAILURE in 30m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b4b204bfdd14e1e89a5ff28489295c4 : FAILURE in 33m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be121ef7a24e4c489c67fec24f023d58 : SUCCESS in 12m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4f8c340c24d49cc9fe50f6cd6137cdb : FAILURE in 27m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c7d7b6e289a4c70b68342a1b4723394 : FAILURE in 24m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79e4956a3db14a2fa3ef1d11f162cbb3 : SUCCESS in 10m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/db05503fb0414df08000dccdebdb30e4 : FAILURE in 42m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0cf416bc58424cdc9df8533ce6aba60e : FAILURE in 2h 14m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3da7e86d554c4b9e9a123b034358ac7e : SUCCESS in 1h 21m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d30d5b6d4036446a8b2bc80881a566d5 : FAILURE in 59m 44s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2f6581f3c5c7437787c8bcb789c1b15d : FAILURE in 59m 29s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d8e3b9217cc84b66bf97a994fe3e8fc9 : FAILURE in 1h 16m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b235bde8997149a1bc6c7e326a85a6bc : FAILURE in 1h 29m 21s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/805da705681d4892b3ddd8b88143f623 : SUCCESS in 49m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7f6d902b6ddd44d4aa0a8e2e9ea9cc9e : SUCCESS in 1h 13m 12s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60f47193f2a24814936b506bca1f2239 : FAILURE in 21m 57s","accounts_in_message":[],"_revision_number":8},{"id":"c8e5845896ba17e3babd0c7dfe04fd4e9003bc22","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-12-15 18:31:14.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2021/Arista_Tempest_Testrefs/changes/97/820897/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"0e4ba6c2d5c0a0fe1c3c8258f4fbfe710aeaa70d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2021-12-16 14:03:31.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"1c942d5cfd81afbe7a859d8624d200d46c1f76df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 14:51:07.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/177eab57be104665b652e9305cdf6397 : SUCCESS in 45m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cdf2c622447c4dab8e24c2e088ca7e17 : SUCCESS in 44m 31s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"cf7b62a76f462ee060b545902da0bdc9eb3c6407","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-16 15:59:02.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/820897/9 : FAILURE in 1h 53m 28s","accounts_in_message":[],"_revision_number":9},{"id":"b871f7799947ee35128edfa4a6a20d82e5c956a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 17:02:12.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b62bf26dcd6c43b4a60d65fad95e5495 : SUCCESS in 43m 42s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3743f4b87d9b4de1b5ab88da8fe4cbc5 : SUCCESS in 1h 09m 19s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7fdcc12ee82f408ab259e2f62a93ab04 : TIMED_OUT in 2h 51m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f27571adbf544eeda8545116b34d76fe : TIMED_OUT in 2h 51m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9659295b0b5f429ba89a069b3336f791 : SUCCESS in 47m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9fa548c2e2145f783099833c310fd2c : SUCCESS in 33m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bcfa39dc3558419887ed060f5ef9ed26 : SUCCESS in 33m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/312f8f9c10be4846a69efdc7d5fcb5b6 : SUCCESS in 11m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57b0c1fd1d7c488b8ddd3f26a4b3e866 : SUCCESS in 20m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8162803d2a014346ad4a11929700b482 : SUCCESS in 22m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/290ac4aead264670bcc5f95961ff779d : SUCCESS in 8m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c5119286b91244aca6494f4a91f5de54 : FAILURE in 45m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2e0846ca74f44ab089207f3e953cf5b4 : FAILURE in 2h 09m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7664c64d7d704a04b1c3f1f9853c1db8 : SUCCESS in 1h 02m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1b878a3b50024745829ba9204ddd37b0 : FAILURE in 1h 07m 05s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9291c3e465fe44408909874d01f2c43b : FAILURE in 1h 05m 42s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a5eee2b0309847a9940831fdbb9911bd : FAILURE in 1h 21m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/be0749a1e9044717a1c32ca77076d29a : FAILURE in 1h 31m 49s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c2fcaccd802448dafca45a139cd4076 : SUCCESS in 59m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/02acda380cd4429f893d8fa4e764670a : SUCCESS in 1h 04m 03s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9a24135231f421a88bbc1f88b07e788 : SUCCESS in 28m 50s","accounts_in_message":[],"_revision_number":9},{"id":"403b64e36c6da13b9878802891bd3431886ab428","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-12-16 18:09:59.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2021/Arista_Tempest_Testrefs/changes/97/820897/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"1882c6948876a161c122f2c4978b221facadaf5d","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2021-12-21 12:13:40.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9d3d5c9daaa103eff93500ad516b25c29a5b26fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 13:11:32.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f37a1fd1f824b55b1455f21d8781fc9 : SUCCESS in 46m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d41c77eedd82463aa516759e842ae8d2 : SUCCESS in 48m 10s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"58bcd69ffa2b400bd7e9b17f672b1b3d12811fa0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-21 14:23:31.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/820897/9 : FAILURE in 2h 08m 16s","accounts_in_message":[],"_revision_number":9},{"id":"3c710bd83ccfc0b013e860ec820a861a466e9892","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 15:11:28.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b1abcb1887d1435c872aaacee30c91d3 : SUCCESS in 36m 13s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a0158bb8cfa644a9b353ce52f4596586 : SUCCESS in 2h 08m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/86ad34ec9e57490d89d8e25edf5d60c3 : FAILURE in 21m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7fe86a49d53b4008a4ca20f6d87167d3 : TIMED_OUT in 2h 52m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d397e01bab744feab7c85f6d704c8418 : SUCCESS in 1h 32m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/703b7493f2144346b8490ca053d1f776 : SUCCESS in 30m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e956927c2b3e448582d2ec52e1efba44 : SUCCESS in 43m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e3fbff4fc9b42abad44ba44675cac7a : SUCCESS in 12m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bb090f53b69493188c8c6e2f2019f9b : SUCCESS in 34m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba8872c37948402fa471c3199a8fc16b : SUCCESS in 21m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34f654a7f83342eb8d611074a433fcc6 : SUCCESS in 9m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1da6ab76cc704b0fac730f9f47f50160 : FAILURE in 51m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/91ebf42fff6441ad841582022d68af82 : TIMED_OUT in 2h 16m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1729a69e199d40858d5f973885678585 : SUCCESS in 1h 21m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d3e82186f8a649a990ab25bc4cc47a2c : FAILURE in 1h 00m 24s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ec1713d160854b61b0e0ab0bc6859900 : FAILURE in 1h 01m 25s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/aa55f95f648d46189255d9e45fea3af2 : FAILURE in 1h 22m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c55c375beaa24b70948f734964ed215b : FAILURE in 1h 31m 27s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/077a31c41c6e41669cd2d69167e36f05 : SUCCESS in 46m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5c955f99dd274de1b8fdd30a23653eae : SUCCESS in 1h 05m 41s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6131453592bc48a091afd185af4a45d4 : SUCCESS in 26m 59s","accounts_in_message":[],"_revision_number":9},{"id":"0b9ce2b8287890f7dcb14f5e98f7aa3d30c6829a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2021-12-22 15:17:56.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"5b9fd2d926f3043b9ae8f5c07d21eced6a81fea6","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2021-12-22 15:18:30.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"95fc43ab6e393e09a19d041f6048a278ae197657","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 15:56:32.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b33ddf1a512c478ab626886d4d9ae18c : SUCCESS in 32m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5efe3f5cea074d4aa9c9f6366d3d941b : SUCCESS in 30m 35s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5e5848d304fb465451f40f6e45374ed5297daa98","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-12-22 16:05:54.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2021/Arista_Tempest_Testrefs/changes/97/820897/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"b22aab4574f82aa3fd7ca13aff81cb63d6bb3db3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-22 17:26:06.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/820897/10 : FAILURE in 2h 06m 16s","accounts_in_message":[],"_revision_number":10},{"id":"7c48afad60b9e0bf455c1e4df22dc7e233cc0582","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 18:14:12.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e26218667442432d82bd04d73af08118 : SUCCESS in 42m 25s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c4d6494559a94c12b309ba99e622dfaf : SUCCESS in 1h 21m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4a0412c67aad44e9afce9a74d9072263 : TIMED_OUT in 2h 50m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/41fe3ec876554957babdbf6bd51b8560 : TIMED_OUT in 2h 52m 50s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/be8fecd0847e4674a214c304f917d6ca : SUCCESS in 1h 01m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0856b110fc3b4ff0acf0e8f130b24fbe : SUCCESS in 35m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65fd255e295b474d8e1a3629be2fbd7b : SUCCESS in 30m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e196af6db664a77a7eb2df4782e8970 : SUCCESS in 33m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54e3a16c4b1f4f2e8dfd8910723dfaff : SUCCESS in 13m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25902bbf00524074878a779605acb0a0 : SUCCESS in 29m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/673c16abb1354d49a1fda94210c2156f : SUCCESS in 24m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f743f49b52aa4fa9b3f97cf9ac409fb1 : SUCCESS in 10m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d62834d8e2aa4764a48f526e8a80cb4d : FAILURE in 47m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b411ac89df5d4856915b4987df3ce790 : FAILURE in 1h 24m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9a8e6895575c42fbbf4335497eabe260 : SUCCESS in 1h 08m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/35275ee297f1446e9ad24152666c5b14 : FAILURE in 1h 06m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/493c4195cbcb4308a49c6cd529e88569 : FAILURE in 59m 56s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/70dad6d0a9544c4c874a317004f8067c : FAILURE in 2h 04m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ee299c0aeb204310bd9798f5029a62ad : FAILURE in 1h 16m 22s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9922f9ae26d6494f866c8ab382379529 : SUCCESS in 1h 02m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6de2f6cd917c4d6fb78d148682034180 : SUCCESS in 1h 33m 30s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ed42f94a500471e80156a12d87f69c6 : SUCCESS in 32m 30s","accounts_in_message":[],"_revision_number":10},{"id":"1d3169691416f4bfbda18b96881b658e47e3df12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2022-01-03 12:43:29.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"553f1aea66538ebf73390f7a8c26323da98859bf","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2022-01-03 12:44:10.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"85163f87467a42ac78776ab100c7fdbef85baf3b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-03 12:47:59.000000000","message":"Patch Set 11:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/97/820897/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"78cff913282c2ff5351daa01f0b5dae00a9bd98b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 13:25:39.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2225944124b947fe9b22b84d0ed78cf2 : FAILURE in 41m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b7b82fc770074ee2bdbbfb89c54e855c : FAILURE in 40m 06s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c7f601e734e3baf58fed76005d8c593951ff282c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 14:56:38.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/61216ee078064fe4a3050034d1b6bbe7 : SUCCESS in 44m 59s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5517464f798f4982af40bb8c0cd8ec49 : SUCCESS in 57m 39s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/853c7511c17d44a09c178853d5c7978e : SUCCESS in 2h 10m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ffa6030b174a4e838aeeeef483eadb5c : SUCCESS in 1h 04m 05s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/aacfff373efa484c99cb91423c0c1c10 : SUCCESS in 57m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5be0573e48454c8385400cf099396e53 : SUCCESS in 36m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f21927c38dd2489597ea921caa0e904b : FAILURE in 40m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52227514769341e6af5d6e1c8dcba117 : FAILURE in 27m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73d85eb1d72344c7ae7b8e1c12f44461 : SUCCESS in 14m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22d7eadf00bf4004bfeaa415e1217ce1 : FAILURE in 27m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9acc024e40d4c488c6ffac2f8b638a3 : FAILURE in 31m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68da2fa003a141c98650f05c86971eea : SUCCESS in 10m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/005f9700be374cacb8cf00b4c81ec2c9 : SUCCESS in 53m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4cbbcd6f275a4c4989a1d9852dd7b157 : SUCCESS in 1h 22m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ad7fa10014ae4bc7b4dd3b070b0c9329 : SUCCESS in 1h 03m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7574772ea6af4e1abd4ae9659fd8396a : SUCCESS in 57m 25s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/efdd3104081644f0b73a4f2dc4e0625c : SUCCESS in 54m 36s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b2f88c666ae3419d8534305a1edd32a3 : SUCCESS in 1h 13m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/021d8ac654674d6b9b1d8bf7365f14bc : SUCCESS in 1h 13m 12s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7de41eaf42a742eb80157b6fee89dc02 : SUCCESS in 57m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/44f4443925ba46d298900b81724eeaf5 : SUCCESS in 1h 13m 39s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9ba7b06a76f48f7babb05b74dc6c97f : FAILURE in 26m 47s","accounts_in_message":[],"_revision_number":11},{"id":"1ad735a3cda17610381fa3d1e40b670f315e6a0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2022-01-03 16:19:22.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"2aa9a1a659edd6f8ae662e53ec42280a7e704953","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-03 16:23:57.000000000","message":"Patch Set 12:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/97/820897/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"2702a02348e3a74e183b23c630a1dd84ddcb9572","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 17:13:58.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/888f843fca26425cb486b4655ea9a855 : SUCCESS in 33m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/452f8661e1bc45859e108eafe9c4d1b9 : SUCCESS in 37m 45s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"4a07c6f827113df3109612730758af97209362da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 18:05:50.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8c2f5be58b8a40b2957ce6f65b4e5357 : SUCCESS in 1h 02m 52s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/19d6aa9970e34892a0726b5b1b21e349 : SUCCESS in 1h 42m 38s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8a84a59cbfa14cadacd3663844a740dd : SUCCESS in 1h 05m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e1eb0ae8ad4f4e95a4cbc206ca6638d7 : SUCCESS in 1h 19m 37s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/53796b17b482483b87f70e488f48d689 : SUCCESS in 56m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4d979b9809564d148b0868df4d4090da : SUCCESS in 32m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/245e60f1d9804ec3b011e9f0f078a805 : SUCCESS in 44m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/409976604596475ea107516be2d601d1 : SUCCESS in 39m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56f1178e0ae94fb5a3ed5e75e95d05ad : SUCCESS in 12m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1df3f100857f42a5beb788cfa9bfcc6c : SUCCESS in 31m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e56755a2e2d2482fab4070bff85365f9 : SUCCESS in 30m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/523b13754e47410087fa0f0077eea98f : SUCCESS in 9m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef4bd40e899e442dbede5f6d6a26d714 : SUCCESS in 1h 03m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9688df68eb4f4352a9b7c25cb720284b : SUCCESS in 1h 21m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/aa4d885c6434424e8d50b1c1aa72c059 : SUCCESS in 1h 00m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5184186990af444baa21f227af5a56b1 : SUCCESS in 1h 01m 52s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4148c4c7e20c431cb3fb214276fbedb6 : SUCCESS in 52m 32s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8e0b4225c9be4ab68b3b9ab54675aa73 : SUCCESS in 1h 03m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f8271dc00bdb4e6194bc7e3ca27a5102 : SUCCESS in 1h 14m 05s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44cfbf71369841a3bbc4d0e1fca4d4a2 : SUCCESS in 1h 10m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ea5629fcf43145a98883d7168d9ef637 : SUCCESS in 1h 13m 57s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b61fc4504b254e9cbbeae6a1d8d68a77 : SUCCESS in 32m 59s","accounts_in_message":[],"_revision_number":12},{"id":"2206570466f4d09671696e2f9eb79fe5d33c526e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-07 12:04:50.000000000","message":"Patch Set 12: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":12},{"id":"9bfa39358f08723eda7c5c90823feb81e350cde6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2022-01-11 14:42:22.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"b07b04fd5342d6539016748d51c19e88f59bcdd2","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2022-01-11 14:43:02.000000000","message":"Patch Set 13:\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"2985d9bc28b60eb1dcfbf2f5cae60c56a335ea27","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-11 15:34:46.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/97/820897/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"a20d6d93df5546eff6a4ee86a73157e8a544b543","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 15:57:11.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/91b9183dc24848e3835a58c399bd7966 : FAILURE in 47m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/71737271ee6f4706993d8dacb0895004 : FAILURE in 48m 30s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"0342c9b453f4308679ae56ad3121e5bbc88dbf29","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-01-11 16:18:27.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/820897/13 : SUCCESS in 1h 35m 13s","accounts_in_message":[],"_revision_number":13},{"id":"ffdcf6e4dbbe8d0c2d202b80b22d01846b1b46b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 18:08:32.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d4724f04341a474fa0dd0cd641e5e4a3 : SUCCESS in 49m 22s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/82e335f74a6745b686ab496e6f3f6ff0 : SUCCESS in 1h 04m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f8aa496ca1fb47eeb116d72bf674c71f : SUCCESS in 1h 02m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b615cbb0547044c197979e07f1d6a5ef : SUCCESS in 51m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a8f7da9ffab547ef90dc8bf1316b4662 : SUCCESS in 52m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2aa54de926c04e7b83b3c6c3622e38ce : SUCCESS in 27m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16548398abb84981aa11f79b9e4e0938 : FAILURE in 32m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80740612ea3349c2b706495f1e5bbaf3 : FAILURE in 33m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a788e0292db34df594c0ac091ca7aff3 : SUCCESS in 13m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/574e3ee760664bfa9bbd2b4564e14baf : FAILURE in 28m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5a3689244e2429b83dc43420a116c10 : FAILURE in 27m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/858ae8e5ed734a6f8327060d04190d7e : SUCCESS in 11m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1c0ee9e307334ebbbf7eb99184166f13 : SUCCESS in 51m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f82d34434ea94f71b940269c96afae34 : SUCCESS in 1h 33m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7cec3b7a156a4db0b110e2fd0e84ca46 : SUCCESS in 1h 08m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/601f37c739594b48b06d8bc1bb2ec6df : SUCCESS in 1h 01m 16s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d7d592bf37374706811fe86ea18bb0a9 : SUCCESS in 1h 00m 49s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a92b29a6832d4414a30a87b20085ee17 : SUCCESS in 1h 21m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4de79713e2ab4d809eadc2260c4ef101 : SUCCESS in 1h 19m 08s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9977cff9e7a24d0588430723de7b8c2c : SUCCESS in 57m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2f62a8515ee1487d9d1be84ee4aca36e : SUCCESS in 1h 12m 48s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/900dab8657cf4ad39ccc45c9212c9d73 : FAILURE in 25m 35s","accounts_in_message":[],"_revision_number":13},{"id":"c982885ca90d6df5c878784fef645fa3c6b5f1a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2022-01-12 11:00:52.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"17ed9b29935263804a3ee0c61bd29d612c341f8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2022-01-12 11:09:02.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"13563523c626f6e6fe9b9d41190fdb737cef2e86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 12:08:31.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fea9aff1374b4d568031e5084344b988 : SUCCESS in 36m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/207fea87dc194d6da38d651a6814e0ab : SUCCESS in 38m 30s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"13ec22ee216f6d99cd062a426b4c6d5399e5bf7d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-01-12 12:57:25.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/820897/15 : SUCCESS in 1h 47m 38s","accounts_in_message":[],"_revision_number":15},{"id":"e0518f1f3b4821ccccbfc87c3d5436a6f1ff2919","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-12 14:25:55.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/97/820897/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"22b452e83e193fabc9cee0a2929da44bc91b7721","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-12 16:15:40.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/97/820897/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"5d01415cebe7cfb002b5012442f83f4ff86f1609","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 18:51:24.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cc95ae6a42674b63ae2b85dee6d6eb85 : SUCCESS in 49m 03s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a056e3fbdd454283b9781ad64b189149 : SUCCESS in 1h 11m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/33cd12a379f3435395e3d63f49fdf311 : SUCCESS in 56m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c8a920c2c4b04f30b87e0293b1d2132b : SUCCESS in 47m 26s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0f89ef9e9c1348b2991b9cf62bbeeb05 : SUCCESS in 49m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/55c1f14c27e44817a8cc844773a976fd : SUCCESS in 36m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf1f89fb7b1e48f7be8c1216c09b98f3 : SUCCESS in 36m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a7141da82d4464497ad756f988945d2 : SUCCESS in 36m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/549d3a6dc0b14d5cbc16dee1fe88a2d3 : SUCCESS in 18m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3eb95e2076c4024a4b93de02221b760 : SUCCESS in 28m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/faf92635b8f24a32ba7244cb30c469cf : SUCCESS in 26m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/964c66e1c08d4a7aace56e6a2bffdee7 : SUCCESS in 10m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/59ddf5c78ea34360803951ef00dbff35 : SUCCESS in 1h 00m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dec2324f271c4281aad0417cdc80c4c6 : SUCCESS in 2h 07m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/059ea61375744abb9e93347739bb50c4 : SUCCESS in 1h 16m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ac1b67ebeaf04f5e89e18696de3db237 : SUCCESS in 57m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/97a579b34a9147ae8ff3c2231e3d3811 : SUCCESS in 59m 33s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9d7925fa45bb4ba6a5b90e039ad8eb0d : SUCCESS in 1h 14m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fea40d5a9be8457ea08d1535629370cb : SUCCESS in 1h 16m 09s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d94f672b49474983bba8c277754fcdb5 : SUCCESS in 59m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/09e4c58ee1c142078a9ed245701f0e88 : SUCCESS in 1h 10m 11s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be7d9eaa4eb144d48cf73ee7a6da2b5b : SUCCESS in 25m 59s","accounts_in_message":[],"_revision_number":15},{"id":"d9aea820fa4aa66d6ace56793c9615bcd025e8a1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-13 14:05:44.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"364d6e5b72188c0a7f9ce2d5ffe7a3b661346661","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-01-14 01:45:37.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"a653290c36e7b7a7773e40298da6da5a80679c84","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-01-19 12:49:24.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"996f52fc075a3becb6a04063671a77ead12dd6d8","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-01-19 18:45:15.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"bfb65fdbe244565d5046566bb124d928a1d6d872","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 18:48:11.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"46696b6fc26b31f098ab694bced333426bac7d8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 23:13:22.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/75994f96e6914834a53dafdfe73c8b98 : SUCCESS in 47m 44s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5cd27e01edfc4997b42a95c068e03777 : SUCCESS in 59m 21s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/678ea83a81eb474ab13bb1f6bb01940a : SUCCESS in 1h 03m 39s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/976b82189d724073b6ffe2857e817a53 : SUCCESS in 50m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f8da01e961944f969250a35f7accb67b : SUCCESS in 51m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b294ff9627a4bb5846383487c718561 : SUCCESS in 37m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb8b4f7dcabb435abe07bb2446f4df6d : SUCCESS in 15m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb19b5aa434b4f7898454aafec4c4876 : SUCCESS in 29m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41dfc5735b364a14856adcaa31eea8b8 : SUCCESS in 25m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/870bf1eec7b64420be3b058beb311466 : SUCCESS in 10m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9b61be74755f458c99f4c983aa63141f : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a1a9cf841d240aead89c836c031da3c : SUCCESS in 9m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eb1163029ed141939eb8c5d60219d28c : SUCCESS in 51m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cdd2faad5d964941a000955f5225bdad : SUCCESS in 1h 37m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e484b03d32b94f6ea435c7f2cfe58d18 : SUCCESS in 1h 32m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4120bf41e68245e39de05c5dac917eaf : SUCCESS in 56m 43s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81d2caad120644f493cc4feb36bf22a8 : SUCCESS in 1h 01m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8629f4a32a1343cb95ee609ece0605f0 : SUCCESS in 22m 57s","accounts_in_message":[],"_revision_number":15},{"id":"3b11c5fac88830f5ed3fac451925469726299fa4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 23:13:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"63beb6285a18a249938935d87b961dc70a897955","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 23:15:58.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/588cbc892d2c45039fb8cf6d842d10de : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"7aae31c9f9ed938760ca0be3c461826b598c7004","revisions":{"6ad2855b4d3c6cb6fe6fa1dd23c02cda37c2ce5a":{"kind":"REWORK","_number":1,"created":"2021-12-07 15:02:22.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/97/820897/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/820897/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/820897/1"}}},"commit":{"parents":[{"commit":"5204e08d648061004b9e2b2201cba04460f0facc","subject":"WIP sleep to raise the chance of crosstalk between dhcp ports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5204e08d648061004b9e2b2201cba04460f0facc"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-10-05 15:02:41.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-12-07 14:57:52.000000000","tz":60},"subject":"A step toward making the dead vlan actually dead","message":"A step toward making the dead vlan actually dead\n\nAll ports plugged into the dead vlan (DEAD_VLAN_TAG 4095 or 0xfff)\nshould not be able to send or receive traffic. We install a flow\nto br-int to drop all traffic of the dead vlan [1]. However before\nthis patch the flow we install looks like:\n\npriority\u003d65535,vlan_tci\u003d0x0fff/0x1fff actions\u003ddrop\n\nWhich is wrong and it usually does not match anything.\n\nAccording to ovs-fields (7) section Open vSwitch Extension VLAN Field,\nVLAN TCI Field [2] (see especially the usage example\nvlan_tci\u003d0x1123/0x1fff) we need to explicitly set the bit 0x1000\nto match the presence of an 802.1Q header.\n\nSetting that bit this flow becomes:\npriority\u003d65535,vlan_tci\u003d0x1fff/0x1fff actions\u003ddrop\n\nwhich is equivalent to:\npriority\u003d65535,dl_vlan\u003d4095 actions\u003ddrop\n\nwhich I believe should match and drop dead vlan traffic.\n\nHowever as discussed in the bug report this is clearly not fixing the\nfull root cause which I still do not understand.\n\nAnyway, I\u0027m uploading this patch because I believe this is already\nfixing something wrong.\n\n[1] https://opendev.org/openstack/neutron/src/commit/ecdc11a56448428f77f5a64fd028f1e4c9644ea3/neutron/plugins/ml2/drivers/openvswitch/agent/openflow/native/br_int.py#L60-L62\n[2] http://www.openvswitch.org/support/dist-docs/ovs-fields.7.html\n\nChange-Id: Ib6b70114efb140cf1393b57ebc350fea4b0a2443\nPartial-Bug: #1930414\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ad2855b4d3c6cb6fe6fa1dd23c02cda37c2ce5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ad2855b4d3c6cb6fe6fa1dd23c02cda37c2ce5a"}]},"branch":"refs/heads/master"},"c55b68c2b752b66201d92a0e85b6f6f38f22bcab":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-12-07 15:03:51.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/97/820897/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/820897/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/820897/2"}}},"commit":{"parents":[{"commit":"9f0bff72ba780fb104a0b04389110daa86276b24","subject":"Merge \"[OVN] Chassis name (OVS system-id) must be a UUID formatted string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f0bff72ba780fb104a0b04389110daa86276b24"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-10-05 15:02:41.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-12-07 15:03:37.000000000","tz":60},"subject":"A step toward making the dead vlan actually dead","message":"A step toward making the dead vlan actually dead\n\nAll ports plugged into the dead vlan (DEAD_VLAN_TAG 4095 or 0xfff)\nshould not be able to send or receive traffic. We install a flow\nto br-int to drop all traffic of the dead vlan [1]. However before\nthis patch the flow we install looks like:\n\npriority\u003d65535,vlan_tci\u003d0x0fff/0x1fff actions\u003ddrop\n\nWhich is wrong and it usually does not match anything.\n\nAccording to ovs-fields (7) section Open vSwitch Extension VLAN Field,\nVLAN TCI Field [2] (see especially the usage example\nvlan_tci\u003d0x1123/0x1fff) we need to explicitly set the bit 0x1000\nto match the presence of an 802.1Q header.\n\nSetting that bit this flow becomes:\npriority\u003d65535,vlan_tci\u003d0x1fff/0x1fff actions\u003ddrop\n\nwhich is equivalent to:\npriority\u003d65535,dl_vlan\u003d4095 actions\u003ddrop\n\nwhich I believe should match and drop dead vlan traffic.\n\nHowever as discussed in the bug report this is clearly not fixing the\nfull root cause which I still do not understand.\n\nAnyway, I\u0027m uploading this patch because I believe this is already\nfixing something wrong.\n\n[1] https://opendev.org/openstack/neutron/src/commit/ecdc11a56448428f77f5a64fd028f1e4c9644ea3/neutron/plugins/ml2/drivers/openvswitch/agent/openflow/native/br_int.py#L60-L62\n[2] http://www.openvswitch.org/support/dist-docs/ovs-fields.7.html\n\nChange-Id: Ib6b70114efb140cf1393b57ebc350fea4b0a2443\nPartial-Bug: #1930414\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c55b68c2b752b66201d92a0e85b6f6f38f22bcab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c55b68c2b752b66201d92a0e85b6f6f38f22bcab"}]},"branch":"refs/heads/master"},"656f696d50276ec868c5444f7f03a7ee555ee273":{"kind":"REWORK","_number":3,"created":"2021-12-08 07:32:51.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/97/820897/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/820897/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/820897/3"}}},"commit":{"parents":[{"commit":"8d647c18ef48f68aa84dd3f76541b0f34c4e692e","subject":"Merge \"Add \"update_network\" implementation to \"L3AgentExtension\" child classes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d647c18ef48f68aa84dd3f76541b0f34c4e692e"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-10-05 15:02:41.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-12-08 07:32:39.000000000","tz":60},"subject":"A step toward making the dead vlan actually dead","message":"A step toward making the dead vlan actually dead\n\nAll ports plugged into the dead vlan (DEAD_VLAN_TAG 4095 or 0xfff)\nshould not be able to send or receive traffic. We install a flow\nto br-int to drop all traffic of the dead vlan [1]. However before\nthis patch the flow we install looks like:\n\npriority\u003d65535,vlan_tci\u003d0x0fff/0x1fff actions\u003ddrop\n\nWhich is wrong and it usually does not match anything.\n\nAccording to ovs-fields (7) section Open vSwitch Extension VLAN Field,\nVLAN TCI Field [2] (see especially the usage example\nvlan_tci\u003d0x1123/0x1fff) we need to explicitly set the bit 0x1000\nto match the presence of an 802.1Q header.\n\nSetting that bit this flow becomes:\npriority\u003d65535,vlan_tci\u003d0x1fff/0x1fff actions\u003ddrop\n\nwhich is equivalent to:\npriority\u003d65535,dl_vlan\u003d4095 actions\u003ddrop\n\nwhich I believe should match and drop dead vlan traffic.\n\nHowever as discussed in the bug report this is clearly not fixing the\nfull root cause which I still do not understand.\n\nAnyway, I\u0027m uploading this patch because I believe this is already\nfixing something wrong.\n\n[1] https://opendev.org/openstack/neutron/src/commit/ecdc11a56448428f77f5a64fd028f1e4c9644ea3/neutron/plugins/ml2/drivers/openvswitch/agent/openflow/native/br_int.py#L60-L62\n[2] http://www.openvswitch.org/support/dist-docs/ovs-fields.7.html\n\nChange-Id: Ib6b70114efb140cf1393b57ebc350fea4b0a2443\nPartial-Bug: #1930414\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/656f696d50276ec868c5444f7f03a7ee555ee273"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/656f696d50276ec868c5444f7f03a7ee555ee273"}]},"branch":"refs/heads/master"},"78a92388f56d5c91a1ad0b3aeb1f56bceb4fcccf":{"kind":"REWORK","_number":4,"created":"2021-12-09 15:38:51.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/97/820897/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/820897/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/820897/4"}}},"commit":{"parents":[{"commit":"2a359891ce87157e91ba715e022e6d09c5ba161b","subject":"Merge \"Improve Router callback system\u0027s publish events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a359891ce87157e91ba715e022e6d09c5ba161b"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-10-05 15:02:41.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-12-09 15:31:01.000000000","tz":60},"subject":"WIP Make the dead vlan actually dead","message":"WIP Make the dead vlan actually dead\n\nAll ports plugged into the dead vlan (DEAD_VLAN_TAG 4095 or 0xfff)\nshould not be able to send or receive traffic. We install a flow\nto br-int to drop all traffic of the dead vlan [1]. However before\nthis patch the flow we install looks like:\n\npriority\u003d65535,vlan_tci\u003d0x0fff/0x1fff actions\u003ddrop\n\nWhich is wrong and it usually does not match anything.\n\nAccording to ovs-fields (7) section Open vSwitch Extension VLAN Field,\nVLAN TCI Field [2] (see especially the usage example\nvlan_tci\u003d0x1123/0x1fff) we need to explicitly set the bit 0x1000\nto match the presence of an 802.1Q header.\n\nSetting that bit this flow becomes:\npriority\u003d65535,vlan_tci\u003d0x1fff/0x1fff actions\u003ddrop\n\nwhich is equivalent to:\npriority\u003d65535,dl_vlan\u003d4095 actions\u003ddrop\n\nwhich should match and drop dead vlan traffic.\n\nHowever there\u0027s a second problem: ovs access ports were designed to\nwork together with the NORMAL action. The NORMAL action considers the\nvlan of an access port, but the openflow pipeline does not. An openflow\nrule does not see the vlan set for an access port, because that vlan\ntag is only pushed to the frame if and when the frame leaves the switch\non a trunk port [3][4].\n\nSo we have to explicitly push the DEAD_VLAN_TAG if we want to keep\nthe concept of the dead vlan.\n\nTODO Do we need the dead vlan? If we have to have per-port flows to\n     get traffic into the dead vlan, than those per-port flows could\n     directly drop the traffic in the first place. But if we completely\n     eliminate the concept of the dead vlan, then a port no longer has\n     a property (stored in ovsdb) marking the port\u0027s dead state - for\n     example to be used when the flows are lost. Maybe we shall keep\n     the dead vlan for this state, but drop the traffic directly?\nTODO Re-think the tables, priorities, flows.\n     The current model is just a hack to test an idea.\nTODO Find all places where we set or clear the dead vlan tag.\n     setting:\n     https://opendev.org/openstack/neutron/src/commit/2a359891ce87157e91ba715e022e6d09c5ba161b/neutron/agent/common/ovs_lib.py#L379-L380\n     https://opendev.org/openstack/neutron/src/commit/2a359891ce87157e91ba715e022e6d09c5ba161b/neutron/plugins/ml2/drivers/openvswitch/agent/ovs_neutron_agent.py#L1361-L1363\n     clearing:\n     https://opendev.org/openstack/neutron/src/commit/2a359891ce87157e91ba715e022e6d09c5ba161b/neutron/plugins/ml2/drivers/openvswitch/agent/ovs_neutron_agent.py#L1225-L1226\nTODO regression avoidance tests\n\n[1] https://opendev.org/openstack/neutron/src/commit/ecdc11a56448428f77f5a64fd028f1e4c9644ea3/neutron/plugins/ml2/drivers/openvswitch/agent/openflow/native/br_int.py#L60-L62\n[2] http://www.openvswitch.org/support/dist-docs/ovs-fields.7.html\n[3] https://mail.openvswitch.org/pipermail/ovs-discuss/2021-December/051647.html\n[4] https://docs.openvswitch.org/en/latest/faq/vlan/\n    see \u0027Q: My OpenFlow controller doesn’t see the VLANs that I expect.\u0027\n\nChange-Id: Ib6b70114efb140cf1393b57ebc350fea4b0a2443\nCloses-Bug: #1930414\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78a92388f56d5c91a1ad0b3aeb1f56bceb4fcccf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78a92388f56d5c91a1ad0b3aeb1f56bceb4fcccf"}]},"branch":"refs/heads/master"},"69e311cb749892f63d1050d8c8921cf61db6eea2":{"kind":"REWORK","_number":5,"created":"2021-12-14 14:00:16.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/97/820897/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/820897/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/820897/5"}}},"commit":{"parents":[{"commit":"2a359891ce87157e91ba715e022e6d09c5ba161b","subject":"Merge \"Improve Router callback system\u0027s publish events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a359891ce87157e91ba715e022e6d09c5ba161b"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-10-05 15:02:41.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-12-14 13:40:42.000000000","tz":60},"subject":"Make the dead vlan actually dead","message":"Make the dead vlan actually dead\n\nAll ports plugged into the dead vlan (DEAD_VLAN_TAG 4095 or 0xfff)\nshould not be able to send or receive traffic. We install a flow\nto br-int to drop all traffic of the dead vlan [1]. However before\nthis patch the flow we install looks like:\n\npriority\u003d65535,vlan_tci\u003d0x0fff/0x1fff actions\u003ddrop\n\nWhich is wrong and it usually does not match anything.\n\nAccording to ovs-fields (7) section Open vSwitch Extension VLAN Field,\nVLAN TCI Field [2] (see especially the usage example\nvlan_tci\u003d0x1123/0x1fff) we need to explicitly set the bit 0x1000\nto match the presence of an 802.1Q header.\n\nSetting that bit this flow becomes:\npriority\u003d65535,vlan_tci\u003d0x1fff/0x1fff actions\u003ddrop\n\nwhich is equivalent to:\npriority\u003d65535,dl_vlan\u003d4095 actions\u003ddrop\n\nwhich should match and drop dead vlan traffic.\n\nHowever there\u0027s a second problem: ovs access ports were designed to\nwork together with the NORMAL action. The NORMAL action considers the\nvlan of an access port, but the openflow pipeline does not. An openflow\nrule does not see the vlan set for an access port, because that vlan\ntag is only pushed to the frame if and when the frame leaves the switch\non a trunk port [3][4].\n\nSo we have to explicitly push the DEAD_VLAN_TAG if we want the dead\nvlan\u0027s drop flow match anything.\n\n[1] https://opendev.org/openstack/neutron/src/commit/ecdc11a56448428f77f5a64fd028f1e4c9644ea3/neutron/plugins/ml2/drivers/openvswitch/agent/openflow/native/br_int.py#L60-L62\n[2] http://www.openvswitch.org/support/dist-docs/ovs-fields.7.html\n[3] https://mail.openvswitch.org/pipermail/ovs-discuss/2021-December/051647.html\n[4] https://docs.openvswitch.org/en/latest/faq/vlan/\n    see \u0027Q: My OpenFlow controller doesn’t see the VLANs that I expect.\u0027\n\nChange-Id: Ib6b70114efb140cf1393b57ebc350fea4b0a2443\nCloses-Bug: #1930414\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69e311cb749892f63d1050d8c8921cf61db6eea2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69e311cb749892f63d1050d8c8921cf61db6eea2"}]},"branch":"refs/heads/master"},"7fee14e75d8c6c05aa27e86f7240fff66060f277":{"kind":"REWORK","_number":6,"created":"2021-12-15 15:19:18.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/97/820897/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/820897/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/820897/6"}}},"commit":{"parents":[{"commit":"1af63e881261ac3a160e9e6d2bea89ac65856cea","subject":"Add Chassis creation wait event in \"TestAgentMonitor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1af63e881261ac3a160e9e6d2bea89ac65856cea"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-10-05 15:02:41.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-12-15 15:18:30.000000000","tz":60},"subject":"Make the dead vlan actually dead","message":"Make the dead vlan actually dead\n\nAll ports plugged into the dead vlan (DEAD_VLAN_TAG 4095 or 0xfff)\nshould not be able to send or receive traffic. We install a flow\nto br-int to drop all traffic of the dead vlan [1]. However before\nthis patch the flow we install looks like:\n\npriority\u003d65535,vlan_tci\u003d0x0fff/0x1fff actions\u003ddrop\n\nWhich is wrong and it usually does not match anything.\n\nAccording to ovs-fields (7) section Open vSwitch Extension VLAN Field,\nVLAN TCI Field [2] (see especially the usage example\nvlan_tci\u003d0x1123/0x1fff) we need to explicitly set the bit 0x1000\nto match the presence of an 802.1Q header.\n\nSetting that bit this flow becomes:\npriority\u003d65535,vlan_tci\u003d0x1fff/0x1fff actions\u003ddrop\n\nwhich is equivalent to:\npriority\u003d65535,dl_vlan\u003d4095 actions\u003ddrop\n\nwhich should match and drop dead vlan traffic.\n\nHowever there\u0027s a second problem: ovs access ports were designed to\nwork together with the NORMAL action. The NORMAL action considers the\nvlan of an access port, but the openflow pipeline does not. An openflow\nrule does not see the vlan set for an access port, because that vlan\ntag is only pushed to the frame if and when the frame leaves the switch\non a trunk port [3][4].\n\nSo we have to explicitly push the DEAD_VLAN_TAG if we want the dead\nvlan\u0027s drop flow match anything.\n\n[1] https://opendev.org/openstack/neutron/src/commit/ecdc11a56448428f77f5a64fd028f1e4c9644ea3/neutron/plugins/ml2/drivers/openvswitch/agent/openflow/native/br_int.py#L60-L62\n[2] http://www.openvswitch.org/support/dist-docs/ovs-fields.7.html\n[3] https://mail.openvswitch.org/pipermail/ovs-discuss/2021-December/051647.html\n[4] https://docs.openvswitch.org/en/latest/faq/vlan/\n    see \u0027Q: My OpenFlow controller doesn’t see the VLANs that I expect.\u0027\n\nChange-Id: Ib6b70114efb140cf1393b57ebc350fea4b0a2443\nCloses-Bug: #1930414\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7fee14e75d8c6c05aa27e86f7240fff66060f277"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7fee14e75d8c6c05aa27e86f7240fff66060f277"}]},"branch":"refs/heads/master"},"40ed9361d62bb7217d2e0f9cb6d244bd54a5438b":{"kind":"REWORK","_number":7,"created":"2021-12-15 15:32:23.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/97/820897/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/820897/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/820897/7"}}},"commit":{"parents":[{"commit":"1af63e881261ac3a160e9e6d2bea89ac65856cea","subject":"Add Chassis creation wait event in \"TestAgentMonitor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1af63e881261ac3a160e9e6d2bea89ac65856cea"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-10-05 15:02:41.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-12-15 15:31:36.000000000","tz":60},"subject":"Make the dead vlan actually dead","message":"Make the dead vlan actually dead\n\nAll ports plugged into the dead vlan (DEAD_VLAN_TAG 4095 or 0xfff)\nshould not be able to send or receive traffic. We install a flow\nto br-int to drop all traffic of the dead vlan [1]. However before\nthis patch the flow we install looks like:\n\npriority\u003d65535,vlan_tci\u003d0x0fff/0x1fff actions\u003ddrop\n\nWhich is wrong and it usually does not match anything.\n\nAccording to ovs-fields (7) section Open vSwitch Extension VLAN Field,\nVLAN TCI Field [2] (see especially the usage example\nvlan_tci\u003d0x1123/0x1fff) we need to explicitly set the bit 0x1000\nto match the presence of an 802.1Q header.\n\nSetting that bit this flow becomes:\npriority\u003d65535,vlan_tci\u003d0x1fff/0x1fff actions\u003ddrop\n\nwhich is equivalent to:\npriority\u003d65535,dl_vlan\u003d4095 actions\u003ddrop\n\nwhich should match and drop dead vlan traffic.\n\nHowever there\u0027s a second problem: ovs access ports were designed to\nwork together with the NORMAL action. The NORMAL action considers the\nvlan of an access port, but the openflow pipeline does not. An openflow\nrule does not see the vlan set for an access port, because that vlan\ntag is only pushed to the frame if and when the frame leaves the switch\non a trunk port [3][4].\n\nSo we have to explicitly push the DEAD_VLAN_TAG if we want the dead\nvlan\u0027s drop flow match anything.\n\n[1] https://opendev.org/openstack/neutron/src/commit/ecdc11a56448428f77f5a64fd028f1e4c9644ea3/neutron/plugins/ml2/drivers/openvswitch/agent/openflow/native/br_int.py#L60-L62\n[2] http://www.openvswitch.org/support/dist-docs/ovs-fields.7.html\n[3] https://mail.openvswitch.org/pipermail/ovs-discuss/2021-December/051647.html\n[4] https://docs.openvswitch.org/en/latest/faq/vlan/\n    see \u0027Q: My OpenFlow controller doesn’t see the VLANs that I expect.\u0027\n\nChange-Id: Ib6b70114efb140cf1393b57ebc350fea4b0a2443\nCloses-Bug: #1930414\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/40ed9361d62bb7217d2e0f9cb6d244bd54a5438b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/40ed9361d62bb7217d2e0f9cb6d244bd54a5438b"}]},"branch":"refs/heads/master"},"06c48a2712cf154ff69b47e0c09f3b7c10074f9b":{"kind":"REWORK","_number":8,"created":"2021-12-15 15:33:48.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/97/820897/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/820897/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/820897/8"}}},"commit":{"parents":[{"commit":"1af63e881261ac3a160e9e6d2bea89ac65856cea","subject":"Add Chassis creation wait event in \"TestAgentMonitor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1af63e881261ac3a160e9e6d2bea89ac65856cea"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-10-05 15:02:41.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-12-15 15:32:56.000000000","tz":60},"subject":"Make the dead vlan actually dead","message":"Make the dead vlan actually dead\n\nAll ports plugged into the dead vlan (DEAD_VLAN_TAG 4095 or 0xfff)\nshould not be able to send or receive traffic. We install a flow\nto br-int to drop all traffic of the dead vlan [1]. However before\nthis patch the flow we install looks like:\n\npriority\u003d65535,vlan_tci\u003d0x0fff/0x1fff actions\u003ddrop\n\nWhich is wrong and it usually does not match anything.\n\nAccording to ovs-fields (7) section Open vSwitch Extension VLAN Field,\nVLAN TCI Field [2] (see especially the usage example\nvlan_tci\u003d0x1123/0x1fff) we need to explicitly set the bit 0x1000\nto match the presence of an 802.1Q header.\n\nSetting that bit this flow becomes:\npriority\u003d65535,vlan_tci\u003d0x1fff/0x1fff actions\u003ddrop\n\nwhich is equivalent to:\npriority\u003d65535,dl_vlan\u003d4095 actions\u003ddrop\n\nwhich should match and drop dead vlan traffic.\n\nHowever there\u0027s a second problem: ovs access ports were designed to\nwork together with the NORMAL action. The NORMAL action considers the\nvlan of an access port, but the openflow pipeline does not. An openflow\nrule does not see the vlan set for an access port, because that vlan\ntag is only pushed to the frame if and when the frame leaves the switch\non a trunk port [3][4].\n\nSo we have to explicitly push the DEAD_VLAN_TAG if we want the dead\nvlan\u0027s drop flow match anything.\n\n[1] https://opendev.org/openstack/neutron/src/commit/ecdc11a56448428f77f5a64fd028f1e4c9644ea3/neutron/plugins/ml2/drivers/openvswitch/agent/openflow/native/br_int.py#L60-L62\n[2] http://www.openvswitch.org/support/dist-docs/ovs-fields.7.html\n[3] https://mail.openvswitch.org/pipermail/ovs-discuss/2021-December/051647.html\n[4] https://docs.openvswitch.org/en/latest/faq/vlan/\n    see \u0027Q: My OpenFlow controller doesn’t see the VLANs that I expect.\u0027\n\nChange-Id: Ib6b70114efb140cf1393b57ebc350fea4b0a2443\nCloses-Bug: #1930414\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/06c48a2712cf154ff69b47e0c09f3b7c10074f9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/06c48a2712cf154ff69b47e0c09f3b7c10074f9b"}]},"branch":"refs/heads/master"},"ac0381f5261a963a0a4e7a6b9e693e2150965417":{"kind":"REWORK","_number":9,"created":"2021-12-16 14:03:31.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/97/820897/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/820897/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/820897/9"}}},"commit":{"parents":[{"commit":"1af63e881261ac3a160e9e6d2bea89ac65856cea","subject":"Add Chassis creation wait event in \"TestAgentMonitor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1af63e881261ac3a160e9e6d2bea89ac65856cea"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-10-05 15:02:41.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-12-16 14:02:41.000000000","tz":60},"subject":"Make the dead vlan actually dead","message":"Make the dead vlan actually dead\n\nAll ports plugged into the dead vlan (DEAD_VLAN_TAG 4095 or 0xfff)\nshould not be able to send or receive traffic. We install a flow\nto br-int to drop all traffic of the dead vlan [1]. However before\nthis patch the flow we install looks like:\n\npriority\u003d65535,vlan_tci\u003d0x0fff/0x1fff actions\u003ddrop\n\nWhich is wrong and it usually does not match anything.\n\nAccording to ovs-fields (7) section Open vSwitch Extension VLAN Field,\nVLAN TCI Field [2] (see especially the usage example\nvlan_tci\u003d0x1123/0x1fff) we need to explicitly set the bit 0x1000\nto match the presence of an 802.1Q header.\n\nSetting that bit this flow becomes:\npriority\u003d65535,vlan_tci\u003d0x1fff/0x1fff actions\u003ddrop\n\nwhich is equivalent to:\npriority\u003d65535,dl_vlan\u003d4095 actions\u003ddrop\n\nwhich should match and drop dead vlan traffic.\n\nHowever there\u0027s a second problem: ovs access ports were designed to\nwork together with the NORMAL action. The NORMAL action considers the\nvlan of an access port, but the openflow pipeline does not. An openflow\nrule does not see the vlan set for an access port, because that vlan\ntag is only pushed to the frame if and when the frame leaves the switch\non a trunk port [3][4].\n\nSo we have to explicitly push the DEAD_VLAN_TAG if we want the dead\nvlan\u0027s drop flow match anything.\n\n[1] https://opendev.org/openstack/neutron/src/commit/ecdc11a56448428f77f5a64fd028f1e4c9644ea3/neutron/plugins/ml2/drivers/openvswitch/agent/openflow/native/br_int.py#L60-L62\n[2] http://www.openvswitch.org/support/dist-docs/ovs-fields.7.html\n[3] https://mail.openvswitch.org/pipermail/ovs-discuss/2021-December/051647.html\n[4] https://docs.openvswitch.org/en/latest/faq/vlan/\n    see \u0027Q: My OpenFlow controller doesn’t see the VLANs that I expect.\u0027\n\nChange-Id: Ib6b70114efb140cf1393b57ebc350fea4b0a2443\nCloses-Bug: #1930414\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac0381f5261a963a0a4e7a6b9e693e2150965417"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac0381f5261a963a0a4e7a6b9e693e2150965417"}]},"branch":"refs/heads/master"},"c788962f31eb1d750ec164c0175ab1e3e1103f71":{"kind":"REWORK","_number":10,"created":"2021-12-22 15:17:56.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/97/820897/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/820897/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/820897/10"}}},"commit":{"parents":[{"commit":"19a20eaca8a51d6c095a68970478de847fb256f5","subject":"Merge \"[OVN] Accept OVS system-id as non UUID formatted string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19a20eaca8a51d6c095a68970478de847fb256f5"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-10-05 15:02:41.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-12-22 15:13:26.000000000","tz":60},"subject":"Make the dead vlan actually dead","message":"Make the dead vlan actually dead\n\nAll ports plugged into the dead vlan (DEAD_VLAN_TAG 4095 or 0xfff)\nshould not be able to send or receive traffic. We install a flow\nto br-int to drop all traffic of the dead vlan [1]. However before\nthis patch the flow we install looks like:\n\npriority\u003d65535,vlan_tci\u003d0x0fff/0x1fff actions\u003ddrop\n\nWhich is wrong and it usually does not match anything.\n\nAccording to ovs-fields (7) section Open vSwitch Extension VLAN Field,\nVLAN TCI Field [2] (see especially the usage example\nvlan_tci\u003d0x1123/0x1fff) we need to explicitly set the bit 0x1000\nto match the presence of an 802.1Q header.\n\nSetting that bit this flow becomes:\npriority\u003d65535,vlan_tci\u003d0x1fff/0x1fff actions\u003ddrop\n\nwhich is equivalent to:\npriority\u003d65535,dl_vlan\u003d4095 actions\u003ddrop\n\nwhich should match and drop dead vlan traffic.\n\nHowever there\u0027s a second problem: ovs access ports were designed to\nwork together with the NORMAL action. The NORMAL action considers the\nvlan of an access port, but the openflow pipeline does not. An openflow\nrule does not see the vlan set for an access port, because that vlan\ntag is only pushed to the frame if and when the frame leaves the switch\non a trunk port [3][4].\n\nSo we have to explicitly push the DEAD_VLAN_TAG if we want the dead\nvlan\u0027s drop flow match anything.\n\nThat means we are adding a flow to push the dead vlan tag from\ndhcp-agent but we are deleting that flow from ovs-agent right after\novs-agent sets the vlan tag of the port to a non-dead vlan. Which is\nugly but we have to keep adding the flow as early as possible if we\nwant to minimize the window until frames can leak onto the dead vlan.\nEven with change there\u0027s a short time window in which the dead vlan\ncould theoroetically leak.\n\n[1] https://opendev.org/openstack/neutron/src/commit/ecdc11a56448428f77f5a64fd028f1e4c9644ea3/neutron/plugins/ml2/drivers/openvswitch/agent/openflow/native/br_int.py#L60-L62\n[2] http://www.openvswitch.org/support/dist-docs/ovs-fields.7.html\n[3] https://mail.openvswitch.org/pipermail/ovs-discuss/2021-December/051647.html\n[4] https://docs.openvswitch.org/en/latest/faq/vlan/\n    see \u0027Q: My OpenFlow controller doesn’t see the VLANs that I expect.\u0027\n\nChange-Id: Ib6b70114efb140cf1393b57ebc350fea4b0a2443\nCloses-Bug: #1930414\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c788962f31eb1d750ec164c0175ab1e3e1103f71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c788962f31eb1d750ec164c0175ab1e3e1103f71"}]},"branch":"refs/heads/master"},"97e1c85aaa6038c01b532f86c13454d82f57a55d":{"kind":"REWORK","_number":11,"created":"2022-01-03 12:43:29.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/97/820897/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/820897/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/820897/11"}}},"commit":{"parents":[{"commit":"cc273b1fe41dac744c2a329ca31a269472124a20","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc273b1fe41dac744c2a329ca31a269472124a20"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-10-05 15:02:41.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2022-01-03 12:42:03.000000000","tz":60},"subject":"Make the dead vlan actually dead","message":"Make the dead vlan actually dead\n\nAll ports plugged into the dead vlan (DEAD_VLAN_TAG 4095 or 0xfff)\nshould not be able to send or receive traffic. We install a flow\nto br-int to drop all traffic of the dead vlan [1]. However before\nthis patch the flow we install looks like:\n\npriority\u003d65535,vlan_tci\u003d0x0fff/0x1fff actions\u003ddrop\n\nWhich is wrong and it usually does not match anything.\n\nAccording to ovs-fields (7) section Open vSwitch Extension VLAN Field,\nVLAN TCI Field [2] (see especially the usage example\nvlan_tci\u003d0x1123/0x1fff) we need to explicitly set the bit 0x1000\nto match the presence of an 802.1Q header.\n\nSetting that bit this flow becomes:\npriority\u003d65535,vlan_tci\u003d0x1fff/0x1fff actions\u003ddrop\n\nwhich is equivalent to:\npriority\u003d65535,dl_vlan\u003d4095 actions\u003ddrop\n\nwhich should match and drop dead vlan traffic.\n\nHowever there\u0027s a second problem: ovs access ports were designed to\nwork together with the NORMAL action. The NORMAL action considers the\nvlan of an access port, but the openflow pipeline does not. An openflow\nrule does not see the vlan set for an access port, because that vlan\ntag is only pushed to the frame if and when the frame leaves the switch\non a trunk port [3][4].\n\nSo we have to explicitly push the DEAD_VLAN_TAG if we want the dead\nvlan\u0027s drop flow match anything.\n\nThat means we are adding a flow to push the dead vlan tag from\ndhcp-agent but we are deleting that flow from ovs-agent right after\novs-agent sets the vlan tag of the port to a non-dead vlan. Which is\nugly but we have to keep adding the flow as early as possible if we\nwant to minimize the window until frames can leak onto the dead vlan.\nEven with change there\u0027s a short time window in which the dead vlan\ncould theoroetically leak.\n\n[1] https://opendev.org/openstack/neutron/src/commit/ecdc11a56448428f77f5a64fd028f1e4c9644ea3/neutron/plugins/ml2/drivers/openvswitch/agent/openflow/native/br_int.py#L60-L62\n[2] http://www.openvswitch.org/support/dist-docs/ovs-fields.7.html\n[3] https://mail.openvswitch.org/pipermail/ovs-discuss/2021-December/051647.html\n[4] https://docs.openvswitch.org/en/latest/faq/vlan/\n    see \u0027Q: My OpenFlow controller doesn’t see the VLANs that I expect.\u0027\n\nChange-Id: Ib6b70114efb140cf1393b57ebc350fea4b0a2443\nCloses-Bug: #1930414\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/97e1c85aaa6038c01b532f86c13454d82f57a55d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/97e1c85aaa6038c01b532f86c13454d82f57a55d"}]},"branch":"refs/heads/master"},"5d387bd7b3d3c6afac3d104a58e9d5c24edbdf00":{"kind":"REWORK","_number":12,"created":"2022-01-03 16:19:22.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/97/820897/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/820897/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/820897/12"}}},"commit":{"parents":[{"commit":"cc273b1fe41dac744c2a329ca31a269472124a20","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc273b1fe41dac744c2a329ca31a269472124a20"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-10-05 15:02:41.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2022-01-03 16:00:22.000000000","tz":60},"subject":"Make the dead vlan actually dead","message":"Make the dead vlan actually dead\n\nAll ports plugged into the dead vlan (DEAD_VLAN_TAG 4095 or 0xfff)\nshould not be able to send or receive traffic. We install a flow\nto br-int to drop all traffic of the dead vlan [1]. However before\nthis patch the flow we install looks like:\n\npriority\u003d65535,vlan_tci\u003d0x0fff/0x1fff actions\u003ddrop\n\nWhich is wrong and it usually does not match anything.\n\nAccording to ovs-fields (7) section Open vSwitch Extension VLAN Field,\nVLAN TCI Field [2] (see especially the usage example\nvlan_tci\u003d0x1123/0x1fff) we need to explicitly set the bit 0x1000\nto match the presence of an 802.1Q header.\n\nSetting that bit this flow becomes:\npriority\u003d65535,vlan_tci\u003d0x1fff/0x1fff actions\u003ddrop\n\nwhich is equivalent to:\npriority\u003d65535,dl_vlan\u003d4095 actions\u003ddrop\n\nwhich should match and drop dead vlan traffic.\n\nHowever there\u0027s a second problem: ovs access ports were designed to\nwork together with the NORMAL action. The NORMAL action considers the\nvlan of an access port, but the openflow pipeline does not. An openflow\nrule does not see the vlan set for an access port, because that vlan\ntag is only pushed to the frame if and when the frame leaves the switch\non a trunk port [3][4].\n\nSo we have to explicitly push the DEAD_VLAN_TAG if we want the dead\nvlan\u0027s drop flow match anything.\n\nThat means we are adding a flow to push the dead vlan tag from\ndhcp-agent but we are deleting that flow from ovs-agent right after\novs-agent sets the vlan tag of the port to a non-dead vlan. Which is\nugly but we have to keep adding the flow as early as possible if we\nwant to minimize the window until frames can leak onto the dead vlan.\nEven with change there\u0027s a short time window in which the dead vlan\ncould theoroetically leak.\n\n[1] https://opendev.org/openstack/neutron/src/commit/ecdc11a56448428f77f5a64fd028f1e4c9644ea3/neutron/plugins/ml2/drivers/openvswitch/agent/openflow/native/br_int.py#L60-L62\n[2] http://www.openvswitch.org/support/dist-docs/ovs-fields.7.html\n[3] https://mail.openvswitch.org/pipermail/ovs-discuss/2021-December/051647.html\n[4] https://docs.openvswitch.org/en/latest/faq/vlan/\n    see \u0027Q: My OpenFlow controller doesn’t see the VLANs that I expect.\u0027\n\nChange-Id: Ib6b70114efb140cf1393b57ebc350fea4b0a2443\nCloses-Bug: #1930414\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5d387bd7b3d3c6afac3d104a58e9d5c24edbdf00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5d387bd7b3d3c6afac3d104a58e9d5c24edbdf00"}]},"branch":"refs/heads/master"},"db8497dda4a7ce5f49de20774b7c92c2a1d0d921":{"kind":"REWORK","_number":13,"created":"2022-01-11 14:42:22.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/97/820897/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/820897/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/820897/13"}}},"commit":{"parents":[{"commit":"9a90af915dade2c7bbb4fbbc98902c954f6a00c3","subject":"Merge \"keepalived \"no_track\" implemented in version 2.0.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a90af915dade2c7bbb4fbbc98902c954f6a00c3"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-10-05 15:02:41.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2022-01-11 14:41:14.000000000","tz":60},"subject":"Make the dead vlan actually dead","message":"Make the dead vlan actually dead\n\nAll ports plugged into the dead vlan (DEAD_VLAN_TAG 4095 or 0xfff)\nshould not be able to send or receive traffic. We install a flow\nto br-int to drop all traffic of the dead vlan [1]. However before\nthis patch the flow we install looks like:\n\npriority\u003d65535,vlan_tci\u003d0x0fff/0x1fff actions\u003ddrop\n\nWhich is wrong and it usually does not match anything.\n\nAccording to ovs-fields (7) section Open vSwitch Extension VLAN Field,\nVLAN TCI Field [2] (see especially the usage example\nvlan_tci\u003d0x1123/0x1fff) we need to explicitly set the bit 0x1000\nto match the presence of an 802.1Q header.\n\nSetting that bit this flow becomes:\npriority\u003d65535,vlan_tci\u003d0x1fff/0x1fff actions\u003ddrop\n\nwhich is equivalent to:\npriority\u003d65535,dl_vlan\u003d4095 actions\u003ddrop\n\nwhich should match and drop dead vlan traffic.\n\nHowever there\u0027s a second problem: ovs access ports were designed to\nwork together with the NORMAL action. The NORMAL action considers the\nvlan of an access port, but the openflow pipeline does not. An openflow\nrule does not see the vlan set for an access port, because that vlan\ntag is only pushed to the frame if and when the frame leaves the switch\non a trunk port [3][4].\n\nSo we have to explicitly push the DEAD_VLAN_TAG if we want the dead\nvlan\u0027s drop flow match anything.\n\nThat means we are adding a flow to push the dead vlan tag from\ndhcp-agent but we are deleting that flow from ovs-agent right after\novs-agent sets the vlan tag of the port to a non-dead vlan. Which is\nugly but we have to keep adding the flow as early as possible if we\nwant to minimize the window until frames can leak onto the dead vlan.\nEven with this change there\u0027s a short time window in which the dead\nvlan could theoretically leak.\n\n[1] https://opendev.org/openstack/neutron/src/commit/ecdc11a56448428f77f5a64fd028f1e4c9644ea3/neutron/plugins/ml2/drivers/openvswitch/agent/openflow/native/br_int.py#L60-L62\n[2] http://www.openvswitch.org/support/dist-docs/ovs-fields.7.html\n[3] https://mail.openvswitch.org/pipermail/ovs-discuss/2021-December/051647.html\n[4] https://docs.openvswitch.org/en/latest/faq/vlan/\n    see \u0027Q: My OpenFlow controller doesn’t see the VLANs that I expect.\u0027\n\nChange-Id: Ib6b70114efb140cf1393b57ebc350fea4b0a2443\nCloses-Bug: #1930414\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/db8497dda4a7ce5f49de20774b7c92c2a1d0d921"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/db8497dda4a7ce5f49de20774b7c92c2a1d0d921"}]},"branch":"refs/heads/master"},"bdc6f5e4338bc921462a396ae3cf60f1a2d9942d":{"kind":"REWORK","_number":14,"created":"2022-01-12 11:00:52.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/97/820897/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/820897/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/820897/14"}}},"commit":{"parents":[{"commit":"9a90af915dade2c7bbb4fbbc98902c954f6a00c3","subject":"Merge \"keepalived \"no_track\" implemented in version 2.0.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a90af915dade2c7bbb4fbbc98902c954f6a00c3"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-10-05 15:02:41.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2022-01-12 10:58:35.000000000","tz":60},"subject":"Make the dead vlan actually dead","message":"Make the dead vlan actually dead\n\nAll ports plugged into the dead vlan (DEAD_VLAN_TAG 4095 or 0xfff)\nshould not be able to send or receive traffic. We install a flow\nto br-int to drop all traffic of the dead vlan [1]. However before\nthis patch the flow we install looks like:\n\npriority\u003d65535,vlan_tci\u003d0x0fff/0x1fff actions\u003ddrop\n\nWhich is wrong and it usually does not match anything.\n\nAccording to ovs-fields (7) section Open vSwitch Extension VLAN Field,\nVLAN TCI Field [2] (see especially the usage example\nvlan_tci\u003d0x1123/0x1fff) we need to explicitly set the bit 0x1000\nto match the presence of an 802.1Q header.\n\nSetting that bit this flow becomes:\npriority\u003d65535,vlan_tci\u003d0x1fff/0x1fff actions\u003ddrop\n\nwhich is equivalent to:\npriority\u003d65535,dl_vlan\u003d4095 actions\u003ddrop\n\nwhich should match and drop dead vlan traffic.\n\nHowever there\u0027s a second problem: ovs access ports were designed to\nwork together with the NORMAL action. The NORMAL action considers the\nvlan of an access port, but the openflow pipeline does not. An openflow\nrule does not see the vlan set for an access port, because that vlan\ntag is only pushed to the frame if and when the frame leaves the switch\non a trunk port [3][4].\n\nSo we have to explicitly push the DEAD_VLAN_TAG if we want the dead\nvlan\u0027s drop flow match anything.\n\nThat means we are adding a flow to push the dead vlan tag from\ndhcp-agent/l3-agent but we are deleting that flow from ovs-agent right\nafter ovs-agent sets the vlan tag of the port to a non-dead vlan. Which\nis ugly but we have to keep adding the flow as early as possible if we\nwant to minimize the window until frames can leak onto the dead vlan.\nEven with this change there\u0027s a short time window in which the dead vlan\ncould theoretically leak.\n\n[1] https://opendev.org/openstack/neutron/src/commit/ecdc11a56448428f77f5a64fd028f1e4c9644ea3/neutron/plugins/ml2/drivers/openvswitch/agent/openflow/native/br_int.py#L60-L62\n[2] http://www.openvswitch.org/support/dist-docs/ovs-fields.7.html\n[3] https://mail.openvswitch.org/pipermail/ovs-discuss/2021-December/051647.html\n[4] https://docs.openvswitch.org/en/latest/faq/vlan/\n    see \u0027Q: My OpenFlow controller doesn’t see the VLANs that I expect.\u0027\n\nChange-Id: Ib6b70114efb140cf1393b57ebc350fea4b0a2443\nCloses-Bug: #1930414\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdc6f5e4338bc921462a396ae3cf60f1a2d9942d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdc6f5e4338bc921462a396ae3cf60f1a2d9942d"}]},"branch":"refs/heads/master"},"7aae31c9f9ed938760ca0be3c461826b598c7004":{"kind":"REWORK","_number":15,"created":"2022-01-12 11:09:02.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/97/820897/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/820897/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/820897/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/820897/15"}}},"commit":{"parents":[{"commit":"9a90af915dade2c7bbb4fbbc98902c954f6a00c3","subject":"Merge \"keepalived \"no_track\" implemented in version 2.0.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a90af915dade2c7bbb4fbbc98902c954f6a00c3"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2021-10-05 15:02:41.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2022-01-12 11:06:12.000000000","tz":60},"subject":"Make the dead vlan actually dead","message":"Make the dead vlan actually dead\n\nAll ports plugged into the dead vlan (DEAD_VLAN_TAG 4095 or 0xfff)\nshould not be able to send or receive traffic. We install a flow\nto br-int to drop all traffic of the dead vlan [1]. However before\nthis patch the flow we install looks like:\n\npriority\u003d65535,vlan_tci\u003d0x0fff/0x1fff actions\u003ddrop\n\nWhich is wrong and it usually does not match anything.\n\nAccording to ovs-fields (7) section Open vSwitch Extension VLAN Field,\nVLAN TCI Field [2] (see especially the usage example\nvlan_tci\u003d0x1123/0x1fff) we need to explicitly set the bit 0x1000\nto match the presence of an 802.1Q header.\n\nSetting that bit this flow becomes:\npriority\u003d65535,vlan_tci\u003d0x1fff/0x1fff actions\u003ddrop\n\nwhich is equivalent to:\npriority\u003d65535,dl_vlan\u003d4095 actions\u003ddrop\n\nwhich should match and drop dead vlan traffic.\n\nHowever there\u0027s a second problem: ovs access ports were designed to\nwork together with the NORMAL action. The NORMAL action considers the\nvlan of an access port, but the openflow pipeline does not. An openflow\nrule does not see the vlan set for an access port, because that vlan\ntag is only pushed to the frame if and when the frame leaves the switch\non a trunk port [3][4].\n\nSo we have to explicitly push the DEAD_VLAN_TAG if we want the dead\nvlan\u0027s drop flow match anything.\n\nThat means we are adding a flow to push the dead vlan tag from\ndhcp-agent/l3-agent but we are deleting that flow from ovs-agent right\nafter ovs-agent sets the vlan tag of the port to a non-dead vlan. Which\nis ugly but we have to keep adding the flow as early as possible if we\nwant to minimize the window until frames can leak onto the dead vlan.\nEven with this change there\u0027s a short time window in which the dead vlan\ncould theoretically leak.\n\n[1] https://opendev.org/openstack/neutron/src/commit/ecdc11a56448428f77f5a64fd028f1e4c9644ea3/neutron/plugins/ml2/drivers/openvswitch/agent/openflow/native/br_int.py#L60-L62\n[2] http://www.openvswitch.org/support/dist-docs/ovs-fields.7.html\n[3] https://mail.openvswitch.org/pipermail/ovs-discuss/2021-December/051647.html\n[4] https://docs.openvswitch.org/en/latest/faq/vlan/\n    see \u0027Q: My OpenFlow controller doesn’t see the VLANs that I expect.\u0027\n\nChange-Id: Ib6b70114efb140cf1393b57ebc350fea4b0a2443\nCloses-Bug: #1930414\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7aae31c9f9ed938760ca0be3c461826b598c7004"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7aae31c9f9ed938760ca0be3c461826b598c7004"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
