)]}'
{"id":"openstack%2Fneutron~821182","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~I03878a2e650d5f0d0167e0f7daa351b12bae8725","project":"openstack/neutron","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-12-09 14:01:11.000000000","reason":"Rodolfo Alonso replied on the change"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2021-12-10 07:21:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I03878a2e650d5f0d0167e0f7daa351b12bae8725","subject":"[OVS] Workaround when OpenFlow controller restarts","status":"MERGED","created":"2021-12-09 13:17:51.000000000","updated":"2021-12-10 07:22:56.000000000","submitted":"2021-12-10 07:21:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821182","cherry_pick_of_change":821211,"cherry_pick_of_patch_set":1,"meta_rev_id":"217bccc0139c2262e0496569ad28f36dc207b837","_number":821182,"virtual_id_number":821182,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-10 07:21:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-12-09 14:01:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-12-09 21:05:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-12-09 21:05:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-09 13:17:51.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-12-09 14:16:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-09 15:11:15.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-12-09 16:15:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-09 21:05:53.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"08a0e3943cad0f92a061f4c92d9b30fd3d3cc3c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-12-09 13:17:51.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"0b36a2f03e5f391b54ffb12ff0058302dbaf2324","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-09 14:01:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cb4c0b710cf5c2751f6c45ee3ab6374d3c9c6738","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 14:16:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/09f5e4798bb7490bad0db0d1cd51feee : SUCCESS in 56m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c7f5e4f8745b4d9baebd5ea1b020c191 : SUCCESS in 56m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"173a1f0bf1282bc4855729add2554850bebe071f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-09 15:11:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/821182/1 : SUCCESS in 1h 52m 42s","accounts_in_message":[],"_revision_number":1},{"id":"56c58af2aab4a64510a622d3a9e73d8f9a3791da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 16:15:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge-wallaby https://zuul.opendev.org/t/openstack/build/aa9a24a1606b407eb7aa20b4a4bebbdf : SUCCESS in 2h 23m 59s\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/d263bc797c9140489e999bcfaab15b79 : SUCCESS in 2h 20m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-wallaby https://zuul.opendev.org/t/openstack/build/f6db8ac76a984cc5bae49947946c18dd : SUCCESS in 1h 36m 55s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/8611915d7b5b49a19d014a0305ceeeb3 : SUCCESS in 2h 46m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81ed4e5936544167b0b1dc0ec3a1664e : SUCCESS in 38m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40bfd1d758b04b458f5f9f57ab49bf1a : SUCCESS in 12m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecb401a7fc404382b6434278107afb1a : SUCCESS in 27m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bd368ad1e2a4e318e81ec9ac2a7ef03 : SUCCESS in 26m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7dc94bfb03354c10a700f6fb6dc93724 : SUCCESS in 26m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b1ec692ed36482e80ba75edb3924806 : SUCCESS in 11m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1d9c803f0d1f4e07aaf4d53b60d25813 : SUCCESS in 51m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/438f3e2b678244b4babfc6b6e7483f28 : SUCCESS in 1h 21m 02s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/544b3cf45ea64d97838547b2226e920d : SUCCESS in 1h 20m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/15b84e98065a4bb3a3821b4e7796000c : SUCCESS in 1h 02m 40s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ca0576e2e87d48b2ae4eafe400b39754 : SUCCESS in 1h 02m 00s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e70d7670b3dd4a00b10ea375861fb469 : SUCCESS in 1h 10m 18s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5f646ccc228a449f9279be17a1fcde3e : SUCCESS in 1h 20m 57s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b6a8e9f193d240728925eb1158e5f332 : SUCCESS in 1h 35m 54s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22129c6ab0b14771856d352a945f3021 : SUCCESS in 1h 03m 06s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6a932043f00b40fab4fdeb7e315e9de9 : SUCCESS in 1h 06m 21s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/99daf947debb470fbfec12e8829dba25 : SUCCESS in 35m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/23051061b7fe430181b769ecc60998cb : SUCCESS in 1h 26m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b326d2aed1e04143b982a503294ca719 : SUCCESS in 1h 16m 33s","accounts_in_message":[],"_revision_number":1},{"id":"02256680aab17976af0153766fa4860f8fb1d01a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-12-09 21:05:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"19a944f7cf791b1f40bfa99ee199e6bb9e5485a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 21:54:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47be553e8118eb6f77ac69ffb1c336051326afa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 00:04:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/349cf93af31a4811ad85a73ce8f7cb7a : SUCCESS in 13m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fd43bb877194452aee30a5ffb2516be : SUCCESS in 29m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b4c893c4c414f3096fb7ad0390f72b5 : SUCCESS in 25m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fdf6a0c0c7946c1bef3198c2318f753 : SUCCESS in 9m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/98dc5f6e4899464d9ac5af28958d77ac : SUCCESS in 38m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b8f2ff7e83c042b79aa905818ece5ffe : SUCCESS in 1h 20m 03s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a1b564b177b244179e63046a049c2b3c : SUCCESS in 50m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ecac7821a7c14e2d9f13f12599cf944e : FAILURE in 32m 31s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d8e82d36642a4c3ab3e505242b776767 : SUCCESS in 1h 51m 23s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1aae50e0205542fcb7ffe0020f959fd7 : SUCCESS in 1h 00m 07s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c7074c45b4cf45ecb0fec552854cdada : SUCCESS in 56m 28s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d5bea142a1d44fcebb4a46444e2e811b : SUCCESS in 34m 02s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4dd51246192f4f439ce6a3358b1199d5 : SUCCESS in 1h 16m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3307d88511a517df500c33e7787c35520e2103e9","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-12-10 02:51:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c2bb68bc1cec67879070c55f5544e23dab0468a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 03:47:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ae8f22f89b64ce1a48a23ac785dd4e9 : SUCCESS in 48m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db5eaa39cbe5446488e0bcd92612a6e4 : SUCCESS in 45m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e962f690552adc84b9f1f8b8793a127ee82e010","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-10 05:13:29.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/821182/1 : SUCCESS in 2h 21m 29s","accounts_in_message":[],"_revision_number":1},{"id":"925082b224114787fee55d20b1f48db29c7ec1b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 05:23:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge-wallaby https://zuul.opendev.org/t/openstack/build/c4be515d01ce498d96de062777b6dabc : SUCCESS in 2h 18m 00s\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/a06e3fa639674e8fbb16ed9d9aeefb90 : SUCCESS in 1h 33m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-wallaby https://zuul.opendev.org/t/openstack/build/6f8f10e06222463d8ad6a2783a41ab93 : SUCCESS in 1h 21m 37s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/3551b36bb521464880acbb96a478a01e : SUCCESS in 1h 55m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/007c4ebbe39742a8b8163c85785bdf3b : SUCCESS in 33m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d15f5173310049d6a9e942250b7e0b2f : SUCCESS in 13m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b53bfc2034fd4d31b30210bb7b7ce36e : SUCCESS in 29m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b36ece338085413e8d1c76dd5e0af78b : SUCCESS in 26m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7bcc306d2c44b1cb1b8263c34285038 : SUCCESS in 28m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55edd09d90e545448fb568e5ecd9b287 : SUCCESS in 11m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9bb3d9fcb34e4249b2bdf7a15a13569d : SUCCESS in 45m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e33a4a39ae8e48669500332a267caf10 : SUCCESS in 1h 16m 10s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/66ef794228ae4ebeb6cc7563c37847db : SUCCESS in 1h 10m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2c541d540e2140ae8788dd292b8a14c9 : SUCCESS in 59m 05s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ee7012c2b00f43c6957a8bb717aa2304 : SUCCESS in 54m 39s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ddaafc39970444959c8ca90222cac143 : SUCCESS in 1h 12m 31s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/636b2a02e35f499e87956e117b28c773 : SUCCESS in 1h 14m 27s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/597b48129bca448d87553624df72c7ea : SUCCESS in 1h 38m 44s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3cda8de4b7514fc8b77ecf5c5d58014c : SUCCESS in 1h 05m 06s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0127f806152f4877b8efd95827c10b7f : SUCCESS in 1h 03m 15s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/fc3700b3c96c412a918e773ec3a04438 : SUCCESS in 38m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/612b6c421d4048ee98503918097d4d00 : SUCCESS in 1h 31m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/638b64241e054d9583f1b3d2ce664055 : SUCCESS in 1h 26m 12s","accounts_in_message":[],"_revision_number":1},{"id":"ea63e1eba4fc89ff8a451cf8cd9c9ed73eae5e6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 05:24:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9876e9fef762ca7c9396aef0078d4666d2af2149","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 07:21:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e14950cbc76342d2bf28c49452ce96e5 : SUCCESS in 12m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/540676dec1cc4e2c86abc2639e8c69b9 : SUCCESS in 24m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcf2f501c8ad4f56b779ee10167c84e4 : SUCCESS in 22m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82be3a746f944f338b30c8d63a52c476 : SUCCESS in 9m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/da188a588b884c5182a0663e57e7461e : SUCCESS in 46m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/715a276671f3468990f3e030cb618ab8 : SUCCESS in 1h 20m 04s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3334f6833c5f47d683dd51a8922e5770 : SUCCESS in 1h 07m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d91e9274a887457ab8d143caf7eab9b8 : SUCCESS in 57m 53s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/16940a587b4d4745a75de5dfdaac44c2 : SUCCESS in 1h 56m 06s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e3be85e2d4e49d08f631507feadc0a6 : SUCCESS in 52m 22s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7ada2479fd744400bf688b9967804bdb : SUCCESS in 54m 31s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1ef2cf520e2b40a298f989c984901310 : SUCCESS in 30m 16s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/cbd0f7aacc164d6ab3c727f176189c56 : SUCCESS in 1h 05m 19s","accounts_in_message":[],"_revision_number":1},{"id":"565836bf2d0a9470731fec6db1abbb2851337e29","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 07:21:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"217bccc0139c2262e0496569ad28f36dc207b837","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 07:22:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bcae4715f2d4ea59398aa93bd74b203 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc0d42e0dfb94b580f98765c06a823d92d07df45","revisions":{"cc0d42e0dfb94b580f98765c06a823d92d07df45":{"kind":"REWORK","_number":1,"created":"2021-12-09 13:17:51.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/82/821182/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/821182/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/821182/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/821182/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/821182/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/821182/1"}}},"commit":{"parents":[{"commit":"82c517d8ff4f80604c4c8f351b0ead6eeaa01439","subject":"Merge \"[OVN] Fix gateway_mtu option should not always be set\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/82c517d8ff4f80604c4c8f351b0ead6eeaa01439"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-10-26 09:21:33.000000000","tz":0},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2021-12-09 13:17:51.000000000","tz":0},"subject":"[OVS] Workaround when OpenFlow controller restarts","message":"[OVS] Workaround when OpenFlow controller restarts\n\nWhen a new OF controller is added or when new OF protocols are added to\na bridge, the OF controller is restarted. When that happens, the os-ken\n``OfctlService`` instance removes the OF controller datapath and waits\nfor an update event that will add a new one. During this time no OF\ncontrollers (datapaths) will be registered and any request will be\nrejected with a ``InvalidDatapath`` exception.\n\nThis patch is a temporary workaround that prevents any command from\nbeing executed just after those two described operations, providing\ntime to OVS to send the OF updates.\n\nConflicts:\n\tneutron/agent/common/ovs_lib.py\n\nRelated-Bug: #1948642\nChange-Id: I03878a2e650d5f0d0167e0f7daa351b12bae8725\n(cherry picked from commit 0194856da19f0ce96b613968ab415821037f18d0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc0d42e0dfb94b580f98765c06a823d92d07df45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc0d42e0dfb94b580f98765c06a823d92d07df45"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
