)]}'
{"id":"openstack%2Fneutron~821599","triplet_id":"openstack%2Fneutron~master~I4594a622339335a8f304ff4827c71cb63729faf3","project":"openstack/neutron","branch":"master","topic":"bug/1799790","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2021-12-20 16:05:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4594a622339335a8f304ff4827c71cb63729faf3","subject":"Add logs for port_bound_to_host","status":"MERGED","created":"2021-12-13 15:49:34.000000000","updated":"2021-12-20 16:08:42.000000000","submitted":"2021-12-20 16:05:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821599-bug/1799790","meta_rev_id":"3bd9be8df5a34ce284610facfb99de74c90e8d67","_number":821599,"virtual_id_number":821599,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-12-14 11:31:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-20 16:05:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-20 13:27:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-12-14 10:55:54.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-14 11:47:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-20 13:27:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-13 16:29:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-13 17:28:36.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-12-13 18:18:28.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-13 18:18:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-14 10:55:54.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-12-14 11:47:20.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2021-12-14 20:35:45.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-20 13:27:06.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"49c92470959e3cd46763b7f63102a32a4e2796f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-12-13 15:49:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7affc9ad9bf082ef6340a096a3ee2c279ebe03ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 16:29:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4853116030594f69861f9d63e18ca2fb : SUCCESS in 34m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2d083079a76e49b8a6046647c25b28a7 : SUCCESS in 34m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a39c730e0f354695992e79109e0efee5d1607ef4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-13 17:28:36.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/821599/1 : SUCCESS in 1h 38m 21s","accounts_in_message":[],"_revision_number":1},{"id":"42729babe8cd5409acb09d58827037e08211afc2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-12-13 18:18:28.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2021/Arista_Tempest_Testrefs/changes/99/821599/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"2a58bc7525ad96cdd1b5bff4ac48e3bdc133d226","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 18:18:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4240abe5ae0f413bafa296d3117098ed : SUCCESS in 44m 25s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b72f32334df046c295ab6b170acdb21a : SUCCESS in 2h 18m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0fd315dc01434be19044230dd14c0d30 : SUCCESS in 2h 09m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ace855f6d5ad4ef6a0b931ee0c79a38e : SUCCESS in 1h 46m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/97f873d1af614eb0967f87d763f9765d : SUCCESS in 2h 22m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/103fe89a06af4656afa4780f901fef95 : SUCCESS in 40m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f15ce0ea20d4bdda9f58e6896548763 : SUCCESS in 35m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aeabc2ce20dc4087ab90d902f9134e3d : SUCCESS in 26m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c60a56a3501e471db06873718e636177 : SUCCESS in 13m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d6ea1ec8c8e4a2896b2f9512cd9b027 : SUCCESS in 21m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0579772c99d403e81c8a0d2dbe8d7b7 : SUCCESS in 25m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ad9940c7d2b4bd095f523457bad9d84 : SUCCESS in 10m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a10deeb9189e47f38c1990657e8f962a : SUCCESS in 46m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/30a08e4cab8646468cd4706fb6c43a0d : SUCCESS in 1h 17m 54s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e7e71d08a7b44e8a93be2d8d4dde4431 : SUCCESS in 1h 02m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ddc16578d8d74c1ca9b80b549b1f021f : SUCCESS in 43m 18s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b41d980603284d34896cf84efa9ebdab : SUCCESS in 55m 33s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6cdc60b8f16f4554b8875c5786968f4a : SUCCESS in 1h 12m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a0848baa05ae4619b411a3a6d5bc46a0 : SUCCESS in 1h 06m 29s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91dce3c2f0844c1793af2aec43f4e7ac : SUCCESS in 38m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1bd6f21578b34fa5a7cb2118460944ad : SUCCESS in 1h 04m 40s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e40afe6d964c499f8b80f943a58367ec : SUCCESS in 27m 35s","accounts_in_message":[],"_revision_number":1},{"id":"be39b3927a16d0f81511fdee62b4e31f5ec3edd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-12-14 10:27:10.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2e8512835c5cba0c5eb6a7e6e464c504a40d60ac","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-12-14 10:55:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"51e1d0f569c05fd2154bddbfd9bba26173a2ff61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 11:13:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a30fda0ca367476492f561a4527d10af : SUCCESS in 33m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f4e3b7c394a5457c8e6279fdb2989bb6 : SUCCESS in 44m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c85ab9d5998a3ae4d4280d662bd314e622558704","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-12-14 11:31:35.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2021/Arista_Tempest_Testrefs/changes/99/821599/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"444b49eba6ece142d7822f5de8f3713f9c3ec18a","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-12-14 11:47:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"520c18f6a0504ec9461c3d69295e3d1ce4e5786c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-14 11:59:16.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/821599/2 : SUCCESS in 1h 31m 25s","accounts_in_message":[],"_revision_number":2},{"id":"52f1d8b68b10569aca70de8f04fbe144c9707047","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 12:37:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/be50c9fd30fe42bc94b966dd38c4ec10 : SUCCESS in 53m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9537af52d8ab455cb4681a1b632418a0 : FAILURE in 19m 19s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/645c087238ae4bbda90fd8afbd48b367 : FAILURE in 30m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5e2ad6b543bb4792b8781f64da8564ec : SUCCESS in 1h 57m 00s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/63589a9e1dc64f84932d15d9f2c955c2 : FAILURE in 25m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/371185fe419e428a8be65d1a5bfc8065 : FAILURE in 18m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b139c861686a47f9a48c00cff99fd268 : SUCCESS in 38m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d950cd785c474a82910e296f2b575aa2 : SUCCESS in 33m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6608671de924863b91aab3d6bc25e55 : SUCCESS in 14m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/647ad6aa3f3e40e093ffc5e9676df7f3 : FAILURE in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5baec6cd1a584f89ba16fef1cd12dbeb : SUCCESS in 21m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44678b1b36854ee3babcf4cb38d9491f : SUCCESS in 11m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d5ef2e2f1fb744269ba1fe3bc6798123 : SUCCESS in 58m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2944882284924363b136ef71e1cf5b45 : SUCCESS in 2h 06m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/38c72f53111b454fbb1f4e1ca824b71f : FAILURE in 18m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2257b21c44a44bdbb3f3c67f077eeef8 : SUCCESS in 1h 06m 00s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/59d4b188883e4409a72d0037f62c89df : FAILURE in 34m 06s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5e644cd2f9e7491eb29e672746d614bf : FAILURE in 22m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a5ae5d8b320347989ec12c55aac118c0 : FAILURE in 27m 46s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73e28672a27844b49de2682de70b500e : SUCCESS in 59m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0c0212c295c547daa71f6061a0836610 : FAILURE in 17m 47s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18e04aa5e4e64ef3ae22d1e691429a6d : SUCCESS in 24m 25s","accounts_in_message":[],"_revision_number":2},{"id":"b8c3290d8af49ece61f210a6a88e693cf234c457","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-14 20:35:45.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/99/821599/2/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/99/821599/2/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"baf4d7d5ac37bba721e4df9c1f4e563af6337b41","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-12-20 10:05:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ae4450244ee33ea3a829f399c02a16d9c490cd2d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 10:58:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d114b1583f9f46efaadb5d1e3c61d4fc : SUCCESS in 35m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/45b8ba59b6464bf2a1f8367af98e72a7 : SUCCESS in 47m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dddb9111c6a8ef74a39e92196f8cf69b1451094b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 11:35:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d693aa10d57f4663af5f9f354bcff022 : SUCCESS in 40m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/29992ee2a3de4864abffc4cbf46f27eb : SUCCESS in 1h 06m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7165f6fe7ac14d08b63c8084eea8f4d1 : SUCCESS in 54m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b940d3da61544ef3b9b8471d4c15ab57 : SUCCESS in 45m 34s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9374913a53ce435e90f654bcb6bc085f : SUCCESS in 56m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6667be988269494e93351ae9e41ed734 : SUCCESS in 48m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/718974c9407f407a8113993feb7695ea : SUCCESS in 30m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a3adf80804f405ba8292db68b44f192 : SUCCESS in 33m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/091146f9aec44ed7a3f24bff918fe62c : SUCCESS in 14m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b78a73beb03841b3b07c1873c5bf8dbe : SUCCESS in 32m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aee6790317a8458a8fc5a22232929106 : SUCCESS in 27m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/734d39b7730345c581a8b70c9247463a : SUCCESS in 13m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4d1aac2063bd40b4b82e5c15f3b703f2 : SUCCESS in 49m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/236e676354384240af6ae6044d6867e7 : SUCCESS in 1h 19m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a9ee3d47a3564a0dbd45be3a54fe6195 : SUCCESS in 1h 08m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7c5cacaa7cc2434db73565b595151a1f : SUCCESS in 1h 03m 29s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/75071b6322dc4beb846e3901d84b9733 : SUCCESS in 57m 48s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b282eb14b8324f98a1dc4efe7fe62253 : SUCCESS in 1h 14m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3cf5b2a137384371808bd1b227d90afe : SUCCESS in 1h 15m 04s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32223145dedc43e393cbc417887a2aa2 : SUCCESS in 1h 06m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3fb0e396f8a2417da7c7c011f2977906 : SUCCESS in 1h 09m 35s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a65f39ed3b75417f9cd5658e7fc18132 : SUCCESS in 26m 41s","accounts_in_message":[],"_revision_number":2},{"id":"cfe17e315092f31b100edba68da5d214dd8762f3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-20 12:30:25.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/821599/2 : SUCCESS in 2h 24m 25s","accounts_in_message":[],"_revision_number":2},{"id":"8315f81a33d2ce7dc2bcec227ba31eec8e6eba2e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-12-20 13:27:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"110d2f6bbb04d6b29451dbc0756fd7888dab55d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 13:29:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0d9fe4f4505065095616009edba7559893928691","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-20 15:38:55.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/99/821599/2/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"7547976f4fe6deb64a12d70f79ea6069e25e5291","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 16:05:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/00eb32ae64d346639f8223a8e3aab461 : SUCCESS in 42m 24s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d8d1e36b2fed4481973258be0168be80 : SUCCESS in 1h 00m 28s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/79e8dab6ba6a41058764acccc00a060c : SUCCESS in 1h 01m 43s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b88e46fd401a48ac9759dbae07c5d448 : SUCCESS in 1h 01m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8b3110226c2a4050a7cb026ac4b98cb2 : SUCCESS in 1h 10m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5ea871bf94b4920b08c8fc2f8a39b77 : SUCCESS in 29m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baf2c731f234420eab6a2a1d7643d29b : SUCCESS in 19m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/182872d90a844d34830c98204433d512 : SUCCESS in 29m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/530532dcbb9b457eac80e78c2ddf3a07 : SUCCESS in 30m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/757a75a888c2432ba96565bac76705a9 : SUCCESS in 14m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ce9a815e06dd4d769c0a1e18f054d502 : SUCCESS in 45m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/489d8d19df1d4667b6a2fb55d59475d6 : SUCCESS in 1h 37m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ad2a6d5e0b1146a0ac8129cfede6acea : SUCCESS in 1h 20m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c46920b2346a4a91801b23c85217dae1 : SUCCESS in 51m 27s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2c5225bcff64a73b6ce5e92b6745aee : SUCCESS in 1h 12m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41019faefa8645df95986bf6e3d13379 : SUCCESS in 30m 49s","accounts_in_message":[],"_revision_number":2},{"id":"c95adbab30923de077181ce78eda7a3a622b498a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 16:05:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3bd9be8df5a34ce284610facfb99de74c90e8d67","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 16:08:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c833cbf273b242998fbff2304b792c29 : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2e0e5d2698fc704c277ec9e27a9016417808f9b7","revisions":{"d0c71999df2cb3783ff3216958844128a918224d":{"kind":"REWORK","_number":1,"created":"2021-12-13 15:49:34.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/99/821599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/821599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/821599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/821599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/821599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/821599/1"}}},"commit":{"parents":[{"commit":"8123cc6ee3faff4b2d13927302b6acbe17b3b814","subject":"Merge \"Fix links for Source code references\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8123cc6ee3faff4b2d13927302b6acbe17b3b814"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-12-13 15:48:06.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-12-13 15:48:06.000000000","tz":60},"subject":"Add logs for port_bound_to_host","message":"Add logs for port_bound_to_host\n\nChange-Id: I4594a622339335a8f304ff4827c71cb63729faf3\nRelated-bug: #1954663\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d0c71999df2cb3783ff3216958844128a918224d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d0c71999df2cb3783ff3216958844128a918224d"}]},"branch":"refs/heads/master"},"2e0e5d2698fc704c277ec9e27a9016417808f9b7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-12-14 10:27:10.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/99/821599/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/821599/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/821599/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/821599/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/821599/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/821599/2"}}},"commit":{"parents":[{"commit":"8123cc6ee3faff4b2d13927302b6acbe17b3b814","subject":"Merge \"Fix links for Source code references\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8123cc6ee3faff4b2d13927302b6acbe17b3b814"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-12-13 15:48:06.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-12-14 10:24:39.000000000","tz":60},"subject":"Add logs for port_bound_to_host","message":"Add logs for port_bound_to_host\n\nChange-Id: I4594a622339335a8f304ff4827c71cb63729faf3\nRelated-bug: #1799790\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e0e5d2698fc704c277ec9e27a9016417808f9b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e0e5d2698fc704c277ec9e27a9016417808f9b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
