)]}'
{"id":"openstack%2Fneutron~821998","triplet_id":"openstack%2Fneutron~master~I03214079f752c7efe6611f2e928f32652fe681bc","project":"openstack/neutron","branch":"master","topic":"bug/1879307","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-12-20 08:09:44.000000000","reason":"Rodolfo Alonso replied on the change"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2021-12-20 09:54:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I03214079f752c7efe6611f2e928f32652fe681bc","subject":"Remove _safe_plug_new method from the interface driver","status":"MERGED","created":"2021-12-16 16:05:17.000000000","updated":"2021-12-20 09:56:37.000000000","submitted":"2021-12-20 09:54:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821998-bug/1879307","meta_rev_id":"a20cd0282c4f4c8cc9aea99c9931f9d2648d06c9","_number":821998,"virtual_id_number":821998,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-12-17 00:24:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-20 09:54:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-12-20 08:09:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-17 07:27:54.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-12-20 08:09:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-16 16:07:20.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-12-16 16:50:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-16 17:49:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-16 18:07:22.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-12-16 19:07:01.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-17 07:27:54.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"6bbed04a8a083eb913383b8c96e9d0b9c50ebf9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-12-16 16:05:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7625af780de76271002d78a726b653c90d96b148","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-16 16:07:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f227b4a74e3d4e8b954ecaca739c486a77f3a87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 16:50:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d16cc50464484712aaa58e9040b7719d : SUCCESS in 30m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8f4bde91fd7c489a9794532c8bfd5095 : SUCCESS in 42m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9fd9d3a44da1de679f717ecd2b52aad91c6a607","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 17:49:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/938ef641f9f042e7bff22dfeab27d4d6 : SUCCESS in 1h 36m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0e59e26ee98742ef8ea6dca04dfff547 : SUCCESS in 1h 08m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6e9e54ea826649d28c06226f1fe109cd : SUCCESS in 1h 06m 35s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6fe6646f4f1f4847a7935f9e4e5aae3b : SUCCESS in 53m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04a54e8d6bb741fdbb613443af3f526e : SUCCESS in 29m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb934506bf4047faab6c971ea2d39e6c : SUCCESS in 27m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ebd589633d84b89a695aa560ce57d86 : SUCCESS in 10m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89be985bac7449258f6aa2d3b821af3e : SUCCESS in 25m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/875ff542c0554aa1bbd57c1bf75a9783 : SUCCESS in 27m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/326ac0a52d0443b891c09f044ddbfce1 : SUCCESS in 9m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5463c153fb74a3eb2c276389f7e8949 : SUCCESS in 12m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7cd42c92461a4b658573075e553c3b81 : SUCCESS in 52m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b5fa8e136c09484e94545a1f9abdf97f : SUCCESS in 1h 21m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c0f274851f2a4af487aa73e511b9b406 : SUCCESS in 1h 07m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8fd9a119b9054acc8b5586e0be2dd969 : SUCCESS in 1h 01m 21s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cfef0d3fe6d1467586f2bc6893b4e057 : SUCCESS in 51m 23s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/77894fa32f26442d93dfd065d1faa399 : SUCCESS in 1h 01m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e3828e1502974cdfb654c19adb1c7e24 : SUCCESS in 1h 16m 29s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1819975fac6409ea65870258acee779 : SUCCESS in 56m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/74f1dfc0be51469f9f2a63dcdba92339 : SUCCESS in 1h 39m 08s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c630f5493b0441588b830e3207ba19e : SUCCESS in 25m 08s","accounts_in_message":[],"_revision_number":1},{"id":"521c1ecdb2487ee17d476feb31c10261d97b4e64","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-16 18:07:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/821998/1 : SUCCESS in 2h 01m 23s","accounts_in_message":[],"_revision_number":1},{"id":"e7219ef6ecf62a22324c4ad5703484ab3d003a5d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-12-16 19:07:01.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2021/Arista_Tempest_Testrefs/changes/98/821998/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"5e4ad5750957fc320a628ce2d50596ac6d43ee05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-12-16 20:57:53.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"03758720b08f3865c68758dbd3364892142afa77","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-12-16 20:58:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0dd4c9a8f109e155fced56d3148e64321a99832a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 21:38:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/54df39eac63e471bae104fa6c14b206b : SUCCESS in 35m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8484d7d36b93463ea3261043a1b11b71 : SUCCESS in 34m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f64c34ed7a2a9a4dcf2a541e0996b6c0fe9d1842","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 22:27:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cb2dda79d984493f88d53768d8db28f9 : SUCCESS in 1h 16m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ebf074a25dd741dbabe980cbb8b9e0ed : SUCCESS in 1h 16m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/079e2fd3e03b4607a2e0077a3990fabc : SUCCESS in 1h 07m 36s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/32b57ba0577641f7ab8ed761e850893a : SUCCESS in 56m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d4626a8eed047a883a8b4e5b58acdc0 : SUCCESS in 35m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b303e78b44949f0917afff5f5d928e0 : SUCCESS in 46m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/215fe8eb08fa49be957e7ebde82bb267 : SUCCESS in 13m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38dc49a0a7df446b8b2a65f85fc3dba1 : SUCCESS in 38m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d38676985444eeda0719af07d43e79e : SUCCESS in 25m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c043b6f0b17d4891918222511cd7aca1 : SUCCESS in 9m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0902bb1d08dc450db5918f9f81454f9f : SUCCESS in 11m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d6dfaa549fe64dd5aedb391414fef27c : SUCCESS in 46m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/88ffa857afd94decbf7a05349be99262 : SUCCESS in 1h 27m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a8ed1bd22a1d49cca99b0eaf735a4432 : SUCCESS in 1h 08m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f7085338a6db43c3b75db1f9bd5f4df3 : SUCCESS in 1h 02m 24s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3f8a997802c146418ef834a885fbdaca : SUCCESS in 58m 31s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f3006771c1934b5a9b3945945c179982 : SUCCESS in 1h 08m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/56fe2625b21742db90cdec6d174f25d4 : SUCCESS in 1h 11m 57s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8276242844445bb89900016e233915c : SUCCESS in 56m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/525a5e4a995c437db7582b24b40bfae7 : SUCCESS in 1h 09m 39s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f99908821f3493d8a992b172cc1b4fc : SUCCESS in 24m 07s","accounts_in_message":[],"_revision_number":2},{"id":"f7c04fe9761a76dff16774b33f3a93a73e8377f7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-16 22:52:00.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/821998/2 : SUCCESS in 1h 53m 19s","accounts_in_message":[],"_revision_number":2},{"id":"f447d408e4d8fc5258f100911a5ea4f7c11d5250","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-12-17 00:24:39.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2021/Arista_Tempest_Testrefs/changes/98/821998/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"afb960e00ec7f51d2a003190c8d575287850a745","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-12-17 07:27:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b59e87aa2d626fb0b2741324f0a0b5602fbecd03","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-20 08:09:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0c2cdb1b94fc8f814edcf6dc57b5ff6578217467","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 08:10:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"68b10b80db23646624a0fe8dadef9f41eceb950d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 09:54:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f94b66da654444af8f9251deb4cb74ce : SUCCESS in 1h 21m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/12a2d624363844db9441a509fcfeb965 : SUCCESS in 1h 10m 08s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6368c99e9e3d4b2aa6306a209768da65 : SUCCESS in 53m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9e0f82c991fb464998b5ac4b392995d6 : SUCCESS in 53m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6207b061bc30465c8563d952951267a5 : SUCCESS in 34m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db153fc1b93043309f903a3362a9897b : SUCCESS in 16m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ed69c1426ba4c7ab43b2b2b803ee646 : SUCCESS in 30m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f1287376c92440ab2b7e02cf9b180f5 : SUCCESS in 25m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0bdd6a1ac854966b5be7cba2dadd8c5 : SUCCESS in 10m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b411bb801e60417ea6a0b9927bff2b6f : SUCCESS in 11m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/682ae4c362394754848be3ddeadec39f : SUCCESS in 37m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc71f8a2a36046638316e8955fdfa34a : SUCCESS in 1h 15m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c4109fb7e0b944a5aa507e22b49ec916 : SUCCESS in 1h 16m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/08883ed94349412b9dd0a0806039c77c : SUCCESS in 58m 55s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/690721348cd14a2a880d1564a6ecd976 : SUCCESS in 54m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3be3e489c12445a4a43738706e8de522 : SUCCESS in 21m 11s","accounts_in_message":[],"_revision_number":2},{"id":"9f1510053fbe1e1045a9f62f721708054c2a4cd7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 09:54:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a20cd0282c4f4c8cc9aea99c9931f9d2648d06c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 09:56:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e4ac8d58f5c4b69b7456d5649927d97 : SUCCESS in 1m 32s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/347e3e580b2f418d9c0c51cf496086e3 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"19b96628378357b9df2b8bb9285f71b7c7b66ca4","revisions":{"09d32a4051b3cd102b820454b04d7395dc872d57":{"kind":"REWORK","_number":1,"created":"2021-12-16 16:05:17.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/98/821998/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/98/821998/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/821998/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/821998/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/821998/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/98/821998/1"}}},"commit":{"parents":[{"commit":"2a359891ce87157e91ba715e022e6d09c5ba161b","subject":"Merge \"Improve Router callback system\u0027s publish events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a359891ce87157e91ba715e022e6d09c5ba161b"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-12-16 15:56:42.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-12-16 16:05:09.000000000","tz":60},"subject":"Remove _safe_plug_new method from the interface driver","message":"Remove _safe_plug_new method from the interface driver\n\nIt was added temporary to have compatybility with 3rd party code\nwhich uses Neutron interface driver but it was said that since\n\"W\" release that old, deprecated way of calling \"plug_new\" method\nwill be removed. Now we are far after \"W\" release so it\u0027s time to\ndo some cleaning there.\n\nRelated-Bug: #1879307\nChange-Id: I03214079f752c7efe6611f2e928f32652fe681bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/09d32a4051b3cd102b820454b04d7395dc872d57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/09d32a4051b3cd102b820454b04d7395dc872d57"}]},"branch":"refs/heads/master"},"19b96628378357b9df2b8bb9285f71b7c7b66ca4":{"kind":"REWORK","_number":2,"created":"2021-12-16 20:57:53.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/98/821998/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/98/821998/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/821998/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/821998/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/821998/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/98/821998/2"}}},"commit":{"parents":[{"commit":"2a359891ce87157e91ba715e022e6d09c5ba161b","subject":"Merge \"Improve Router callback system\u0027s publish events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a359891ce87157e91ba715e022e6d09c5ba161b"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-12-16 15:56:42.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-12-16 20:57:52.000000000","tz":0},"subject":"Remove _safe_plug_new method from the interface driver","message":"Remove _safe_plug_new method from the interface driver\n\nIt was added temporary to have compatybility with 3rd party code\nwhich uses Neutron interface driver but it was said that since\n\"W\" release that old, deprecated way of calling \"plug_new\" method\nwill be removed. Now we are far after \"W\" release so it\u0027s time to\ndo some cleaning there.\n\nRelated-Bug: #1879307\nChange-Id: I03214079f752c7efe6611f2e928f32652fe681bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19b96628378357b9df2b8bb9285f71b7c7b66ca4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19b96628378357b9df2b8bb9285f71b7c7b66ca4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
