)]}'
{"id":"openstack%2Fneutron~822098","triplet_id":"openstack%2Fneutron~master~Iba90b44331fdc63273fd3d19c583a24b5295c0ac","project":"openstack/neutron","branch":"master","topic":"bug/1955121","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-12-21 10:58:50.000000000","reason":"Rodolfo Alonso replied on the change"},"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2021-12-20 10:56:41.000000000","reason":"Harald Jensås replied on the change"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2021-12-21 13:32:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iba90b44331fdc63273fd3d19c583a24b5295c0ac","subject":"Don\u0027t fail subnet validation if gw_ip is actually not changed","status":"MERGED","created":"2021-12-17 11:37:11.000000000","updated":"2021-12-21 13:40:06.000000000","submitted":"2021-12-21 13:32:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"822098-bug/1955121","meta_rev_id":"31769accfcfcc7fcea76fbad705782bfe1cbe188","_number":822098,"virtual_id_number":822098,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-12-20 10:38:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-21 13:32:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-12-21 10:58:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-20 10:56:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":1,"date":"2021-12-21 10:50:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-12-20 21:37:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-12-21 10:58:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-17 12:15:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-17 12:15:46.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-12-17 13:18:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-17 13:27:12.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-17 13:46:21.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-12-17 13:50:08.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2021-12-17 15:44:24.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2021-12-20 21:37:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-12-21 10:50:25.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"8eec17286bdc4b527e49a8d26827f42fda4f6425","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-12-17 11:37:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a20a5ed602481e7b5909167793d8db07f9c6490","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 12:15:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5214400d38c745b1841248bf5b398ee9 : SUCCESS in 33m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a733e76ddd864639aae0347f5ca7372b : SUCCESS in 32m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c82709fbbf4d35ebb5093f6642ddfd3c19e26d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 13:18:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/99ac6cebb1164ba3bdce7ce6bac5e621 : SUCCESS in 35m 30s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/be35408dfb5d44d0ac2b92e9b31539c9 : SUCCESS in 1h 16m 48s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e894932e8df745ccb1b184d1adda8143 : SUCCESS in 1h 14m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c480714d5c3641988ba894e52531ba0a : SUCCESS in 1h 14m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/363e17edb1954b2392981b9ca8177d28 : SUCCESS in 1h 25m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f0565bf28ed24b76adb61ec8c1108f10 : SUCCESS in 37m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ddca848ab634b729e9b08434260e8d7 : SUCCESS in 30m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12eba7e73fd14f599473b60dce8455fa : SUCCESS in 34m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b49365971ef46f69fca49c29a789819 : SUCCESS in 14m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db85a0026e7646769a55e898bfc36f44 : SUCCESS in 30m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c7d92fe4b5e4a839852a7be9d7299d6 : SUCCESS in 25m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7602efe6a914a8980f4fea3c0289672 : SUCCESS in 10m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dd55608f2dbc428786291b37c4cbe9eb : SUCCESS in 49m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e7c3b33178824e8ab544453a7ecce076 : SUCCESS in 1h 34m 46s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7ea986ab91e84a2cbe29d570234352b2 : SUCCESS in 1h 11m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f86789a81b0841c8aa4d662b9b6076be : SUCCESS in 52m 25s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/36b0c966bd104d59b53c1d9362c00d60 : SUCCESS in 1h 02m 03s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f13f2b7b3dbd4feca3d11aa62197a532 : SUCCESS in 1h 17m 46s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d4404c60d8e94df0b7d01788d09ad769 : SUCCESS in 1h 20m 12s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41bd4956fca94c469b2927d94bbd84db : SUCCESS in 1h 03m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7a3e9e4896934f879a61776d01962a8f : SUCCESS in 1h 23m 47s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d85245cc50d4c56bb96b93b6ec3b7c8 : SUCCESS in 29m 14s","accounts_in_message":[],"_revision_number":1},{"id":"b42b0e889983a45dd607c37c67a50d95ce8fc666","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-12-17 13:27:12.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2021/Arista_Tempest_Testrefs/changes/98/822098/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"8829d820f5f9b6e586ffea1c6612a08105bfe316","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-17 13:46:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/822098/1 : SUCCESS in 2h 08m 30s","accounts_in_message":[],"_revision_number":1},{"id":"00f967ff3d5f31430279b8f5189a576e2fe31d08","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2021-12-17 13:50:08.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"578d38aafed2fdecfe88528e0bc93c6af5532793","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2021-12-17 15:44:24.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ceba470ce74267569caa272c341a61fca4247b90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-12-20 09:09:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"602ec79c672f15ca1181e626b97dd9ffbddb04e8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-12-20 09:09:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1f487cda50e35917f3db947ef4a27c53c5835725","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 09:51:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d81280bb1cb4d33b24755cbf7118cc6 : SUCCESS in 31m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cebf90d40aae4dde8c9d50dae94d1964 : SUCCESS in 39m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2c479aa5436d622011e9a5218f20dac24337da42","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-12-20 10:38:58.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2021/Arista_Tempest_Testrefs/changes/98/822098/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"8a8d2eaf8362bdf76611e4ee0a8d67f2e584c868","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 10:40:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0856aa0d6f574827ada955f41a8db28a : SUCCESS in 41m 36s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/89d29905c87440019dbb770f108380d3 : SUCCESS in 1h 12m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a42de085ed344ccd8c91e2d9bf77f0d7 : SUCCESS in 57m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8d96150a0ac24ebc84feec77afa848c6 : SUCCESS in 1h 12m 24s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a49320abe9d24affb55593b05b8d5c64 : SUCCESS in 1h 27m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9dea649181274d218856ec0dae723a2a : SUCCESS in 34m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c997d920caf8442199ca0af1ae0dec17 : SUCCESS in 30m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5600e823bc7d435abb1a460a5b18f64f : SUCCESS in 35m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9632f630813b438db6d4152eb17fa757 : SUCCESS in 12m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a0c0c2a76fa4f9aae195283b67e0735 : SUCCESS in 30m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55740a9cc98d45d186a324323cc98e75 : SUCCESS in 29m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a718492c3d6c4492bb781b0f6684d257 : SUCCESS in 11m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/87e02a7ef6e540759b46e9f70803f238 : SUCCESS in 44m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bbbb11dd35974d0a8f69fb8fbd89d6b3 : SUCCESS in 1h 28m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/57883117435944d5898b2d19f8b6b825 : SUCCESS in 1h 05m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9bacfffc738d4357bbdf2aa5a93784c2 : SUCCESS in 58m 21s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9ff5fc8a49e44353bf4e3b60df0c09e5 : SUCCESS in 55m 05s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c2b36df3d5e6417b97e6af2fa702ed21 : SUCCESS in 1h 10m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c1eb1f927e5c4753b71662e080a0ad7f : SUCCESS in 1h 15m 16s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7eeda5fb08be4b1093181e48f81669cd : SUCCESS in 59m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cca41ddc988244c3abec704a4e201179 : SUCCESS in 1h 03m 12s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/385f4a32bb054ba2b15a906952c1373f : SUCCESS in 20m 57s","accounts_in_message":[],"_revision_number":2},{"id":"da0717d7d8bb4c263c5ee6168e16d5a5f7e25df4","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2021-12-20 10:56:41.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9bbe46e7ff109aad52d289b6b8c45804980dc82d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-20 11:04:32.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/822098/2 : SUCCESS in 1h 54m 07s","accounts_in_message":[],"_revision_number":2},{"id":"40f5fd7cddcbeb4153a1ba7acbded50e2ac96825","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-12-20 21:37:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e7c0a119faa1af38ca98e0fc532a3a5ad2dcd7d9","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-12-21 10:50:25.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"237ba2ff535688bc7858cd02d446f2dd5535b3e6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-21 10:58:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1a7ed3ea10cb4e269b14d0320f616cc7e3752744","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 10:59:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"127e9524f482835f933a7ac27875e3cfd501d911","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 13:32:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b5c6f26c2a8144ef8ca5cdef5a00bd58 : SUCCESS in 51m 16s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/947efbd420424218af790b6b8a16a3f7 : SUCCESS in 1h 18m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/af49f73b4348463391e0f84684a82624 : SUCCESS in 1h 34m 51s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ff8149871ba34f39bf48a7b17023a3fd : SUCCESS in 1h 18m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bb03a0f803f24664a300ff368e798f6c : SUCCESS in 1h 23m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92ebf595622c4c0bae4383ac23dd497b : SUCCESS in 27m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75a08449414b4fa89412b6fec5c1b58f : SUCCESS in 13m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a1e5942832f451984e740b0bf2e1cb9 : SUCCESS in 30m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e18b33228621473496fee9f16834544e : SUCCESS in 33m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fa0e64db79a4e938bd88e8ce4fd4000 : SUCCESS in 11m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ab30fc080a484de7bbdadd7472d70bc5 : SUCCESS in 49m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/325afc28397740f9841dbd9fae5e06ca : SUCCESS in 1h 34m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2cb794c62b084748af84a16e769819a2 : SUCCESS in 1h 15m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/512ad4f99cd7468d8cae2d876f9a2a1a : SUCCESS in 1h 00m 32s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a93f50c904a480faa05e09210c58332 : SUCCESS in 1h 00m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/644fa40e821c46b4ae2ed898522387d4 : SUCCESS in 30m 42s","accounts_in_message":[],"_revision_number":2},{"id":"50a171b67c7c6f0efed8c9051784f458eccf513c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 13:32:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"31769accfcfcc7fcea76fbad705782bfe1cbe188","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 13:40:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38d26c4af4044d5c962a732a35469a67 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6809bed632b1c155bf370b78353e6b0fe059b3f0","revisions":{"d755c38838bf23457578d9f44416bf250ce874cd":{"kind":"REWORK","_number":1,"created":"2021-12-17 11:37:11.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/98/822098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/98/822098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/822098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/822098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/822098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/98/822098/1"}}},"commit":{"parents":[{"commit":"69ccb24be4d3f2244d0b14d80ea3c6311b48dcc9","subject":"Merge \"Fix stackalytics\u0027 link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69ccb24be4d3f2244d0b14d80ea3c6311b48dcc9"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-12-17 11:34:07.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-12-17 11:34:07.000000000","tz":60},"subject":"Don\u0027t fail subnet validation if gw_ip is actually not changed","message":"Don\u0027t fail subnet validation if gw_ip is actually not changed\n\nIn subnet update API call Neutron checks if gateway_ip was send to be\nupdated and if so, it checkes if old gateway_ip isn\u0027t already allocated\nto some router port. If it\u0027s already used, Neutron returns 409 response.\nThis is valid behaviour but sometimes, some automation tools may do\nsubnet update request and pass the same gateway ip as already used by\nthe subnet. In such case, as gateway_ip is actually not changed Neutron\nshould not raise exception in that validation.\n\nCloses-Bug: #1955121\nChange-Id: Iba90b44331fdc63273fd3d19c583a24b5295c0ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d755c38838bf23457578d9f44416bf250ce874cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d755c38838bf23457578d9f44416bf250ce874cd"}]},"branch":"refs/heads/master"},"6809bed632b1c155bf370b78353e6b0fe059b3f0":{"kind":"REWORK","_number":2,"created":"2021-12-20 09:09:24.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/98/822098/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/98/822098/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/822098/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/822098/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/822098/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/98/822098/2"}}},"commit":{"parents":[{"commit":"69ccb24be4d3f2244d0b14d80ea3c6311b48dcc9","subject":"Merge \"Fix stackalytics\u0027 link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69ccb24be4d3f2244d0b14d80ea3c6311b48dcc9"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-12-17 11:34:07.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-12-20 09:08:37.000000000","tz":60},"subject":"Don\u0027t fail subnet validation if gw_ip is actually not changed","message":"Don\u0027t fail subnet validation if gw_ip is actually not changed\n\nIn subnet update API call Neutron checks if gateway_ip was send to be\nupdated and if so, it checkes if old gateway_ip isn\u0027t already allocated\nto some router port. If it\u0027s already used, Neutron returns 409 response.\nThis is valid behaviour but sometimes, some automation tools may do\nsubnet update request and pass the same gateway ip as already used by\nthe subnet. In such case, as gateway_ip is actually not changed Neutron\nshould not raise exception in that validation.\n\nCloses-Bug: #1955121\nChange-Id: Iba90b44331fdc63273fd3d19c583a24b5295c0ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6809bed632b1c155bf370b78353e6b0fe059b3f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6809bed632b1c155bf370b78353e6b0fe059b3f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
