)]}'
{"id":"openstack%2Fneutron~822293","triplet_id":"openstack%2Fneutron~stable%2Fxena~I78dd012764c7bd7a29aeb8d97c00b627d7723aeb","project":"openstack/neutron","branch":"stable/xena","topic":"bug/1950686","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-12-21 23:11:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I78dd012764c7bd7a29aeb8d97c00b627d7723aeb","subject":"Do not announce any DNS resolver if \"0.0.0.0\" or \"::\" provided","status":"MERGED","created":"2021-12-20 08:44:46.000000000","updated":"2021-12-21 23:13:36.000000000","submitted":"2021-12-21 23:11:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822293-bug/1950686","meta_rev_id":"8a3976c8b53bdea796014f2e1b24d5dff49e9985","_number":822293,"virtual_id_number":822293,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-21 23:11:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-12-21 12:55:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-12-21 21:29:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-12-21 21:29:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-20 09:24:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-20 10:16:11.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-12-20 10:25:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-21 12:55:59.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-12-21 21:29:18.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"a5086cfcda72c06929b25c7ed8ff007398ae0329","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-20 08:44:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17e6147d50aaec78140d597b980177839c7d6c3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 09:24:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bacbf037e0a349d5a923e2fcd0b38979 : SUCCESS in 29m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f767258f4e44f06a3843d69b4d65b60 : SUCCESS in 37m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d33dbd23f4a2aab233bfac13612f1749c445ad2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-20 10:16:11.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/822293/1 : SUCCESS in 1h 30m 39s","accounts_in_message":[],"_revision_number":1},{"id":"296adc41c7188ec510730c53df158fef219a0646","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 10:25:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/12f8a72fcba74d3daad383e03cc7ac4b : SUCCESS in 35m 23s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/20052a8febda430cb2ca1b1353cc1985 : SUCCESS in 1h 19m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/258cf8f456bf44259e8af515c7d25db0 : SUCCESS in 1h 09m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a96741462f7f41b89997bdc3db3e5a33 : SUCCESS in 1h 10m 31s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/246d9aa17a2c42d2934b84931af08401 : SUCCESS in 1h 03m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d65bc1b7bdb74b0b84c8a04ecb80965d : SUCCESS in 41m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81edfc9acb91440bb767686aede47a90 : SUCCESS in 37m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b79be0ad75247378ea4e8453056ee06 : SUCCESS in 10m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41d1488511084c12ab249ebebe5208ee : SUCCESS in 21m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07c99299a43340d3b9b0ce46b6242e7a : SUCCESS in 27m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa4848269b8e453cb8b883748c38c6ea : SUCCESS in 21m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5f0569f8ec346a5881ed2b7a4f9214a : SUCCESS in 9m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/81315df6ea8948688903106c34a4769e : SUCCESS in 41m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5bc53914a26f4756ba7907b450eb9718 : SUCCESS in 1h 35m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f8694a79adb04d1ca9fe9849cb8c518c : SUCCESS in 1h 11m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b3ea758b8d754b609c5d2c1259d28ab8 : SUCCESS in 1h 01m 06s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/daedd75980a54ff5bd3d6f07abab9999 : SUCCESS in 59m 59s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cfb66f310b5f4779bc89b5cb5da2fab8 : SUCCESS in 51m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/073305bbd2f34648879ac78405edf282 : SUCCESS in 1h 10m 36s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abafb5a3d4f8465c978d09b408374026 : SUCCESS in 57m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/51a052d52703466c922284e1a0aaac27 : SUCCESS in 1h 03m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cafeec7da1e12d2f9e4307493aa31d45dcbf8b67","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-12-21 12:55:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f418752bce7a5830600b1f523edef1be9956cd9b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-12-21 21:29:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"90a1edf4a2fa6fdac9787100fbcee7500d110cb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 21:30:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f57d0061ee2944b297ed0e36a42edaaf70b44943","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 23:11:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/05d2b83a066a4f41bf34220a10ece0cd : SUCCESS in 42m 38s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/840a2f0ea84d4d33a19e12897d0b275a : SUCCESS in 55m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/72868a27db62455ca2493108fd70944b : SUCCESS in 1h 04m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1475f4476e08486ebf2a44e5a266ad2f : SUCCESS in 53m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ba7bb75da1574ef9ba3db3ec07694fa8 : SUCCESS in 45m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0cd8827268541ffb0393fd8781c4433 : SUCCESS in 15m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eef1f9cc327444b29f44b2f1e67ae5d5 : SUCCESS in 24m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c302b0761ec420aa61b96748574a083 : SUCCESS in 22m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27e779aab2354db5b665b63ee76a8658 : SUCCESS in 11m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b1e2a794950c4f1ca99e701a37271b80 : SUCCESS in 48m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/80659293380348a1942212b68a274633 : SUCCESS in 1h 40m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6b3d438b39a04dbe95448b06de9dfe24 : SUCCESS in 1h 02m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2ada8feac1384d149fd70ab3c857944f : SUCCESS in 57m 38s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/175d4aa8c6aa4620a80c22e22ebc3b13 : SUCCESS in 50m 32s","accounts_in_message":[],"_revision_number":1},{"id":"372c56f434dcf720efaf28813d5777649eceec73","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 23:11:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8a3976c8b53bdea796014f2e1b24d5dff49e9985","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 23:13:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05132cfb320d4c3587903a92fc1faedb : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c73fc94204d432e111e1a7c5ba8b779c5fb0b3b6","revisions":{"c73fc94204d432e111e1a7c5ba8b779c5fb0b3b6":{"kind":"REWORK","_number":1,"created":"2021-12-20 08:44:46.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/93/822293/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/822293/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/822293/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/822293/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/822293/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/822293/1"}}},"commit":{"parents":[{"commit":"e6953e217c731559724971a371e81d2b6f9837e0","subject":"Merge \"[OVS] Workaround when OpenFlow controller restarts\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6953e217c731559724971a371e81d2b6f9837e0"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-12-07 11:42:53.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-12-20 08:44:18.000000000","tz":0},"subject":"Do not announce any DNS resolver if \"0.0.0.0\" or \"::\" provided","message":"Do not announce any DNS resolver if \"0.0.0.0\" or \"::\" provided\n\nThe DHCP server should not announce any DNS resolver at all on the\nsubnet if \"0.0.0.0\" (IPv4) or \"::\" (IPv6) are configured as DNS\nname servers in any subnet.\n\nhttps://docs.openstack.org/neutron/latest/admin/config-dns-res.html\n\nCloses-Bug: #1950686\nChange-Id: I78dd012764c7bd7a29aeb8d97c00b627d7723aeb\n(cherry picked from commit a416f8b0ab810f34ea274183df590a73d150a137)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c73fc94204d432e111e1a7c5ba8b779c5fb0b3b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c73fc94204d432e111e1a7c5ba8b779c5fb0b3b6"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
