)]}'
{"id":"openstack%2Fneutron~822673","triplet_id":"openstack%2Fneutron~stable%2Fxena~Ib5de7e65839f52c35c43801969e3f0c16dead5bb","project":"openstack/neutron","branch":"stable/xena","topic":"bug/1954751-stable/xena","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-01-10 21:05:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib5de7e65839f52c35c43801969e3f0c16dead5bb","subject":"[Functional] Fix expected number of the enqueue_state_change calls","status":"MERGED","created":"2021-12-22 14:13:05.000000000","updated":"2022-01-10 21:07:36.000000000","submitted":"2022-01-10 21:05:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822673-bug/1954751-stable/xena","cherry_pick_of_change":822051,"cherry_pick_of_patch_set":1,"meta_rev_id":"faed7bbe9b7e7841ecf2e8444ca8a93f52020db8","_number":822673,"virtual_id_number":822673,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-10 21:05:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-01-10 14:02:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-12-22 16:19:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-01-10 14:38:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2021-12-23 08:04:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-01-10 14:38:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-22 14:54:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-22 15:46:28.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-12-22 15:52:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-22 16:19:16.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-12-23 08:04:04.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-01-10 14:02:20.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-01-10 14:38:04.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"43fdefe018cdf7b484b779427f3db8af7b9025cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-12-22 14:13:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ddcd4fdcdc6eb361d8f4495abd75718c45afcedf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 14:54:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cabe23749dee4c7499ead8f51af44425 : SUCCESS in 35m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b88dfe62fb394c74b1d4556e9c1ee217 : SUCCESS in 36m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9277866f4fccebfaab9d302a28ee163c220e53eb","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-22 15:46:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/822673/1 : SUCCESS in 1h 32m 40s","accounts_in_message":[],"_revision_number":1},{"id":"54dde512a540b93d33a71c871eea1716bc1c40d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 15:52:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/970d78d50e39455a900ac37341180287 : SUCCESS in 38m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70a007da6cbe4045a3b65e8678fd98b7 : SUCCESS in 14m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54f623c2d83d4f9fa46c43c07deda6e8 : SUCCESS in 20m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/824eb06bb66d4972bef540f1e9ed8927 : SUCCESS in 42m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d6c47f1326d4733a878a9cb3fbf9373 : SUCCESS in 26m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c00689962504d19a493e37fdb7ccdc3 : SUCCESS in 9m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d6f0f37d9b184324a0ea7f89663f9bf6 : FAILURE in 56m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa23e59d5a714df4995f41306f4816ce : SUCCESS in 1h 32m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5623440437154cd19f5fe46cb0fd1614 : SUCCESS in 1h 20m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/79411590319c46bb9365c8603907a625 : SUCCESS in 1h 23m 09s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3e4a92d2fd0641a79116981c2a8de15d : SUCCESS in 1h 07m 48s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bce8a8b94b7d4ce1aaf2641660d09b7b : SUCCESS in 1h 34m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3171d40abd9940cdbb8945d74d8e2c8c : SUCCESS in 1h 16m 51s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7288cbd1fb18499d89ccd093eb188aa1 : SUCCESS in 55m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d1dcc276163145ea8cf097c4accf22ab : SUCCESS in 1h 14m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f141abb0702eb996c9ef977d63a369990f231f1","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-12-22 16:19:16.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"603c6daf79346dbab43086bf9b9b1e2d549a84e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 17:04:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/175e677591f349568d73c8092a178eae : SUCCESS in 39m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/67ab4ee24a524206b9c6de576c8d8434 : SUCCESS in 31m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59aede1facea5c23caef2cc2e1769157a1e50222","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-22 18:08:42.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/822673/1 : FAILURE in 1h 48m 30s","accounts_in_message":[],"_revision_number":1},{"id":"72c5bccf5b7e3a05fb425944023529774c7dc5e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 18:11:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de385097d6b0476681c653bd3d511cdf : SUCCESS in 31m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2219e97fc027463ab93c091a0252fbce : SUCCESS in 16m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/793a9af47a264966af4fca3ff66e6b09 : SUCCESS in 32m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fbc02ef73f34d1c9203cdd8ce5d1c6f : SUCCESS in 24m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce31031f88ca402487c654d9f76c23ce : SUCCESS in 23m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea46f20073ce4580a94d43cf324b339c : SUCCESS in 12m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d0557452e8548768a48e5a293cf4658 : SUCCESS in 49m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5962e29e06374fa4becfa811b8d67c31 : SUCCESS in 1h 39m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2dd2227542ce4603a886e10adfe9407c : SUCCESS in 1h 10m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f69c5cea965541dd993f89f528dbd609 : SUCCESS in 1h 10m 59s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1cfe765ebc194768a8d43e3843aa64bf : SUCCESS in 1h 02m 36s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1a644747d81b4962a778694aade83c97 : SUCCESS in 1h 11m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4a71a4c149bc4439a951e8d63a6a4d9c : SUCCESS in 1h 06m 56s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d96d7efe38484f6d900b15d27df03645 : SUCCESS in 50m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/11fc229df5f84a7f8c0f775488836e09 : SUCCESS in 1h 05m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd145911c3368dc62aaf0a5efd74e657d5ad8bf6","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-12-23 08:04:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"28b7c5605d16c43667aa55cc4dc151859b3158e2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-10 14:02:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"49eb69ed2b563fba190a75cc57431beec1419b6a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-01-10 14:38:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ce5c7f46930bb586437bd86fdd677df414b884f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 14:39:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d7530d637c0c3e8f56d85a3666c91e6cdded40e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 21:05:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a4c432b345e4f7e85b5f939359b0723 : SUCCESS in 13m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7cf20055ce6468fb092d1d4088ce4a2 : SUCCESS in 28m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f766a2c9747848c7a6b2b45b05c7c02c : SUCCESS in 22m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d1adb81471d434a8fa20d30ecf41bd7 : SUCCESS in 10m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0fceca6ef0914dcfa07fea8a1c53ec8f : SUCCESS in 45m 13s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17340c00884541b9a43f29827726e24a : SUCCESS in 59m 23s","accounts_in_message":[],"_revision_number":1},{"id":"bf1d711b311802d3fcdf03d7ee3dc616c4c8e8c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 21:05:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"faed7bbe9b7e7841ecf2e8444ca8a93f52020db8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 21:07:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/167bd1c027e54c32b522497084c0e59e : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6265ab319e11006cc707890b2d5fb5ab2cfabb94","revisions":{"6265ab319e11006cc707890b2d5fb5ab2cfabb94":{"kind":"REWORK","_number":1,"created":"2021-12-22 14:13:05.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/73/822673/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/822673/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/822673/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/822673/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/822673/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/822673/1"}}},"commit":{"parents":[{"commit":"d394ccd307d04feb42e5e57c8eb7a199b60ddcfe","subject":"Merge \"Do not announce any DNS resolver if \"0.0.0.0\" or \"::\" provided\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d394ccd307d04feb42e5e57c8eb7a199b60ddcfe"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-12-16 21:25:27.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-12-22 14:13:05.000000000","tz":0},"subject":"[Functional] Fix expected number of the enqueue_state_change calls","message":"[Functional] Fix expected number of the enqueue_state_change calls\n\nIn the HA router\u0027s keepalived state change monitor tests, it was\nexpected that enqueue_state_change method will be called 3 or 4 times.\nBut after some changes in the keepalived_state_change monitor which were\ndone some time ago, it may be now that it will be called just 2 or 3\ntimes:\n- 2 when initial status will be \"primary\" and it will be just\n  transition to \"backup\",\n- 3 when initial status will be \"backup\", then it will transition to\n  \"primary\" and finally to \"backup\" again.\n\nTo reflect those 2 possibilities, test was changed that it will expect\n2 or 3 calls and will check only that last 2 will be always transition\nto \"primary\" and then to \"backup\".\n\nAdditionally this patch adds some extra logging in that test so it will\nbe easier to check what was going on in that test.\n\nCloses-Bug: #1954751\nChange-Id: Ib5de7e65839f52c35c43801969e3f0c16dead5bb\n(cherry picked from commit c6a6c5ae127a97856c880da3f82257977a06a571)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6265ab319e11006cc707890b2d5fb5ab2cfabb94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6265ab319e11006cc707890b2d5fb5ab2cfabb94"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
