)]}'
{"id":"openstack%2Fneutron~822709","triplet_id":"openstack%2Fneutron~stable%2Fussuri~Ib5de7e65839f52c35c43801969e3f0c16dead5bb","project":"openstack/neutron","branch":"stable/ussuri","topic":"bug/1954751-stable/ussuri","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-01-10 22:31:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib5de7e65839f52c35c43801969e3f0c16dead5bb","subject":"[Functional] Fix expected number of the enqueue_state_change calls","status":"MERGED","created":"2021-12-22 14:16:39.000000000","updated":"2022-01-10 22:34:25.000000000","submitted":"2022-01-10 22:31:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822709-bug/1954751-stable/ussuri","meta_rev_id":"cc1c03cd44f69a356f104b983a528a9927f6f658","_number":822709,"virtual_id_number":822709,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-10 22:31:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-01-10 14:02:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-12-22 16:19:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-12-23 08:07:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-01-10 15:30:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-22 16:19:44.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-12-22 16:26:09.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2021-12-22 18:09:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-23 08:07:29.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-01-10 14:02:38.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"f8a390125deeb7a26024d79c8c448cdd79b619fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-12-22 14:16:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eaf0a7f5cf90c2d73bae763aebd74c2836adabe5","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-12-22 16:19:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a7b34230c52b0922077b59c99fe187a868faf26a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-12-22 16:26:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/822709/1 : SUCCESS in 2h 08m 46s","accounts_in_message":[],"_revision_number":1},{"id":"7bbaae2e57b1bd1ff41b370f72fdc78f96477163","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 18:09:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd4bf7717550408db6e947bd7deb2bfd : SUCCESS in 33m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eea2f319efec439c8c5581cececbeef6 : SUCCESS in 13m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21a8a75c5c854f2184d441fe7aabea06 : SUCCESS in 25m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1fe8388407a54212959b9ad8d10aef75 : SUCCESS in 27m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9774f31bf2db4782bf7ccdbc8cc21465 : SUCCESS in 25m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5922cc220ba046b294327b6d436e3ff3 : SUCCESS in 10m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b00ab46cd6244c4cb31fe04b8185fad9 : SUCCESS in 56m 50s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/453a7916ae754ca69d43a0da60f70377 : SUCCESS in 1h 22m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b98ec9c0b2f4b81ac3866e0c5243c11 : SUCCESS in 1h 21m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/76c89fe9cab74adcbc832b24e0a372a2 : SUCCESS in 1h 29m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/66e60d274e1944c8b91842f7834a9b54 : FAILURE in 37m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/19ebd75dc619432cb762a54d85cfc0c2 : FAILURE in 40m 52s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/45835aeec3664611899bc2a7048b5474 : SUCCESS in 1h 09m 00s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/4cdcc2333b434e339814c6989e9381ef : SUCCESS in 55m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0fbaed8ac65346209e2ecda6e75ed359 : SUCCESS in 1h 07m 49s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2da651c5406d4f85b9d535ba1f33b471 : SUCCESS in 1h 38m 27s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9fd665553cd945a0bc69dd05ca9c2e9f : SUCCESS in 1h 18m 15s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eabcd4d796cc44a4b778b41019301539 : SUCCESS in 1h 07m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c81e4d75d5a84320a30b7dec52944698 : SUCCESS in 1h 04m 38s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8de7029edd174609b0a0b92b9b3c5b7f : SUCCESS in 1h 45m 38s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ba6f593fa044d768dcc6349ed7653a7 : SUCCESS in 1h 07m 58s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cd3c83ae87f746f5ad79dc621bf252db : SUCCESS in 1h 20m 08s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/96cb20a0fc5040fe9be6d9c365e00a13 : SUCCESS in 46m 38s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e152595596b94b5dbd3c2680838b14b2 : SUCCESS in 1h 17m 15s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9277d74be8b44a1abdc59bc8ab799c85 : SUCCESS in 58m 24s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e3def33a781746d58700ea083cddf998 : SUCCESS in 3h 45m 17s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0ebf14dbf4a94e009f8af53782cb1006 : SUCCESS in 1h 26m 34s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3073860a852c4fa88e104b69f6bed045 : SUCCESS in 1h 40m 23s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2256109d3c834b719fb88ec878b3610a : SUCCESS in 2h 11m 11s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/491fd975a3d1427bbed74c0cb1d9fd13 : FAILURE in 1h 37m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c2ba446b2223abc5e296828c86c8a267c6988eb","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-12-23 08:07:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3113d6c0190bb284a751d9eda35e6c9cb99b3abe","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-03 08:20:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"185136b87ae2b1af32ea12dadf0e25ed76fe27ad","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-01-03 11:12:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d7f420d35ff4553d949076edcbc3c0f2d98d4f7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 12:11:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cfec946eb624a92a16276d1d09815d0 : SUCCESS in 31m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e09a4d85752942d7bb94b9f13a46b5d3 : SUCCESS in 11m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/485eefece0a642808b47b064b09d8318 : SUCCESS in 36m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c0224c44a9b84d098d0fe099210cbd01 : SUCCESS in 28m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d2370f0c1f94ecb912d7d764fe9eef4 : SUCCESS in 23m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39b3579e89104331b878ed0d76cc0ecb : SUCCESS in 10m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/29329777f42d4f96abdff0bceabaaa32 : SUCCESS in 54m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/91fd1e93e249471aa58ae7bcc163f0f2 : SUCCESS in 1h 19m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/823d54a5311447c38734454ab8a3cdfb : SUCCESS in 1h 21m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/54651045339446fcba55681b2612819f : SUCCESS in 1h 21m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3d2cdaed0a044231bbe1f370b8e5e51b : SUCCESS in 1h 08m 31s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5af643603dc84b76b0a3a6b84f11f00c : SUCCESS in 1h 11m 13s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1b27d4ba4e2248d1afabcafa60ae2228 : SUCCESS in 1h 23m 49s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/bb1672682c574d6a8ee3e1d52a6d108b : SUCCESS in 1h 03m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b6de883d52cd47269fd23f9dc820fb43 : SUCCESS in 1h 19m 48s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0f9a0716ad5240a8be9cd45fb872a232 : SUCCESS in 1h 35m 32s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0a2aedacabd1482baad3902a4c8a15f2 : SUCCESS in 1h 20m 11s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3064ad09379e491cba68368a78fbd928 : SUCCESS in 1h 14m 36s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e5b6f938c1f6485c93b49577babcf1ba : SUCCESS in 1h 05m 11s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6c2ae9cb579645e7b8f3cb1f10e58842 : SUCCESS in 1h 42m 08s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23b7fa91853f425db791d51870d2e2e9 : SUCCESS in 1h 02m 25s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c004933af8b146d4b9b892f1635e520f : SUCCESS in 1h 20m 50s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/772e8d872abc4bd7ab41faa4231ada0c : SUCCESS in 47m 13s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3e0393cd01114f6ebe43f52aeb971393 : SUCCESS in 1h 17m 00s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7d9a6486bb084d03b0bcdf72f39095e3 : SUCCESS in 1h 00m 47s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/613201188d1d4318bf5dddaf752b7020 : SUCCESS in 3h 43m 57s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a6077539f5854fd48b5c645e6e01da0b : SUCCESS in 1h 24m 27s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f8075e5c43644be1845bf6249a58a982 : SUCCESS in 1h 46m 23s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ee268d2577a946e68a3d0ecd094ab956 : SUCCESS in 2h 07m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/072ce07df64f48df999377ed83449470 : FAILURE in 1h 39m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0ff3832075e98104bb1dfb216df63cbbd9dad109","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-10 14:02:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7a522cfe2adef07fd40b445ef5f3f26eeb1795ed","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-01-10 15:30:59.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c9496f99f403d1060c13b3ae586368bc0716af03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 15:32:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"626db8577881c966d6828a2ce6c10a44483c2d62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 22:31:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/655383f89266491fb5c1729e8faeb17f : SUCCESS in 13m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86fb407b0df5482cb15fcca3ac419d5d : SUCCESS in 21m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12194573ff8f4b7c983370037ebd59f6 : SUCCESS in 29m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6507633e1f84f2498be67d59c76bd3c : SUCCESS in 13m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4a9139b40cec42908aebfd1291747047 : SUCCESS in 55m 20s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/df7f0cba02ce48af9034b649c7775271 : SUCCESS in 1h 21m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/56c1b921dd7c498eb7c0df26747e0e98 : SUCCESS in 1h 15m 52s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b242a035103c461e907e07e5c1644d5b : SUCCESS in 1h 07m 43s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/76c412b76ac44035a4af5c3efb375d54 : SUCCESS in 1h 12m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b25b5cf93687454fb4d7b156fc469009 : SUCCESS in 1h 15m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a2bb26d5ef9447b0b03715c927e158b0 : SUCCESS in 1h 17m 09s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2cbbca5149a1407e8e6146915f11b3d9 : SUCCESS in 1h 23m 29s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2741d75653c54cafa58570c9ad3d9291 : SUCCESS in 1h 20m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0a92ebcf93764e79954d4de936ec34c1 : SUCCESS in 1h 21m 04s","accounts_in_message":[],"_revision_number":1},{"id":"7b6ec247786cc4132475a24781de73e4a901d215","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 22:31:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cc1c03cd44f69a356f104b983a528a9927f6f658","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 22:34:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a1865e9593149b3b4facde460543a47 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bed8c6d58028b7d5a426dbd474dccbe34048fd63","revisions":{"bed8c6d58028b7d5a426dbd474dccbe34048fd63":{"kind":"REWORK","_number":1,"created":"2021-12-22 14:16:39.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/09/822709/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/822709/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/822709/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/822709/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/822709/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/822709/1"}}},"commit":{"parents":[{"commit":"607dacdd180faaada54987483877bf89a10f319f","subject":"Merge \"Do no use \"--strict\" for OF deletion in TRANSIENT_TABLE\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/607dacdd180faaada54987483877bf89a10f319f"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-12-16 21:25:27.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-12-22 14:15:58.000000000","tz":60},"subject":"[Functional] Fix expected number of the enqueue_state_change calls","message":"[Functional] Fix expected number of the enqueue_state_change calls\n\nIn the HA router\u0027s keepalived state change monitor tests, it was\nexpected that enqueue_state_change method will be called 3 or 4 times.\nBut after some changes in the keepalived_state_change monitor which were\ndone some time ago, it may be now that it will be called just 2 or 3\ntimes:\n- 2 when initial status will be \"primary\" and it will be just\n  transition to \"backup\",\n- 3 when initial status will be \"backup\", then it will transition to\n  \"primary\" and finally to \"backup\" again.\n\nTo reflect those 2 possibilities, test was changed that it will expect\n2 or 3 calls and will check only that last 2 will be always transition\nto \"primary\" and then to \"backup\".\n\nAdditionally this patch adds some extra logging in that test so it will\nbe easier to check what was going on in that test.\n\nConflicts:\n    neutron/tests/functional/agent/l3/test_ha_router.py\n\nCloses-Bug: #1954751\nChange-Id: Ib5de7e65839f52c35c43801969e3f0c16dead5bb\n(cherry picked from commit c6a6c5ae127a97856c880da3f82257977a06a571)\n(cherry picked from commit 533ca2c99c454c171f2b8d6577a39bb7be6376e6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bed8c6d58028b7d5a426dbd474dccbe34048fd63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bed8c6d58028b7d5a426dbd474dccbe34048fd63"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
