)]}'
{"id":"openstack%2Fneutron~823564","triplet_id":"openstack%2Fneutron~stable%2Fxena~Ib90704a85c9ed659d71ac8ed42b1b542bd2496a2","project":"openstack/neutron","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2022-01-07 09:55:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib90704a85c9ed659d71ac8ed42b1b542bd2496a2","subject":"[OVN] Update check_for_mcast_flood_reports() to check for mcast_flood","status":"MERGED","created":"2022-01-05 17:04:38.000000000","updated":"2022-01-07 09:57:18.000000000","submitted":"2022-01-07 09:55:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823564","meta_rev_id":"4edb1267b39b35e84df8cdc96cf07e6e06557d5a","_number":823564,"virtual_id_number":823564,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-07 09:55:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-01-07 08:37:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-01-06 10:56:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-01-07 08:37:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-05 18:05:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-05 18:23:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-06 10:56:23.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-01-07 08:37:12.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"b4b1a75665cb16677212524eb6f04fdb78219139","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-01-05 17:04:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63f7287604e9934ac95f5efacc8535cc435e18aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 18:05:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77a0d3b7cd5b4fe3b8b1b22741301d84 : SUCCESS in 50m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/710c8e794a0549cab8343ce14f7e4332 : SUCCESS in 48m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26bdac8ab051712ebc7feae66ac987cbff11bfcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 18:23:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a44af54a15f44171adb5cc04e4d75cdd : SUCCESS in 59m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/458d6187f681446c9cdcb9c072dbf7c9 : SUCCESS in 1h 16m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6243cb5e0a464eefaa0b22aad48265e4 : SUCCESS in 45m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33e0a4e8ce43465a92916cd884f922b9 : SUCCESS in 12m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/012283ec864b4972b94d0146116467be : SUCCESS in 23m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6fb3c0ccb76f4043b1cf135d2ab770cb : SUCCESS in 25m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4c411af27de40fd93b6014e204d8090 : SUCCESS in 29m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d608540dbe214f5cbc29770e7cfe7fba : SUCCESS in 11m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e0159f0a93af4865bf8cbc4a38782c50 : SUCCESS in 50m 12s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8817143634274099951e20ec58f43eaa : SUCCESS in 50m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/07518d6984de49aba7870e0b2065b746 : SUCCESS in 1h 10m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a10b3ab2a94bdf84e42f4a048bd5e2028f4310d1","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-01-06 10:56:23.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1c069fde9f92956a54da3cb295f4d5e35431680b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-07 08:37:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"14c04f7fcd843a7138ff470f1d12e4eb0966406a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 08:37:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7b0d22bcbd238248feaa9513f0a1d55a2bc27ef8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 09:55:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/293ed0359c374324a944951e47465ee5 : SUCCESS in 43m 28s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/119a6f2e438d43cc98ebb289b3d29c4c : SUCCESS in 1h 16m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb3e7e9b1afd4e98a22bb8702f99b3a1 : SUCCESS in 10m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2add9a53f234e0ab50bdd3febec5fc9 : SUCCESS in 28m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14695d6279454312a777bc7fc4beff7a : SUCCESS in 28m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9383c5beefd24db190883684f86575a8 : SUCCESS in 9m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c4ba07b82dc642c6bda2e8793161fcf5 : SUCCESS in 54m 30s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/451dc7c42f0e44f39c7f65fe9a6f436b : SUCCESS in 1h 03m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5520118434af0eb6fa394c21f64a7d0781a928cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 09:55:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4edb1267b39b35e84df8cdc96cf07e6e06557d5a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 09:57:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80e660029c2a47909172207d70450c89 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"09f713d5d50307eeead575d377fdb38711691045","revisions":{"09f713d5d50307eeead575d377fdb38711691045":{"kind":"REWORK","_number":1,"created":"2022-01-05 17:04:38.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/64/823564/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/823564/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/823564/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/823564/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/823564/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/823564/1"}}},"commit":{"parents":[{"commit":"484b881c6cf62fd795cfdb32209e456befb6e1f7","subject":"Merge \"Add Chassis creation wait event in \"TestAgentMonitor\"\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/484b881c6cf62fd795cfdb32209e456befb6e1f7"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-10-28 13:17:20.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2022-01-05 16:57:42.000000000","tz":-180},"subject":"[OVN] Update check_for_mcast_flood_reports() to check for mcast_flood","message":"[OVN] Update check_for_mcast_flood_reports() to check for mcast_flood\n\nThe patch [0] changed the value of the \"mcast_flood\" option for the\nlocalnet ports in OVN to \"false\" to avoid multicast packet duplication\nbut, the maintenance task check_for_mcast_flood_reports() was never\nupdated to account for that change.\n\nSo, localnet ports created prior to [0] would continue to have the\n\"mcast_flood\" value as \"true\" and new ports would have it as \"false\".\nThis patch fixes the maintenance task.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/797418\n\nChange-Id: Ib90704a85c9ed659d71ac8ed42b1b542bd2496a2\nCloses-Bug: #1949081\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit d02fb560aa35c4931d40000e8ca4c0e4105299b7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/09f713d5d50307eeead575d377fdb38711691045"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/09f713d5d50307eeead575d377fdb38711691045"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
