)]}'
{"id":"openstack%2Fneutron~823565","triplet_id":"openstack%2Fneutron~stable%2Fussuri~Ib90704a85c9ed659d71ac8ed42b1b542bd2496a2","project":"openstack/neutron","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2022-01-07 10:30:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib90704a85c9ed659d71ac8ed42b1b542bd2496a2","subject":"[OVN] Update check_for_mcast_flood_reports() to check for mcast_flood","status":"MERGED","created":"2022-01-05 17:06:10.000000000","updated":"2022-01-07 10:32:18.000000000","submitted":"2022-01-07 10:30:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823565","meta_rev_id":"0847e08c65ab2922518bf7011e839438b95aa695","_number":823565,"virtual_id_number":823565,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-07 10:30:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-01-07 08:37:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-01-06 10:57:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-01-07 08:37:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-05 20:23:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-06 10:57:18.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-01-07 08:37:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"9e5bf068efc3560d4ff3cc1c1bbfdb43f79cea93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-01-05 17:06:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05f8225eacf8a9d867cda4e577ad5eb28fc4d7fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 20:23:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/f1f539637a2f40f4ac825cc5ace792f2 : SUCCESS in 51m 01s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/a813e9211d804fa49baee24918956198 : SUCCESS in 3h 03m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd8039990c1b419888d99dc0a00fe67e : SUCCESS in 38m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e15f58fb1cce40d38df15e44ad429b2e : SUCCESS in 12m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35e1c52a204b4f4ab1d240b81580347a : SUCCESS in 27m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3031e0415ffd4cf29e59e072833a3ae8 : SUCCESS in 31m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb70897e66d74ec8a57263cd170faa2a : SUCCESS in 29m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4272f7db4e534e36bc1fa1c53dc77162 : SUCCESS in 12m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/964dbff6d36544159bcaea655b1cb379 : FAILURE in 56m 24s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/16a4a670656a406eb9f8d4e1ed8744a6 : SUCCESS in 1h 24m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f94ca98436a745188c6737bae6210345 : SUCCESS in 1h 21m 43s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e83b0efab4a54673bf070316f3d6b926 : SUCCESS in 1h 32m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6d52c89338fe4a099e577d436c260cbd : SUCCESS in 1h 25m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b96ce405249943d7bca581eae9416f5f : SUCCESS in 1h 14m 38s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/cf06c14c3970462a8c04894b6e7a7b8b : SUCCESS in 1h 12m 15s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7a025e0c61af45ea94bea293cc4a0c7c : SUCCESS in 1h 05m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b1b37e908ed740f1b039455a7f9e4eb3 : SUCCESS in 1h 18m 47s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/255aeb3a304945b7b0f78299afbae5df : SUCCESS in 1h 27m 27s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8bed9e18e7c04dfeaf496150d28dbada : FAILURE in 1h 20m 09s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eb4ca5a859da414e9edd24fe6c48f836 : SUCCESS in 1h 14m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/41020b2d7f4549e6a4e36d1bbdabd4ef : SUCCESS in 56m 05s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ed582055152442aa828d08245c9a05eb : SUCCESS in 1h 25m 16s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae48eee43b994542b97f0033ee371c55 : SUCCESS in 1h 10m 19s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f9748af5500543a4854213d2993700fb : SUCCESS in 1h 16m 27s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/89dd3beeca1943d895f32756abe43a8e : SUCCESS in 36m 23s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/9e4467701183490ba389a83d8032f162 : SUCCESS in 1h 27m 54s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c329a16e18d34c6cb9bcda50beac7d40 : SUCCESS in 47m 50s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f0b757ed9edb450a9b5256780e68cd82 : SUCCESS in 3h 14m 23s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6f762e753aa24dc984cf1e66dbbaa096 : SUCCESS in 1h 28m 53s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2d56820d6ae749869bbdc203e564f4c0 : SUCCESS in 1h 25m 49s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/9e55f0d980eb4aa8a121ba2f83cdf7a4 : FAILURE in 1h 40m 05s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4e2fdbdd35db4f58964af6a3beed66a5 : FAILURE in 1h 36m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d85a886dbcf428efc36507b64cdb8d7ac52c645","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-01-06 10:57:18.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"228c6744f1a72f440708db90d7fb2d73b263a632","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 15:13:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/d335aa4bfcf34c3090290d2412c17124 : SUCCESS in 45m 57s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/df2f71b7358848df96b4f952cb6159a7 : SUCCESS in 1h 46m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c4eeca7610341b6b1aa4b7efcc7b40e : SUCCESS in 41m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2049b388bee4322afb1ecff1e4e6140 : SUCCESS in 14m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14e65df8c77a42ebaa6a3a70dc6678cc : SUCCESS in 31m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1d24a29bc46d4b3194101df44afb63ff : SUCCESS in 30m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc6cc1e453af4a4ea8fc95468f5b98c6 : SUCCESS in 24m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2fb121a3463468fa7d5b1237a67fd90 : SUCCESS in 9m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7b52fd9c18354bc8a8ea8c2fa178b4fd : SUCCESS in 54m 55s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/263ee69ec73b46c9a4c956d50a776435 : SUCCESS in 1h 23m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/44941d91d7664787be9cea939d4255c9 : SUCCESS in 1h 21m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9a21f990661a42ffaebb451d4292b834 : SUCCESS in 1h 20m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c2031d23fa1348b88bb297643062624f : SUCCESS in 1h 10m 29s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/eca35d46d11b4778813d4b088425a524 : SUCCESS in 1h 14m 58s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0982f783574145bbbaf6559516d5d235 : SUCCESS in 1h 06m 46s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/1ca87d626da64b6f8fa0040017acfd09 : SUCCESS in 1h 04m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/058ccdb1a86944c98f9e1cdbd6856c16 : SUCCESS in 1h 06m 34s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6aa74964d61e43638c6ce7f245b69a65 : SUCCESS in 1h 32m 36s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/907ccac2a7a144d8bb59a655026c764f : SUCCESS in 1h 18m 54s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3b560686833644af819a21894c50defb : SUCCESS in 1h 07m 58s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4917cc97af6c4214bc40da24c0568510 : SUCCESS in 59m 40s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b6b3bfd92e9d4e4b8b7552dd46394218 : SUCCESS in 1h 37m 46s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5a4be5051524f7bbb3e897dbca853eb : SUCCESS in 1h 00m 11s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ccb484c1eb844164af4eae8841a58587 : SUCCESS in 1h 22m 13s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f5516b3b428042f0b5539a506a0133c7 : SUCCESS in 35m 48s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7233741a13e5468691cdf0d1b70d1c16 : SUCCESS in 1h 15m 16s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8e5cbd0fd714410abd02ac3c76780206 : SUCCESS in 50m 34s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7e47c95a735044a78dcbbd1248a69bb1 : SUCCESS in 3h 34m 05s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6b1025eb486042c2a7d304b41d8d412d : SUCCESS in 1h 18m 37s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/50cecd24458941c1849095671fa93302 : SUCCESS in 1h 41m 18s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/4f07627b6a29447ab080f8641fe8c088 : SUCCESS in 2h 06m 11s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ee5606330228460cb6ffb25ee0dadda9 : FAILURE in 1h 44m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f3a0c3c0cbe96c1c6a32821f8fb01849d619eac","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-07 08:37:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"53b954f1b49f519dcb29af48e62b2dd741edbef6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 08:38:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3c2675f0e85568caf2a486b92bb86b098d9c5425","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 10:30:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/06fcb94527a240b4b5d9f4fb9f0056ee : SUCCESS in 41m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1752695388c4167a40511cc3df55a64 : SUCCESS in 12m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6b1d2ecd44b4f9cab585aee0426ffaf : SUCCESS in 21m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c54fa6bae56f4091afcb872dddeb47da : SUCCESS in 23m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee5856dd60c241cf9df9dd0315c66a0f : SUCCESS in 14m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8e74517973ce4376895a45def732f3b5 : SUCCESS in 47m 57s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8c05a504da6840769fe4996950ddb70d : SUCCESS in 1h 25m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7b8564a7a62e4bb693eaebc2063002e3 : SUCCESS in 1h 02m 01s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6c5c6542818e47b4aa7a32dc20fbcd10 : SUCCESS in 1h 09m 42s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a63e65100a8d4df5946f976af558be4c : SUCCESS in 1h 00m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f34128d8a2e54411a29948de750a3367 : SUCCESS in 1h 11m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/309836fad0434fbf83e646236c9bd183 : SUCCESS in 1h 08m 42s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6e027ea6255a4b279c743119416c8b46 : SUCCESS in 1h 26m 30s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4932bb84622423c93b58bb533fac7f5 : SUCCESS in 56m 35s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/21645ab3d49b444c9ae454ee59d098c5 : SUCCESS in 1h 06m 49s","accounts_in_message":[],"_revision_number":1},{"id":"a56fb4ecbc870fe23709248241c4ea662319f5c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 10:30:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0847e08c65ab2922518bf7011e839438b95aa695","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 10:32:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22add5da955a40fa9b1d120870c84dae : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3113930ac76f33afc3b6becab2081419a6a1bf22","revisions":{"3113930ac76f33afc3b6becab2081419a6a1bf22":{"kind":"REWORK","_number":1,"created":"2022-01-05 17:06:10.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/65/823565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/823565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/823565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/823565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/823565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/823565/1"}}},"commit":{"parents":[{"commit":"7bca2cc526b8ea391a1da627f39b5496a329504d","subject":"Merge \"[OVN] Accept OVS system-id as non UUID formatted string\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7bca2cc526b8ea391a1da627f39b5496a329504d"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-10-28 13:17:20.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2022-01-05 17:06:07.000000000","tz":-180},"subject":"[OVN] Update check_for_mcast_flood_reports() to check for mcast_flood","message":"[OVN] Update check_for_mcast_flood_reports() to check for mcast_flood\n\nThe patch [0] changed the value of the \"mcast_flood\" option for the\nlocalnet ports in OVN to \"false\" to avoid multicast packet duplication\nbut, the maintenance task check_for_mcast_flood_reports() was never\nupdated to account for that change.\n\nSo, localnet ports created prior to [0] would continue to have the\n\"mcast_flood\" value as \"true\" and new ports would have it as \"false\".\nThis patch fixes the maintenance task.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/797418\n\nChange-Id: Ib90704a85c9ed659d71ac8ed42b1b542bd2496a2\nCloses-Bug: #1949081\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit d02fb560aa35c4931d40000e8ca4c0e4105299b7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3113930ac76f33afc3b6becab2081419a6a1bf22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3113930ac76f33afc3b6becab2081419a6a1bf22"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
