)]}'
{"id":"openstack%2Fneutron~823593","triplet_id":"openstack%2Fneutron~stable%2Fvictoria~I3c106ace74b5c6e4ed0cb7e827baf5d6595ec5d0","project":"openstack/neutron","branch":"stable/victoria","topic":"bug/1914394-ovn-metadata-networks-stable/victoria","attention_set":{},"removed_from_attention_set":{"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2022-01-06 11:28:53.000000000","reason":"Lucas Alvares Gomes replied on the change"},"13686":{"account":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"last_update":"2022-01-07 17:50:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3c106ace74b5c6e4ed0cb7e827baf5d6595ec5d0","subject":"[OVN] Fix RowNotFound exception while waiting for metadata networks","status":"MERGED","created":"2022-01-06 06:23:41.000000000","updated":"2022-01-07 17:52:59.000000000","submitted":"2022-01-07 17:50:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823593-bug/1914394-ovn-metadata-networks-stable/victoria","cherry_pick_of_change":773893,"cherry_pick_of_patch_set":2,"meta_rev_id":"b7444435a28827771ef4b9657c640e8007417d86","_number":823593,"virtual_id_number":823593,"owner":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-07 17:50:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2022-01-06 11:28:53.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2022-01-07 13:54:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2022-01-07 13:54:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-06 06:23:41.000000000","updated_by":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2022-01-06 10:01:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-07 13:54:03.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"2acb15d6db9ef1d0973beef5570c13c9cf085dac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-01-06 06:23:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2e36a15f5b45b6a4fd9a6491411f011663723e52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 10:01:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/82788e9e9a2d4121868dcd085cbab9c7 : SUCCESS in 43m 53s\n- neutron-tempest-plugin-scenario-ovn-victoria https://zuul.opendev.org/t/openstack/build/c7102fa6cb1e456fb155e8fbf2df62a5 : SUCCESS in 1h 28m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a988de142e554891b47565e4f0ab24d6 : SUCCESS in 32m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02b25c3bcb764d66bdaf63721b763c86 : SUCCESS in 11m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f02051b381f84a438d6bdfd8258283e4 : SUCCESS in 21m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c32e6697ea474531aa2ff8a419bb9cb5 : SUCCESS in 21m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10c75e0ceb964d4581ba08cab3a0d2a3 : SUCCESS in 9m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5c5d81c8f6944679a572277acd5a6910 : SUCCESS in 1h 15m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b79cc7d679649b3a309bb3826252b9c : SUCCESS in 1h 21m 42s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5f933810c3ce4c51a9df1d8077643fc3 : SUCCESS in 1h 17m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a224e7cb5e0b449bb498558a2fe320a7 : SUCCESS in 1h 05m 17s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/03841887c7cc4617a1b2cfa41ff54c28 : SUCCESS in 57m 06s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/5275d9773e4f4c98a81497bcbf346eeb : SUCCESS in 1h 15m 53s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d64b155d61834719bf8aebd675f768c4 : SUCCESS in 57m 27s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/47306763d6384a6e8c934bdd7a935b70 : SUCCESS in 1h 07m 38s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2520406a1b284e9783023e545cd80182 : SUCCESS in 1h 23m 16s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/001224b774154adb933a825b237510c4 : SUCCESS in 1h 12m 58s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6cec17a71a3443858520f235c510576a : SUCCESS in 1h 06m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a513e1008d7a4248bfabdf96690beda9 : SUCCESS in 1h 19m 53s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fe3a7b16fe95412e9c7c6aafc2f39e0c : SUCCESS in 1h 26m 46s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48c270e78eb841faa832e99b0f169bac : SUCCESS in 1h 12m 08s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8612207ffefd4459bfd5eba587409420 : SUCCESS in 1h 19m 50s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e24cf2b73dc84293b54f955591105d67 : FAILURE in 32m 00s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/14bd996e7813464b8b9f59ed3d7fa9cb : SUCCESS in 1h 16m 31s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d8cd71910ea64030aed68beef8bb24ab : SUCCESS in 3h 35m 14s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/be93a9d4876145aab9d45a6b7c784601 : SUCCESS in 1h 20m 51s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b9175316213c4b878a7d78c589e4ac4e : SUCCESS in 1h 42m 43s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/915c8ddfae8b490bb10fc83e95d28323 : SUCCESS in 2h 02m 33s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/30b993e03dbf4e059fcca24ff50ecde7 : FAILURE in 2h 53m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"50db96146560d4262ef6c891347fbdbf1bc778f8","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-01-06 11:28:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5d6663f3d942699bd3b1594fb28994f88829650e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-01-07 13:54:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b7ef9a5b2726086d4aca95230e33b4dd4e6d227c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 13:54:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fdeb66b14aa1157de84a21b87f2a5dbc07cab4e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 17:50:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/600c4d34aa934f6d901a6f1695fe679a : SUCCESS in 47m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d852cdf34642467289aef7068631ad67 : SUCCESS in 13m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/992fb2b57b2f4d2d993c1f8b59c941a2 : SUCCESS in 25m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c311d93578c449e1bcefd02920267ac7 : SUCCESS in 26m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36b7182e379a4b10b7d379b4e8dbf168 : SUCCESS in 8m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a07027f003044b95850c673ebbe9f46e : SUCCESS in 1h 12m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fe82b6bd6cc049c683ae8da06f14efb3 : SUCCESS in 1h 19m 17s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/eb6308cebd914195a18b9be1140d51c7 : SUCCESS in 1h 02m 16s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e89be8d0f35a4d1ca6a3756a13c0d1b1 : SUCCESS in 59m 54s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c452e1f47e814c049ab529446e385ed8 : SUCCESS in 1h 06m 54s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/00bdcc7f790b44a184d640dbfd76df28 : SUCCESS in 1h 17m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/86a1cf37359344d6b25e9736675fd56a : SUCCESS in 59m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a40f32c38ec1455a8af9c2fd89e3dfb4 : SUCCESS in 56m 58s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b223f3a295c74d19bdfd5abb155d12e4 : SUCCESS in 1h 33m 07s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81cb9d843a874230a8097e2a2c94a38a : SUCCESS in 1h 05m 04s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d59480ee9ae4423abbafbd2383cc9122 : SUCCESS in 1h 14m 26s","accounts_in_message":[],"_revision_number":1},{"id":"d93276696bc5cf7201d045336a591cb0410e7ca1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 17:50:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b7444435a28827771ef4b9657c640e8007417d86","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 17:52:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cce63b881fa4e8d9e23bd950451e5c9 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cbf3fe098bb44605174b8e582d56814fd36632c7","revisions":{"cbf3fe098bb44605174b8e582d56814fd36632c7":{"kind":"REWORK","_number":1,"created":"2022-01-06 06:23:41.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/93/823593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/823593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/823593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/823593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/823593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/823593/1"}}},"commit":{"parents":[{"commit":"124215ccd63e64412051c66fb28e6898c7e2fa10","subject":"Merge \"[OVN] Accept OVS system-id as non UUID formatted string\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/124215ccd63e64412051c66fb28e6898c7e2fa10"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2021-02-03 14:00:13.000000000","tz":0},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-01-06 06:23:41.000000000","tz":0},"subject":"[OVN] Fix RowNotFound exception while waiting for metadata networks","message":"[OVN] Fix RowNotFound exception while waiting for metadata networks\n\nIn the set_port_status_up() the OVN driver waits for the metadata to be\nprovisioned (15 seconds) [0] prior to sending the event to Nova notifying\nthat the provisioning of the port is done (network-vif-plugged). But\nthere could be a race condition while trying to get that information\nwhich results in a RowNotFound being raise in the waiting loop.\n\nOnce that happens, the exception is bubbled up and the OVN driver end up\nnot sending the event to Nova and the instance will fail to deploy (it\nwill be stuck in BUILD state until it times out).\n\nThis patch changes the logic of the method looking for the metadata\nnetwork information to not raise RowNotFound so that the waiting loop\ncan iteract again [0] until the information is available.\n\n[0]\nhttps://github.com/openstack/neutron/blob/cbd72e2f4846ec64ff6e6ef24099a8e90ddebf31/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py#L1124\n\nChange-Id: I3c106ace74b5c6e4ed0cb7e827baf5d6595ec5d0\nCloses-Bug: #1914394\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit b618d98541599ad0ef73be41a3edd83d1fc75a56)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cbf3fe098bb44605174b8e582d56814fd36632c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cbf3fe098bb44605174b8e582d56814fd36632c7"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
