)]}'
{"id":"openstack%2Fneutron~823596","triplet_id":"openstack%2Fneutron~stable%2Fvictoria~If03a2ad2475cdb390c4388d6869cd0b2a0555eb7","project":"openstack/neutron","branch":"stable/victoria","topic":"bug/1951872-stable/xena-stable/wallaby-stable/victoria","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-01-12 10:39:02.000000000","reason":"Change was submitted"},"13686":{"account":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"last_update":"2022-01-12 10:39:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If03a2ad2475cdb390c4388d6869cd0b2a0555eb7","subject":"[OVN] Add reverse DNS records","status":"MERGED","created":"2022-01-06 06:45:34.000000000","updated":"2022-01-12 10:49:54.000000000","submitted":"2022-01-12 10:39:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823596-bug/1951872-stable/xena-stable/wallaby-stable/victoria","cherry_pick_of_change":823595,"cherry_pick_of_patch_set":1,"meta_rev_id":"fa1f2adc0a283c01178ca1a00571bb139f1e5458","_number":823596,"virtual_id_number":823596,"owner":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-12 10:39:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-01-11 13:12:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-01-12 08:22:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-01-12 08:22:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-06 06:45:34.000000000","updated_by":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-01-06 10:56:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-11 13:12:11.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-01-12 08:22:52.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"5e34bc80d57d6dd78cd4c819b4697dd40c02ff43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-01-06 06:45:34.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"cb542001d8e13d0a8a132c45d43043123c6ddf99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 10:56:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/f62dcf3010fe4a9592bf347d24907891 : SUCCESS in 47m 21s\n- neutron-tempest-plugin-scenario-ovn-victoria https://zuul.opendev.org/t/openstack/build/dc05419e53954a4db46b363b23332aa6 : SUCCESS in 1h 04m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f9484d41a7e4015b3aee53e0bab4c40 : SUCCESS in 32m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0e2df4397264772a630af0f1ea06102 : SUCCESS in 14m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad421c0db16f4d5f9648a12853ec5e42 : SUCCESS in 25m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba599e63372f433694c91a5eed648057 : SUCCESS in 26m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2eaf2c15a9a6415592f858b5122c14c9 : SUCCESS in 9m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a91faa6bf670425bab0bfbe8138352a4 : SUCCESS in 1h 15m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c427e9057594c8fb278873bc445abbb : SUCCESS in 1h 21m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/773844062c5441d1832a53e4fedd1fe9 : SUCCESS in 1h 23m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a558177194214de7a4b54f7e17605067 : SUCCESS in 1h 00m 27s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0e5e0ae8d226466f81d133fcad213426 : SUCCESS in 56m 27s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/ea5542cc6e844422a7f19b468d00c7be : SUCCESS in 1h 16m 25s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0d7f93cdccd84998854a08c6834745e6 : SUCCESS in 59m 05s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/8c5a7cc088a042228fb3a122279fd2fa : SUCCESS in 1h 04m 46s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c4a616a7a1234868a089439b8e1b5a92 : SUCCESS in 1h 13m 23s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/526bfa1fb345406ea45909bfabdbf3eb : SUCCESS in 1h 41m 24s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7ea882c9ef894638973d369d932b0ac1 : SUCCESS in 1h 04m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4084b73362dd45778a0d93bfe21a3d88 : SUCCESS in 1h 13m 21s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6b34b8c03260475f8aa51b04a436031f : SUCCESS in 1h 35m 30s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d829b7bfb7ca4ac0af6306af077c8652 : SUCCESS in 1h 15m 14s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a19009d7a8844480864b9f13775e50eb : SUCCESS in 1h 12m 50s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ce42a973e6f84e6ab881fc6dc003d990 : FAILURE in 36m 37s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/9c2a4b76c6874ab9bfd36f002c2378a3 : SUCCESS in 1h 16m 50s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5b2545d441c440d8860c6792972a9424 : SUCCESS in 4h 02m 11s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a790ed94ec7643ee806f956ca8fcc647 : SUCCESS in 1h 36m 47s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/004f55f3ae5846f492a6f5b90584d981 : SUCCESS in 1h 32m 46s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e84895ba8e5d44a2a8a36602a43b8f12 : SUCCESS in 2h 23m 31s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4ad4fe6e918e432ea236e7afdabe67fd : FAILURE in 35m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"929c600fab3f5069cdb375e5da3660279f1e31b0","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-01-11 13:12:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"441606bb05ae00695ca1ed51d0985cc2f07df566","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-12 08:22:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1f236de4ea785fa1d98cb0f5fd8aaefe2d1f29db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 08:23:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b87a18871521a3fe407aba874c6b9b72ae51dbd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 10:39:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/8fdb196449d6432b8c14e8c2a7da1705 : SUCCESS in 1h 02m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41163971bb5f413eb63158eac94dc427 : SUCCESS in 15m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6eafdc296d4e4279922d75e73b4204a0 : SUCCESS in 25m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60726be39e0b4ea18649c05f42a42651 : SUCCESS in 24m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80ea494d237a4c39a100b73bc3f7ff02 : SUCCESS in 11m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a33bcbec9ba645b48bf9aef11a9107e7 : SUCCESS in 1h 24m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/96225de15022489da8e45ea22893021d : SUCCESS in 1h 24m 44s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/ca49ac4df7924843b72d57d8b14d2240 : SUCCESS in 1h 15m 30s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/962fa470880b434daad1412dc7656632 : SUCCESS in 1h 07m 40s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7b4e91ef64a74e2d98e9ab3cae6122b8 : SUCCESS in 59m 38s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/32c8098be550418dac6d5985d2fb3ff1 : SUCCESS in 1h 14m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/925b23f6e5c04ff385c8cc17f34e3840 : SUCCESS in 1h 04m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5cf59b8a57214b9b91dbbae25168ae41 : SUCCESS in 1h 05m 15s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a8116bfc78f14663a86f8f00339ff3b1 : SUCCESS in 1h 33m 45s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/764a90eb12f3423dbba9a58321fda741 : SUCCESS in 1h 07m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5b73c653745a47529965314b982f942b : SUCCESS in 1h 10m 11s","accounts_in_message":[],"_revision_number":1},{"id":"c7b8c860443254e070991a809ee276ef1854c015","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 10:39:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fa1f2adc0a283c01178ca1a00571bb139f1e5458","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 10:49:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5c6fc20ed524abeb3c4a338deb14894 : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3664aa1bb421cc1d7add0c2c5209af27b384ec1b","revisions":{"3664aa1bb421cc1d7add0c2c5209af27b384ec1b":{"kind":"REWORK","_number":1,"created":"2022-01-06 06:45:34.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/96/823596/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/823596/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/823596/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/823596/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/823596/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/823596/1"}}},"commit":{"parents":[{"commit":"124215ccd63e64412051c66fb28e6898c7e2fa10","subject":"Merge \"[OVN] Accept OVS system-id as non UUID formatted string\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/124215ccd63e64412051c66fb28e6898c7e2fa10"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-12-20 07:37:25.000000000","tz":330},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-01-06 06:45:34.000000000","tz":0},"subject":"[OVN] Add reverse DNS records","message":"[OVN] Add reverse DNS records\n\nPTR DNS requests support is available since\novn-21.06.0[1].\n\nThis patch adds/removes required enteries for each ip\naddress of the port in DNS NB table, For example for ip\naddresses \"10.0.0.4 fd5a:cdd8:f382:0:f816:3eff:fe5b:bb6\"\nand fqdn \"vm1.ovn.test.\" following enteries are added:-\n- 4.0.0.10.in-addr.arpa\u003d\"vm1.ovn.test\"\n- 6.b.b.0.b.5.e.f.f.f.e.3.6.1.8.f.0.0.0.0.2.8.3.f.8.d.d.c.a.5.d.f.ip6.arpa\u003d\"vm1.ovn.test\"\n\n[1] https://github.com/ovn-org/ovn/commit/82a4e44\n\nCloses-Bug: #1951872\nChange-Id: If03a2ad2475cdb390c4388d6869cd0b2a0555eb7\n(cherry picked from commit 5b597d6e4ee647502d3f2aecd46de63f409317b4)\n(cherry picked from commit 5b30b14ad726e41793c76ecd0fca8ee66cf51607)\n(cherry picked from commit 591570c5f4069234bef460274aa5442b60eed5de)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3664aa1bb421cc1d7add0c2c5209af27b384ec1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3664aa1bb421cc1d7add0c2c5209af27b384ec1b"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
