)]}'
{"id":"openstack%2Fneutron~823970","triplet_id":"openstack%2Fneutron~master~Id421c5fb36e50f7fdc5a84e04ad6a654c03e4244","project":"openstack/neutron","branch":"master","topic":"bug/1956770","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-01-11 13:56:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id421c5fb36e50f7fdc5a84e04ad6a654c03e4244","subject":"keepalived \"no_track\" implemented in version 2.0.3","status":"MERGED","created":"2022-01-10 11:52:33.000000000","updated":"2022-01-11 13:59:14.000000000","submitted":"2022-01-11 13:56:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"823970-bug/1956770","meta_rev_id":"6d2c8bec38f1477c062b3015e49f548a6c9ac3e0","_number":823970,"virtual_id_number":823970,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-01-10 12:40:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-11 13:56:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-11 12:13:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2022-01-10 16:27:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-11 12:13:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-10 12:37:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-10 12:40:35.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-10 13:51:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-10 16:27:46.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2022-01-11 03:50:02.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-11 12:13:23.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"3e4931e3d80c8a37d6427f8df04c9375a365bf7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-10 11:52:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95c346297bc2a124a48400212973017627eb2534","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 12:37:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/571e5954e1e244a5a2a1713fdf6f64c2 : SUCCESS in 40m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff4d06df1548452fa400c5a9ed5104f5 : SUCCESS in 38m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b7b9c253a4fdfa76925c7e43b927b9bf17df2c6c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-10 12:40:35.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/70/823970/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"26ea237f717dc86c9f68759b5d7c039481f0bd3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 13:51:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c104ee771b634a6ca8a0879134b19f89 : SUCCESS in 47m 38s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/642efb28f05541e0b1b870fce511e445 : SUCCESS in 1h 25m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e8fe2e510ddf4a5cafaa4e0aead3407d : SUCCESS in 1h 18m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d092e62103d34c94bc85ce63eb1cc9e6 : SUCCESS in 1h 09m 37s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/402b4da4e0c443d8854ac8105b0b17e4 : SUCCESS in 1h 16m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14b9d0bc04924eacac58f7893ddd39f9 : SUCCESS in 31m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/823dc2af3ab448a5ac4c81bbcd5cb2b7 : SUCCESS in 29m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c02092e394fc4719be19d4d990e66817 : SUCCESS in 12m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c6b2b711ed34b788c140cf375e808d6 : SUCCESS in 26m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a89d8a95b7314c6d8f96ef3b339c9aa2 : SUCCESS in 24m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/906a3306e4c64adf8da283a81b96f0b8 : SUCCESS in 10m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/42d52dfec34345ab8aedb3557849bc8e : SUCCESS in 43m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fd8058ae4a02487da135e06fee9a6148 : SUCCESS in 1h 23m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0a1c84c623a941fbb34a451600c7e732 : SUCCESS in 1h 04m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a1bf89b3f5be4e51ae622189b8edbd09 : SUCCESS in 56m 40s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4003acc1d1c4432196cb6ba95a18809c : SUCCESS in 54m 09s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3b8c3fcf8931441989d55de44b6bad48 : SUCCESS in 1h 16m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f9059ebccd694e9d8f897f66a6eab021 : SUCCESS in 1h 19m 40s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1489f1fdd2ce44119ac891624bea370e : SUCCESS in 59m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/98ae446767ed4e21935f4bc4fc691d12 : SUCCESS in 1h 52m 15s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62cfe449af7c4e3f8fac0381097d83c7 : SUCCESS in 25m 09s","accounts_in_message":[],"_revision_number":1},{"id":"ee9c1cf1ba1fdf0a2db08b428847716fc7f1b772","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-01-10 16:27:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8e15ac9d3777ec4583321b64730c7515a742d6e1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-11 03:50:02.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/70/823970/1/SRIOV-neutron-macvtap : SUCCESS\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/70/823970/1/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"38474771022c98c15ec730cfe38cc50eb7ae5aa2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-01-11 12:13:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f57f52bc5328abc3fe79812e91401c6eab61ccab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 12:14:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c9e57bf075de5e946edb045ff5c47757c67e578","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 13:56:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/71d9caf5d9e1487ea3c6aabf76c256db : SUCCESS in 51m 29s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9e7109d3016c4c9c8193ada1bc0284d4 : SUCCESS in 1h 01m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/93b6481c5493474fb9736fbb91c44598 : SUCCESS in 59m 07s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7daf97746ca6452e8fa79a96f00ece40 : SUCCESS in 48m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cbe8809cdd324ec2823a7483dc268155 : SUCCESS in 44m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f1442a61f44458a83a18c79ef3267d7 : SUCCESS in 30m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18c58409ff2440649bf5beb83950e33a : SUCCESS in 14m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5aae7ca455ed417a9c5f5fed5d9050e5 : SUCCESS in 24m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/000a74dd5c794bc5a47550641ad976e1 : SUCCESS in 25m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac2c1072d4fd45ea9855570e067f28de : SUCCESS in 9m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e4d7ed9d04594cf190b6ddc3021a8202 : SUCCESS in 52m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4323dcca5389419892bcd5c83642a6ed : SUCCESS in 1h 26m 59s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/78fe8409b309479fa57c36c1bfcfb564 : SUCCESS in 1h 38m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/422ee4f484ed420580b01592255a748c : SUCCESS in 1h 02m 01s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0fb53a61c4c47d8b5f4b7ee86303895 : SUCCESS in 55m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/510de577d65c47eebaa6fa5f29339305 : SUCCESS in 23m 07s","accounts_in_message":[],"_revision_number":1},{"id":"26fe9e669200b3b1c07bf2ecf8f71e7e260c6d5c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 13:56:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6d2c8bec38f1477c062b3015e49f548a6c9ac3e0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 13:59:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ff8a1f7d891428784b04a905ec2766e : SUCCESS in 2m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03d08dee3844710f09f15bb34f557d86666530da","revisions":{"03d08dee3844710f09f15bb34f557d86666530da":{"kind":"REWORK","_number":1,"created":"2022-01-10 11:52:33.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/70/823970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/70/823970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/70/823970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/70/823970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/70/823970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/70/823970/1"}}},"commit":{"parents":[{"commit":"52c46782015a195fa17e237ca46eab887b8ec1ee","subject":"Merge \"Add extra debug log with current and new port statuses in ML2 plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/52c46782015a195fa17e237ca46eab887b8ec1ee"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-01-10 11:50:35.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-01-10 11:50:35.000000000","tz":0},"subject":"keepalived \"no_track\" implemented in version 2.0.3","message":"keepalived \"no_track\" implemented in version 2.0.3\n\nFix the version when keepalived implemented \"no_track\" feature [1].\n\n[1]https://github.com/acassen/keepalived/blob/master/ChangeLog#L2722\n\nCloses-Bug: #1956770\nChange-Id: Id421c5fb36e50f7fdc5a84e04ad6a654c03e4244\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/03d08dee3844710f09f15bb34f557d86666530da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/03d08dee3844710f09f15bb34f557d86666530da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
