)]}'
{"id":"openstack%2Fneutron~825073","triplet_id":"openstack%2Fneutron~master~If7cf55f2c12b388fc34fa942b6b0bf70cb9eb866","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-03-16 09:22:26.000000000","reason":"Rodolfo Alonso replied on the change"},"30380":{"account":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"last_update":"2022-03-25 18:50:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If7cf55f2c12b388fc34fa942b6b0bf70cb9eb866","subject":"[OVN] Reschedule router GW chassis when AZ updated","status":"MERGED","created":"2022-01-18 09:27:09.000000000","updated":"2022-03-25 18:59:57.000000000","submitted":"2022-03-25 18:50:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"825073","meta_rev_id":"20ca79e4c8735a9d1f652902d1efde8b4bb6fe7b","_number":825073,"virtual_id_number":825073,"owner":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-03-14 01:19:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-25 18:50:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-03-16 09:22:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-25 15:49:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-03-16 10:41:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-25 15:49:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-18 09:50:40.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-01-18 10:54:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-18 11:06:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-18 14:24:38.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-01-18 14:29:29.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-18 22:35:28.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-16 10:41:14.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2022-03-25 15:49:52.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"f846e94372cbef6d2eed8a10776558cac6dc4f0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-01-18 09:27:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ccf87a1fe39d678828432646f4a584a4a0e9f28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 10:54:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b54f9c40de3a4215b348c0231cc652c6 : FAILURE in 38m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9d1b6fe5bc364911ba745ac730c6ddca : FAILURE in 49m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb562b911f47554bfafca6c30f19bef9a59ae20b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 11:06:47.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/790f8000496b4132bb8cbf30440a94ca : SUCCESS in 47m 17s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e4b76434b56a400c8930d2d20f4a2cfa : SUCCESS in 1h 18m 48s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6fdc232ab0614226aa10ff4a874fd13a : SUCCESS in 1h 16m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/29c3ec05d99e4c278e7fd7aeba905c8b : SUCCESS in 47m 37s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5e2e4173608247308c81f42d8ca29efa : SUCCESS in 1h 06m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/44b4b9f93d244d2da9342859ad4e76dd : SUCCESS in 29m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e04266cd44240679db70771316fd81b : FAILURE in 35m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8c4c4cf093a4f4292e676f82b6dd28f : FAILURE in 42m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16ebd3de747f4f12823f3c1c8ec96117 : FAILURE in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/619c0007f7ea4b8aa5bcf7fa9851d1ec : FAILURE in 30m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb2670f7a316495eb5a9a5911381d13b : FAILURE in 26m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8e2d977fb474a8da73438bfc4cbcb6b : SUCCESS in 10m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6179d5688ebc4af9864c883021c4a7f5 : SUCCESS in 50m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8585374a94e142ddac18b7a6c7f3cdd3 : SUCCESS in 1h 23m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2be19f60d2084f3291113f30e416d527 : SUCCESS in 1h 07m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/977fe2f4c9334249bd16f45aec8f8619 : SUCCESS in 1h 02m 53s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f623af240d3041408ec9d07c3e5aeea3 : SUCCESS in 1h 00m 48s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/89af3ee2d0584a10b676a84ac91c3933 : SUCCESS in 1h 16m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/94c063a51d644262903b1307d8181ea4 : SUCCESS in 1h 11m 44s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f7a24b5e95040a78e4be349c48a2cb2 : SUCCESS in 48m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c4157761baa647e892f2dfbae9c92f98 : SUCCESS in 59m 03s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f006df89d13497a9e5a3ce80671321b : FAILURE in 27m 56s","accounts_in_message":[],"_revision_number":1},{"id":"de53f90801c01a49be4244090980fccde831aff9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-01-18 11:29:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d8c4abc1072bc24737a93e9d5339e97caae8a247","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 12:30:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/165d2b57c55e4d1ca5f9dc74ee428e14 : FAILURE in 46m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dfa776cef63143fd90e49d9a4990cba6 : FAILURE in 34m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"69ddecc755337e8f096fea48df9954515ed975db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 13:24:58.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a6cbba44178f4f88a3dfd59ff8252494 : SUCCESS in 54m 06s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7dc6c23cd8c14c938c0189a54e0b9f62 : SUCCESS in 1h 07m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/bace1cd1552744a08649fd7ecf9bba76 : SUCCESS in 1h 03m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/59c923087f0b478b96a14b949b8fa0df : SUCCESS in 58m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/80219af1e09d4d20bada10412d6c724d : SUCCESS in 1h 04m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/42e65b63075949a68d2548cb912f3ddf : SUCCESS in 32m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/233aee878ccf4595b0c15a6bfd13e35e : FAILURE in 41m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f688a2a81ca48f995f5d25678e47796 : FAILURE in 33m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39a98169f50d4dc0ae6924fcfdbcdd2f : FAILURE in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1bc2f225f4847eea8044cf4018176bd : FAILURE in 26m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ac8f23a7e0e4a69918fc803c720337d : FAILURE in 27m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2551f1677d184740b9ef8ed1d9aed9ec : SUCCESS in 9m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0afdad8b29a049c88780df374b0ba38d : SUCCESS in 46m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d404044785d24bfe891837e95dc8d354 : SUCCESS in 1h 41m 28s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5985e4fdfaab40ce9278751284422517 : SUCCESS in 1h 08m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/015d8849bdcc44b8802ede59ef6e7522 : SUCCESS in 57m 51s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a798d0c014df474eaa3ac3b3a4cf55b5 : SUCCESS in 56m 43s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/854c4817a9e74d6e9aa904921f30d4b4 : SUCCESS in 1h 24m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/683fac460c1d40d7acce3d6d496df00b : SUCCESS in 1h 10m 30s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/52f19ccbf3f7488893c4811a8cd9a797 : SUCCESS in 45m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b84a78de7f744d0482654e11748c2737 : SUCCESS in 59m 17s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f036b63245f94cf59c29307d43d3afec : FAILURE in 24m 24s","accounts_in_message":[],"_revision_number":2},{"id":"6121295687260c05a55ec183c155f8c5c90dd7c1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-01-18 14:24:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/825073/2 : SUCCESS in 1h 51m 38s","accounts_in_message":[],"_revision_number":2},{"id":"9bfc0704152141c404be95c163181556366a739f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-18 14:29:29.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/73/825073/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"0448b023eb16177e6885c8b933bcea38390d222e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-18 20:36:55.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/73/825073/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"1e7c0b6e8b6ba9b2dc67c69aa989e558aec56fa3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-18 22:35:28.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/73/825073/1/SRIOV-neutron-direct : SUCCESS\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/73/825073/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"6eb6e6482858b2c30dd8dbbe72dcf3d4a05a0868","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-01-19 02:55:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"994d28dad8ec40b92ca7024cbb5f46da9a003e25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 03:57:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb6bf525e0c04f1cabd972e9b69fa77e : FAILURE in 40m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/82177d02325c46b0ad051a4dc8c21472 : FAILURE in 44m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9f25b30fbb682c1af05bc7607fe985d988f99aaa","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-19 03:58:28.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/73/825073/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"811f5842ce78023e33c624a33aacc3c3227c9bc5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-01-19 04:47:00.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/825073/3 : FAILURE in 1h 50m 28s","accounts_in_message":[],"_revision_number":3},{"id":"7086146f22b6400293387733dfd588d75fab2eb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 05:00:33.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/20a7d68082c949908ef77cfc6ed28767 : SUCCESS in 49m 24s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cffaffda8e62484f96ec524cf2ed2053 : SUCCESS in 58m 32s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6beeb7ad4d2a4635adeb47fdf611b81d : SUCCESS in 55m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/812268f216bc406994646f362e3f05ac : SUCCESS in 1h 00m 35s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5e6b6cd71ee944c9832a5dbc081f47af : SUCCESS in 55m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/60504c9e0a6d4bbe9da51b1ab978eb9f : SUCCESS in 29m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e6df9e5d8de485c85b8295f44411609 : FAILURE in 29m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbfde43c86964057a9eb98d70e80e394 : FAILURE in 29m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60a96d84b4a04002831c1e8299d713df : FAILURE in 9m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7850d6b13cb74401818122267a0bbe6c : FAILURE in 26m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d43724de3fe34ad38f8be2f242f8915c : FAILURE in 28m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/831c937592e440548adfa569d88fbe4d : SUCCESS in 11m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b565ac85d76a44339d4ff33641f43b8b : SUCCESS in 46m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f274dd87d74944649e7026eb77be1194 : SUCCESS in 1h 38m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/51d50a85f2e94236b9e8c6d3085f3bee : SUCCESS in 1h 24m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/974008bd9fc14adc9af62b9235152933 : SUCCESS in 1h 00m 43s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6475acc27ec8439190cb271378febca3 : SUCCESS in 49m 24s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f614307e699b430aad8e55cfca5c8c82 : SUCCESS in 1h 04m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/88b2b67f0475453d8a312a632c1160cd : SUCCESS in 1h 00m 03s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bc18d15c72f44c181038eb3bf1ac6c0 : SUCCESS in 1h 28m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d20a46a26f18477885fc2beb52950332 : SUCCESS in 1h 14m 27s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca6d420cd9e74e8d86c2079eb2e213c3 : FAILURE in 26m 30s","accounts_in_message":[],"_revision_number":3},{"id":"d789bd30d9926bf3ce9cfa1fbff2ec7020fb2392","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-01-19 07:35:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4e2eecef8e399bc715aff020a827fe98a883b016","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-19 08:28:40.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/73/825073/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"ef1d7ecc152de65e4c3cca6f7facd50d36d4afb1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 08:59:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/70da3d44f80d4767acf309738c2518f8 : TIMED_OUT in 1h 21m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b4696eb741ff4909b37dbc2c0fc8a3fe : FAILURE in 37m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e446cbc9f4f0bea4a4ade15b4e522a88d2deaa09","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-01-19 09:12:07.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/825073/4 : SUCCESS in 1h 36m 26s","accounts_in_message":[],"_revision_number":4},{"id":"fe5d8e17eb4a4d50cab22ba8256c477b6e875059","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 09:35:10.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c6d98f1a0b7e4797a6a26a5fe5217bb4 : SUCCESS in 44m 50s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/995c4a39cbf5481180578a84afd59e12 : SUCCESS in 1h 20m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/53a8f8877e79429eaed6a1c8e037f3fb : SUCCESS in 1h 11m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9e5b851b329f41f28d0d3e72d8eb8255 : SUCCESS in 1h 07m 21s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/369265eb80234d12a25ce615daa7ec7b : SUCCESS in 1h 28m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b80ce7f4dd1b4d5f8f7b41723a9a1785 : SUCCESS in 39m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b59ac71bfee74f81b8ed4be553da2812 : FAILURE in 36m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1b86ea266784d2680c34fdf651eb96e : FAILURE in 33m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80be0d54814f46cb8fc63e72285723fd : FAILURE in 7m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e0db71b98e348909510dd19ad87729d : FAILURE in 21m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21eeb77f8e6f4d6f9e0b777017d86930 : FAILURE in 24m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afbd0a2a9954461daca81128bcb5fc21 : SUCCESS in 8m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6d3854e3292446896e4591454dc4dfb : FAILURE in 50m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8c64112967e84f4290413e03b4704ab9 : SUCCESS in 1h 57m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e6df4498955b41f78455237648a88000 : SUCCESS in 1h 08m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dc4136801cf84abca45c63cab73d3f50 : SUCCESS in 58m 02s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b008b38cd0634ac8903efd6f75b1c7f7 : SUCCESS in 1h 10m 19s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/57baeb4d2f244ee4bb6afc7507b8b4a9 : SUCCESS in 1h 26m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1c0deb2c3a304f90ba045fc939c57dd5 : SUCCESS in 1h 11m 55s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca85feef0843460585770d3312e48f48 : SUCCESS in 59m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b359ea49d9424e888fe7e50ded9257b2 : SUCCESS in 1h 13m 20s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25bd19b6bc674f6a9d01b42739a73994 : FAILURE in 34m 14s","accounts_in_message":[],"_revision_number":4},{"id":"64b0367d9d329d23208d384b1db6e63ba1dd3e22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-01-19 10:38:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"60665fb7422971ea4b9237e1f94175304ae64216","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 11:21:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/68e8389a651046ea94640abbc6a03005 : FAILURE in 31m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/93fba81ee17d480f8b1bf7283b151533 : FAILURE in 35m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ecea2e707703dcaf7987465e1af96b3a6608c984","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 12:08:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/54765b5c0fa4469d847ed7dc6068305b : SUCCESS in 41m 28s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/08f03a4d6bc6479f8162701411fa02bc : SUCCESS in 54m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d6146144dc044efca8ad2c06a1f84c47 : SUCCESS in 52m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/48d519c2a68b4d2dbfc0742f6ea32dfe : SUCCESS in 50m 08s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/acd7806107cc46b79a6d384c85c86fdc : SUCCESS in 48m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1b531da126cb4e02a4999b474b267f2f : SUCCESS in 29m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b9a0010ce29484e838f1b870aa192be : FAILURE in 43m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/438be3c92fcc4fcbb7decad6bf474845 : FAILURE in 34m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09a38b33a52247259f6428e5d7c636fb : SUCCESS in 12m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed0525a7b0d84ad7bc73568956baa10b : FAILURE in 24m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cad3e5cce3624a46b32ba734de255565 : FAILURE in 25m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a73bd850647e49f3a6d7d5f8d89f12a4 : SUCCESS in 8m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/497b8e4b75084e9db924f5c1ba06410b : SUCCESS in 41m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cdb7ce2b650342bc9d287e53918ba82b : SUCCESS in 1h 13m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5f040f6852d444c59f38f164b7b607e9 : SUCCESS in 1h 02m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f9a882832fdc44f4b436e2bd8d95d029 : SUCCESS in 54m 09s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/acc109e9cebe4f778d8fcdeed42e7fcb : SUCCESS in 46m 24s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9f9fe368fb1247b18de8ebb350b853f8 : SUCCESS in 1h 26m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/864cbf36079a493aa8c6f4a9519c4f36 : SUCCESS in 1h 00m 51s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07a670954273410b944b1a956508678a : SUCCESS in 1h 10m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e35f6bcbd0834b4b9f65504629e8fa51 : SUCCESS in 1h 12m 42s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a75f0a3bdc0a4ac69791dd2977c7f5a0 : FAILURE in 26m 54s","accounts_in_message":[],"_revision_number":5},{"id":"c7d092aae245b54f74659e29fe38325e4186da38","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-01-19 12:33:11.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/825073/5 : SUCCESS in 1h 53m 41s","accounts_in_message":[],"_revision_number":5},{"id":"f70274b0a7e7ee4e329b33cd868138ead4f2068b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-19 14:17:41.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/73/825073/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"e77014be49f524b4623701dfbaf88445d39c32e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-01-20 05:35:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"96aaeb1402ca5a0819bf6a495246bf4e07ddbe58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 06:18:07.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/05e0a55e775d40f7be47fbed481264d7 : SUCCESS in 32m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/541b6af973534a0b810e3900f08db463 : SUCCESS in 40m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"32ae31328e50c5b5dc1260e59f1a62ab59034f2c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-20 06:32:06.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/73/825073/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"8f5e81a1ceca92d2f6da4eba7e83fee168e3fc75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 07:08:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/02627e0dd92245db83ab9455cad81040 : SUCCESS in 39m 14s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/184afd855fed48d592af9a73a52e7663 : SUCCESS in 55m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/41f803a033374d1bb22e741027bf632d : SUCCESS in 1h 11m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7208f58e17e7440ab620195b4a0fdb0b : SUCCESS in 45m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f1ccce42974a46c5991d1b0fb7fd3ada : SUCCESS in 1h 07m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0630d420959949918a6283e17d60384c : SUCCESS in 25m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7514dcfb0e304c8fa8a5f40f4951de02 : SUCCESS in 30m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58f4634b857b4cd0b8746f6bd60b190b : SUCCESS in 38m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/405000fe2e174176976e7bba0ffff539 : SUCCESS in 14m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5f8003d6e9e48ae889992d57c54cdf9 : SUCCESS in 43m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15c86ce83de949d7a341cb57692cadb2 : SUCCESS in 26m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6809430e8e914108a4496f2db8224aef : SUCCESS in 9m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c51151d07db2447394d14b043d159ea3 : SUCCESS in 55m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e4354e0387b4494e89c7b3607e1ef52c : SUCCESS in 1h 25m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/83f5e7a88113497293c9e666dd12ad85 : SUCCESS in 1h 13m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/da61dafa6c76407684effcf8d2703bb1 : SUCCESS in 47m 49s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1703a28dd4a9414a8c9403c24f3a1076 : SUCCESS in 1h 08m 01s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7d84106db17e4fe68469b83915f2e1ce : SUCCESS in 1h 06m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fd792b74730b4594835bb5e970581b6b : SUCCESS in 1h 25m 40s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a558536358bf45e18f078ca7997822b1 : SUCCESS in 50m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c111425375cb4cefb963b379bfb61e0a : SUCCESS in 1h 06m 32s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4828f6004f724d718e30a656697e46d0 : SUCCESS in 26m 43s","accounts_in_message":[],"_revision_number":6},{"id":"4fdb06ef45d4ab15463e03b9f0f34c1c679e17ee","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-01-20 07:23:37.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/825073/6 : SUCCESS in 1h 47m 14s","accounts_in_message":[],"_revision_number":6},{"id":"628aef0f447d03445b0d66215fdc793ac9514cb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-01 05:56:34.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4feb657061b6e8adadaff535e3c243277de07d0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 06:42:48.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e734b0627ad64b0bae0f5ce8452a0034 : FAILURE in 43m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bfd19b0e056f43c2a1747705c88c07b6 : FAILURE in 43m 56s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ba706ae60d58c3f0584d4ff1840b32f6baff5440","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-01 07:18:54.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/73/825073/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"212d132ff44c413df3dd9157feeb0c62d218f083","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 07:25:48.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3c98e0eb23ad4fc5a57f8b37a439e427 : SUCCESS in 44m 21s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3a43c0ed381642e69e9384a1d947cd62 : SUCCESS in 1h 14m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d3fbd95c0d8941d1adcbf7cfd62b3ae9 : SUCCESS in 1h 15m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c461d778c8ed4fbe8ccdee0bd66b53a9 : SUCCESS in 58m 14s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9162584f259b47d1ab3e81447087c660 : SUCCESS in 53m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4ad7db2ea9d14cfda74d1e7631fc2e8c : SUCCESS in 30m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b088f7e691a2417b9d38d8a9dbbbf380 : FAILURE in 35m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aabbd920a04145c3a434bac5dfb99cb2 : FAILURE in 32m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/971d780adfb9431fb69ab493df83c2f5 : SUCCESS in 15m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b83c989419924df5977e22305b09d9bd : FAILURE in 29m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c6faf28ca4a4d9693689d746b520598 : FAILURE in 24m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9c6f858220841f892807dd8f597b73b : SUCCESS in 11m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/754267d1e9ff482894eb5c71437d1297 : FAILURE in 53m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/964554f4a69743c19ce14f9c9b33bf7a : SUCCESS in 1h 26m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2041371063f24335b1fc54ffb51710ff : SUCCESS in 1h 12m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4e140c3b434a481490ccbcbd45a24936 : SUCCESS in 52m 41s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/97726ce48a094479867a7f739352c64a : SUCCESS in 1h 00m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8cc81e8c9598416589fd5a07bda7f374 : SUCCESS in 1h 15m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a7758cc058f643d3bae3f15399443211 : SUCCESS in 1h 15m 25s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65d9ce162f024538b55942e8ee194bb0 : SUCCESS in 52m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/aec6177ce5fd4e84884f6b8fcacb8c62 : SUCCESS in 57m 06s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3779b0160a24554fbfe91ee10c719a632a2064bb","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-01 07:42:40.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/825073/7 : SUCCESS in 1h 45m 24s","accounts_in_message":[],"_revision_number":7},{"id":"da17d43e82fcdb9b365f88d9520436cf86498529","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-01 12:47:36.000000000","message":"Patch Set 7:\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b3ba941c0d2d80da6dd9310d01116a17b38e5eaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-02 02:29:52.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ddfc0a7ad2fd8233b80acb071f77795fa3b9f88e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-02 03:18:48.000000000","message":"Patch Set 8:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/73/825073/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"036e4cea1e4f0947079d792b124db087d46066c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 03:31:23.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ac2f01fc31c433c9f42c0ca625aef73 : FAILURE in 55m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/47d4c2e002a940069eac25d6d3037fc5 : FAILURE in 56m 18s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"bf2f997b8fb6a0768fc5ba4bad886a534f2f7eb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 04:05:34.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0bc15e6f077b46ef8709897f48e1be9b : SUCCESS in 44m 52s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/dc5c37908cb64d8d816c7dd1a415aa24 : SUCCESS in 1h 00m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c6bf69b668d74ea4871e356b29b392fd : SUCCESS in 58m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/347be03f555046a59c5c556db0a2ed29 : SUCCESS in 50m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/93c40b4d245349a6b61c431e175decd7 : SUCCESS in 46m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d434696eb4f34f8ebb4477ac8a8f7e3e : SUCCESS in 31m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e13f9a525d264b46a69f6b8484c69fc4 : FAILURE in 31m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30afa49cb2224616a7b06997d39a6c97 : FAILURE in 30m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8e086a787564ee2a3a1bdc347998576 : SUCCESS in 16m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cc637744a73443aab3f6fecaad307ce : FAILURE in 25m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c689486999744a987e74a65712fe2ae : FAILURE in 24m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b5509c6b18f4814b4baf5d85c73a391 : SUCCESS in 11m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/62cc28e1cc22424d872fe85a98267adb : FAILURE in 39m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/35db2cb8d40648cbab6866cc7c2eebd8 : SUCCESS in 1h 19m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/25e0d719ccca421296d4f1dba6eac856 : SUCCESS in 57m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/db309a52b5be4d81a062c904993d6fb7 : SUCCESS in 44m 15s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/34ad5228f1a541b98e7ad5ae81c95e64 : SUCCESS in 56m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7c4aea818c8947588db61ceecf018c01 : SUCCESS in 1h 14m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d58f55e4839941c4a578861794b7f4b7 : SUCCESS in 1h 16m 58s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1c53344382d414e9d1ad396c893c12a : SUCCESS in 48m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/decdf40b47864366addb940c3f10e469 : SUCCESS in 1h 04m 14s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7dd3ff1bea3039920e0687c3e15e64540a6672c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-02 05:54:10.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2d4b72e7cea183ec96452c03bb9cd499eb4594b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 06:41:15.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/53bf590ed72d4095b53820b29a664693 : FAILURE in 43m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/12c22221e49f4b82afd449eeeea85030 : FAILURE in 45m 00s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c1709b2a6eb7fdf75ad58ae840adf12d71ef14dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 07:23:00.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/96c464de6b79473abedb9aa43a276e02 : SUCCESS in 38m 59s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2466dfa924eb47738b09db6e2da8fa39 : SUCCESS in 1h 16m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/83e0cd4c447140d886ad308c7c197f1f : SUCCESS in 1h 12m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4db1517191454107b11f7dd6cbf875a4 : SUCCESS in 46m 35s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1f9c6912b1484d95accf4393b149101a : SUCCESS in 1h 17m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ad1769670868498ab91dd063f7cd235b : SUCCESS in 40m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c14a79116d384829b855e6ac91e53d19 : FAILURE in 32m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa2b6027b2a34c5192b3fc49e3f226d6 : FAILURE in 35m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e93fb52560ef4780a12f67a2ea55260b : SUCCESS in 14m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8655cb3e4ee4f8c888a4a72faf0388c : FAILURE in 36m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1ffa5fe9f8a4298ae5c1161907e9ca0 : FAILURE in 24m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e0605916c0949c997b437b6b7fdb406 : SUCCESS in 10m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3ae422dc5ab54af5aef711bcb19ebc2d : FAILURE in 56m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/08af35c9f0ce420682227d44e4e80fd1 : SUCCESS in 1h 26m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e50f0be2bea0423ca9926068f84c54fd : SUCCESS in 1h 01m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9227f64ff4f04029abf265a8fe6d1f3f : SUCCESS in 59m 11s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/23e0a5e074524bf4b21f87df6218dc24 : SUCCESS in 51m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a64ded739f88450a9bbb77e5b514e132 : SUCCESS in 1h 06m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d1f3488377884b15b0aa24a2feb91b00 : SUCCESS in 1h 17m 01s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f7e6d95e4914e308f084a7a9ebbcfaf : SUCCESS in 51m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/feaf3d11a1c94b33ba849fab6c8dac6d : SUCCESS in 1h 21m 50s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"04579636d82ea4a2b59578dbf31f1a32cad09fb3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-02 07:32:00.000000000","message":"Patch Set 9:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/73/825073/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"e66265f631713a0b861246d5232ea8f80737ba91","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-02 07:42:57.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/825073/9 : FAILURE in 1h 48m 04s","accounts_in_message":[],"_revision_number":9},{"id":"236a35af4b3b13bf3a2d8a9d4ccfada6c7c3c205","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-02 08:45:48.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"721849be661de416b0a188d984a47efd0d8bdf74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 09:47:20.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b867cd2766540f29b288ab5d1b85bb8 : SUCCESS in 53m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/296768b1510a45f6bdc5f09b992af81b : SUCCESS in 52m 19s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"cdbe2e8ea2e199bb54393fb85fbcf2285bf545f6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-02 10:17:36.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/825073/10 : SUCCESS in 1h 31m 03s","accounts_in_message":[],"_revision_number":10},{"id":"5c2550dde2d113d116de9a542bb3861d0f90fbc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 10:32:32.000000000","message":"Patch Set 10: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6876e3bbfe20415bb6f699897c2d8db5 : SUCCESS in 46m 41s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2455b3411c494e009dc9a65304a713b3 : SUCCESS in 58m 52s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/996e9c6af1264bf2943c90a49e26c2be : SUCCESS in 1h 02m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/42d00485b230467cb6575312c7a55ef4 : SUCCESS in 48m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2477c8c806b14b4d9f699a42a01d40fd : SUCCESS in 48m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/502b1b12929e43c09b363ac1a17c69db : SUCCESS in 28m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdacb7e624314fc1856e00d77bc7a262 : SUCCESS in 45m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ecf34cc4f33451ca631d9f2a14371bf : SUCCESS in 34m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1a9d2ee8aea4f29b5d06b09b9b2ed71 : FAILURE in 9m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32bc478004e54138b8e7d160db0f3da3 : SUCCESS in 35m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/857ca31bdb2f4eda83d576ccde2db452 : SUCCESS in 36m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/690bfa6386df4f1a8a437e2b349459bc : SUCCESS in 10m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/055f5ae382ba49e1b4339f4924a5c321 : FAILURE in 50m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b71c1aabb7554c788fb71f4fbec3761e : SUCCESS in 1h 44m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1c3e55a1f834418ea75ca921ff9e9328 : SUCCESS in 1h 08m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ca5bf5b4f0c240159311f59fed408a62 : SUCCESS in 41m 35s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/48b62dbbb217429d9eeed8204f0e1725 : FAILURE in 1h 02m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a3f7df26af9a49f698d37d9603964735 : SUCCESS in 1h 24m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2b0d193bf628440483acf6b28cdf9309 : FAILURE in 1h 01m 41s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0758bbaec204ffc89c6eae615a60796 : SUCCESS in 43m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9e386aca304c4fbaace8bf3653b21d7d : SUCCESS in 1h 20m 59s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"827753c2912eed3144182198530cbd631b3353fb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-02 10:38:33.000000000","message":"Patch Set 10:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/73/825073/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"24104b0b5509a2bb41d663425ef8cf4c3523bac3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-02 11:16:56.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"e32580c72785368ccabb1c17fb8a954a3c3a67a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 12:04:43.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2629d304b6a4c1c9b6531dbf044af9c : SUCCESS in 45m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5ef37849abc14944b269557e31b8bda2 : SUCCESS in 44m 55s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ec4bacbd8e1fa5eb60a5164760acfc20c7e17b5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 12:42:20.000000000","message":"Patch Set 11: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d6bcd93d413a4498a86e95ecae5b8276 : SUCCESS in 50m 42s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b076e64d108e4eecaf315e1562b2589a : SUCCESS in 1h 14m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5d727947316a44da8904b8cc6e34862d : SUCCESS in 1h 03m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2250cafb84f74c77976a1c696a90663f : SUCCESS in 58m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6faa70d8a28e47eb91972aadbfeb829e : SUCCESS in 1h 03m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/08608c2a3f39490fa9124c4fe5b55ef0 : SUCCESS in 31m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3ae183accd14f7db1754b6298955452 : SUCCESS in 35m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c13f90d9218444faebc6e432f9b5782 : SUCCESS in 28m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/114fc7b8c005480ab151c5cb91a27c65 : FAILURE in 9m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a527951e1fb4a0bbad52f7b9fb0d26b : SUCCESS in 26m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef70f64d54fb4f48b387efce4b1be295 : SUCCESS in 25m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6354a1d244541838706b98ded7b7541 : SUCCESS in 11m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/40f6c6e4cd0f4748a3457765c6234f12 : SUCCESS in 42m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/50e87115b47b451f8e4294755c7a367c : SUCCESS in 1h 22m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c0a8867c53ec4ac1820d21616ae53ea7 : SUCCESS in 1h 08m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0edbd29673564faeaf835054aa121d70 : SUCCESS in 1h 00m 32s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/82ae4750675e484abc308f5215ed1d2b : SUCCESS in 54m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fbb7ab89e58d4ba6bbe49f53876b41d8 : SUCCESS in 1h 11m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f9c41f04c9cc4a1fa1bc457548378e7d : SUCCESS in 1h 12m 41s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a051d5d7efe4fac8bc0fcd816d2913a : SUCCESS in 55m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fabd4a50b8d2474abba225f1a00ab318 : SUCCESS in 1h 05m 28s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a46e897a9a39239212b5fb577b050460ca210a66","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-02 13:06:11.000000000","message":"Patch Set 11:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/73/825073/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"0b8e189edaf9a75fb12feb0306743b5e0a8fb154","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-02 13:32:35.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/825073/11 : SUCCESS in 2h 14m 54s","accounts_in_message":[],"_revision_number":11},{"id":"31f643924a6a1ffa716b73c01bdb3ef08213a443","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-03 00:30:19.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"271969565abbbb2e4615df2437e1e451f91364fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 01:13:53.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e03d575ff14845938e6729eef569b57e : SUCCESS in 40m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/48caf06d636c47758cdd6fa334492325 : SUCCESS in 41m 40s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"466fbaa274c8d2c45a734ec83c101cc24642a376","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-03 01:34:45.000000000","message":"Patch Set 12:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/73/825073/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"76267d4774d625373682d8d5a147898ba2ed65d2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-03 02:23:39.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/825073/12 : SUCCESS in 1h 52m 36s","accounts_in_message":[],"_revision_number":12},{"id":"41ec2c2c94bddc4e3cba5c014de9430ba4e258c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 02:34:18.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fe36711c1f8745eaaf87c391eb0f17dd : SUCCESS in 51m 48s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c53dc51f1ab8419f9b86d299e6c39cc1 : SUCCESS in 1h 11m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/182b14f40ab049278a77fdc72944c8b6 : SUCCESS in 1h 07m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/34a5cf8eb3ce4d3091f37ea00a382f0f : SUCCESS in 1h 05m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c1520e8e41d14f5da2723e582feb3abf : SUCCESS in 1h 07m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c1d65ab589cb426880abef87505098c2 : SUCCESS in 33m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d94aad6e44e4e009d26d0f478bcf7f0 : SUCCESS in 40m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3200e36321fe473c8384adfa949797d1 : SUCCESS in 49m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14534ed9223b43d79fe4ff5227995f7f : SUCCESS in 12m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46dcb965bd8f4a6ab02ada407b548e3a : SUCCESS in 43m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/245f4064d52c45fea396038e6a6b358d : SUCCESS in 21m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba4f0ec7d51c45d583a4718fd34122e9 : SUCCESS in 10m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/721249c7a2dc48ccbf6783ff7f84fb5e : FAILURE in 57m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ea852f2ef2174e94a28ab5c98a173098 : SUCCESS in 2h 01m 50s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6f8575da843544b3b8e876e1dce2bc2f : SUCCESS in 1h 22m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6498258a2b7f48578224986eb8df0344 : SUCCESS in 1h 00m 38s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bfb8b2ae48b24f3bb2f5cff66df58fbf : SUCCESS in 51m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/84097016765e490a901d493e80109cc9 : SUCCESS in 1h 09m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/232cb22c5b72421e98d5986129fd16fa : SUCCESS in 1h 06m 26s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d7b2b16c69747818e13a6d0b24ae80f : SUCCESS in 1h 09m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3fa93230a23d4d559126cdbaff27630d : SUCCESS in 1h 05m 40s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"94c5d8889231af371525461226a72b52a502308d","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-03 02:36:21.000000000","message":"Patch Set 12:\n\n(6 comments)","accounts_in_message":[],"_revision_number":12},{"id":"5f8453c784de0f8198f03deb425825a64abcb8b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 04:07:08.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5413a99011b94cb7b3fb694c06b4ffd0 : SUCCESS in 42m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4977fd5e9a624bbcad769995cd821e28 : SUCCESS in 1h 13m 49s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"942bb62ce2e9f4319553b199ce0bb74e943cebdd","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-03 04:25:01.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/825073/12 : SUCCESS in 1h 48m 08s","accounts_in_message":[],"_revision_number":12},{"id":"ab3075d48afc2f0bb871aae1bb669ee08f5d4f07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 04:38:43.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/913e8b7751e9481c97fced21dee099df : SUCCESS in 50m 07s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9e4ff7eb540e432b961674d6d14cb389 : SUCCESS in 1h 01m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/093b434c954142f492356b447592aa2e : SUCCESS in 1h 05m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e8ec0ace944a4c6793296d61ebb665ac : SUCCESS in 51m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0352b04bc70f478cbe93dafb11f1e230 : SUCCESS in 53m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/665b37aa7dcf429f838be27b4fd3e711 : SUCCESS in 32m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9adac8d391e34eefb79952fa4684ca4f : SUCCESS in 42m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3f78f22677444019bb81807a89ef9f4 : SUCCESS in 37m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/243e77d489af492683d33c9f578f3e5a : SUCCESS in 16m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f89129f38244868a10a3ef5190f9b40 : SUCCESS in 29m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3299e4fdabf44244883d4e31fbe27883 : SUCCESS in 25m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/254a0a5a8ec2494091984f7590f6e519 : SUCCESS in 11m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ba5835884fbd4bceb51eff2074a297a1 : FAILURE in 53m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/245e48abc7f74f13a53ba9d9fad5bfb2 : SUCCESS in 1h 44m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/18905df743db492390fa842c8babb924 : SUCCESS in 1h 23m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7e7495ffe4394c03a8639d489e09bbd9 : SUCCESS in 45m 25s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e66f36e79e7d4665bebd589b4fc6a594 : SUCCESS in 59m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/964685b0896c45f1a3421bd18cc116a1 : SUCCESS in 1h 11m 57s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1a81ffe5a58042cc90cd460fbf1a9f95 : SUCCESS in 1h 16m 37s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f95ea663c693443dbd3d37548a05156a : SUCCESS in 46m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6d5f8b2cb22b48ec830907a8837ee297 : SUCCESS in 1h 24m 43s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"87b580dc0d67c8f7cfbbe3af00d3fbad2fec5a96","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-03 05:39:35.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"86fef554245ee5ea1e9db91284e59c7df60466e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 06:54:29.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/12e19af205554ea39b6a8d8f74bedd0b : SUCCESS in 57m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/73c4dfb668054634aa1ceb30676d5a25 : SUCCESS in 56m 09s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"917aa409ad60c990e3596ede693dc70278eef28a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-03 07:12:17.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/825073/12 : SUCCESS in 1h 32m 08s","accounts_in_message":[],"_revision_number":12},{"id":"39b1868d5bf2871a0753b55e6c693e85fd62f0ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 07:17:31.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7f4b927e9f5c4e66ab174264504ccb69 : SUCCESS in 56m 19s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7655cf8def3e4d20bd9ce2289f481bad : SUCCESS in 1h 05m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/64c5bdf127a048a99b2129f684daa521 : SUCCESS in 1h 22m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/aef0684349914796bd3efdea3958bcb7 : SUCCESS in 58m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7d660778a2714560b331aaec7d7d5ec9 : SUCCESS in 1h 03m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1d30ea53df89424b92e7fbe9e2d9746a : SUCCESS in 26m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a139d3447652499c8a06cab666e059c4 : SUCCESS in 43m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b528d852f8804747aed772505367b9cc : SUCCESS in 38m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/904ebe990eb3484fb22f9c297f02d8c4 : SUCCESS in 18m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13ab2269337e4c8f800f2ec5d192d969 : SUCCESS in 29m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1edd22fa59dd4b98ba5ca9fc4d3e4488 : SUCCESS in 33m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65665b0fb33a4602bdeeeaab5fb9f5be : SUCCESS in 10m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/78a381d88bdc477baeedc1e36995e485 : SUCCESS in 58m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d81c331dfff447dd8abe96de0d37f106 : SUCCESS in 1h 32m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/dd33dc5605ea470e8ce33371863b3f0c : SUCCESS in 1h 15m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2aca718993a74911a6ffed0837c9e162 : SUCCESS in 59m 18s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5f606290d9114a1592c357b989c0536f : SUCCESS in 1h 03m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/986380f704b44e8c96bb05d3dcf04d50 : SUCCESS in 1h 25m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8ad5c776b5384aefb219c14239e2d254 : SUCCESS in 1h 22m 15s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04f55072138043f694741870964fd94e : SUCCESS in 57m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/344ad157a88840c4bba87f5134dc77f9 : SUCCESS in 1h 10m 49s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"502f94185ed62ed3541913095cd11814ffb0eec2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-07 02:46:41.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"7d2d561126138f89bf9de8c18d4ba567522f84aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 04:24:06.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d094071a80204be79f07ef11dfb78ada : SUCCESS in 1h 07m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/547fabc70ef34af99f16661f89aec0a4 : SUCCESS in 1h 03m 52s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"3ecfbbe61dfdfa77b11a84754d37102ef1bd2c4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 04:24:21.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fdd70972c7de44e0b2d6830e470a99c1 : SUCCESS in 46m 43s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/59870e40ab2f4d58beca237e7b6c5dd9 : SUCCESS in 1h 02m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8e425c1fd90344f4a4a5bcd6a01396dd : SUCCESS in 56m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c6483548a02a4598828829a16a2b327f : SUCCESS in 54m 17s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7209cdacb696464fbbd77bee893057f9 : SUCCESS in 55m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/21a877b1b05a458790bc4fd29213f4a4 : SUCCESS in 34m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13edf9d698714edbbfc2d8728606c4dc : SUCCESS in 38m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07b56d97107045f2839572f7707814ab : SUCCESS in 36m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/338206cf372841f99471842b938e61be : SUCCESS in 16m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3dd5a60e645a4664a600b1df94831108 : SUCCESS in 24m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28ecdf4b4e144a1d84897ad5c2a1cce1 : SUCCESS in 28m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de3c85aaf1134194b90347d97c4e8cf4 : SUCCESS in 13m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/967964db3ffa4cc8a9148940d2c33c60 : SUCCESS in 49m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/619027a855bf4354a6c3bd6c2d33eb2d : SUCCESS in 1h 29m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/12ed8f4fd68b4169b940cb0be2cbcf5b : SUCCESS in 1h 05m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9e170c2a619746e18684e7c6160677ef : SUCCESS in 42m 18s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/40a7e6511b3f4bf0a71ad23c7481c45b : SUCCESS in 45m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a58c67743d234a60b74c4b21b5c34792 : SUCCESS in 1h 18m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7762bc07338a460ea8cb0bf40d9f87d2 : SUCCESS in 1h 14m 25s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56a697698e7c4b57bc085cd922ed3b03 : SUCCESS in 1h 13m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e081cdef143e4203816babc011cabd32 : SUCCESS in 1h 14m 33s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"75bd3ed105107ca3243f6a3dc590fe49a054ccf7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-07 04:38:25.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/825073/13 : SUCCESS in 1h 51m 00s","accounts_in_message":[],"_revision_number":13},{"id":"e92eefe307b7ccf4a3fc01b1c5daf566eac54e36","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-08 04:35:05.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/73/825073/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"d37ac1a7e9ed94578a3bf8ec571c2a7ea75c64ca","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-12 12:11:27.000000000","message":"Patch Set 13: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"e246a53a07c97bab88fc9dc097b2c8b1b8a6abf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-14 00:33:23.000000000","message":"Patch Set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"ed1e7e0da1510fd1cc6e089f0d0a8f8dd9183e32","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-14 00:34:35.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"382e68b91a64358eab813e59fb4a5573de92ff1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 01:19:13.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b35f85c52564a729269d29db5a01c32 : SUCCESS in 44m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/21a11384dd4b43a9b3485821a8d88426 : SUCCESS in 43m 58s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"3f65bdaff995ad4d777657e87e10f8c266966069","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-14 01:19:32.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/73/825073/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"9487825271a31cab991445d242a3550e2d4c1933","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 02:10:54.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6027b3686dee4003abb9d3386f0760a4 : SUCCESS in 45m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/98d07baf20f9476b990705ece5803f94 : POST_FAILURE in 1h 18m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/eb5a32a58d474303a6ea60d084e92d9a : SUCCESS in 1h 14m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eba74aa0ecfb43ae88ac3aae6355a5d7 : SUCCESS in 1h 07m 40s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/43a3d0228f9b4ace87d54171101f9810 : SUCCESS in 1h 20m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cca0b3cec67a44e4abe73f1a759fb559 : SUCCESS in 31m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbf3ea80a8a741c6bc8552c2863c6921 : SUCCESS in 30m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c235ab468d74258b7e67fafc2c443cc : SUCCESS in 27m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8e7673a296a43c3a40dd6f5aa355967 : SUCCESS in 12m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88a73432a6f94e029b2fd34438448667 : SUCCESS in 37m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfede0c547f94dc286f2f95039343806 : SUCCESS in 38m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80954b84cd2f4d198844d3a082de7d0a : RETRY_LIMIT in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04acdc149d5a40718f6f6ecc1ec74d98 : SUCCESS in 9m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e030b57dba644dc09594236796ec1608 : SUCCESS in 39m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/19c730e821d04e4b80fbf65b018cfd23 : SUCCESS in 59m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a17a596cfbca43ac9bb8b1dcab985346 : SUCCESS in 1h 34m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/93f9ccb733154f7eb38cfe5922db5f2b : SUCCESS in 56m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1329a3773442476f8ea6e5fb6f35ea87 : SUCCESS in 41m 15s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7fbef67d342440ad9929665c3efd6485 : SUCCESS in 42m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/69165018b3de44d3a0d2bfdd89484794 : SUCCESS in 55m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e813fe7aa65542fb9668ec9011192b3f : SUCCESS in 52m 27s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fee152f1aa45428d801c283fbbb9e513 : SUCCESS in 54m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b21fb1eade24484bb10e9ccb73d9eabb : SUCCESS in 1h 00m 18s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"cd691f90c411450b518ae65b97ef89ed83bc9bac","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-03-14 02:24:42.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"8058a679994f416312dfbe0815e6401c78995125","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-14 02:26:26.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/825073/14 : SUCCESS in 1h 52m 18s","accounts_in_message":[],"_revision_number":14},{"id":"3cc876feb42e7ac8c490960b9dff9fa88c01e65b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 04:20:24.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e6436277f66f4f60be7f9606a39c880e : FAILURE in 1h 20m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c1e742f27d9a4ec48df50f81dcf1c5ba : SUCCESS in 56m 40s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"f154dc6b8724378a1410166ef66556398b251c2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 04:41:36.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4de63ae190cd4403a7f3057a771771d2 : SUCCESS in 51m 57s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/497b2b5a805244979fa1d4ce1b96d303 : SUCCESS in 59m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4bc0573d308a49f9b3327a755ceee54f : SUCCESS in 1h 04m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fdcf438adacd46cdb30c7e5f2c9fbbdd : SUCCESS in 53m 37s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/43079b3bd8af4645b29f31ac88c2ddb7 : SUCCESS in 58m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4abd06bd65e7417ea6df4e709cf909eb : SUCCESS in 34m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/852ef8eed68b49d59b358fc9cfa87faa : SUCCESS in 39m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70c11e749111418597716831af597c30 : SUCCESS in 38m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9cb19fd34ed48efa4d1b0820ec8fb4f : SUCCESS in 16m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d30c9d0db0242008ae35b7e3acfdc3b : SUCCESS in 31m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efc94ab5b79e4d58951e1b5ec66185a7 : SUCCESS in 29m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6acbedcd07a14d649a18667153e416df : RETRY_LIMIT in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b6e2ad13e1540839e1e00f6cd3a2efa : SUCCESS in 11m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbfb46d4ba604ff98a8da957b8392989 : SUCCESS in 30m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c9a2597681f843b889c44823bff7fcee : SUCCESS in 52m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/573568cec1d14fa2b577f68e6a8cac6d : SUCCESS in 2h 07m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0f9204bf652e46c6a6f9e1c8e6212f5a : SUCCESS in 1h 08m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6245f25baa7246c6a72c3b095b7c3d21 : SUCCESS in 54m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f67aca985b83499fa0af31430014621c : SUCCESS in 1h 00m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f3cdb7b7a65346cf84c3021976f61c42 : SUCCESS in 1h 17m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/41034d54abc44a6dacf352acad4410a0 : SUCCESS in 1h 21m 54s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4eff448524ac43c88a5f5d252f51a2d7 : SUCCESS in 58m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b87cc9116a804409974ca601bdc03436 : SUCCESS in 1h 12m 49s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"c03ba08ead493e877f6175c8e7c9e46d7ea77faf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-16 09:22:26.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"f062d9a6529a38bc4739c87c4d33e22c26f81c02","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-03-16 10:41:14.000000000","message":"Patch Set 14: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"8eb09119ad9b94104b83b0a0f54d92eaec98540a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-25 15:49:52.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"cc95f2747b839efe09212e199c3b21e1beb049c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 15:51:58.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"9e1737ac3ddf015850468fbb730cd1586f937411","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 18:50:54.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/af7fd548c08541eab4746ef390f79f65 : SUCCESS in 52m 54s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6bd1e6f983084e40bcc021ae19976bca : SUCCESS in 1h 32m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/408cf006358f4234ba59fef1f07116a7 : SUCCESS in 1h 06m 12s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/dd2a5a758eec49559c79c7ce1c1420e5 : SUCCESS in 1h 05m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f0e14522fa8d4a1d87fa308ff60b54a5 : SUCCESS in 1h 03m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c70c6d8be1e44bf94bd623075fe96e4 : SUCCESS in 36m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6831f0c64d3454391f3782987afa46a : SUCCESS in 18m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d537933afed4909bc3aef2617fae1fc : SUCCESS in 38m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6760c5f55a2746abbe7c68eea3b7fb1c : SUCCESS in 28m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a99ff92163004a4c9cb65daef203a694 : SUCCESS in 11m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4e6f167d4d504b15b8a1614126fbca32 : SUCCESS in 58m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ea69b1024bff4e57ae2099edbd71adcb : SUCCESS in 1h 26m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1b9a9c8210e548c1b96a4dfd8ec698bb : SUCCESS in 1h 30m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c005f34aefed47759abaef962a3149fe : SUCCESS in 1h 00m 12s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b0a77119df3e4c6bb2413e03f27640b7 : SUCCESS in 58m 07s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0be988fe62ae4078b53c79cd9f678c51 : SUCCESS in 1h 03m 54s","accounts_in_message":[],"_revision_number":14},{"id":"d82aa941f8429dd2740a2cae820f1dfe61657071","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 18:50:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"20ca79e4c8735a9d1f652902d1efde8b4bb6fe7b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 18:59:57.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09ca861221c24c6db8b4f6b33f660729 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"b5253b224bcf6cec428f18a831798d0955c61797","revisions":{"8675e57de2df514d5a4da589b27857f302acfa0d":{"kind":"REWORK","_number":1,"created":"2022-01-18 09:27:09.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/73/825073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/825073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/825073/1"}}},"commit":{"parents":[{"commit":"a9730576b1d86e995ce4cf0db165a9830e2ba800","subject":"Merge \"Exclude router gw subnet ports from port list while port delete.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a9730576b1d86e995ce4cf0db165a9830e2ba800"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-01-18 09:25:06.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-01-18 09:25:25.000000000","tz":480},"subject":"[ovn]Reschduler router gateway chassis when chassis\u0027s available zone updated","message":"[ovn]Reschduler router gateway chassis when chassis\u0027s available zone updated\n\nwhen chassis\u0027s available zone updated, triger rescheduler router gateway\nchassis.\n\nCloses-bug: #1958225\n\nChange-Id: If7cf55f2c12b388fc34fa942b6b0bf70cb9eb866\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8675e57de2df514d5a4da589b27857f302acfa0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8675e57de2df514d5a4da589b27857f302acfa0d"}]},"branch":"refs/heads/master"},"156444a22e36ab5d71ce267f716d6a1e54e4aef5":{"kind":"REWORK","_number":2,"created":"2022-01-18 11:29:47.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/73/825073/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/825073/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/825073/2"}}},"commit":{"parents":[{"commit":"a9730576b1d86e995ce4cf0db165a9830e2ba800","subject":"Merge \"Exclude router gw subnet ports from port list while port delete.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a9730576b1d86e995ce4cf0db165a9830e2ba800"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-01-18 09:25:06.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-01-18 11:29:23.000000000","tz":480},"subject":"[WIP][ovn]Reschduler router gateway chassis when chassis\u0027s available zone updated","message":"[WIP][ovn]Reschduler router gateway chassis when chassis\u0027s available zone updated\n\nwhen chassis\u0027s available zone updated, triger rescheduler router gateway\nchassis.\n\nCloses-bug: #1958225\n\nChange-Id: If7cf55f2c12b388fc34fa942b6b0bf70cb9eb866\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/156444a22e36ab5d71ce267f716d6a1e54e4aef5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/156444a22e36ab5d71ce267f716d6a1e54e4aef5"}]},"branch":"refs/heads/master"},"f107f55e04f853b3a96ff898b27f63eef9b6abe6":{"kind":"REWORK","_number":3,"created":"2022-01-19 02:55:50.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/73/825073/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/825073/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/825073/3"}}},"commit":{"parents":[{"commit":"a9730576b1d86e995ce4cf0db165a9830e2ba800","subject":"Merge \"Exclude router gw subnet ports from port list while port delete.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a9730576b1d86e995ce4cf0db165a9830e2ba800"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-01-18 09:25:06.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-01-19 02:55:23.000000000","tz":480},"subject":"[WIP][ovn]Reschduler router gateway chassis when chassis\u0027s available zone updated","message":"[WIP][ovn]Reschduler router gateway chassis when chassis\u0027s available zone updated\n\nwhen chassis\u0027s available zone updated, triger rescheduler router gateway\nchassis.\n\nCloses-bug: #1958225\n\nChange-Id: If7cf55f2c12b388fc34fa942b6b0bf70cb9eb866\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f107f55e04f853b3a96ff898b27f63eef9b6abe6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f107f55e04f853b3a96ff898b27f63eef9b6abe6"}]},"branch":"refs/heads/master"},"d3f1cca3b71e7ba09985e5d5c818a7a93dd09104":{"kind":"REWORK","_number":4,"created":"2022-01-19 07:35:01.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/73/825073/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/825073/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/825073/4"}}},"commit":{"parents":[{"commit":"305bab4f2393560dde781279183d1d7ec10967f3","subject":"Merge \"Ensure subports transition to DOWN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/305bab4f2393560dde781279183d1d7ec10967f3"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-01-18 09:25:06.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-01-19 07:34:54.000000000","tz":480},"subject":"[WIP][ovn]Reschduler router gateway chassis when chassis\u0027s available zone updated","message":"[WIP][ovn]Reschduler router gateway chassis when chassis\u0027s available zone updated\n\nwhen chassis\u0027s available zone updated, triger rescheduler router gateway\nchassis.\n\nCloses-bug: #1958225\n\nChange-Id: If7cf55f2c12b388fc34fa942b6b0bf70cb9eb866\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d3f1cca3b71e7ba09985e5d5c818a7a93dd09104"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d3f1cca3b71e7ba09985e5d5c818a7a93dd09104"}]},"branch":"refs/heads/master"},"30b5e3366e1eb0a226a448a94b2e340c5c0c9c5b":{"kind":"REWORK","_number":5,"created":"2022-01-19 10:38:40.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/73/825073/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/825073/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/825073/5"}}},"commit":{"parents":[{"commit":"305bab4f2393560dde781279183d1d7ec10967f3","subject":"Merge \"Ensure subports transition to DOWN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/305bab4f2393560dde781279183d1d7ec10967f3"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-01-18 09:25:06.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-01-19 10:38:17.000000000","tz":480},"subject":"[WIP][ovn]Reschduler router gateway chassis when chassis\u0027s available zone updated","message":"[WIP][ovn]Reschduler router gateway chassis when chassis\u0027s available zone updated\n\nwhen chassis\u0027s available zone updated, triger rescheduler router gateway\nchassis.\n\nCloses-bug: #1958225\n\nChange-Id: If7cf55f2c12b388fc34fa942b6b0bf70cb9eb866\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/30b5e3366e1eb0a226a448a94b2e340c5c0c9c5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/30b5e3366e1eb0a226a448a94b2e340c5c0c9c5b"}]},"branch":"refs/heads/master"},"599f4d45fa60a79f266f1b3cd25ca9c64d4c9801":{"kind":"REWORK","_number":6,"created":"2022-01-20 05:35:42.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/73/825073/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/825073/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/825073/6"}}},"commit":{"parents":[{"commit":"305bab4f2393560dde781279183d1d7ec10967f3","subject":"Merge \"Ensure subports transition to DOWN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/305bab4f2393560dde781279183d1d7ec10967f3"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-01-18 09:25:06.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-01-20 02:21:31.000000000","tz":480},"subject":"[ovn]Reschduler router gateway chassis when chassis\u0027s available zone updated","message":"[ovn]Reschduler router gateway chassis when chassis\u0027s available zone updated\n\nwhen chassis\u0027s available zone updated, triger rescheduler router gateway\nchassis.\n\nCloses-bug: #1958225\n\nChange-Id: If7cf55f2c12b388fc34fa942b6b0bf70cb9eb866\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/599f4d45fa60a79f266f1b3cd25ca9c64d4c9801"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/599f4d45fa60a79f266f1b3cd25ca9c64d4c9801"}]},"branch":"refs/heads/master"},"e72e5bb084b4a388aa57ed1396550f919c97ece6":{"kind":"REWORK","_number":7,"created":"2022-03-01 05:56:34.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/73/825073/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/825073/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/825073/7"}}},"commit":{"parents":[{"commit":"3f0b82f84aa0c5636230692cf7efa1867a6b73c2","subject":"Merge \"Use the \"connectivity\" property of \"MechanismDriver\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f0b82f84aa0c5636230692cf7efa1867a6b73c2"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-01-18 09:25:06.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-03-01 05:54:46.000000000","tz":480},"subject":"[ovn]Reschduler router gateway chassis when chassis\u0027s available zone updated","message":"[ovn]Reschduler router gateway chassis when chassis\u0027s available zone updated\n\nwhen chassis\u0027s available zone updated, triger rescheduler router gateway\nchassis.\n\nCloses-bug: #1958225\n\nChange-Id: If7cf55f2c12b388fc34fa942b6b0bf70cb9eb866\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e72e5bb084b4a388aa57ed1396550f919c97ece6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e72e5bb084b4a388aa57ed1396550f919c97ece6"}]},"branch":"refs/heads/master"},"813fcdbc67d57a7ecf3bf2909fb9d0abc2404d09":{"kind":"REWORK","_number":8,"created":"2022-03-02 02:29:52.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/73/825073/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/825073/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/825073/8"}}},"commit":{"parents":[{"commit":"3f0b82f84aa0c5636230692cf7efa1867a6b73c2","subject":"Merge \"Use the \"connectivity\" property of \"MechanismDriver\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f0b82f84aa0c5636230692cf7efa1867a6b73c2"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-01-18 09:25:06.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-03-02 02:29:14.000000000","tz":480},"subject":"[ovn]Rescheduler router gateway chassis","message":"[ovn]Rescheduler router gateway chassis\n\nwhen chassis\u0027s available zone updated, triger rescheduler router\ngateway chassis.\n\nCloses-bug: #1958225\n\nChange-Id: If7cf55f2c12b388fc34fa942b6b0bf70cb9eb866\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/813fcdbc67d57a7ecf3bf2909fb9d0abc2404d09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/813fcdbc67d57a7ecf3bf2909fb9d0abc2404d09"}]},"branch":"refs/heads/master"},"47a12cff1e1bc821a60020ccd341cd17139726ef":{"kind":"REWORK","_number":9,"created":"2022-03-02 05:54:10.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/73/825073/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/825073/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/825073/9"}}},"commit":{"parents":[{"commit":"3f0b82f84aa0c5636230692cf7efa1867a6b73c2","subject":"Merge \"Use the \"connectivity\" property of \"MechanismDriver\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f0b82f84aa0c5636230692cf7efa1867a6b73c2"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-01-18 09:25:06.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-03-02 05:53:26.000000000","tz":480},"subject":"[ovn]Rescheduler router gateway chassis","message":"[ovn]Rescheduler router gateway chassis\n\nwhen chassis\u0027s available zone updated, triger rescheduler router\ngateway chassis.\n\nCloses-bug: #1958225\n\nChange-Id: If7cf55f2c12b388fc34fa942b6b0bf70cb9eb866\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/47a12cff1e1bc821a60020ccd341cd17139726ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/47a12cff1e1bc821a60020ccd341cd17139726ef"}]},"branch":"refs/heads/master"},"84e607b7a9c2ce21fbc128b4151b0f4a290d4668":{"kind":"REWORK","_number":10,"created":"2022-03-02 08:45:48.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/73/825073/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/825073/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/825073/10"}}},"commit":{"parents":[{"commit":"3f0b82f84aa0c5636230692cf7efa1867a6b73c2","subject":"Merge \"Use the \"connectivity\" property of \"MechanismDriver\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f0b82f84aa0c5636230692cf7efa1867a6b73c2"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-01-18 09:25:06.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-03-02 08:45:01.000000000","tz":480},"subject":"[ovn]Rescheduler router gateway chassis","message":"[ovn]Rescheduler router gateway chassis\n\nwhen chassis\u0027s available zone updated, triger rescheduler router\ngateway chassis.\n\nCloses-bug: #1958225\n\nChange-Id: If7cf55f2c12b388fc34fa942b6b0bf70cb9eb866\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84e607b7a9c2ce21fbc128b4151b0f4a290d4668"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84e607b7a9c2ce21fbc128b4151b0f4a290d4668"}]},"branch":"refs/heads/master"},"016578a9264fe2070f0aa2c0647b5d3b43ddbb6f":{"kind":"REWORK","_number":11,"created":"2022-03-02 11:16:56.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/73/825073/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/825073/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/825073/11"}}},"commit":{"parents":[{"commit":"3f0b82f84aa0c5636230692cf7efa1867a6b73c2","subject":"Merge \"Use the \"connectivity\" property of \"MechanismDriver\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f0b82f84aa0c5636230692cf7efa1867a6b73c2"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-01-18 09:25:06.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-03-02 11:16:09.000000000","tz":480},"subject":"[ovn]Rescheduler router gateway chassis","message":"[ovn]Rescheduler router gateway chassis\n\nwhen chassis\u0027s available zone updated, triger rescheduler router\ngateway chassis.\n\nCloses-bug: #1958225\n\nChange-Id: If7cf55f2c12b388fc34fa942b6b0bf70cb9eb866\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/016578a9264fe2070f0aa2c0647b5d3b43ddbb6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/016578a9264fe2070f0aa2c0647b5d3b43ddbb6f"}]},"branch":"refs/heads/master"},"bb9e5dc993ac951697c4eca3dd36c5889d22af14":{"kind":"REWORK","_number":12,"created":"2022-03-03 00:30:19.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/73/825073/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/825073/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/825073/12"}}},"commit":{"parents":[{"commit":"3f0b82f84aa0c5636230692cf7efa1867a6b73c2","subject":"Merge \"Use the \"connectivity\" property of \"MechanismDriver\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f0b82f84aa0c5636230692cf7efa1867a6b73c2"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-01-18 09:25:06.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-03-03 00:29:55.000000000","tz":480},"subject":"[ovn]Rescheduler router gateway chassis","message":"[ovn]Rescheduler router gateway chassis\n\nwhen chassis\u0027s available zone updated, triger rescheduler router\ngateway chassis.\n\nCloses-bug: #1958225\n\nChange-Id: If7cf55f2c12b388fc34fa942b6b0bf70cb9eb866\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bb9e5dc993ac951697c4eca3dd36c5889d22af14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bb9e5dc993ac951697c4eca3dd36c5889d22af14"}]},"branch":"refs/heads/master"},"1dcfa4363f868eaec0687cc07625788d16b63bea":{"kind":"REWORK","_number":13,"created":"2022-03-07 02:46:41.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/73/825073/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/825073/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/825073/13"}}},"commit":{"parents":[{"commit":"3f0b82f84aa0c5636230692cf7efa1867a6b73c2","subject":"Merge \"Use the \"connectivity\" property of \"MechanismDriver\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f0b82f84aa0c5636230692cf7efa1867a6b73c2"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-01-18 09:25:06.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-03-07 02:46:16.000000000","tz":480},"subject":"[ovn]Rescheduler router gateway chassis","message":"[ovn]Rescheduler router gateway chassis\n\nwhen chassis\u0027s available zone updated, triger rescheduler router\ngateway chassis.\n\nCloses-bug: #1958225\n\nChange-Id: If7cf55f2c12b388fc34fa942b6b0bf70cb9eb866\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1dcfa4363f868eaec0687cc07625788d16b63bea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1dcfa4363f868eaec0687cc07625788d16b63bea"}]},"branch":"refs/heads/master"},"b5253b224bcf6cec428f18a831798d0955c61797":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2022-03-14 00:33:23.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/73/825073/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/825073/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/825073/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/825073/14"}}},"commit":{"parents":[{"commit":"3f0b82f84aa0c5636230692cf7efa1867a6b73c2","subject":"Merge \"Use the \"connectivity\" property of \"MechanismDriver\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f0b82f84aa0c5636230692cf7efa1867a6b73c2"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-01-18 09:25:06.000000000","tz":480},"committer":{"name":"ZhouHeng","email":"zhouhenglc@inspur.com","date":"2022-03-14 00:33:23.000000000","tz":0},"subject":"[OVN] Reschedule router GW chassis when AZ updated","message":"[OVN] Reschedule router GW chassis when AZ updated\n\nwhen chassis\u0027s available zone updated, triger rescheduler router\ngateway chassis.\n\nCloses-bug: #1958225\n\nChange-Id: If7cf55f2c12b388fc34fa942b6b0bf70cb9eb866\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5253b224bcf6cec428f18a831798d0955c61797"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5253b224bcf6cec428f18a831798d0955c61797"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
