)]}'
{"id":"openstack%2Fneutron~825076","triplet_id":"openstack%2Fneutron~stable%2Fxena~Iacc7d4e543ac261a6f566b49834b6865f6eded95","project":"openstack/neutron","branch":"stable/xena","topic":"bug/1957794","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-01-26 21:44:44.000000000","reason":"Change was submitted"},"30463":{"account":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"last_update":"2022-01-26 21:44:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iacc7d4e543ac261a6f566b49834b6865f6eded95","subject":"Exclude router gw subnet ports from port list while port delete.","status":"MERGED","created":"2022-01-18 09:53:55.000000000","updated":"2022-01-26 21:47:25.000000000","submitted":"2022-01-26 21:44:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"825076-bug/1957794","meta_rev_id":"79f0891dfa78e48a659d6a553f982c18ab970661","_number":825076,"virtual_id_number":825076,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-26 21:44:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-01-26 16:18:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-01-20 07:35:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2022-01-26 16:18:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"}],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-18 09:53:55.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"state":"REVIEWER"},{"updated":"2022-01-18 11:04:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-18 11:37:31.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-01-18 11:49:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-20 07:35:24.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-01-26 16:18:43.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"b95d7cf28735ea1d94a80ad63bf1b10167b58b9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-18 09:53:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65e57172c2f9b2e8c34e999ed7ba0eefcb5ef4dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 11:04:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd8de35e8a714ec6aebcfccd795c7b64 : SUCCESS in 59m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cea3f2ad76ba40d595f72c48f6d73b11 : SUCCESS in 42m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7fb46643329d86ccd4504656419132bae96fd490","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-01-18 11:37:31.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/825076/1 : SUCCESS in 1h 32m 38s","accounts_in_message":[],"_revision_number":1},{"id":"708e36b079eeb0ec46ad605a21fd44809e96199c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 11:49:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e89bf5d9d2064ca29b96378adfddb4a6 : SUCCESS in 48m 35s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/183466b6b48847e5b0e76eb2784e9f5b : SUCCESS in 1h 18m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9ebff63a824c44a9999167876889b073 : SUCCESS in 1h 14m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ec9804e8bae74f11abfd185c9518d7ac : SUCCESS in 1h 08m 17s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/78e475c544cd41b6945e7452458ec4f5 : SUCCESS in 1h 14m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/558bb626238143d283a40501c786ed84 : SUCCESS in 37m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acf1d47d960a4888ad3cf123384e3baf : SUCCESS in 37m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08c7bc18e4274f51ace625ab276179a6 : SUCCESS in 12m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/760d86d08722481795c5fa072413b84d : SUCCESS in 31m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84281a1c0d1b4f1889860f5c80051e55 : SUCCESS in 27m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c609c1199624f8c8d734443adde50e1 : SUCCESS in 23m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/205626760c3547f686b8abc93f21f587 : SUCCESS in 10m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/edce497f3e584a41ba5a27403a88a51a : SUCCESS in 51m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7540a197da274658b1dd92d9ad5033c6 : SUCCESS in 1h 52m 28s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/819636e68e334f9b969350fbf0c2908a : SUCCESS in 1h 14m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb91afcf2fc54895907c93d8189f3248 : SUCCESS in 57m 08s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d94931d5b2d74f5ba1db7953699e0423 : SUCCESS in 1h 00m 15s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b644a853d7c342e2bfdb372a24f4ebda : SUCCESS in 1h 08m 05s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ff3dae1cdd534f77b2e4f3f805b8cae3 : SUCCESS in 1h 06m 11s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe94ea13f01b4e9ea02768ca08e85583 : SUCCESS in 56m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1e5cb2e96f13427ab18c0e1da8f91a67 : SUCCESS in 1h 08m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"116215955d552c78156dc307919f6175affa3910","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-20 07:35:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ecd6bcf2b69948c54c5175794df89b183ee1644c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-01-26 16:18:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"32c70f6efb47bd64bea7b42777e29a8cafb08753","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 16:20:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5bf2d98f3f7176e675add81f2b2041434b99e22f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 21:44:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/62b55c5e5ad243f68a399f5d76d311f9 : SUCCESS in 45m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9fbb4d9ac72c4367aca63f9a40d4e78d : SUCCESS in 1h 06m 47s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5498b2d5dd174a3c9572606984671bc4 : SUCCESS in 1h 20m 18s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/571baa5202c245338583fe1d9497a8bc : SUCCESS in 56m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2623a4503d404c1a95b4eef8e2e2b3f5 : SUCCESS in 1h 14m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ace3b083755413084a2602cab3542f1 : SUCCESS in 16m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86e4d0e2b0b34951a6f5a9e8cc213c49 : SUCCESS in 30m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/510e5177c1d2404c8d0639cb52b4dd15 : SUCCESS in 25m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cd661806f104770bec05ecff00708ff : SUCCESS in 9m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/813d70879c2c4aa79a336bf9f262bafa : SUCCESS in 51m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef900c5775bd42dabd099380ef797931 : SUCCESS in 1h 46m 59s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c7653fa8a2a14a81a604b92c392fc514 : SUCCESS in 1h 20m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8af2787602de4c45ad6c62db41ea97ad : SUCCESS in 1h 08m 58s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/901e4de9ac3642d585f8c9c7360f1be6 : SUCCESS in 1h 00m 03s","accounts_in_message":[],"_revision_number":1},{"id":"5eb6a15466d3743359ff05e9e0136d8e20201b65","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 21:44:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"79f0891dfa78e48a659d6a553f982c18ab970661","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 21:47:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85df1eea9c994389945cf1a6c4e726ed : SUCCESS in 1m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"488a68cb49d3c076113aefd77b82d2c3374602ea","revisions":{"488a68cb49d3c076113aefd77b82d2c3374602ea":{"kind":"REWORK","_number":1,"created":"2022-01-18 09:53:55.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/76/825076/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/825076/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/825076/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/825076/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/825076/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/825076/1"}}},"commit":{"parents":[{"commit":"97180b01837638bd0476c28bdda2340eccd649af","subject":"Register cleanup hooks for the object change handlers only when needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/97180b01837638bd0476c28bdda2340eccd649af"}]}],"author":{"name":"labedz","email":"krzysztof.tomaszewski@ovhcloud.com","date":"2022-01-10 16:09:25.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-01-14 20:57:52.000000000","tz":0},"subject":"Exclude router gw subnet ports from port list while port delete.","message":"Exclude router gw subnet ports from port list while port delete.\n\nDuring port delete Neutron checks if host still handle some ports\nfrom subnets that are used by router.\n\nIt may happen that the same subnet that is used for router gateway\nport may be used for another VM port. Then we skip notification\nto update L3 agent about related to router port deletion.\n\nRelated-Bug: #1957794\nChange-Id: Iacc7d4e543ac261a6f566b49834b6865f6eded95\n(cherry picked from commit 4c6c48956ee8d9458924334f6650ca29e753a1ef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/488a68cb49d3c076113aefd77b82d2c3374602ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/488a68cb49d3c076113aefd77b82d2c3374602ea"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
