)]}'
{"id":"openstack%2Fneutron~826099","triplet_id":"openstack%2Fneutron~master~I1a5f705064f90f422fc0ca971d79135ac3ccfc9f","project":"openstack/neutron","branch":"master","topic":"bug/1932154","attention_set":{},"removed_from_attention_set":{"13686":{"account":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"last_update":"2022-01-27 18:14:59.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-01-26 11:20:27.000000000","reason":"Slawek Kaplonski replied on the change"}},"hashtags":[],"change_id":"I1a5f705064f90f422fc0ca971d79135ac3ccfc9f","subject":"[OVN] Add unit test for binding profile validation","status":"MERGED","created":"2022-01-24 15:17:55.000000000","updated":"2022-01-27 19:10:48.000000000","submitted":"2022-01-27 18:14:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826099-bug/1932154","meta_rev_id":"2caf31f48a59dc0735028bae08b8b0b4159de5f6","_number":826099,"virtual_id_number":826099,"owner":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-27 18:14:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-01-27 17:11:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-26 11:20:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-01-27 17:11:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-24 15:48:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-24 16:36:44.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-01-24 17:02:07.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-24 17:14:01.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-01-27 17:11:42.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"9fff88536800c76da47296aea23cb8ed82730eba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-01-24 15:17:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c2fcfcb8b3d97a2c7b9e16f329f9d2de2e96e8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 15:48:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7513f515e3aa4dc49b8969b74c04b37e : SUCCESS in 28m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/374f6e3ab1ce40669f987aef0e0ed3d8 : SUCCESS in 26m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb576db6ceba4dbc8e32ff85c9712202 : SUCCESS in 12m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/934d7b857b1a445c88b78e22a6985395 : SUCCESS in 27m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f67a942b11347238adc42aa583f2585 : SUCCESS in 21m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/629083514d304f3fa23c2f6005fbabf3 : SUCCESS in 9m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b29ad41ff6c4b019d2a5bd31bff694c : SUCCESS in 21m 38s","accounts_in_message":[],"_revision_number":1},{"id":"801dcc5523ef27b087d9736de6d912fcd8a7812e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 16:10:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/66804cb541c644fc841ce6ea1689d8e4 : SUCCESS in 44m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ab3d71a99b53449ab056a61c44d9cb9f : SUCCESS in 45m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"21a44381706db47ff171c6938ecc454df9aa01bc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-24 16:36:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a2af79f4ca0fc5aba86a012aee5f9dfdf5fbf4b5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-24 17:02:07.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/99/826099/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"315c59df38d3ec892151b83faea8e1a319862d72","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-01-24 17:14:01.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/826099/1 : SUCCESS in 1h 54m 28s","accounts_in_message":[],"_revision_number":1},{"id":"1bf4de3db08ea8e190283cf9ab7651e2bbc314f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-01-25 15:30:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"87d8d13b0744a67e5c7f4315c5c399540f2d59c8","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-01-25 15:44:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"68c2ae9b4dd885eaea9e0b47cd7a369d9892ed3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 16:14:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54fbfd5fc2984c64913c27180dfac7e3 : SUCCESS in 34m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0519a89bec944103b3144d714f3829ce : SUCCESS in 28m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db06edbb38f149ceaa29ff2b8f875667 : SUCCESS in 11m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6024012fd2aa4f5796b0fb48b4ee63c3 : SUCCESS in 24m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a032574df7843f99c6ae0455679b419 : SUCCESS in 26m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab11d9b6026b4f3a8cbd037e0c2ea75f : SUCCESS in 8m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/face46c6b5d044748b05fdd3bf03713f : SUCCESS in 21m 21s","accounts_in_message":[],"_revision_number":2},{"id":"c238108447cdc03aae58d49ee70efb26b04231fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 16:43:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0606f69bf1c84b16b0c1780717794846 : SUCCESS in 33m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1458a4b105d34b94baf2a402a687cabb : SUCCESS in 1h 02m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3d59f0b543d3371ab0db1c44d364f36ef0152619","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-01-25 17:20:21.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/826099/2 : SUCCESS in 1h 48m 45s","accounts_in_message":[],"_revision_number":2},{"id":"2bab2cd923dc1acb8a0e4d68e1e6a0cb136205f3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-26 11:20:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a8e078a0c1079e3c41db325375f2213ea9775238","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-27 17:11:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c346323356ea9c6803cec673703af71226081058","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 17:16:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"82e00086f669804ca511f273280091bd8ad56ce0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 18:14:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2501182057f64ad2acd7406fff686218 : SUCCESS in 33m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/648ef2740bbc4194849d451980c0d20f : SUCCESS in 15m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fa6426a09424e99956b21cab8b57ced : SUCCESS in 24m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d05f9cf7f65e4945a80faad24bbcaeae : SUCCESS in 26m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d0c313e263f4e50a2a6d459e6456d68 : SUCCESS in 11m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bde304b0f04a4dfd9b0122d107610c73 : SUCCESS in 27m 03s","accounts_in_message":[],"_revision_number":2},{"id":"08c167b31b0b360d0fd5854223a1101c288243ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 18:14:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2caf31f48a59dc0735028bae08b8b0b4159de5f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 19:10:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81eb6cb1338c44a5a827f7b84e9866b5 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1a7f4d5b038e396578644c74a120048f77595a47","revisions":{"fc0e1460970f10c0e08f5ba863f23d56d72bc414":{"kind":"REWORK","_number":1,"created":"2022-01-24 15:17:55.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/99/826099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/826099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/826099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/826099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/826099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/826099/1"}}},"commit":{"parents":[{"commit":"24c802711a939f532842c1d7b95a1afe004e809a","subject":"Merge \"Improve DHCP RPC handler\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24c802711a939f532842c1d7b95a1afe004e809a"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-01-24 10:09:41.000000000","tz":60},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-01-24 15:17:36.000000000","tz":60},"subject":"[OVN] Add unit test for binding profile validation","message":"[OVN] Add unit test for binding profile validation\n\nThe ``validate_and_get_data_from_binding_profile`` helper function\nin ``neutron.common.ovn.utils`` does currently not have unit tests.\n\nTo be able to safely modify this function in a upcoming patch we\nadd unit tests separately.\n\nPartial-Bug: #1932154\nChange-Id: I1a5f705064f90f422fc0ca971d79135ac3ccfc9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc0e1460970f10c0e08f5ba863f23d56d72bc414"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc0e1460970f10c0e08f5ba863f23d56d72bc414"}]},"branch":"refs/heads/master"},"1a7f4d5b038e396578644c74a120048f77595a47":{"kind":"REWORK","_number":2,"created":"2022-01-25 15:30:29.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/99/826099/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/826099/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/826099/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/826099/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/826099/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/826099/2"}}},"commit":{"parents":[{"commit":"4015f9a1d89a9ec6538c68a722bcf65da31be19a","subject":"Merge \"Create a PeriodicWorker for DbQuotaNoLockDriver clean up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4015f9a1d89a9ec6538c68a722bcf65da31be19a"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-01-24 10:09:41.000000000","tz":60},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-01-25 15:30:11.000000000","tz":60},"subject":"[OVN] Add unit test for binding profile validation","message":"[OVN] Add unit test for binding profile validation\n\nThe ``validate_and_get_data_from_binding_profile`` helper function\nin ``neutron.common.ovn.utils`` does currently not have unit tests.\n\nTo be able to safely modify this function in a upcoming patch we\nadd unit tests separately.\n\nPartial-Bug: #1932154\nChange-Id: I1a5f705064f90f422fc0ca971d79135ac3ccfc9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a7f4d5b038e396578644c74a120048f77595a47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a7f4d5b038e396578644c74a120048f77595a47"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
