)]}'
{"id":"openstack%2Fneutron~826545","triplet_id":"openstack%2Fneutron~master~Id9c906b2d141c3bedd80fb5f868190f8a4b66f54","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-03-01 14:21:24.000000000","reason":"Slawek Kaplonski replied on the change"},"9531":{"account":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"last_update":"2022-03-02 00:56:55.000000000","reason":"LIU Yulong replied on the change"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2022-03-01 14:34:39.000000000","reason":"Oleg Bondarev replied on the change"},"30463":{"account":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"last_update":"2022-03-02 02:30:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id9c906b2d141c3bedd80fb5f868190f8a4b66f54","subject":"Don\u0027t set HA ports down while L3 agent restart.","status":"MERGED","created":"2022-01-26 23:26:43.000000000","updated":"2022-03-02 02:34:01.000000000","submitted":"2022-03-02 02:30:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"826545","meta_rev_id":"4c23cdcbc2d498d5213c3709ace532f0a9d993fe","_number":826545,"virtual_id_number":826545,"owner":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-03-01 15:59:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-02 02:30:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-01 14:32:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-03-02 00:56:55.000000000","_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-01 14:34:39.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-03-02 00:56:55.000000000","_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-26 23:48:26.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-01-27 00:29:01.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-27 00:56:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-27 01:31:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-31 15:40:15.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-03-01 14:24:59.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"CC"},{"updated":"2022-03-01 14:34:39.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-03-01 14:44:39.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"CC"},{"updated":"2022-03-02 00:56:55.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"}],"messages":[{"id":"ac51baf6278046b8b8c9f6090f582848562c7561","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"date":"2022-01-26 23:26:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51227f95f75dce4dc901b13ad58cc07231570c53","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"date":"2022-01-26 23:27:19.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"ffae0bc990997d2316a11103ca2c43014c25e685","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"date":"2022-01-26 23:30:12.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d63e3c2681172e556bfa88363258c63124531b8d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"date":"2022-01-26 23:34:41.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"455c0202b26d34bfa6e74ff53cfbf75929036dcf","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-01-26 23:48:26.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/826545/3 : FAILURE in 12m 54s","accounts_in_message":[],"_revision_number":3},{"id":"a7db9090e0048bc60a77b1a5f73cabb3d0ae191b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"date":"2022-01-27 00:13:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fcf8194201d9c25e38eb2fa505f334c283eb4a6d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"date":"2022-01-27 00:15:04.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"f2dcd9f226ea6506fa62640ddbb23ffe80d6465f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-01-27 00:27:31.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/826545/4 : FAILURE in 12m 55s","accounts_in_message":[],"_revision_number":4},{"id":"4ef9976f89346a6a4a09ad694ba66349afe31a37","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-27 00:29:01.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/45/826545/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"0b185f53aec1eb1e27fcc56d2320084aa06b8645","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 00:56:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d40e26c8c5943119d6256c6a06e27d3 : FAILURE in 33m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2a727c94e5064330b339e07a0714b9e6 : FAILURE in 39m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e0ccb3ad617d73b3938609bf170faf11dc252a6c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-27 01:16:24.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/45/826545/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"0be8da59bbb380a02d0790826d74bb4ede162230","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 01:31:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/11dffd8efe5a467aa1579f7496a830b9 : SUCCESS in 1h 01m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9b9d0ada33564d60bcf44ee9e3047369 : SUCCESS in 1h 13m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a1869d6ee59f448694c946ec41ad2bfe : SUCCESS in 1h 00m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/953c79a05a334c6e966ae7f3678a94fb : FAILURE in 42m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7533925d8524919a848d96c1f810bbe : FAILURE in 35m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/597da1d3ea3f4c45a60072ab2bb1e79b : SUCCESS in 15m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8c3045c15f946d9a41c4dad88590256 : FAILURE in 22m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f9fdb9966034b8db5644f6d1fbb70d8 : FAILURE in 22m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c49e0d2b823f4189be87895b7770ff35 : SUCCESS in 9m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6f88711261c34b48be0c35041664cae2 : SUCCESS in 40m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/858baec15236444897c4b31f4d2c0733 : FAILURE in 1h 13m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7f5fa932760a4f89a1a6250dec96d4c2 : SUCCESS in 59m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7600b9067b7b4747b4a7d49a9678c35a : SUCCESS in 47m 52s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/14d1f5ef7f464914bd9acce61852588c : SUCCESS in 51m 56s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b06e74fc42fa415b96b524dac6cd3256 : SUCCESS in 1h 03m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/35d16b627efd4f57baab0eff043138ac : SUCCESS in 58m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6c24ce8cfa0400784040cc40d91d53e : FAILURE in 21m 47s","accounts_in_message":[],"_revision_number":4},{"id":"a7aaa4af9cab3d4c1c1aaa73812a83485558fd45","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-27 02:04:53.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/45/826545/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"ed7564f9372026b78cc7ef7295a74903f86ba16a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-27 03:07:28.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/45/826545/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"228918ef00194821f2fa911146b159e792dc24c2","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"date":"2022-01-27 09:16:26.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"d6fdae258759491225e1d7392d99105966a83e58","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"date":"2022-01-31 15:27:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"98bea99033c7feb53c7e031f5c3a7cbe0b09fdb8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-31 15:40:15.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a3816bd8cffe9d2e39a73bc278a5f2b202936adc","author":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"date":"2022-01-31 15:56:22.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"a699dd40bad52712f044b20853cc92a18b48f18b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-01-31 16:36:01.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2022/Arista_Tempest_Testrefs/changes/45/826545/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"7ccde9e950eb054885c784874d249c0670896b9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 16:45:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe8e152b3bfa42ef924ff07d716a74e6 : SUCCESS in 55m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4d1dedb863b14f7fbf2e289e768d3c00 : SUCCESS in 52m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"62b1d63dd2eeee69a5d1a4026265e767de975d71","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-01-31 17:01:03.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/826545/5 : SUCCESS in 1h 32m 15s","accounts_in_message":[],"_revision_number":5},{"id":"57ee56db345a0664b3eb65b36d271a0df8c1f4f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 17:35:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9d3d194db93244dab563759b3ca0ffe5 : SUCCESS in 55m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c474e15b36b04b6892dc3f4cef765a2f : SUCCESS in 1h 01m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/04a178190ebd4a588a6b07f16449f346 : SUCCESS in 47m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8150f63a2a7f47229387eb6adfa3c312 : SUCCESS in 32m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/639ede4a04b44e7a9ce4966cd34db552 : SUCCESS in 31m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10c7bf5b7dbc4e7c91d949e122518d46 : SUCCESS in 14m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2060872ff0ce4e3287fec11551fcf68f : SUCCESS in 24m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a2ff9b16ff54041b8498796e812c78f : SUCCESS in 24m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71adc3417f6b4ac18aefde860824de25 : SUCCESS in 11m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ffff96bf3f404e5eb5bd4e966d31ccb2 : SUCCESS in 44m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/36a79dd90c514fe6be13faba4a76eb15 : SUCCESS in 1h 44m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f00c50aada2f44d1924c4c6d5d67004b : SUCCESS in 1h 40m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6182850dd6994b068ba0577de2562f0a : SUCCESS in 59m 39s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cff83cccce994ca3b8c8db31f644c5e7 : SUCCESS in 54m 24s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/943045f9f0b1456e8010f6905cd5925b : SUCCESS in 1h 17m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/12409dbd15744adb8bac7b6736fc7675 : SUCCESS in 1h 10m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e30634f2e9a64a53913694dcd130b96e : SUCCESS in 26m 49s","accounts_in_message":[],"_revision_number":5},{"id":"7ba15a4825f7a23d58f8d027b7a1054f488acd60","author":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"date":"2022-02-01 08:53:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"85e7daaf3396f639e7ccf4434fdeffd570f9b4e2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-02-17 10:12:03.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"38b65bdc44af79380a36bd5d2959bc9288733973","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"date":"2022-02-28 23:58:41.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2c024ae8fd159010681a7acd3f09bc6993f450a1","author":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"date":"2022-03-01 00:00:48.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c4a9a947fb57626b2cc8388f4bfab4340dadfb80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"date":"2022-03-01 00:04:13.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"5a1874da4f1e4ff23a4f583189db0c8d90dece61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 00:50:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/46f2f3a2e043421c9fa0af997bd50a89 : SUCCESS in 42m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/68526c80f406483494233a7382190cf4 : SUCCESS in 44m 37s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"79b5133ee9e8ed4c0a34a1e0de54e0cf8d277b6b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-01 00:54:03.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/45/826545/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"6de03076885418a16809f29afe168d8ae2e754d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 01:39:01.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6ad50aa94131441a95c10c0c14603b9b : SUCCESS in 1h 08m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/39ba1900efbb443b9e5c61b00d6dfe7f : SUCCESS in 1h 14m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/16ae8be340f5431b9a20e88d7d5403b6 : SUCCESS in 1h 08m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c3dfdb8c02b45739da9a6a0119be4fe : SUCCESS in 28m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8f42424111f41329fa805872ce09d1c : SUCCESS in 24m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/966bbf71891c476bab54784f2d7f2100 : FAILURE in 8m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/535b3a393f3741d4b1f3858d6aec6a51 : SUCCESS in 23m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/baab8b6f7a9d4b3d9054388dcd0e01bb : SUCCESS in 22m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcae5e8ac0614cfcab07d2404f7213ba : SUCCESS in 10m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/94c786b2cdb64a67bfc0979ad4b712f3 : SUCCESS in 48m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6f688fb6e4194a029e6ebc6265c448cf : SUCCESS in 1h 28m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/667e260578c540a1bf0587c8bfb10c12 : SUCCESS in 1h 03m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cfb39004153f4f1fa9dd06a7b43393e5 : SUCCESS in 50m 39s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/237eb91c262a441db1d285b36ab4ac76 : SUCCESS in 54m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/15722c9e7b394c10a6f5c47db389f486 : SUCCESS in 58m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fc99fbf9cfd94c70b45b8b027d2b5243 : SUCCESS in 1h 04m 24s","accounts_in_message":[],"_revision_number":7},{"id":"92f2cdedfc7337451ec2a491d993cd4aa81d92b3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-01 01:42:39.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/45/826545/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"d5ad2536311ce5f41fef04653ad495040ad5164d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-01 02:06:30.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/826545/7 : SUCCESS in 2h 01m 30s","accounts_in_message":[],"_revision_number":7},{"id":"c272ca66a7ae40f58e7dfd8041cfcecb31e5a5b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"date":"2022-03-01 08:43:57.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4eca9b1778f14bb56388690ffabeca587a0a2e25","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-01 09:24:12.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/45/826545/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"0b8b0f65d083f5aad91738ba55d4ff506c79728b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 09:31:56.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9ace29ba9cb44ecab8a81d6beb010b55 : SUCCESS in 46m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7cc152fc911b488f8979dec254451486 : SUCCESS in 46m 16s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b51a7018d9134864dae6fcaa8fec682571199e90","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-01 10:17:57.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/826545/8 : SUCCESS in 1h 33m 16s","accounts_in_message":[],"_revision_number":8},{"id":"11d792e8eb4469f63a73671c23df4e6090684eb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 10:25:44.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ae29daaa68554a81b1d0cbf72ab045b1 : SUCCESS in 1h 28m 32s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2f444dc73cec4344ab19c44678b1969f : SUCCESS in 1h 18m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7669d9eec9db4fbf90f99abb0ded84e9 : SUCCESS in 51m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa6cb6e1e3c1442d80d223041ced8167 : SUCCESS in 23m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/237be2bb99e348bca6465bfeeb81d8d5 : SUCCESS in 32m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab762f30a9ce42ad8b9195316424ace2 : SUCCESS in 13m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ccbc51150b854d1299250d459856afb9 : SUCCESS in 33m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7417e23a9f554415a004b4669b0da266 : SUCCESS in 24m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c4f0d6cb91f44cdb642bb0f02b4b5c3 : SUCCESS in 10m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4a7937ab34f34ce88bfe475931640087 : SUCCESS in 51m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6b4fe02d0b040a8a464a6f9d458b00a : SUCCESS in 1h 27m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/64ff8aa5ad01405f9ef47387a2e31a74 : SUCCESS in 1h 01m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b1ee9b00c6704f80899b6b2bdd7a3727 : SUCCESS in 1h 03m 18s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3006c22cd4374adba9a1ffeabaf8f7d1 : SUCCESS in 52m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/39f40592073146d89e0a5caf9f1799d0 : SUCCESS in 1h 34m 53s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9c5f56ca60ad454db62690912ff03eae : SUCCESS in 1h 21m 51s","accounts_in_message":[],"_revision_number":8},{"id":"b4fff385b01f2de25230f79eaa83ba124c6fda3c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-01 14:05:57.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b8bbd1b15353d3374c42312ba6eccb1c9b39e979","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"date":"2022-03-01 14:20:36.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"4d5ff55f46396d2e916db76a4bf122ad6c0e3b98","author":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"date":"2022-03-01 14:21:17.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c48ffaf8ee4dbd99da3fe72b39a5b80aa8ceda27","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-01 14:21:24.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"3e756021025e8493af2a0bc6023823d2869723aa","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-03-01 14:24:59.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ad387d794a9fd6e20d65a675fdb9102aa89ee198","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"date":"2022-03-01 14:27:42.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"f62c1f0b21a4325d1441549b58d2b6dee4b2b130","author":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"date":"2022-03-01 14:28:30.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d8fb05a97e3194b22c53decc19cc04756e335f3b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-01 14:32:46.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"21033c679bd7a967ac54c2408a0643fcfed1de61","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-03-01 14:34:39.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"b24c65f2e75c6e4d96cf78793834a2bb871c57d3","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-03-01 14:44:39.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"8240bd797d5035a1d4e1a49efd7d673eb899ee0d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-01 15:10:28.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/45/826545/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"7f6431cc0ac00600eeb3cf09f8457c21f2e7d420","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 15:14:47.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a58fd6be823b4c1185a5d418c1515a82 : SUCCESS in 43m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f4fb90277444f4f939eb7f90a5df5b5 : SUCCESS in 45m 01s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"3aa071409a6e981dcfd68a25ea9dd8f932ffcfea","author":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"date":"2022-03-01 15:33:14.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"510cbdfcc785642d83803c86a00f90a5bac54713","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-01 15:59:56.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/45/826545/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"daa6c0b508d3db629e2f75db3002776a5b88dac6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-01 16:02:10.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/826545/10 : SUCCESS in 1h 33m 41s","accounts_in_message":[],"_revision_number":10},{"id":"bc100b00cd3d3ffd65115a550e352b7a9bf22692","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 16:04:16.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0b6c660b117a436e92ed5ce82758d2fc : SUCCESS in 1h 26m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/eddd2a927be34a019d3000727ca8607f : SUCCESS in 1h 25m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f18241c0dc1a4ac2823cae5ce0a19872 : SUCCESS in 1h 21m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfd0c63668c04eadb2f9acf547a4f59b : SUCCESS in 36m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0881c5b1ecd64b0c8b869f8e7332fddd : SUCCESS in 39m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dfa58f40f414e48823e50eda3f04167 : SUCCESS in 16m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f22bb7e3f45e45eaa65c57c22f246644 : SUCCESS in 33m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/101893a1d38c4889b23ab6b558d4983e : SUCCESS in 26m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90bedc4982994cf78befcae78c745bd9 : SUCCESS in 10m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/234b81737b3b4d96a97f74bdbca070a0 : FAILURE in 50m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/024de1c79327466f843f221a8a416386 : SUCCESS in 1h 31m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6bc44fc89c2645698a1742b331bd9b0b : SUCCESS in 1h 11m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cf7bb800fd084bf0af89bb04745e01d9 : SUCCESS in 57m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/de12e0e8c28e4b718b147c89d39095a7 : SUCCESS in 57m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0979a76e63cf4a34992ca6e94044b979 : SUCCESS in 1h 16m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d8a0618d8e9f4b13b42abbeb56c10109 : SUCCESS in 1h 15m 21s","accounts_in_message":[],"_revision_number":10},{"id":"ed21db4426e9c482f36eb5378516628dfb5d2ab8","author":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"date":"2022-03-01 16:16:17.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"cc6b57c70d8b2c08f9fa2f56eab3c28002d588dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 17:02:13.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e8f032dcd89047b889e13ca9bed9bbd7 : SUCCESS in 45m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3cd0e2889bab481fb3c497aa65d7e5f0 : SUCCESS in 43m 53s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c3ef2ebacf1abb849cabe82bc4ce28c2b7172226","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 17:58:00.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c1c90fc7e0834afc866ba39eb674eba0 : SUCCESS in 1h 17m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5bdac4e625d7413e9d7522c9720e7f74 : SUCCESS in 1h 09m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/51f96734fad44150bc7fcd1cb56579e9 : SUCCESS in 54m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0eec52b8872544a191d009ad2d4c2e47 : SUCCESS in 45m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3afcf13def054ab5b0d3c71be0cce9de : SUCCESS in 32m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1137b47bf8ab47b8b1911657023f62b7 : SUCCESS in 12m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67740d5c3b7543f9979312e563a5117c : SUCCESS in 24m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c200a8587a8d460a9230c68c1a9a984d : SUCCESS in 31m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe5d05962e9749ffa970fb48a5a9e07a : SUCCESS in 9m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/186ebd2e30a9443f884b1858a260b136 : SUCCESS in 47m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d739d23d69884298a4a8f32aca8dfe8a : SUCCESS in 1h 38m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/57f9a776ec3049179ec5f22a1b0ce490 : SUCCESS in 1h 12m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9fd92dbfba0c4e9bab97dfa3f39315db : SUCCESS in 52m 35s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7fc6942155d14bbeb43191af2d7febec : SUCCESS in 54m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/033a5309f664431bb16a4881af30237f : SUCCESS in 1h 17m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f5ee7698549b4492b5d9e77fb303dd66 : SUCCESS in 1h 10m 10s","accounts_in_message":[],"_revision_number":10},{"id":"11f70b5ea5211e6990ce4e3cc6bfcd2118e15c55","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-01 18:08:51.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/826545/10 : SUCCESS in 1h 51m 50s","accounts_in_message":[],"_revision_number":10},{"id":"5d137c11e6b1969bc5ee0ec842f38a62004af329","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-03-02 00:56:55.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"88a30b682bad30f6fd648a3eb25c25288ff2b81e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 00:57:42.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"343665c5f8d36a979ac97ec48ec2263a617cbd53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 02:30:03.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/05d6b76f64644544a89f38ec80be4312 : SUCCESS in 1h 05m 38s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2562fbe1b4c7450ba6558a84ff997dcd : SUCCESS in 1h 02m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b665a1ed5871407795bfa50b56c3e1ec : SUCCESS in 55m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd2a6955503445cfb70b88fe7d73c596 : SUCCESS in 36m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f9d831270f84b62a9fae51c055de242 : SUCCESS in 12m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7c5d2d9a1ba4487abb8c036a6d3d4bf : SUCCESS in 33m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c2b6131defa47b2ae5b60a60df3d4da : SUCCESS in 31m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ce63d1ee9e242b9843cf7910c4c7c37 : SUCCESS in 9m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ab4abcb8d5724aff8a0754024d85899f : SUCCESS in 48m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9bbb86c9787949acaadbc75457184068 : SUCCESS in 1h 21m 01s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/997f8c72a16f476c9c0931154700db6e : SUCCESS in 1h 02m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/13eb7839aff24c968d53f99efd7a316b : SUCCESS in 44m 07s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7025083321974b928c7dd99289ce9d96 : SUCCESS in 52m 52s","accounts_in_message":[],"_revision_number":10},{"id":"0c83a25e5a40cd1b664c3f85d4fa45a72dd5a3c0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 02:30:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"4c23cdcbc2d498d5213c3709ace532f0a9d993fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 02:34:01.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0214c2aa655042d5887fcd73a3d9d5f8 : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"f430cd00725f8303f5313cb7784c9aed4b982e62","revisions":{"b5603dd0e7aa76ecfb96f9b1751bfa3414aaaeef":{"kind":"REWORK","_number":1,"created":"2022-01-26 23:26:43.000000000","uploader":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"ref":"refs/changes/45/826545/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/826545/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/826545/1"}}},"commit":{"parents":[{"commit":"473f4db1d666a93c8483b4013eed8ce836825e88","subject":"Merge \"Local IP internal documentation and release note\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/473f4db1d666a93c8483b4013eed8ce836825e88"}]}],"author":{"name":"labedz","email":"krzysztof.tomaszewski@ovhcloud.com","date":"2022-01-26 23:13:40.000000000","tz":60},"committer":{"name":"labedz","email":"krzysztof.tomaszewski@ovhcloud.com","date":"2022-01-26 23:23:35.000000000","tz":60},"subject":"Don\u0027t set HA ports down while L3 agent restart.","message":"Don\u0027t set HA ports down while L3 agent restart.\n\nBecause of the fix for bug[1] L3 agent puts all it\u0027s HA ports\ndown during initialization phase. Unfortunately such operation\ncan break already working L3 communication. Rewiring ha-* port\nfrom down state to up can takes few seconds and some VRRP packages\ncould be lost then. That triggers keepalived on other node so router\nHA state change may be triggered.\n\nThis change prevents putting HA ports down when during initialization\nphase L3 agent finds already configured own net namespaces. Existance\nof such net namespace is a good proof that there is a network\nconfiguration existing so host wasn\u0027t rebooted so most probably it is\njust agent restart.\n\n[1] https://bugs.launchpad.net/neutron/+bug/1597461\n\nChange-Id: Id9c906b2d141c3bedd80fb5f868190f8a4b66f54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5603dd0e7aa76ecfb96f9b1751bfa3414aaaeef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5603dd0e7aa76ecfb96f9b1751bfa3414aaaeef"}]},"branch":"refs/heads/master"},"34aaec16f7147600fbf7a7d76368f7a5f97d5560":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-01-26 23:30:12.000000000","uploader":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"ref":"refs/changes/45/826545/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/826545/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/826545/2"}}},"commit":{"parents":[{"commit":"473f4db1d666a93c8483b4013eed8ce836825e88","subject":"Merge \"Local IP internal documentation and release note\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/473f4db1d666a93c8483b4013eed8ce836825e88"}]}],"author":{"name":"labedz","email":"krzysztof.tomaszewski@ovhcloud.com","date":"2022-01-26 23:13:40.000000000","tz":60},"committer":{"name":"labedz","email":"krzysztof.tomaszewski@ovhcloud.com","date":"2022-01-26 23:28:50.000000000","tz":60},"subject":"Don\u0027t set HA ports down while L3 agent restart.","message":"Don\u0027t set HA ports down while L3 agent restart.\n\nBecause of the fix for bug[1] L3 agent puts all it\u0027s HA ports\ndown during initialization phase. Unfortunately such operation\ncan break already working L3 communication. Rewiring ha-* port\nfrom down state to up can takes few seconds and some VRRP packages\ncould be lost then. That triggers keepalived on other node so router\nHA state change may be triggered.\n\nThis change prevents putting HA ports down when during initialization\nphase L3 agent finds already configured own net namespaces. Existance\nof such net namespace is a good proof that there is a network\nconfiguration existing so host wasn\u0027t rebooted so most probably it is\njust agent restart.\n\n[1] https://bugs.launchpad.net/neutron/+bug/1597461\n\nRelated-Bug: #1959151\nChange-Id: Id9c906b2d141c3bedd80fb5f868190f8a4b66f54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/34aaec16f7147600fbf7a7d76368f7a5f97d5560"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/34aaec16f7147600fbf7a7d76368f7a5f97d5560"}]},"branch":"refs/heads/master"},"fc576f01b11d5d80002b7b0541aab6e73d1b610f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-01-26 23:34:41.000000000","uploader":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"ref":"refs/changes/45/826545/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/826545/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/826545/3"}}},"commit":{"parents":[{"commit":"473f4db1d666a93c8483b4013eed8ce836825e88","subject":"Merge \"Local IP internal documentation and release note\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/473f4db1d666a93c8483b4013eed8ce836825e88"}]}],"author":{"name":"labedz","email":"krzysztof.tomaszewski@ovhcloud.com","date":"2022-01-26 23:13:40.000000000","tz":60},"committer":{"name":"labedz","email":"krzysztof.tomaszewski@ovhcloud.com","date":"2022-01-26 23:34:03.000000000","tz":60},"subject":"Don\u0027t set HA ports down while L3 agent restart.","message":"Don\u0027t set HA ports down while L3 agent restart.\n\nBecause of the fix for bug[1] L3 agent puts all it\u0027s HA ports\ndown during initialization phase. Unfortunately such operation\ncan break already working L3 communication. Rewiring ha-* port\nfrom down state to up can takes few seconds and some VRRP packages\ncould be lost then. That triggers keepalived on other node so router\nHA state change may be triggered.\n\nThis change prevents putting HA ports down when during initialization\nphase L3 agent finds already configured own net namespaces. Existance\nof such net namespace is a good proof that there is a network\nconfiguration existing so host wasn\u0027t rebooted so most probably it is\njust agent restart.\n\n[1] https://bugs.launchpad.net/neutron/+bug/1597461\n\nCloses-Bug: #1959151\nChange-Id: Id9c906b2d141c3bedd80fb5f868190f8a4b66f54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc576f01b11d5d80002b7b0541aab6e73d1b610f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc576f01b11d5d80002b7b0541aab6e73d1b610f"}]},"branch":"refs/heads/master"},"d52eb7ae7cf76a4c450d85bf6af45c682625381e":{"kind":"REWORK","_number":4,"created":"2022-01-27 00:13:55.000000000","uploader":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"ref":"refs/changes/45/826545/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/826545/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/826545/4"}}},"commit":{"parents":[{"commit":"473f4db1d666a93c8483b4013eed8ce836825e88","subject":"Merge \"Local IP internal documentation and release note\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/473f4db1d666a93c8483b4013eed8ce836825e88"}]}],"author":{"name":"labedz","email":"krzysztof.tomaszewski@ovhcloud.com","date":"2022-01-26 23:13:40.000000000","tz":60},"committer":{"name":"labedz","email":"krzysztof.tomaszewski@ovhcloud.com","date":"2022-01-27 00:13:11.000000000","tz":60},"subject":"Don\u0027t set HA ports down while L2 agent restart.","message":"Don\u0027t set HA ports down while L2 agent restart.\n\nBecause of the fix for bug[1] L3 agent puts all it\u0027s HA ports\ndown during initialization phase. Unfortunately such operation\ncan break already working L3 communication. Rewiring ha-* port\nfrom down state to up can takes few seconds and some VRRP packages\ncould be lost then. That triggers keepalived on other node so router\nHA state change may be triggered.\n\nThis change prevents putting HA ports down when during initialization\nphase L3 agent finds already configured own net namespaces. Existance\nof such net namespace is a good proof that there is a network\nconfiguration existing so host wasn\u0027t rebooted so most probably it is\njust agent restart.\n\n[1] https://bugs.launchpad.net/neutron/+bug/1597461\n\nCloses-Bug: #1959151\nChange-Id: Id9c906b2d141c3bedd80fb5f868190f8a4b66f54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d52eb7ae7cf76a4c450d85bf6af45c682625381e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d52eb7ae7cf76a4c450d85bf6af45c682625381e"}]},"branch":"refs/heads/master"},"5121be53adff490a35a2c91d938e6708b67b61fe":{"kind":"REWORK","_number":5,"created":"2022-01-31 15:27:43.000000000","uploader":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"ref":"refs/changes/45/826545/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/826545/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/826545/5"}}},"commit":{"parents":[{"commit":"473f4db1d666a93c8483b4013eed8ce836825e88","subject":"Merge \"Local IP internal documentation and release note\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/473f4db1d666a93c8483b4013eed8ce836825e88"}]}],"author":{"name":"labedz","email":"krzysztof.tomaszewski@ovhcloud.com","date":"2022-01-26 23:13:40.000000000","tz":60},"committer":{"name":"labedz","email":"krzysztof.tomaszewski@ovhcloud.com","date":"2022-01-31 15:23:47.000000000","tz":60},"subject":"Don\u0027t set HA ports down while L2 agent restart.","message":"Don\u0027t set HA ports down while L2 agent restart.\n\nBecause of the fix for bug[1] L3 agent puts all it\u0027s HA ports\ndown during initialization phase. Unfortunately such operation\ncan break already working L3 communication. Rewiring ha-* port\nfrom down state to up can takes few seconds and some VRRP packages\ncould be lost then. That triggers keepalived on other node so router\nHA state change may be triggered.\n\nThis change prevents putting HA ports down when during initialization\nphase L3 agent finds already configured own net namespaces. Existance\nof such net namespace is a good proof that there is a network\nconfiguration existing so host wasn\u0027t rebooted so most probably it is\njust agent restart.\n\n[1] https://bugs.launchpad.net/neutron/+bug/1597461\n\nCloses-Bug: #1959151\nChange-Id: Id9c906b2d141c3bedd80fb5f868190f8a4b66f54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5121be53adff490a35a2c91d938e6708b67b61fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5121be53adff490a35a2c91d938e6708b67b61fe"}]},"branch":"refs/heads/master"},"80a684f360e8ecd56fa44eff7b5df1811bbb161e":{"kind":"REWORK","_number":6,"created":"2022-02-28 23:58:41.000000000","uploader":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"ref":"refs/changes/45/826545/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/826545/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/826545/6"}}},"commit":{"parents":[{"commit":"473f4db1d666a93c8483b4013eed8ce836825e88","subject":"Merge \"Local IP internal documentation and release note\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/473f4db1d666a93c8483b4013eed8ce836825e88"}]}],"author":{"name":"labedz","email":"krzysztof.tomaszewski@ovhcloud.com","date":"2022-01-26 23:13:40.000000000","tz":60},"committer":{"name":"labedz","email":"krzysztof.tomaszewski@ovhcloud.com","date":"2022-02-28 23:55:34.000000000","tz":60},"subject":"Don\u0027t set HA ports down while L2 agent restart.","message":"Don\u0027t set HA ports down while L2 agent restart.\n\nBecause of the fix for bug[1] L3 agent puts all it\u0027s HA ports\ndown during initialization phase. Unfortunately such operation\ncan break already working L3 communication. Rewiring ha-* port\nfrom down state to up can takes few seconds and some VRRP packages\ncould be lost then. That triggers keepalived on other node so router\nHA state change may be triggered.\n\nThis change prevents putting HA ports down when during initialization\nphase L3 agent finds already configured own net namespaces. Existance\nof such net namespace is a good proof that there is a network\nconfiguration existing so host wasn\u0027t rebooted so most probably it is\njust agent restart.\n\n[1] https://bugs.launchpad.net/neutron/+bug/1597461\n\nCloses-Bug: #1959151\nChange-Id: Id9c906b2d141c3bedd80fb5f868190f8a4b66f54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80a684f360e8ecd56fa44eff7b5df1811bbb161e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80a684f360e8ecd56fa44eff7b5df1811bbb161e"}]},"branch":"refs/heads/master"},"a053c5b24e7d0e18122215864f46562e559d979c":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2022-03-01 00:04:13.000000000","uploader":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"ref":"refs/changes/45/826545/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/826545/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/826545/7"}}},"commit":{"parents":[{"commit":"473f4db1d666a93c8483b4013eed8ce836825e88","subject":"Merge \"Local IP internal documentation and release note\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/473f4db1d666a93c8483b4013eed8ce836825e88"}]}],"author":{"name":"labedz","email":"krzysztof.tomaszewski@ovhcloud.com","date":"2022-01-26 23:13:40.000000000","tz":60},"committer":{"name":"Krzysztof Tomaszewski","email":"krzysztof.tomaszewski@ovhcloud.com","date":"2022-03-01 00:04:13.000000000","tz":0},"subject":"Don\u0027t set HA ports down while L2 agent restart.","message":"Don\u0027t set HA ports down while L2 agent restart.\n\nBecause of the fix for bug[1] and issue with linux_utils\nget_process_count_by_name() L3 agent puts all it\u0027s HA ports down\nduring initialization phase. Unfortunately such operation can break\nalready working L3 communication. Rewiring ha-* port from down state to\nup can takes few seconds and some VRRP packages could be lost then.\nThat triggers keepalived on other node so router HA state change\nmay be triggered.\n\nThis change prevents putting HA ports down when during initialization\nphase L3 agent finds already configured own net namespaces. Existance\nof such net namespace is a good proof that there is a network\nconfiguration existing so host wasn\u0027t rebooted so most probably it is\njust agent restart.\n\n[1] https://bugs.launchpad.net/neutron/+bug/1597461\n\nCloses-Bug: #1959151\nChange-Id: Id9c906b2d141c3bedd80fb5f868190f8a4b66f54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a053c5b24e7d0e18122215864f46562e559d979c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a053c5b24e7d0e18122215864f46562e559d979c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"bedf60fc5448a013cf4fdabc9936830aa3bea448":{"kind":"REWORK","_number":8,"created":"2022-03-01 08:43:57.000000000","uploader":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"ref":"refs/changes/45/826545/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/826545/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/826545/8"}}},"commit":{"parents":[{"commit":"473f4db1d666a93c8483b4013eed8ce836825e88","subject":"Merge \"Local IP internal documentation and release note\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/473f4db1d666a93c8483b4013eed8ce836825e88"}]}],"author":{"name":"labedz","email":"krzysztof.tomaszewski@ovhcloud.com","date":"2022-01-26 23:13:40.000000000","tz":60},"committer":{"name":"labedz","email":"krzysztof.tomaszewski@ovhcloud.com","date":"2022-03-01 08:40:53.000000000","tz":60},"subject":"Don\u0027t set HA ports down while L2 agent restart.","message":"Don\u0027t set HA ports down while L2 agent restart.\n\nBecause of the fix for bug[1] and issue with linux_utils\nget_process_count_by_name() L3 agent puts all it\u0027s HA ports down\nduring initialization phase. Unfortunately such operation can break\nalready working L3 communication. Rewiring ha-* port from down state to\nup can takes few seconds and some VRRP packages could be lost then.\nThat triggers keepalived on other node so router HA state change\nmay be triggered.\n\nThis change prevents putting HA ports down when during initialization\nphase L3 agent finds already configured own net namespaces. Existance\nof such net namespace is a good proof that there is a network\nconfiguration existing so host wasn\u0027t rebooted so most probably it is\njust agent restart.\n\n[1] https://bugs.launchpad.net/neutron/+bug/1597461\n\nCloses-Bug: #1959151\nChange-Id: Id9c906b2d141c3bedd80fb5f868190f8a4b66f54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bedf60fc5448a013cf4fdabc9936830aa3bea448"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bedf60fc5448a013cf4fdabc9936830aa3bea448"}]},"branch":"refs/heads/master"},"1c37880135fe15c98a19aa02e8237fa2e6315d9c":{"kind":"REWORK","_number":9,"created":"2022-03-01 14:20:36.000000000","uploader":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"ref":"refs/changes/45/826545/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/826545/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/826545/9"}}},"commit":{"parents":[{"commit":"473f4db1d666a93c8483b4013eed8ce836825e88","subject":"Merge \"Local IP internal documentation and release note\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/473f4db1d666a93c8483b4013eed8ce836825e88"}]}],"author":{"name":"labedz","email":"krzysztof.tomaszewski@ovhcloud.com","date":"2022-01-26 23:13:40.000000000","tz":60},"committer":{"name":"labedz","email":"krzysztof.tomaszewski@ovhcloud.com","date":"2022-03-01 14:20:08.000000000","tz":60},"subject":"Don\u0027t set HA ports down while L2 agent restart.","message":"Don\u0027t set HA ports down while L2 agent restart.\n\nBecause of the fix for bug[1] and issue with linux_utils\nget_process_count_by_name() L3 agent puts all it\u0027s HA ports down\nduring initialization phase. Unfortunately such operation can break\nalready working L3 communication. Rewiring ha-* port from down state to\nup can takes few seconds and some VRRP packages could be lost then.\nThat triggers keepalived on other node so router HA state change\nmay be triggered.\n\nThis change prevents putting HA ports down when during initialization\nphase L3 agent finds already configured own net namespaces. Existance\nof such net namespace is a good proof that there is a network\nconfiguration existing so host wasn\u0027t rebooted so most probably it is\njust agent restart.\n\n[1] https://bugs.launchpad.net/neutron/+bug/1597461\n\nCloses-Bug: #1959151\nChange-Id: Id9c906b2d141c3bedd80fb5f868190f8a4b66f54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c37880135fe15c98a19aa02e8237fa2e6315d9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c37880135fe15c98a19aa02e8237fa2e6315d9c"}]},"branch":"refs/heads/master"},"f430cd00725f8303f5313cb7784c9aed4b982e62":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2022-03-01 14:27:42.000000000","uploader":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"ref":"refs/changes/45/826545/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/826545/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/826545/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/826545/10"}}},"commit":{"parents":[{"commit":"473f4db1d666a93c8483b4013eed8ce836825e88","subject":"Merge \"Local IP internal documentation and release note\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/473f4db1d666a93c8483b4013eed8ce836825e88"}]}],"author":{"name":"labedz","email":"krzysztof.tomaszewski@ovhcloud.com","date":"2022-01-26 23:13:40.000000000","tz":60},"committer":{"name":"Krzysztof Tomaszewski","email":"krzysztof.tomaszewski@ovhcloud.com","date":"2022-03-01 14:27:42.000000000","tz":0},"subject":"Don\u0027t set HA ports down while L3 agent restart.","message":"Don\u0027t set HA ports down while L3 agent restart.\n\nBecause of the fix for bug[1] and issue with linux_utils\nget_process_count_by_name() L3 agent puts all it\u0027s HA ports down\nduring initialization phase. Unfortunately such operation can break\nalready working L3 communication. Rewiring ha-* port from down state to\nup can takes few seconds and some VRRP packages could be lost then.\nThat triggers keepalived on other node so router HA state change\nmay be triggered.\n\nThis change prevents putting HA ports down when during initialization\nphase L3 agent finds already configured own net namespaces. Existance\nof such net namespace is a good proof that there is a network\nconfiguration existing so host wasn\u0027t rebooted so most probably it is\njust agent restart.\n\n[1] https://bugs.launchpad.net/neutron/+bug/1597461\n\nCloses-Bug: #1959151\nChange-Id: Id9c906b2d141c3bedd80fb5f868190f8a4b66f54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f430cd00725f8303f5313cb7784c9aed4b982e62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f430cd00725f8303f5313cb7784c9aed4b982e62"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
