)]}'
{"id":"openstack%2Fneutron~827488","triplet_id":"openstack%2Fneutron~master~I401d121b69bcf479cba5b9d151bcd65dabf5aea6","project":"openstack/neutron","branch":"master","topic":"bug/1959176","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-06-15 14:59:24.000000000","reason":"Change was abandoned"},"23567":{"account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"last_update":"2022-03-21 10:53:49.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I401d121b69bcf479cba5b9d151bcd65dabf5aea6","subject":"Decrease OS_TEST_TIMEOUT to 240 for dsvm-fullstack","status":"ABANDONED","created":"2022-02-02 12:21:26.000000000","updated":"2022-06-15 14:59:24.000000000","total_comment_count":12,"unresolved_comment_count":2,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"16ac1410d6f8f182b9f26d9ef9b93d01ea9d117e","_number":827488,"virtual_id_number":827488,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-03-21 12:47:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-03-21 13:00:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2022-02-02 12:32:22.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-02-02 12:34:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-02-02 12:48:57.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-02-02 13:26:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-02 14:19:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-02 14:26:47.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-02-03 07:55:05.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2022-02-16 18:32:35.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-17 10:10:33.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-03-16 23:25:59.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"b6873e4645b4f34670d24684d6c9aba7837363a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-02 12:21:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"071a317b15df95cbd3e8d29f242abd655897bb1f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-02 12:32:22.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0314d905f5af26d355f2c3a1c059acc34bfa0b6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-02-02 12:34:58.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"96f1b86ff37c9f36b6f49b174a649082785a04ce","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-02-02 12:48:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eca62c018716917a910baf73dc372abb870eb815","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 13:26:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/75c4c455e068458889e6a1bb14ff0d6a : SUCCESS in 51m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0c5294e568a14e728cbfb58fed679372 : SUCCESS in 52m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"72b88ac4d9e7ea05683d51d633a7af5c1c7a2818","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 14:19:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c4b5f1098cf43c2ac9ce6e428ef2960 : SUCCESS in 37m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1067616ef9649dc9a842959283d2cae : SUCCESS in 30m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a84fd3165da4b60823c204ac4bfd794 : SUCCESS in 14m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/334527550ca14d6faa7578f2b7886121 : SUCCESS in 32m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66e3914d0fc34163b0211ba3f4c1dd95 : SUCCESS in 25m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7700a54ea2542de9132d24cd9f0bd79 : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6f920a3dae5487991387069d738af21 : SUCCESS in 10m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef4b74f1d455442e8308ed5106c90dfa : SUCCESS in 46m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8361ce8c96544f1bb67c8addef11fdea : FAILURE in 1h 45m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8b8c9a02bf54f1a911a600f29b8c2f6 : SUCCESS in 22m 40s","accounts_in_message":[],"_revision_number":1},{"id":"5e5e95ddadff49bf1064319962d3ffcb20165c81","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-02 14:26:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/827488/1 : SUCCESS in 1h 55m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b02f0aefef8575c74228e9624894250cdc47f9a0","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2022-02-03 07:55:05.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b47d2515921272322f94e9f8a4b83a8531f2b1da","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2022-02-03 07:55:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6ae5a59cf2cc7aa5addedded9cddd1fe80ff5a25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 08:43:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f177ca4bc8254c5191191f2afbe8fdef : SUCCESS in 45m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/26e1e55c612542398beb13d70e2f316d : SUCCESS in 46m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"73c2bf024aa17afb342fb5e25a25c36d150fe482","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 09:07:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/090d638e246648eda0581b8760903f0a : SUCCESS in 34m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f77a9cfa3fa6494d82aa6895a386bf4d : SUCCESS in 32m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dc88a18ca7c453a9ddc8670cf3169ab : SUCCESS in 11m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5ac1cfd06054f62af7c3f5a01caa903 : SUCCESS in 32m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01458c6a734947b9990776f38225773d : SUCCESS in 22m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87749e2c44604f1ba26ab1eef085d5da : SUCCESS in 10m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14ad34780372483fb98b37fc7145210b : SUCCESS in 9m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3c0241be950a41aba9b4d67f638135c5 : SUCCESS in 49m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3f3226fc06ef46a487cb15df6aca70cc : FAILURE in 1h 10m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/650fc22d20db447eab5db06a0b448531 : SUCCESS in 22m 17s","accounts_in_message":[],"_revision_number":1},{"id":"6faa2f3a7eec87dadb549e078d75556a5cab7442","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-03 09:29:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/827488/1 : SUCCESS in 1h 33m 23s","accounts_in_message":[],"_revision_number":1},{"id":"a8d8b7498138aaa6566304892736fcf484949c71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-16 17:36:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"31ec19ad170e53305b47b708736971118a9d9658","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-16 17:48:42.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"2ed04ab766f5109913226cdae10acd5566e990dc","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-16 17:50:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e8a920dfc93a9c0fd5060f48c86c1bfbacfd6db5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-02-16 18:32:35.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2022/Arista_Tempest_Testrefs/changes/88/827488/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"51790555c22d18fbc85a1a662aa6889c088f1932","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 18:39:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee512e9af09b48029350b16f43d0c549 : SUCCESS in 43m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/30b91d6e8fe34e3eb14264212f20c1c4 : SUCCESS in 43m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"156e604822ff3194c6caefb658a376cf904d83a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 19:17:09.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/776efc388da548c2a538f1d2ff591935 : SUCCESS in 1h 21m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/21af3ec0aa0a47679662b87ea4231593 : SUCCESS in 1h 19m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/91f4146edfb44911b659b758a328d764 : SUCCESS in 44m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e8105a81b4e64a80a9d17332d42a358f : SUCCESS in 1h 01m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44e011268dc541f1912c87802a1e9531 : SUCCESS in 40m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d93fcc73f4cb44d4a4fff3fa82ee7265 : SUCCESS in 33m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb5234809b95499090effc87d39f7f9c : FAILURE in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61980d76c6d048b1ae86e646c2ec52fb : SUCCESS in 30m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b86f9cf6580340dcb9a5dce0a0543831 : SUCCESS in 23m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/228cbc30aee345eea1f71e75d1227c9f : SUCCESS in 8m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/784a80efaf6c4665bd70edb918f61871 : SUCCESS in 9m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d5e8186b4d546948dd3ba14e8860c0a : SUCCESS in 47m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/653ba2b7b60c4a889d8f0fa6134237f8 : SUCCESS in 1h 13m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f389336bd117458cbd813d2d6435df65 : SUCCESS in 1h 01m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bc820b58368947598fd591bcdb9883f2 : SUCCESS in 57m 17s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dc3f01e30d9a460fbe0aa7f526852ee2 : SUCCESS in 42m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c0881d38232143c0b8ecc4ecb1373001 : SUCCESS in 1h 06m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f31ac9e3636e40c58754c85225035686 : SUCCESS in 1h 12m 20s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c15a6697b6a4facb5cc169a9a7a7d9d : SUCCESS in 37m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/91c9cba520d241ada46e2bf1da71da43 : SUCCESS in 1h 02m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e2cc752dad745ab4305d2b59fe7dc9a06f617263","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-02-16 19:34:05.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2022/Arista_Tempest_Testrefs/changes/88/827488/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"936ffdbeab96364c2cf788fd58b9cee6e1b74765","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-16 19:37:00.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/827488/3 : FAILURE in 1h 47m 34s","accounts_in_message":[],"_revision_number":3},{"id":"9e6d7acb98970f7be20f7985f9f65a0d0505a35d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-17 07:35:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7752cdf3a98051c389df1af6751be2c9ac2b0cc5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-02-17 08:35:04.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2022/Arista_Tempest_Testrefs/changes/88/827488/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"f072128b64fb34c8934bd3e63f036179aa70c1e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 09:04:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a654bdb62fd840d0855dba32a7cce7ae : SUCCESS in 56m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a75c2e91facb4b1394d5c58d99e1dd6e : SUCCESS in 50m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e8d37b1f5bd48ec58f86cda5ad8e11febb8b43d2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-17 09:06:58.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/827488/4 : FAILURE in 1h 31m 04s","accounts_in_message":[],"_revision_number":4},{"id":"86f642e60366af531334c0f71a7d486d408811c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 09:19:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8f481dde574f46e88de247047e2c7daf : SUCCESS in 1h 41m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/368d905a0e3640f0a1a192cd923779e6 : SUCCESS in 1h 14m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e7bf53e187644275bc02461ec0db1ec9 : SUCCESS in 1h 03m 12s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/bf4952d4b9f84ca2afbbd5a1e1846f90 : SUCCESS in 1h 07m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb4e2797e675425489e34090e5a1cdb3 : SUCCESS in 41m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38927e179c0d4317a50113e00fa57c25 : SUCCESS in 34m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eafef68f710c4c7b89f9771931a79b67 : SUCCESS in 14m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89957ae0e0854891935cd35b30320d6a : SUCCESS in 30m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73dabe6c5c1048d6b0971694420a5890 : SUCCESS in 29m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bee2768a112462097f2b19beed1d687 : SUCCESS in 9m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5be08b39853414d94bb6f65bfd8dcf8 : SUCCESS in 6m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7acaf03609f34bcc80f2139b1b7a7f83 : SUCCESS in 54m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1885d066fa0e431abb15a2323b7d3e0e : FAILURE in 1h 26m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5d946646003749ddab6233ea50ac497d : SUCCESS in 1h 22m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/be12d88f2c39449eb1f461dd794a606e : SUCCESS in 58m 36s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/02da1bedb1994c4a96f09be461196308 : SUCCESS in 57m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e0c156b0dddb47fca0babcb30b01d5b9 : SUCCESS in 1h 16m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/929c78b8b10a40a5b0c05e9a293fd7e8 : SUCCESS in 1h 05m 17s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7be10c3f07f84301a87dc72dd268d6e3 : SUCCESS in 55m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/005b4c5bf8ae4bcfab7a999c2adc5f72 : SUCCESS in 1h 05m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c56d651414c1c1e1edefea7d5da266bcfff619ca","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-17 09:35:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"40728f6cde06f43609bf8192a83f55524ba285c1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-02-17 10:10:33.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"331e5a652d6ff5c9898542d425b3713a1659c32a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 10:58:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7af255cce2cc46ea9f04542556bbeb32 : SUCCESS in 58m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/260f63216fa448738c242a63b318fbd3 : SUCCESS in 1h 08m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c162b066308e4e51b9a15444a29ddabc : SUCCESS in 44m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1b299e8f35954efca56fcd9b00b7bdf1 : SUCCESS in 1h 01m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c92485f1992647379cb23620c371d7f6 : SUCCESS in 37m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88d752cf7a92467fbc3a7be4032a0c52 : SUCCESS in 42m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dad0e56c7224bdba97e3ada4c046df9 : SUCCESS in 14m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/554e1208e570454aa9f96d84a4f7f3ea : SUCCESS in 29m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9dfd30e475c94dea9a38a0cad452796e : SUCCESS in 25m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9ebcee93c734ec3be7d5b2fcc8d6cee : SUCCESS in 9m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d33a732d53ea41b1b2fcf041ba307648 : SUCCESS in 9m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c3eaf08761cb4581a86630370fb0112c : SUCCESS in 59m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0eba16d337794b0e9fed1491bf5c8a07 : SUCCESS in 1h 21m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b990a70ce8f147c88055d421c7d69854 : SUCCESS in 1h 06m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2ddad72a787a41558c7f650346788819 : SUCCESS in 55m 44s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c3710c8871124baaa323386c2055b9c4 : SUCCESS in 58m 42s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d42b65c4d0ad4e7ba2772260675633aa : SUCCESS in 1h 18m 48s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1515140be5e4454181dd694c41551626 : SUCCESS in 1h 03m 25s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5be5259a12b44b6cb0de619dc1c2770f : SUCCESS in 47m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/50c1cc0702a64c58adec0e58dea1b687 : SUCCESS in 1h 14m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"387dcaefcf777f1583589c25e3e2c4a026b8595c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-17 11:33:56.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/827488/4 : FAILURE in 1h 58m 15s","accounts_in_message":[],"_revision_number":4},{"id":"2cd40c2132fe2326dc0d6b95a3cef068b63dad20","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 15:49:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e17cdf665aed45b9a30869aa261b2c91 : SUCCESS in 1h 03m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e1390eb571824b6988e85443adb52596 : SUCCESS in 1h 02m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"80dab702df3132496c4ba8cb2a7e6543ee4ce86b","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2022-03-08 07:56:41.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0ba7bbd46259033fcf79b8eb9882c82179af57ba","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-03-10 07:12:51.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"cafceedfee6b6726ab9dfaa226060a99a641c2fb","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-03-16 23:25:59.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2c9586093c3abcd410f5e0941727218371f61440","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 23:26:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9b57f0290b864b3a4333b6921823d0067d8a5a37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 01:25:47.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b8f1c5de3ace4995bba74d81dbfa4ea9 : SUCCESS in 1h 02m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/50b733f782fd497e98b500c72274803b : SUCCESS in 1h 04m 50s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b6420452de5c4c26b047dfda6af2d434 : SUCCESS in 52m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2d3baefbe43945e2b41e570848045664 : SUCCESS in 57m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e07c00b5c9f74b7caa2a64c525728f2a : SUCCESS in 30m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6a80f4e9f0d48809920e8b61f52e9b7 : SUCCESS in 14m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cc765985a504547b22ec19e3a2fa8fb : SUCCESS in 22m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81b42e837b594c2c8d9f8a2bd25e1a69 : SUCCESS in 25m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b8c302969724f16ae733526fe425246 : SUCCESS in 11m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/764cb8d0e180470bbeed9f430d79e638 : SUCCESS in 10m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c73a050989b547738dbc945f66adfd9d : SUCCESS in 47m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/26fe45284d5b4798a43fccee1031898d : FAILURE in 1h 15m 10s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/74780ee141c2465986190c12d620df23 : SUCCESS in 1h 08m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/152e711862e747fcafea9775eeab1b4b : SUCCESS in 51m 50s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f5ad49b283134e3199370f5a0d7d9f26 : SUCCESS in 52m 22s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b32b15179934f258611aa332c43cbde : SUCCESS in 49m 35s","accounts_in_message":[],"_revision_number":4},{"id":"cb9a166f5aad7df4f118b685d20e3acea94097f4","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-03-17 05:26:23.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4ace9be3fa7a5dadddfd290b05bc0e9806d1133a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-03-17 13:01:21.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"3831d0a003c4089a8266eea2c75bc0cd23749077","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-03-17 13:11:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5fb64a33b40cdbdcee6baa1787f707c8f726bcef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-03-18 10:25:01.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"fbed680fa6c4bb6f9101fe5fdbe885c6c3bf3668","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-18 11:19:29.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/88/827488/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"f6fb642feb727ee6c1ed226b8067255aa971341e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 11:21:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/586cb0c11c334b3c8f91920cbd866fe2 : SUCCESS in 54m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4ac52fba15ff45fcafd542d4ae1ffaf9 : SUCCESS in 53m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"acb28589fbe478434901f0cd9cb71e682646577c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 12:44:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/89dce549f9594b338034f69ecc3deaf9 : SUCCESS in 41m 16s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9afe5088e7ef4f0290537eed82fdb8a5 : SUCCESS in 1h 08m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a42b810f64d2496c9ec3bc4db8203381 : SUCCESS in 1h 53m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3e541a9fea5141db96bbff164707b7da : SUCCESS in 52m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/78152d43352c4fcd8c622da097c04a80 : SUCCESS in 51m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1d0522089dd9403f84c6f9c51d566c6c : SUCCESS in 28m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad8f121b37cb43ba914205ce54031625 : SUCCESS in 34m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72df9de2735f40e18483b21c5a14fa81 : SUCCESS in 30m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8094b12b42154c7e80b03524dd26a7bb : SUCCESS in 16m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f47078e95a4642c5a8733db32abe0986 : SUCCESS in 23m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa44659e6ad9406aaa5c97a50c30dc2c : SUCCESS in 27m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb8e62f037354a748262843959fadae4 : SUCCESS in 30m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5f36380ced14f1f8381e6b4e6c9dba1 : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eed29cb26a17405dbc5948511754b291 : SUCCESS in 12m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6c14a8aab4e049c6a292076913905662 : FAILURE in 1h 10m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6ae2a119e427406d9da1bfd75441176a : FAILURE in 1h 36m 00s\n- neutron-fullstack-with-uwsgi_2 https://zuul.opendev.org/t/openstack/build/8b303e401faa42a89a763530f6c154a2 : FAILURE in 1h 29m 01s\n- neutron-fullstack-with-uwsgi_3 https://zuul.opendev.org/t/openstack/build/c8afbb7d64814530a0fcc6f423ffb0d9 : FAILURE in 2h 08m 18s\n- neutron-fullstack-with-uwsgi_4 https://zuul.opendev.org/t/openstack/build/5042257434dd4e99baf45d99df8083e6 : FAILURE in 1h 19m 06s\n- neutron-fullstack-with-uwsgi_5 https://zuul.opendev.org/t/openstack/build/db23ca8ed75b4cdda362c99e1fe78284 : FAILURE in 2h 09m 42s\n- neutron-fullstack-with-uwsgi_6 https://zuul.opendev.org/t/openstack/build/cff79c6c1c564171b49001883d39c2e7 : FAILURE in 1h 41m 32s\n- neutron-fullstack-with-uwsgi_7 https://zuul.opendev.org/t/openstack/build/8c23386656a8432ea5e8b7c4b94fae6a : FAILURE in 1h 36m 47s\n- neutron-fullstack-with-uwsgi_8 https://zuul.opendev.org/t/openstack/build/e4457852e86f436287f150fdbd530ebd : FAILURE in 1h 43m 12s\n- neutron-fullstack-with-uwsgi_9 https://zuul.opendev.org/t/openstack/build/db316c20134d4f169415469b53ec5664 : FAILURE in 2h 11m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a404a30d242b457f83b9f311ac41faed : SUCCESS in 1h 08m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/06d5e9722d57444db474fb8fb9f3c42e : SUCCESS in 1h 15m 59s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/18e1f563e7a8495ca323496ae62cc100 : SUCCESS in 55m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/716a2d5359444f609a91fd2ca670893f : SUCCESS in 1h 10m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9753565f78ec462281dec2b113c14613 : SUCCESS in 1h 08m 54s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86636be6aaf14d61a44610fef2ab8a69 : SUCCESS in 56m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/51a34b35b4aa4c83ae108a54d1786085 : SUCCESS in 1h 27m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"25d34ebdb483e178d3c756d44091de63fa2d7a8c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-03-18 12:47:57.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8e591ce41d5c72520006aa58da60b5efa79542a8","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-18 13:15:43.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/827488/5 : SUCCESS in 2h 49m 59s","accounts_in_message":[],"_revision_number":5},{"id":"d7731d95723c9d814689bd3675c8b8378b21978d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-03-18 16:38:16.000000000","message":"Patch Set 5: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"65aee81b0917885d0af06d267c9e32d0a4883869","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 18:32:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/714de9c045ff43fd83765a791e6765a5 : SUCCESS in 45m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a2e61abc379a4118871f88a5dbbf2936 : SUCCESS in 47m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1da64f4e7b5e985a5a6813219cf2f04d6c2d7b35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 18:38:55.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a148469063c941f68d05c33fa3bb1d20 : SUCCESS in 1h 02m 15s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/35dd6e42ca294751a1186b196a0c18fc : SUCCESS in 1h 30m 46s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/924d3ab90f964f0eb4492b04152c6224 : SUCCESS in 1h 34m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/92d1b0fad96e4223973d1de3d0277eee : SUCCESS in 1h 23m 43s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2a88079a44064a53ab129f1384f7bb64 : SUCCESS in 1h 36m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/357f0678aac54c4aa05aebde78ae315d : SUCCESS in 39m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8338a5ac88ea4f7697ccf0e032e0048c : SUCCESS in 48m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c2ba78b53224f22baab4017f3ce24c0 : SUCCESS in 36m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7ecc131aaba4a9b8a3c0f6282fbedae : SUCCESS in 15m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9d04c4015c7454eb32d9aac165f3d91 : SUCCESS in 26m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62689c4af94d47adbed9fb32cfd060a3 : SUCCESS in 29m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90e931bf8cb646b1b01641e2ea2b9cfb : SUCCESS in 32m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c7bbda6c1ff4b15835914d1681f7c09 : SUCCESS in 10m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e12fa3c1c658488bbc8869ff6a20d97e : SUCCESS in 10m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/622250b20cf945129e30402d858fce53 : SUCCESS in 55m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1303e14960544969a63e54506ea905bd : FAILURE in 1h 44m 23s\n- neutron-fullstack-with-uwsgi_2 https://zuul.opendev.org/t/openstack/build/cbda7f541ec24ad4aa8da9fadd0ceb06 : FAILURE in 1h 39m 54s\n- neutron-fullstack-with-uwsgi_3 https://zuul.opendev.org/t/openstack/build/f8715edefdab41a082bed30274af4df5 : FAILURE in 1h 44m 27s\n- neutron-fullstack-with-uwsgi_4 https://zuul.opendev.org/t/openstack/build/867b913592c24779be8fbdae5125cd48 : FAILURE in 1h 45m 35s\n- neutron-fullstack-with-uwsgi_5 https://zuul.opendev.org/t/openstack/build/c3f58dcca0e442e29f8388af91447d95 : FAILURE in 1h 41m 13s\n- neutron-fullstack-with-uwsgi_6 https://zuul.opendev.org/t/openstack/build/02a48a773187430b88293b389c86d16f : FAILURE in 1h 43m 41s\n- neutron-fullstack-with-uwsgi_7 https://zuul.opendev.org/t/openstack/build/2bdf048d8f8a4730b3b223c8988cb040 : FAILURE in 1h 42m 51s\n- neutron-fullstack-with-uwsgi_8 https://zuul.opendev.org/t/openstack/build/cbbb24b72237479c9b5f8df49026de73 : FAILURE in 1h 43m 50s\n- neutron-fullstack-with-uwsgi_9 https://zuul.opendev.org/t/openstack/build/638f9517dd86493bae0ed9d0504994f5 : FAILURE in 1h 39m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b94674b22bc947e7baf0fc1a8f45fd8e : SUCCESS in 1h 07m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/18723d6d480f4ed58e19c6875e1d405c : SUCCESS in 1h 00m 51s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/78020ec9fbab4c659e13e06b2c5bd5b5 : SUCCESS in 53m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/abe531682f094701983d2bb9c93fb2a4 : SUCCESS in 1h 19m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fe7733f323144a7596f3e31bf2190eec : SUCCESS in 1h 16m 54s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/defd013045fb4592a5c6397c67fdc913 : SUCCESS in 54m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/10f73eec0dc7401eae458197adeff3d5 : SUCCESS in 1h 07m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"af76c21ef4c6ab49888a74d3d1ab3831e386d43f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-18 19:18:19.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/827488/5 : SUCCESS in 2h 39m 29s","accounts_in_message":[],"_revision_number":5},{"id":"bb2ae0128de3473b6d2e87c8047d1dc9da508861","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-03-21 10:53:42.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"99406d09baec991aeeba1a59f99ffde028444f06","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-03-21 10:53:49.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":6},{"id":"9b2fa1565b54d2c0c0448fb121c8b151ab04b80b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-21 12:31:18.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/827488/6 : SUCCESS in 1h 36m 50s","accounts_in_message":[],"_revision_number":6},{"id":"3db7c9709d36ba9080579ac0b1a201ab22f1fbdc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 12:37:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/76f8ae8845134311979fd7840de87bca : FAILURE in 1h 19m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/89ba032a9dd84778aa15ceb765e2ebad : FAILURE in 1h 19m 27s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ae7424d4c9f32b111727e8f549060c8beb499ca6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-21 12:47:24.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/88/827488/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"13e22bb010bf935b1c9946b5b79401dbc0d12aed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 13:00:47.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6c3c7d185f8948b1bac6ea9998dd7423 : SUCCESS in 46m 55s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/32f759921abf45b0a5f5cf8beb0f09d3 : FAILURE in 2h 01m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b65cf9037cea43fcb28ae18e78ae74fe : SUCCESS in 1h 07m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9a8ad81bfa7a49ca9b7358267b611509 : SUCCESS in 49m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ea7f06b234e14ef3a81792107bed447b : SUCCESS in 47m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dc2a2b8560a84d56af2c54bb9beccbc2 : SUCCESS in 29m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0320200c92e448db9c8a7161c4036083 : SUCCESS in 29m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a1ee73fad3e41d885824fc56dcb4793 : SUCCESS in 30m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa0cad402cc5452fb3020d5c7ea45c36 : SUCCESS in 12m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6082e655f9b48e1a40c050af50867e9 : SUCCESS in 22m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af43193fe8e74277a52ce017d4f721d6 : SUCCESS in 22m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7f845ca9f7e47f9bf93869a2cbd28db : SUCCESS in 28m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1131b26609d6468d86de173ae1dffdc2 : SUCCESS in 11m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47b6646f5e65436d86a211cde4939537 : SUCCESS in 10m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6debabc47d7c4303b69f70835fe55c85 : SUCCESS in 48m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7695b2b95b5f4ca3a503641687ab2c7a : FAILURE in 1h 36m 52s\n- neutron-fullstack-with-uwsgi_2 https://zuul.opendev.org/t/openstack/build/37c26dbbaa92415492fdb3b4213fc1b7 : FAILURE in 1h 45m 16s\n- neutron-fullstack-with-uwsgi_3 https://zuul.opendev.org/t/openstack/build/6d10326a0ad746db8412dd21a3c2e7b5 : FAILURE in 1h 59m 27s\n- neutron-fullstack-with-uwsgi_4 https://zuul.opendev.org/t/openstack/build/6834127720bc4409aae2b5e672f7606e : FAILURE in 1h 56m 26s\n- neutron-fullstack-with-uwsgi_5 https://zuul.opendev.org/t/openstack/build/baebd7bd63b14485a7d363ad2a8e079b : FAILURE in 1h 50m 52s\n- neutron-fullstack-with-uwsgi_6 https://zuul.opendev.org/t/openstack/build/f60df2af31cc4df6b3da5ec3e33fd165 : FAILURE in 1h 52m 30s\n- neutron-fullstack-with-uwsgi_7 https://zuul.opendev.org/t/openstack/build/1dce17339d6840b4801a3421460cdd1c : FAILURE in 1h 57m 05s\n- neutron-fullstack-with-uwsgi_8 https://zuul.opendev.org/t/openstack/build/9d3a6ceadd3044f79e1bcb4f7f9ce0b3 : FAILURE in 1h 55m 39s\n- neutron-fullstack-with-uwsgi_9 https://zuul.opendev.org/t/openstack/build/b9a6e0888b504daa8fa51583f2a5eec4 : FAILURE in 1h 44m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/929072d3559c43668e5818ad9d90b540 : SUCCESS in 1h 25m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0f9276fac6b0410d80d2e43e3739f149 : SUCCESS in 1h 00m 15s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/918bca8302244682b7a99c9f4d873e95 : SUCCESS in 1h 05m 49s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0178e7920469402b8b07ecc363360a17 : SUCCESS in 1h 21m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/517e5003842146648cb8bc7f9589f80c : SUCCESS in 1h 17m 44s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2509952f50604ba594a32459ab969056 : SUCCESS in 1h 06m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d54d63fc2c7c4ce19a0a3578ddac7330 : SUCCESS in 1h 17m 17s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"16ac1410d6f8f182b9f26d9ef9b93d01ea9d117e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-06-15 14:59:24.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c546edc0fb3d346215405bef6a3c8d910d8e67c0","revisions":{"a2b9f07ca4a6e89aeaffbc503584e2fa1845d86d":{"kind":"REWORK","_number":1,"created":"2022-02-02 12:21:26.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/88/827488/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/827488/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/827488/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/827488/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/827488/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/827488/1"}}},"commit":{"parents":[{"commit":"e7b70521d0e230143a80974e7e4795a2acafcc9b","subject":"Merge \"Use elevated context to update router\u0027s external gateway\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e7b70521d0e230143a80974e7e4795a2acafcc9b"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-02-02 12:16:25.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-02-02 12:16:25.000000000","tz":60},"subject":"Remove OS_TEST_TIMEOUT\u003d600 from dsvm-fullstack","message":"Remove OS_TEST_TIMEOUT\u003d600 from dsvm-fullstack\n\nOS_TEST_TIMEOUT was set to 600 due to old db teardown issues (see [0]).\nThose are not affecting fullstack now, but the long timeout causes\njob timeouts, it\u0027s time to use the default 180s for test timeout.\n\n[0]: https://bugs.launchpad.net/neutron/+bug/1541742\n\nChange-Id: I401d121b69bcf479cba5b9d151bcd65dabf5aea6\nRelated-Bug: #1959176\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a2b9f07ca4a6e89aeaffbc503584e2fa1845d86d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a2b9f07ca4a6e89aeaffbc503584e2fa1845d86d"}]},"branch":"refs/heads/master"},"066f812b4a46a9d232f061a8b05d2e48015388aa":{"kind":"REWORK","_number":2,"created":"2022-02-16 17:36:38.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/88/827488/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/827488/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/827488/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/827488/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/827488/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/827488/2"}}},"commit":{"parents":[{"commit":"f324ddf769b9d9b68530a191dcf2e419ff2c25d4","subject":"Merge \"Make sure \"dead vlan\" ports cannot transmit packets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f324ddf769b9d9b68530a191dcf2e419ff2c25d4"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-02-02 12:16:25.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-02-16 17:35:35.000000000","tz":60},"subject":"Remove OS_TEST_TIMEOUT\u003d600 from dsvm-fullstack","message":"Remove OS_TEST_TIMEOUT\u003d600 from dsvm-fullstack\n\nOS_TEST_TIMEOUT was set to 600 due to old db teardown issues (see [0]).\nThose are not affecting fullstack now, but the long timeout causes\njob timeouts, it\u0027s time to use the default 180s for test timeout.\n\nFullstack test test_ha_router_restart_agents_no_packet_lost fails with\nsmaller timeout, a possible fix is to add fixtures.TimeoutException\nto _kill_process_and_wait, as during fullstack execution processes are\nstarted as Fixtures.\n\n[0]: https://bugs.launchpad.net/neutron/+bug/1541742\n\nChange-Id: I401d121b69bcf479cba5b9d151bcd65dabf5aea6\nRelated-Bug: #1959176\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/066f812b4a46a9d232f061a8b05d2e48015388aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/066f812b4a46a9d232f061a8b05d2e48015388aa"}]},"branch":"refs/heads/master"},"39a8ec24c5a41b69acf56e02a41051f9680813db":{"kind":"REWORK","_number":3,"created":"2022-02-16 17:48:42.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/88/827488/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/827488/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/827488/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/827488/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/827488/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/827488/3"}}},"commit":{"parents":[{"commit":"f324ddf769b9d9b68530a191dcf2e419ff2c25d4","subject":"Merge \"Make sure \"dead vlan\" ports cannot transmit packets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f324ddf769b9d9b68530a191dcf2e419ff2c25d4"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-02-02 12:16:25.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2022-02-16 17:48:42.000000000","tz":0},"subject":"Remove OS_TEST_TIMEOUT\u003d600 from dsvm-fullstack","message":"Remove OS_TEST_TIMEOUT\u003d600 from dsvm-fullstack\n\nOS_TEST_TIMEOUT was set to 600 due to old db teardown issues (see [0]).\nThose are not affecting fullstack now, but the long timeout causes\njob timeouts, it\u0027s time to use the default 180s for test timeout.\n\nFullstack test test_ha_router_restart_agents_no_packet_lost fails with\nsmaller timeout, a possible fix is to add fixtures.TimeoutException\nto _kill_process_and_wait, as during fullstack execution processes are\nstarted as Fixtures.\n\n[0]: https://bugs.launchpad.net/neutron/+bug/1541742\n\nChange-Id: I401d121b69bcf479cba5b9d151bcd65dabf5aea6\nRelated-Bug: #1959176\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/39a8ec24c5a41b69acf56e02a41051f9680813db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/39a8ec24c5a41b69acf56e02a41051f9680813db"}]},"branch":"refs/heads/master"},"abc03d755ef9e4a19b159793384f3d3d3102e7c0":{"kind":"REWORK","_number":4,"created":"2022-02-17 07:35:11.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/88/827488/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/827488/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/827488/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/827488/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/827488/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/827488/4"}}},"commit":{"parents":[{"commit":"f324ddf769b9d9b68530a191dcf2e419ff2c25d4","subject":"Merge \"Make sure \"dead vlan\" ports cannot transmit packets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f324ddf769b9d9b68530a191dcf2e419ff2c25d4"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-02-02 12:16:25.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-02-17 07:34:08.000000000","tz":60},"subject":"Remove OS_TEST_TIMEOUT\u003d600 from dsvm-fullstack","message":"Remove OS_TEST_TIMEOUT\u003d600 from dsvm-fullstack\n\nOS_TEST_TIMEOUT was set to 600 due to old db teardown issues (see [0]).\nThose are not affecting fullstack now, but the long timeout causes\njob timeouts, it\u0027s time to use the default 180s for test timeout.\n\nFullstack test test_ha_router_restart_agents_no_packet_lost fails with\nsmaller timeout, a possible fix is to add fixtures.TimeoutException\nto _kill_process_and_wait, as during fullstack execution processes are\nstarted as Fixtures.\n\n[0]: https://bugs.launchpad.net/neutron/+bug/1541742\n\nChange-Id: I401d121b69bcf479cba5b9d151bcd65dabf5aea6\nRelated-Bug: #1959176\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/abc03d755ef9e4a19b159793384f3d3d3102e7c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/abc03d755ef9e4a19b159793384f3d3d3102e7c0"}]},"branch":"refs/heads/master"},"b6b363330eca93602a9776f8e0d2182eaed37c3f":{"kind":"REWORK","_number":5,"created":"2022-03-18 10:25:01.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/88/827488/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/827488/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/827488/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/827488/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/827488/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/827488/5"}}},"commit":{"parents":[{"commit":"f962fff7f9155cdb5aed0211837f14efd1d57d38","subject":"Merge \"[OVN] Remove maintenance \"check_for_localnet_legacy_port_name\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f962fff7f9155cdb5aed0211837f14efd1d57d38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-02-02 12:16:25.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-03-18 10:07:04.000000000","tz":60},"subject":"Decrease OS_TEST_TIMEOUT to 240 for dsvm-fullstack","message":"Decrease OS_TEST_TIMEOUT to 240 for dsvm-fullstack\n\nOS_TEST_TIMEOUT was set to 600 due to old db teardown issues (see [0]).\nThose are not affecting fullstack now, but the long timeout causes\njob timeouts. test_ha_router_restart_agents_no_packet_lost execution\ntime is around 180sec, so the OS_TEST_TIMEOUT (180sec) is short for\nfullstack execution, here I changed it to 240.\n\nFullstack test test_ha_router_restart_agents_no_packet_lost fails with\nsmaller timeout, a possible fix is to add fixtures.TimeoutException\nto _kill_process_and_wait, as during fullstack execution processes are\nstarted as Fixtures.\n\n[0]: https://bugs.launchpad.net/neutron/+bug/1541742\n\nChange-Id: I401d121b69bcf479cba5b9d151bcd65dabf5aea6\nRelated-Bug: #1959176\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6b363330eca93602a9776f8e0d2182eaed37c3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6b363330eca93602a9776f8e0d2182eaed37c3f"}]},"branch":"refs/heads/master"},"c546edc0fb3d346215405bef6a3c8d910d8e67c0":{"kind":"REWORK","_number":6,"created":"2022-03-21 10:53:42.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/88/827488/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/827488/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/827488/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/827488/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/827488/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/827488/6"}}},"commit":{"parents":[{"commit":"f962fff7f9155cdb5aed0211837f14efd1d57d38","subject":"Merge \"[OVN] Remove maintenance \"check_for_localnet_legacy_port_name\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f962fff7f9155cdb5aed0211837f14efd1d57d38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-02-02 12:16:25.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2022-03-21 10:53:42.000000000","tz":0},"subject":"Decrease OS_TEST_TIMEOUT to 240 for dsvm-fullstack","message":"Decrease OS_TEST_TIMEOUT to 240 for dsvm-fullstack\n\nOS_TEST_TIMEOUT was set to 600 due to old db teardown issues (see [0]).\nThose are not affecting fullstack now, but the long timeout causes\njob timeouts. test_ha_router_restart_agents_no_packet_lost execution\ntime is around 180sec, so the OS_TEST_TIMEOUT (180sec) is short for\nfullstack execution, here I changed it to 300.\n\nFullstack test test_ha_router_restart_agents_no_packet_lost fails with\nsmaller timeout, a possible fix is to add fixtures.TimeoutException\nto _kill_process_and_wait, as during fullstack execution processes are\nstarted as Fixtures.\n\n[0]: https://bugs.launchpad.net/neutron/+bug/1541742\n\nChange-Id: I401d121b69bcf479cba5b9d151bcd65dabf5aea6\nRelated-Bug: #1959176\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c546edc0fb3d346215405bef6a3c8d910d8e67c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c546edc0fb3d346215405bef6a3c8d910d8e67c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
