)]}'
{"id":"openstack%2Fneutron~827615","triplet_id":"openstack%2Fneutron~master~Ida7c3ac974058769eb7ff901715e9b17a998bf79","project":"openstack/neutron","branch":"master","topic":"neutron-ci-improvements","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-02-03 12:45:29.000000000","reason":"Rodolfo Alonso replied on the change"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-02-03 15:53:05.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2022-02-03 15:53:05.000000000","reason":"Change was submitted"},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2022-02-03 15:53:05.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-02-03 15:53:05.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-02-03 12:38:16.000000000","reason":"Lajos Katona replied on the change"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2022-02-03 13:47:54.000000000","reason":"Oleg Bondarev replied on the change"}},"hashtags":[],"change_id":"Ida7c3ac974058769eb7ff901715e9b17a998bf79","subject":"Update irrelevant-files to skip run of untouched jobs","status":"MERGED","created":"2022-02-03 09:45:03.000000000","updated":"2022-02-03 16:04:30.000000000","submitted":"2022-02-03 15:53:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827615-neutron-ci-improvements","meta_rev_id":"514bc975e3170018969dcfe37fe2c2d4090de923","_number":827615,"virtual_id_number":827615,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-03 15:53:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-02-03 12:47:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-02-03 13:19:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-03 13:47:54.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-03 13:47:54.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-03 10:18:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-03 10:20:33.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-02-03 10:20:33.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2022-02-03 10:20:33.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-02-03 10:20:33.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-02-03 10:20:33.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-02-03 10:20:33.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-02-03 11:22:23.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"}],"messages":[{"id":"8b54d7f1e3ec150d474c8ea126527c7ac4edef94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-03 09:45:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e38651cef945680cb38f7d3f2800ede0f14c52cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 10:18:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c39a9e0e1f924c4b908630a74d15fc5c : SUCCESS in 31m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1a669fc843848f78cf9958df4a84e7c : SUCCESS in 27m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b3145449d7e4121a050ad8e4f8ba5d8 : SUCCESS in 12m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cacb2a3e1bc4d21ada44c3397c69658 : SUCCESS in 21m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1205292060e45e1848a06fb6aa6ab6a : SUCCESS in 22m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a96650412ee54cdcbd1eb03de5c97d84 : SUCCESS in 9m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8d2342ae05f4d739f338235a77a4837 : SUCCESS in 27m 31s","accounts_in_message":[],"_revision_number":1},{"id":"c0715e12e0ff242512c7c4b8628606bb9d7afc00","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 10:23:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b056cd940b804a89ba9f284aac11c4b4 : SUCCESS in 35m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/08ee027c6d954f9a91b74e33bf2ae03d : SUCCESS in 37m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d5540256a6ad3f9fb90ffe482d497495b9ebba2c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-03 11:22:23.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/827615/1 : FAILURE in 1h 36m 32s","accounts_in_message":[],"_revision_number":1},{"id":"93ca9f419706a82b794b6859f04d2e318dbf3b0e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-03 12:38:16.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7cf2a7915c0efcb2925e36e5eeb30c53be218553","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-02-03 12:45:29.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cffe3324dc48d3d38141c3e054b9d022fb7b37df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-03 12:47:10.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5fda88b4056c1140788d899b2e48ae1b9acc50ef","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-02-03 12:47:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"11a6a482780851e7faf74c4920b7527f74ac4cc7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-03 13:19:52.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7680155e4fed2fe72b041db10b3237696385e305","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 13:35:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46ea43ad22484b5389c697101fba9cc5 : SUCCESS in 41m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea47014ebe3f40cf9d6ac9d8f2a2ce0c : SUCCESS in 35m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7358433e8ef547daad644987a49e7ba9 : SUCCESS in 16m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33d9794d49ce4d4aa8ad6ebf0a84334b : SUCCESS in 26m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/581330ef609144e4a87a86bec2545253 : SUCCESS in 25m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e28dd0c7a8bc4ce5a6f6845d7b063134 : SUCCESS in 9m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b971bf2286d44c0dbe8711801a71bfe8 : SUCCESS in 30m 28s","accounts_in_message":[],"_revision_number":2},{"id":"f55a42dfe1a7c3686284e3c5d7f570c6a3d4681f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 13:42:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a03d517c5cb4a8fb1a56c2277d7454a : SUCCESS in 50m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/70c7748a45094460b852fbe920788e50 : SUCCESS in 44m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8f52e884fe6c6cdda2dd193e345b5b92194232bc","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-02-03 13:47:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5a44cb5ffc25720ee0aa497913f487ab1ce0a4d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 13:48:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"49c5fd2de1c4fa1a6ef34260a2ad3c44af34caaa","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-03 14:37:07.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/827615/2 : SUCCESS in 1h 49m 04s","accounts_in_message":[],"_revision_number":2},{"id":"be4283f18cacd5b46504994bd25ada363337686f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 15:53:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82110dc62b5d43fcb010b6c17560cc11 : SUCCESS in 34m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81615abd10714b5aa524b3b65e25a6ab : SUCCESS in 15m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93d4a786c4f84ff0bbf3ca8a03440d59 : SUCCESS in 33m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24d63befe31549afaa954b11156d1db1 : SUCCESS in 26m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d81cb580ee544e4b91756056fc54899d : SUCCESS in 10m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d7f426b14674dddb9ff088c579b531b : SUCCESS in 27m 14s","accounts_in_message":[],"_revision_number":2},{"id":"c5461a8293a89b4e0b2b76d3133fa7593817c543","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 15:53:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"514bc975e3170018969dcfe37fe2c2d4090de923","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 16:04:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f220ea9237f4490f9fb122ea7899cf8e : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ca8b40403f7e1ab1f383da01b4a18b066fd35b93","revisions":{"f599d3fac41ce61957faa08f40fb8ef3eca2dbfd":{"kind":"REWORK","_number":1,"created":"2022-02-03 09:45:03.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/15/827615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/827615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/827615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/827615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/827615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/827615/1"}}},"commit":{"parents":[{"commit":"1a16c35f305bf0964638cd134c2c66183daebe11","subject":"Merge \"\"L3AgentExtensionsManager\" check loaded extensions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a16c35f305bf0964638cd134c2c66183daebe11"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-03 09:42:52.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-03 09:42:55.000000000","tz":330},"subject":"Update irrelevant-files to skip run of untouched jobs","message":"Update irrelevant-files to skip run of untouched jobs\n\nCurrently when a job config is updated, all other\njobs are triggered unnecessary.\n\nThis patch updates irrelevant files to ignore changes\nin all job config files except zuul.d/project.yaml(as\nthis contains the list of jobs to run not their definitions).\nWith this only the jobs whose definitions are updated(along with\nit\u0027s child jobs) will be triggered because by default\nmatch-on-config-updates is set to true for a job[1] and\nthat ignores files or irrelevant-files for a job inclusion.\n\n[1] https://zuul-ci.org/docs/zuul/latest/config/job.html#attr-job.match-on-config-updates\nChange-Id: Ida0fc8924f94038962d0548d2c305f88f6e056cd\n\nChange-Id: Ida7c3ac974058769eb7ff901715e9b17a998bf79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f599d3fac41ce61957faa08f40fb8ef3eca2dbfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f599d3fac41ce61957faa08f40fb8ef3eca2dbfd"}]},"branch":"refs/heads/master"},"ca8b40403f7e1ab1f383da01b4a18b066fd35b93":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-02-03 12:47:10.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/15/827615/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/827615/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/827615/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/827615/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/827615/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/827615/2"}}},"commit":{"parents":[{"commit":"1a16c35f305bf0964638cd134c2c66183daebe11","subject":"Merge \"\"L3AgentExtensionsManager\" check loaded extensions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a16c35f305bf0964638cd134c2c66183daebe11"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-03 09:42:52.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2022-02-03 12:47:09.000000000","tz":0},"subject":"Update irrelevant-files to skip run of untouched jobs","message":"Update irrelevant-files to skip run of untouched jobs\n\nCurrently when a job config is updated, all other\njobs are triggered unnecessary.\n\nThis patch updates irrelevant files to ignore changes\nin all job config files except zuul.d/project.yaml(as\nthis contains the list of jobs to run not their definitions).\nWith this only the jobs whose definitions are updated(along with\nit\u0027s child jobs) will be triggered because by default\nmatch-on-config-updates is set to true for a job[1] and\nthat ignores files or irrelevant-files for a job inclusion.\n\n[1] https://zuul-ci.org/docs/zuul/latest/config/job.html#attr-job.match-on-config-updates\n\nChange-Id: Ida7c3ac974058769eb7ff901715e9b17a998bf79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca8b40403f7e1ab1f383da01b4a18b066fd35b93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca8b40403f7e1ab1f383da01b4a18b066fd35b93"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
