)]}'
{"id":"openstack%2Fneutron~828103","triplet_id":"openstack%2Fneutron~master~I5b5e7957cfe8020001777fd40e038eaafb5fb894","project":"openstack/neutron","branch":"master","topic":"bug/1932154","attention_set":{},"removed_from_attention_set":{"13686":{"account":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"last_update":"2022-02-22 18:54:08.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-02-21 13:15:53.000000000","reason":"Slawek Kaplonski replied on the change"},"24824":{"account":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"last_update":"2022-02-22 18:54:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5b5e7957cfe8020001777fd40e038eaafb5fb894","subject":"[OVN] Off-path SmartNIC DPU Documentation","status":"MERGED","created":"2022-02-07 10:29:50.000000000","updated":"2022-02-22 18:59:35.000000000","submitted":"2022-02-22 18:54:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":30,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828103-bug/1932154","meta_rev_id":"60bf77938865e7531541ba02255ffd40b60c4562","_number":828103,"virtual_id_number":828103,"owner":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-22 18:54:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2022-02-21 13:15:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-02-22 16:47:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-02-22 16:47:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-02-18 19:58:58.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-07 10:45:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-07 12:00:07.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-02-08 17:39:05.000000000","updated_by":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"reviewer":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"state":"CC"},{"updated":"2022-02-16 14:50:43.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-02-18 19:58:58.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"1261e67109557d41a1416de49e4e9a9f7308886b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-02-07 10:29:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfa7bc97b9f0f037bd942a4dc979189209fb2dd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 10:45:55.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da55c0322fe2473184e37c6b05dd4692 : SUCCESS in 11m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b74639455f8465da7bc4e71aad8426f : FAILURE in 7m 09s","accounts_in_message":[],"_revision_number":1},{"id":"2880e08ce7b6a25fe283e31eeb2d18ad04ba0fc6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-07 12:00:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/828103/1 : SUCCESS in 1h 28m 34s","accounts_in_message":[],"_revision_number":1},{"id":"cce6d8001274c6b8e61c291371085b8f0ed39c53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-02-07 15:32:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bb7fa64c6fee30769206ab7bb77e2e84ec764059","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 15:54:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f13a2f751f747a7ba713a2bd53feb99 : SUCCESS in 11m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36371904735b4e7f862ad278905a5f75 : SUCCESS in 8m 43s","accounts_in_message":[],"_revision_number":2},{"id":"f3d5660a8f760ad6702b1bf818b31edf40fde375","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-02-07 16:34:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ea9ae3dcc523a92eade4b804ba2867b81f9f8be4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-02-07 16:51:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"694e62add3126f1ba330c667d81dc3b553780892","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 17:14:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1811f9b34d974ba5b1a007175eec1745 : SUCCESS in 15m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2251ca7de44947649a04c81783c827c1 : SUCCESS in 10m 17s","accounts_in_message":[],"_revision_number":4},{"id":"183a14b99edbe0c5ab7dad9a53815c1b0d7a724d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-07 18:40:20.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/828103/4 : SUCCESS in 1h 47m 33s","accounts_in_message":[],"_revision_number":4},{"id":"16b3c6c98c688043cc49086a5344c23c4cef8342","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-02-08 06:14:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f9908bad232e9cc9a1e22138cf780578c1ead80e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 06:36:24.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a741da175af84530bedbc6fe5019b226 : SUCCESS in 15m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8abf8253298440386745fe150f6734e : FAILURE in 6m 09s","accounts_in_message":[],"_revision_number":5},{"id":"727e921d207be0e15f963d5375ed40a6db9af53c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-08 07:47:11.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/828103/5 : SUCCESS in 1h 30m 38s","accounts_in_message":[],"_revision_number":5},{"id":"e8d17ae1984acd2c35d83ec69ff829ec91970a93","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-02-08 17:39:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cf6f1aa71efa6ae3533b687cad477815f5658619","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-02-09 03:46:36.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4eecefeee196941e477ac7356085e33dac33be2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 04:00:58.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b82792e2b65c48118ec4ab580b60f613 : SUCCESS in 11m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed880b4f932348dba5ae6261e4cec446 : FAILURE in 6m 23s","accounts_in_message":[],"_revision_number":5},{"id":"150658dec849d3be1efdcd63d832998a26c6c91a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-09 05:15:35.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/828103/5 : FAILURE in 1h 28m 08s","accounts_in_message":[],"_revision_number":5},{"id":"b9e7fbcd2d79bc3098565c9dd909827e243f4d60","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-02-09 07:25:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5e2bdadcbf3562592ae970e10c2d2936f518049a","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-02-09 18:11:34.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"083289fce7c8cf9e1e4d3cc686147ab7b4f0d32a","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-02-09 18:39:20.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"d425dd0b583923d270fbe8fad2aedddade2528e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-02-14 09:34:15.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"4be28793f111f717005add42abd57af70f51bba9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 09:48:14.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ceb64ffbe7b4d778d3d633f858de768 : SUCCESS in 11m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90783287fefa4355b3ae3188e757dc4c : FAILURE in 5m 36s","accounts_in_message":[],"_revision_number":6},{"id":"f6cc4620b254d68e6bdb6155ac125efb5b455582","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-14 11:28:00.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/828103/6 : FAILURE in 1h 51m 42s","accounts_in_message":[],"_revision_number":6},{"id":"e0546703dd8b1a9faacb56ff9f7b1d73c4350e04","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-02-15 09:59:54.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"757f7422c015560d280d5467df4f9671fdadbaa3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-02-15 10:03:03.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0daa97023ec7a118893e4ae2f79eecb90f534263","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 10:18:16.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/349e24345df540a1a8221279362af928 : SUCCESS in 11m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efc9cffcf45344c396361131c1424ca5 : FAILURE in 7m 14s","accounts_in_message":[],"_revision_number":7},{"id":"749c16c827b5f3361cdbb31849f245a037075f6f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-15 11:53:57.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/828103/7 : SUCCESS in 1h 49m 48s","accounts_in_message":[],"_revision_number":7},{"id":"91ef69732b9829aa2797bc88ecbae68125658b51","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-02-16 14:50:43.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"ad11748c15f30397988118974ac71bcc5b41e4d8","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-02-16 15:07:16.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"f611237aa11d1a7e92c75f438d9394b2ed3f8619","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-02-16 18:37:51.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"0f1fca338542abd1179eedf719c788432ef48095","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 19:01:26.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4579c1ec8bb47fa894c4b5517faff87 : SUCCESS in 13m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05f5f748ec764111bf7ab9d07966b811 : FAILURE in 5m 44s","accounts_in_message":[],"_revision_number":8},{"id":"a48a26e62644d62b21b68ba0cdc9403d2d0edb97","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-02-16 20:03:57.000000000","message":"Patch Set 8:\n\n(7 comments)","accounts_in_message":[],"_revision_number":8},{"id":"a43494a5732d8e2374590c574b14f48d1b78b8d5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-16 20:27:23.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/828103/8 : SUCCESS in 1h 48m 28s","accounts_in_message":[],"_revision_number":8},{"id":"f0e39bbb6437a1b5e47ff89b0d986aa96ff9c023","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-02-17 12:01:11.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"937c2f5009bc07a254172a80e522bbdef0c4b7d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 12:17:30.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c66d8d033e1344f3aa9aa7486c59875b : SUCCESS in 14m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60141a4b12ac4cbe8827c85c3a85ce0f : SUCCESS in 10m 31s","accounts_in_message":[],"_revision_number":8},{"id":"0304c389bfd9a11c9eb79fafdd75dd7d7bfaf217","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-17 13:36:06.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/828103/8 : SUCCESS in 1h 34m 04s","accounts_in_message":[],"_revision_number":8},{"id":"92081e3e5116676e2bd4a50611a5e57faf71639d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-02-17 13:57:07.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"a28a1f74af881df0c50e232717f0bd52852eb9b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 14:13:57.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b771ead233c643c89a80d7f1f6626df9 : SUCCESS in 12m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5cb0a06e87a406bb729091108b6b076 : SUCCESS in 9m 02s","accounts_in_message":[],"_revision_number":9},{"id":"49b11083bff5829aa8154aac36fea958adf1cbd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-02-17 14:27:13.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"58e746fb18a6d3a4fdf445718c2f10e9a1f02e02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 14:41:04.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3894895df3ad4adda6239dd98cda027f : SUCCESS in 11m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c27a6b98fe9a4783b37bf3a3c1112529 : SUCCESS in 8m 32s","accounts_in_message":[],"_revision_number":10},{"id":"9ac8ec32b34e4d66e0ddd99f2b372b6cc66b9136","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-17 15:03:34.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/828103/10 : FAILURE in 34m 58s","accounts_in_message":[],"_revision_number":10},{"id":"3e742a1a777b35471dc34ea806650da4136c7cb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-02-18 06:18:46.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"15df73689f040218db57bdf7b22f770c8153aa08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 06:33:07.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08dd022c50dd4488b37bf274d641de27 : SUCCESS in 10m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b180bbd35f644b35a427217849f47e8e : SUCCESS in 9m 12s","accounts_in_message":[],"_revision_number":11},{"id":"279d1774510aadb5bb38c9da0a0e63b5a721f9e1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-18 07:51:41.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/828103/11 : SUCCESS in 1h 31m 44s","accounts_in_message":[],"_revision_number":11},{"id":"47f00c02fdef1e5b21a029710cac81210609cbf3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-18 19:58:58.000000000","message":"Patch Set 11: Review-Priority+1","accounts_in_message":[],"_revision_number":11},{"id":"38d1b0525d498a1c70e0f0a0a6f1c9f33ba5dc1c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-02-21 13:15:53.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"0e799979b360b1e499b240b41e3bcd29505da98b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-22 16:47:07.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"25ca92e79c8774d267076b2e798254292d5c15e0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-22 16:47:21.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"af44c72ea2110ca1767d4390ed95fb42a964081c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 16:49:46.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"aa367b6edb786eccd8bccb7a1670e2f0be8a6d7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 18:54:07.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edb416c61fbd4379ac5690490dac747f : SUCCESS in 14m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f48b4f24819c40079716d38f970aab93 : SUCCESS in 10m 04s","accounts_in_message":[],"_revision_number":11},{"id":"6766795d0dd7be74e432435285190aa737b921a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 18:54:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"60bf77938865e7531541ba02255ffd40b60c4562","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 18:59:35.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfa1d7fc897f4aa8a18fe5f550cb61b1 : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"0c9c60d5ca53bd415fac76ef8aa5ae17f92704e7","revisions":{"eeef00b23b698e7431727875d77c3a2101cc3eda":{"kind":"REWORK","_number":1,"created":"2022-02-07 10:29:50.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/03/828103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/03/828103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/03/828103/1"}}},"commit":{"parents":[{"commit":"522c89f7ba46f9e91a083b317f4da6cf8bb29f7d","subject":"[OVN] Off-path SmartNIC DPU Port Binding with OVN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/522c89f7ba46f9e91a083b317f4da6cf8bb29f7d"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-02-07 10:27:21.000000000","tz":60},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-02-07 10:28:36.000000000","tz":60},"subject":"WIP: Off-path SmartNIC DPU Documentation","message":"WIP: Off-path SmartNIC DPU Documentation\n\nCloses-Bug: #1932154\nChange-Id: I5b5e7957cfe8020001777fd40e038eaafb5fb894\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eeef00b23b698e7431727875d77c3a2101cc3eda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eeef00b23b698e7431727875d77c3a2101cc3eda"}]},"branch":"refs/heads/master"},"62e4302bbceb984b300de9e8173fa6f4673e3239":{"kind":"REWORK","_number":2,"created":"2022-02-07 15:32:38.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/03/828103/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/03/828103/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/03/828103/2"}}},"commit":{"parents":[{"commit":"522c89f7ba46f9e91a083b317f4da6cf8bb29f7d","subject":"[OVN] Off-path SmartNIC DPU Port Binding with OVN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/522c89f7ba46f9e91a083b317f4da6cf8bb29f7d"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-02-07 10:27:21.000000000","tz":60},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-02-07 15:28:35.000000000","tz":60},"subject":"WIP: Off-path SmartNIC DPU Documentation","message":"WIP: Off-path SmartNIC DPU Documentation\n\nCloses-Bug: #1932154\nChange-Id: I5b5e7957cfe8020001777fd40e038eaafb5fb894\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62e4302bbceb984b300de9e8173fa6f4673e3239"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62e4302bbceb984b300de9e8173fa6f4673e3239"}]},"branch":"refs/heads/master"},"b981bf69fdc9fa01c9faec1999ba18178a0c846e":{"kind":"REWORK","_number":3,"created":"2022-02-07 16:34:53.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/03/828103/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/03/828103/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/03/828103/3"}}},"commit":{"parents":[{"commit":"522c89f7ba46f9e91a083b317f4da6cf8bb29f7d","subject":"[OVN] Off-path SmartNIC DPU Port Binding with OVN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/522c89f7ba46f9e91a083b317f4da6cf8bb29f7d"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-02-07 10:27:21.000000000","tz":60},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-02-07 16:33:58.000000000","tz":60},"subject":"WIP: Off-path SmartNIC DPU Documentation","message":"WIP: Off-path SmartNIC DPU Documentation\n\nCloses-Bug: #1932154\nChange-Id: I5b5e7957cfe8020001777fd40e038eaafb5fb894\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b981bf69fdc9fa01c9faec1999ba18178a0c846e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b981bf69fdc9fa01c9faec1999ba18178a0c846e"}]},"branch":"refs/heads/master"},"a072031fe299e268bab6272d271f6496b31873d9":{"kind":"REWORK","_number":4,"created":"2022-02-07 16:51:36.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/03/828103/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/03/828103/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/03/828103/4"}}},"commit":{"parents":[{"commit":"522c89f7ba46f9e91a083b317f4da6cf8bb29f7d","subject":"[OVN] Off-path SmartNIC DPU Port Binding with OVN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/522c89f7ba46f9e91a083b317f4da6cf8bb29f7d"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-02-07 10:27:21.000000000","tz":60},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-02-07 16:50:36.000000000","tz":60},"subject":"WIP: Off-path SmartNIC DPU Documentation","message":"WIP: Off-path SmartNIC DPU Documentation\n\nCloses-Bug: #1932154\nChange-Id: I5b5e7957cfe8020001777fd40e038eaafb5fb894\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a072031fe299e268bab6272d271f6496b31873d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a072031fe299e268bab6272d271f6496b31873d9"}]},"branch":"refs/heads/master"},"41a9acf896af422d775e0d60ee59bbaa4454a5c1":{"kind":"REWORK","_number":5,"created":"2022-02-08 06:14:39.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/03/828103/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/03/828103/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/03/828103/5"}}},"commit":{"parents":[{"commit":"037dfae2914c192d48ca418e62fc1f1e9ad7b506","subject":"[OVN] Off-path SmartNIC DPU Port Binding with OVN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/037dfae2914c192d48ca418e62fc1f1e9ad7b506"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-02-07 10:27:21.000000000","tz":60},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-02-08 06:13:32.000000000","tz":60},"subject":"WIP: Off-path SmartNIC DPU Documentation","message":"WIP: Off-path SmartNIC DPU Documentation\n\nCloses-Bug: #1932154\nChange-Id: I5b5e7957cfe8020001777fd40e038eaafb5fb894\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41a9acf896af422d775e0d60ee59bbaa4454a5c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41a9acf896af422d775e0d60ee59bbaa4454a5c1"}]},"branch":"refs/heads/master"},"bdbee6b8cba0d8412233ac016d25e85021fd3f18":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-02-14 09:34:15.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/03/828103/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/03/828103/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/03/828103/6"}}},"commit":{"parents":[{"commit":"78e7a908cd5bf07bb44af93f11533dea1ae4795e","subject":"[OVN] Off-path SmartNIC DPU Port Binding with OVN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78e7a908cd5bf07bb44af93f11533dea1ae4795e"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-02-07 10:27:21.000000000","tz":60},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-02-14 09:33:26.000000000","tz":60},"subject":"WIP: Off-path SmartNIC DPU Documentation","message":"WIP: Off-path SmartNIC DPU Documentation\n\nCloses-Bug: #1932154\nChange-Id: I5b5e7957cfe8020001777fd40e038eaafb5fb894\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdbee6b8cba0d8412233ac016d25e85021fd3f18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdbee6b8cba0d8412233ac016d25e85021fd3f18"}]},"branch":"refs/heads/master"},"b94c4386a630abeaf0de7b1bb12a31e9239a1aed":{"kind":"REWORK","_number":7,"created":"2022-02-15 10:03:03.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/03/828103/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/03/828103/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/03/828103/7"}}},"commit":{"parents":[{"commit":"ba4bf413291d786ce7edca7baeeb9a4441ca4f2a","subject":"[OVN] Off-path SmartNIC DPU Port Binding with OVN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba4bf413291d786ce7edca7baeeb9a4441ca4f2a"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-02-07 10:27:21.000000000","tz":60},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2022-02-15 10:02:30.000000000","tz":180},"subject":"WIP: Off-path SmartNIC DPU Documentation","message":"WIP: Off-path SmartNIC DPU Documentation\n\nCloses-Bug: #1932154\nCo-Authored-By: Dmitrii Shcherbakov \u003cdmitrii.shcherbakov@canonical.com\u003e\nChange-Id: I5b5e7957cfe8020001777fd40e038eaafb5fb894\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b94c4386a630abeaf0de7b1bb12a31e9239a1aed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b94c4386a630abeaf0de7b1bb12a31e9239a1aed"}]},"branch":"refs/heads/master"},"742ee42d435fedac1ded645302022136eaa7fc48":{"kind":"REWORK","_number":8,"created":"2022-02-16 18:37:51.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/03/828103/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/03/828103/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/03/828103/8"}}},"commit":{"parents":[{"commit":"ba4bf413291d786ce7edca7baeeb9a4441ca4f2a","subject":"[OVN] Off-path SmartNIC DPU Port Binding with OVN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba4bf413291d786ce7edca7baeeb9a4441ca4f2a"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-02-07 10:27:21.000000000","tz":60},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2022-02-16 16:49:09.000000000","tz":180},"subject":"Off-path SmartNIC DPU Documentation","message":"Off-path SmartNIC DPU Documentation\n\nCloses-Bug: #1932154\nCo-Authored-By: Dmitrii Shcherbakov \u003cdmitrii.shcherbakov@canonical.com\u003e\nChange-Id: I5b5e7957cfe8020001777fd40e038eaafb5fb894\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/742ee42d435fedac1ded645302022136eaa7fc48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/742ee42d435fedac1ded645302022136eaa7fc48"}]},"branch":"refs/heads/master"},"93759ac94ad8d1762b45d7333170ae44405e2261":{"kind":"REWORK","_number":9,"created":"2022-02-17 13:57:07.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/03/828103/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/03/828103/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/03/828103/9"}}},"commit":{"parents":[{"commit":"d6eb8f70be1e701589645631c436e46592a98c30","subject":"[OVN] Off-path SmartNIC DPU Port Binding with OVN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6eb8f70be1e701589645631c436e46592a98c30"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-02-07 10:27:21.000000000","tz":60},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-02-17 13:56:34.000000000","tz":60},"subject":"[OVN] Off-path SmartNIC DPU Documentation","message":"[OVN] Off-path SmartNIC DPU Documentation\n\nCloses-Bug: #1932154\nCo-Authored-By: Dmitrii Shcherbakov \u003cdmitrii.shcherbakov@canonical.com\u003e\nChange-Id: I5b5e7957cfe8020001777fd40e038eaafb5fb894\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/93759ac94ad8d1762b45d7333170ae44405e2261"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/93759ac94ad8d1762b45d7333170ae44405e2261"}]},"branch":"refs/heads/master"},"75a7eb1571674e38abbcea659f903b3e10583cac":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2022-02-17 14:27:13.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/03/828103/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/03/828103/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/03/828103/10"}}},"commit":{"parents":[{"commit":"2e9829e2e74790696e49991618ac666e55ac9610","subject":"[OVN] Off-path SmartNIC DPU Port Binding with OVN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e9829e2e74790696e49991618ac666e55ac9610"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-02-07 10:27:21.000000000","tz":60},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-02-17 14:26:41.000000000","tz":60},"subject":"[OVN] Off-path SmartNIC DPU Documentation","message":"[OVN] Off-path SmartNIC DPU Documentation\n\nCloses-Bug: #1932154\nCo-Authored-By: Dmitrii Shcherbakov \u003cdmitrii.shcherbakov@canonical.com\u003e\nChange-Id: I5b5e7957cfe8020001777fd40e038eaafb5fb894\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/75a7eb1571674e38abbcea659f903b3e10583cac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/75a7eb1571674e38abbcea659f903b3e10583cac"}]},"branch":"refs/heads/master"},"0c9c60d5ca53bd415fac76ef8aa5ae17f92704e7":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2022-02-18 06:18:46.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/03/828103/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/03/828103/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/828103/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/03/828103/11"}}},"commit":{"parents":[{"commit":"7d64d0c116cf3f5ec2a35fb2ffe08dde61343ad8","subject":"[OVN] Off-path SmartNIC DPU Port Binding with OVN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d64d0c116cf3f5ec2a35fb2ffe08dde61343ad8"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-02-07 10:27:21.000000000","tz":60},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-02-18 06:18:29.000000000","tz":60},"subject":"[OVN] Off-path SmartNIC DPU Documentation","message":"[OVN] Off-path SmartNIC DPU Documentation\n\nCloses-Bug: #1932154\nCo-Authored-By: Dmitrii Shcherbakov \u003cdmitrii.shcherbakov@canonical.com\u003e\nChange-Id: I5b5e7957cfe8020001777fd40e038eaafb5fb894\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c9c60d5ca53bd415fac76ef8aa5ae17f92704e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c9c60d5ca53bd415fac76ef8aa5ae17f92704e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
