)]}'
{"id":"openstack%2Fneutron~828155","triplet_id":"openstack%2Fneutron~stable%2Fxena~Ie670fba4b3afe427747732d2c3948d92311e960e","project":"openstack/neutron","branch":"stable/xena","topic":"bug/1958363","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-02-15 11:58:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie670fba4b3afe427747732d2c3948d92311e960e","subject":"Use a thread local variable to store the Nova Notifier enable flag","status":"MERGED","created":"2022-02-07 14:06:27.000000000","updated":"2022-02-15 12:00:52.000000000","submitted":"2022-02-15 11:58:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828155-bug/1958363","meta_rev_id":"2dd6981b6de9e0f75d4a1fc8d181ed11e20219ea","_number":828155,"virtual_id_number":828155,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-15 11:58:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-02-10 14:35:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-02-15 10:08:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-02-10 15:35:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-02-15 10:08:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-07 15:11:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-07 15:48:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-07 15:57:33.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-02-10 14:35:38.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-02-10 15:35:56.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-02-15 10:08:55.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"23c53f0a51d2189b5c5c3f1017558075168cc397","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-02-07 14:06:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9db229c3d228eb8dccf0c85c4a77ba45d7002f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 15:11:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/355e0637c77645c29ae8c1077b2d3e2e : TIMED_OUT in 1h 02m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2e083c7ef8b4f6d943edd06898c97ac : TIMED_OUT in 1h 02m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3737edfeace61ac9dac5b0d33ca181d6ad315df4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 15:48:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/74ceda64ee9649f59866dd984e385def : SUCCESS in 38m 39s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/1438f7b3b8654cea849f0584f00738e4 : SUCCESS in 1h 28m 06s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/c3197505e2ca4a93bf756aa9fe18441c : SUCCESS in 1h 16m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/0fb2f8cf48a8444b971230868e3101b6 : SUCCESS in 1h 08m 29s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/3d7c1dea9b0845b0ae4115cb42e5857a : SUCCESS in 1h 00m 30s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/26a56f9d996745b6829092c78df3b294 : SUCCESS in 24m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdf05d571c944247be0231f113170bc2 : SUCCESS in 35m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e2405e18f884c26b38eafcff4085b3a : SUCCESS in 12m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f06483f12aef48c79667f02743a291b4 : SUCCESS in 23m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d94e8ff4e9954e42b1a20ea013ddad2d : SUCCESS in 21m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ed280568aa34e3dae3862230c9a895c : SUCCESS in 21m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c333a857aaf64ba7b5dc0f2531795966 : SUCCESS in 11m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5035ccaefbc94058ab3759bf40321011 : SUCCESS in 48m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2f08825a402c4540bf87fe41a65d02e1 : SUCCESS in 1h 37m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bb46e571f52d4cfc8ce3ee8bcd571eed : SUCCESS in 1h 17m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5ad5abd8a58642d6a2e8b3bff656e4b6 : SUCCESS in 51m 59s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e217c1f54c934e71ba16f935b8f5cd90 : SUCCESS in 57m 51s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cfac41ae039a4900a181dc2c8156710d : SUCCESS in 1h 01m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5bead92be8194517b9b2e495204d4851 : SUCCESS in 1h 03m 39s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/740642e2b3934836aa9c93c3fbc986b5 : SUCCESS in 50m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/af16df6c033c474da32dc724673794cc : SUCCESS in 1h 02m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4f77fb796b3db87c1062945ba73bca4365bc099","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-07 15:57:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/828155/1 : SUCCESS in 1h 50m 22s","accounts_in_message":[],"_revision_number":1},{"id":"13ad7f9dde755334c7334cc506b1c43bc30dddb8","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-02-10 14:35:38.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fdaf7b4a47fd6d3d360e99566d6d9aec0c2c4dc4","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-02-10 15:35:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"917bd01be425ff1cdb8d9a8e46cf529c867d2df5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-02-15 10:08:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2c81da1c5f75248950e24a3aa310c02e4ff17c36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 10:09:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3e4706fea548e5de66e870585846cbebeff620db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 11:58:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/e5064069466e488cbbf4fb6f32df769d : SUCCESS in 37m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d156214847a48f582d1971deb905147 : SUCCESS in 13m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4b935dca1fb421fb4a8fbf158aa4c01 : SUCCESS in 25m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e845d297353748208cd706343342f8cd : SUCCESS in 25m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d84f0d25aeb540a093e77795e3919711 : SUCCESS in 11m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/00d1e45b5c5f41c9a9298f694790f513 : SUCCESS in 53m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/edc08fb409f84e6a8e755802beccc0df : SUCCESS in 1h 46m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/de71ad8fbef7497f8d84a5de39a8db16 : SUCCESS in 59m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6486de7e5b46461492e92a0ffe3b0ee0 : SUCCESS in 1h 14m 15s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a540a41aea4943e6aa3ed6e510d2b1ad : SUCCESS in 46m 11s","accounts_in_message":[],"_revision_number":1},{"id":"f4034de0538ad264ba4f2f837614e8ec306ef78d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 11:58:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2dd6981b6de9e0f75d4a1fc8d181ed11e20219ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 12:00:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a15bfeff1db49d18354b4814cbf0045 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68458c369e69605d4c3a75eb3cb21cbbbd68cb87","revisions":{"68458c369e69605d4c3a75eb3cb21cbbbd68cb87":{"kind":"REWORK","_number":1,"created":"2022-02-07 14:06:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/828155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/828155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/828155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/828155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/828155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/828155/1"}}},"commit":{"parents":[{"commit":"d317bedb7d41e09a22a12891735abc5df51c0d19","subject":"Update irrelevant-files for non scenario jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d317bedb7d41e09a22a12891735abc5df51c0d19"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-01-29 10:10:22.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-02-03 19:40:51.000000000","tz":0},"subject":"Use a thread local variable to store the Nova Notifier enable flag","message":"Use a thread local variable to store the Nova Notifier enable flag\n\nThe Nova Notifier can be called simultaneously by several RPC\ncallbacks from the agents (DHCP, L2), trying to update the provisioning\nstatus of a port. In order to handle each context notifier enable flag,\na thread local variable is used. This will isolate the flag update\nif two entities inform at the same time and one RPC callback is attended\nduring the processing of the other one.\n\nThis patch also removes the debug messages added to debug this issue.\n\nCloses-Bug: #1958363\n\nConflicts:\n    neutron/notifiers/nova.py\n\nChange-Id: Ie670fba4b3afe427747732d2c3948d92311e960e\n(cherry picked from commit 507989fc622a6633926c4288827f482bae6054dd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/68458c369e69605d4c3a75eb3cb21cbbbd68cb87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/68458c369e69605d4c3a75eb3cb21cbbbd68cb87"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
