)]}'
{"id":"openstack%2Fneutron~828512","triplet_id":"openstack%2Fneutron~stable%2Fqueens~I94eb9c6e5be792a9a09f824247eb0aee63f6bda8","project":"openstack/neutron","branch":"stable/queens","topic":"remove_sphinx_job","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-02-15 10:32:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I94eb9c6e5be792a9a09f824247eb0aee63f6bda8","subject":"[stable][queens]: remove publish-openstack-sphinx-docs template","status":"MERGED","created":"2022-02-09 10:28:27.000000000","updated":"2022-02-15 10:32:24.000000000","submitted":"2022-02-15 10:32:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828512-remove_sphinx_job","meta_rev_id":"db009d3836ef424a4fba32c0baa0f29e3f78126e","_number":828512,"virtual_id_number":828512,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-15 10:32:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-02-15 08:46:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-02-09 15:10:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-02-15 08:46:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-09 12:36:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-09 15:08:48.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2022-02-09 15:10:16.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-02-15 08:46:13.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"c7bfab8af56724aa7ac83f65a8fa607b22119071","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-09 10:28:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f493826d06c0cbf7a1d126dcbd64c105d40eae0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 12:36:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/bfff96261bd643b8830b47b5efae43ce : SUCCESS in 43m 37s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/1d20697bdd26444695074b08ecc362c2 : SUCCESS in 45m 43s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/572188e0f9d540e5a6248c71698cb413 : FAILURE in 50m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5818f301b6d4d479f31ed0df1e7e6fa : SUCCESS in 10m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ce30ebdd89f640528af8cfc62c4f708d : SUCCESS in 22m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f671b66a50e94b46be2e800fdc79b0fd : SUCCESS in 24m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/31079ae0ddd848c4bd0d3b38d50f840c : SUCCESS in 1h 08m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c1a2a2e571534a329ffb45f74c17fd6e : SUCCESS in 53m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab38d91987b54b83a5d1a1f1ab6d928c : SUCCESS in 1h 11m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/904b2e02a9964cd1b778fe38a6704082 : SUCCESS in 38m 27s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/6c56aba55c7443fa84374d18b4636c1c : SUCCESS in 1h 51m 29s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/094f1e0fa209441b8b559e63fc5e5259 : SUCCESS in 1h 23m 39s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/748b1402f1ca4a63baef1d263667fae7 : SUCCESS in 1h 59m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/733e3a4f015b4ebc8ddceb365b4b5e76 : SUCCESS in 39m 18s","accounts_in_message":[],"_revision_number":1},{"id":"af53d16ff1fd11cc6b68129d10ab8003585a3559","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-09 12:57:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"64905fd8423a8115d729d060f106470faa5a71ac","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-02-09 15:08:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2ee768ccbb5d54f2e4a8723a4a6dec5d605d8aae","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-02-09 15:10:16.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e88dc8974bcf497005ea9fab04309b595da3ca4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 15:20:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/2f6db23b02564269aa3b881982119b5b : SUCCESS in 56m 41s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/f37a53c81d2d4be29224d7492cc030c7 : SUCCESS in 1h 05m 19s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/aeac066a32a443198451e655d8be4e95 : FAILURE in 1h 07m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ca209eb8eca4172aef3dccedbd83650 : SUCCESS in 13m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/02456001a779444b8dc2ce3cbe37a9f7 : SUCCESS in 25m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c69d568d89a243048a8ce65b1576d93b : SUCCESS in 34m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0dabaa62b8ca438989afe20f42128184 : SUCCESS in 1h 39m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ffe142b7b9234b628bcc3a2ecd7329dc : SUCCESS in 56m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e085c33276ce4992be3d76491ec014b6 : SUCCESS in 1h 30m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fdd0b590fe294861bb53646542797f2d : SUCCESS in 47m 36s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/cc6108a32d62440d877286780fb93416 : SUCCESS in 1h 58m 14s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1c7d54850e6641bd96958471ff25516f : SUCCESS in 1h 44m 47s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a563e9ab961a4a2d9df02fbdc2f6df01 : SUCCESS in 1h 33m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4443e5e02d384391866caed0c668d210 : SUCCESS in 34m 48s","accounts_in_message":[],"_revision_number":1},{"id":"2ce5a69380d2a8a2c0a815aeaab0b20aa5244b80","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-02-09 15:44:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af2685810cc6e6799a6a35e27313e2af4c63a3f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 18:10:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/fa11ab6a0e444055bb8e819a0b9d90fa : SUCCESS in 44m 10s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/bd01711c57a04679b162844185aa49ef : SUCCESS in 49m 02s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/72cd8898c1dc467988f3bfaad98af046 : FAILURE in 51m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5853c6e77ec54883a02a63d142ff9833 : SUCCESS in 11m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/efbb4097e96a4fe6944117c648a58ba1 : SUCCESS in 30m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9988f8f5d51e41c5b0756dc73d35c822 : SUCCESS in 30m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f34613eeedcd4c2e86f14bcd981e2313 : SUCCESS in 1h 14m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/058060ef498547079a2d6a97307f25dd : SUCCESS in 54m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d2f93eb2d770494683d336f20a2d76cc : SUCCESS in 1h 20m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7a9f9b8cfa31447e86c85fa475b53f26 : SUCCESS in 51m 51s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/669e9079a851429f931dd1e42e81f537 : SUCCESS in 1h 30m 49s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d1248195a648484dbbd7b4ab06b83cc7 : SUCCESS in 1h 42m 03s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/df800ae91f524465905c5e35205a6fef : SUCCESS in 1h 12m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40a1e69258da4917b0d4a00a7e978eb8 : SUCCESS in 30m 21s","accounts_in_message":[],"_revision_number":1},{"id":"1d27b67fda3b186814280bbad3a83f46c2a36e42","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-10 09:11:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5f840a94baed5ce6ecc93a1d0bdc678084f83bb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 11:12:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/9d4f069115064ad6a55650003df4a766 : SUCCESS in 38m 25s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/65ecf9a1e6ea429b9ee59b7d353c4c6a : SUCCESS in 50m 56s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/4535f0d0fe14452ca30c5383c38f1e62 : FAILURE in 51m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bf436b0e30d4e698bc40371f8c95ab6 : SUCCESS in 10m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/81da018916c54c57aaa40f8df70bab32 : SUCCESS in 23m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/86b6e23a70064348944721e705641952 : SUCCESS in 25m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a2a16119703342c290deebb21c35c79c : FAILURE in 1h 03m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/553d11baeacd456bbbafc2700ab34356 : SUCCESS in 47m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/152be21456a74943890a47219778f17c : SUCCESS in 1h 06m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b3ada3635f5c48ea854f1b3eb75e3697 : SUCCESS in 33m 02s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/87bc54c0fb59483ea7cfe5f8313e8266 : SUCCESS in 1h 31m 25s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/df12e534e36f412d8e07c07b4c8f1755 : SUCCESS in 1h 22m 33s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/dde4ab353719483e8fcd6094ba983a1d : SUCCESS in 1h 35m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d64b13c10134e8d848bcfc5ca87ffc5 : SUCCESS in 31m 13s","accounts_in_message":[],"_revision_number":1},{"id":"07d472141a35ed06885fce161ffa03f4c46b3258","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-02-10 12:41:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3d7ff510c2dd5aaa2d578a0b121ebccfed1d577a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 14:18:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/77baa18f14ea455db88ca6281bf9fafe : SUCCESS in 43m 44s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/25e0c8177d0f4b99b840193a9cc94e57 : SUCCESS in 51m 37s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/de8038dc0fe6473d823520f2deafa6ba : FAILURE in 52m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/982f20bfd9ad461e82a28e8ad8a0a91c : SUCCESS in 11m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3efb74de08c74893b9a32275d19d69be : SUCCESS in 24m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a181c92b8a204e59a21e558e9986927d : SUCCESS in 27m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/faad15b37a1240bc8f35e975398c473e : SUCCESS in 1h 09m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/34382e0f0db542aa87bd89a23fe37d1f : SUCCESS in 52m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e98fc396ba6c4e41972607d2e4e00a7a : SUCCESS in 1h 12m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/42b52e59b0e94cd98ee2445a083bc54a : SUCCESS in 39m 01s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/023bab9301684fd6bf4c6138538a08be : SUCCESS in 1h 26m 08s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/6f7efec09e684368ab7d4e9f6b18b97e : SUCCESS in 1h 26m 07s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d6b277092352457f92b58e255bdbb1da : SUCCESS in 1h 03m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/201bd7a671a44ddbb662b94af489d07d : SUCCESS in 30m 38s","accounts_in_message":[],"_revision_number":1},{"id":"79c971256b0f545ba66f2519de386e6c2826b5e3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-14 08:53:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"94a7a098765c6e56e5f34a6354ca8a5d54f2d6a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 10:30:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/4ab4235a19c3405b96f42c77f45a2017 : SUCCESS in 42m 08s\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/4140dc1ebcb34bf4812cbcac4aea4875 : SUCCESS in 48m 34s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/c8f6cd5cea434ae7b51c45b622f985d4 : SUCCESS in 57m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef904e6db3b04404b4f8a34328e2bccd : SUCCESS in 12m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/025f72957f96469bb272a361ffa4f1a8 : SUCCESS in 27m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c86df7ba540f4e9d87141a0163bee5fb : SUCCESS in 22m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ce73083e2d5a4cbd86420a536ad3124a : SUCCESS in 1h 29m 15s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6896646182404dbdaabf73bcf8953d04 : SUCCESS in 57m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/66def6bf84014c428e254c0607abd656 : SUCCESS in 1h 14m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fe5091f1b89c4bbcbdc80255f8d9b909 : SUCCESS in 33m 31s\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/8344297e093e48e7bc96eb14b0713127 : SUCCESS in 1h 31m 14s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/401cbad663584a7ebc7e670d63590044 : SUCCESS in 1h 13m 35s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f3a280ed1674476cb386991c57018f73 : SUCCESS in 1h 09m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89842634d7274996b79352eb3c6c2098 : SUCCESS in 24m 57s","accounts_in_message":[],"_revision_number":1},{"id":"111632c46ee821eb9261bac22f2f0a66c49f4e5b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-02-15 08:46:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"edb23ce1034985f6f09bbd1d5dda5592cc7e74de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 08:47:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ce34a9c27c3a9b688162927d7c52e3a401b4e1a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 10:32:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/3a0e29ba29f44a36a1b25e01c14ab1d6 : SUCCESS in 38m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60299ac67c864230bb22cf058535a8ae : SUCCESS in 10m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2f06b70efe6544ffa6e6d8a4600e5519 : SUCCESS in 22m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1ea6e53abf874f24899fc89b840a8177 : SUCCESS in 23m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d966f1f82cc44cdc851460181a8149e6 : SUCCESS in 1h 37m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8d83464ed1084f39b4b2ad7181117b36 : SUCCESS in 1h 06m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e22004577bdf48f38d517765b2d5e021 : SUCCESS in 1h 21m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/97ab5efa327d489eb660bc43790788eb : SUCCESS in 42m 48s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1c9865f2402e405c9f21a068dba5040c : SUCCESS in 1h 38m 52s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/213251bc29ee40e48124848c81fe3085 : SUCCESS in 1h 31m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed3dfc9daa244301b3b46466bf01de22 : SUCCESS in 35m 53s","accounts_in_message":[],"_revision_number":1},{"id":"db009d3836ef424a4fba32c0baa0f29e3f78126e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 10:32:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f17abe2f9e8fa213bc118d8ef0f70477bf855e96","revisions":{"f17abe2f9e8fa213bc118d8ef0f70477bf855e96":{"kind":"REWORK","_number":1,"created":"2022-02-09 10:28:27.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/12/828512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/12/828512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/828512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/828512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/12/828512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/12/828512/1"}}},"commit":{"parents":[{"commit":"5af1e9cdc33b65b2fd4c7eeee3ece17aca3332dd","subject":"Merge \"Do no use \"--strict\" for OF deletion in TRANSIENT_TABLE\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5af1e9cdc33b65b2fd4c7eeee3ece17aca3332dd"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-02-09 10:24:38.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-02-09 10:24:38.000000000","tz":60},"subject":"[stable][queens]: remove publish-openstack-sphinx-docs template","message":"[stable][queens]: remove publish-openstack-sphinx-docs template\n\nSphinx build jobs on older branches (stable/queens, stable/pike) are\nfailing due to missing python 2.7 support for it, see [1], till an\nOpenstack global solution is accepted, let\u0027s remove the job template.\n\n[1]: http://lists.openstack.org/pipermail/openstack-discuss/2022-February/027106.html\n\nChange-Id: I94eb9c6e5be792a9a09f824247eb0aee63f6bda8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f17abe2f9e8fa213bc118d8ef0f70477bf855e96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f17abe2f9e8fa213bc118d8ef0f70477bf855e96"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
