)]}'
{"id":"openstack%2Fneutron~828739","triplet_id":"openstack%2Fneutron~master~Iabaee4e556afb3dc75a82d99dc4a597fe4d7dd21","project":"openstack/neutron","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-02-25 09:11:33.000000000","reason":"Rodolfo Alonso replied on the change"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-02-24 17:41:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iabaee4e556afb3dc75a82d99dc4a597fe4d7dd21","subject":"Refactor session \"is_active\" handling for sqlalchemy-20","status":"MERGED","created":"2022-02-10 16:20:08.000000000","updated":"2022-02-25 09:11:33.000000000","submitted":"2022-02-24 17:41:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"828739-sqlalchemy-20","meta_rev_id":"88778e59a2d83f985a9450f69e97a32fab710e70","_number":828739,"virtual_id_number":828739,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-02-24 10:47:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-24 17:41:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-24 15:57:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-02-24 14:58:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-24 15:57:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-10 17:15:43.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-10 17:56:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-10 18:01:13.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2022-02-10 18:08:36.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-02-15 11:27:37.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"CC"},{"updated":"2022-02-24 14:58:43.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-02-24 15:57:15.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-02-25 07:21:13.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"CC"}],"messages":[{"id":"80943473346ec5564bee9a035a8b83e42e106dae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-02-10 16:20:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac5b955482b5763e893eba631cf429efcfee55b6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-02-10 17:15:43.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2022/Arista_Tempest_Testrefs/changes/39/828739/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"d992da33339da2a46e24b30ba6ed2b49d900b839","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 17:56:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c6330a803c6e40fc8a3cd067128c833e : SUCCESS in 47m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8df7bbda156f4cfa97a6497dd3b29600 : SUCCESS in 54m 28s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/beb24fa11af0466797261f831f8b8bef : SUCCESS in 56m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d054aa257216492692a32c4e691f5c8f : SUCCESS in 48m 45s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0845d080948b4229b7d533dea5f2eded : SUCCESS in 57m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/294779c0d75940d68a6904763254aa67 : SUCCESS in 30m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/530ac1ce7c034c1ca4a4c3a7df580b65 : FAILURE in 30m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/639f48985e4d4f43aa922c0998c10593 : FAILURE in 29m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81d7e78faa6d4a7aa4e43643f9076c9f : SUCCESS in 14m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5eb6ed66395b41318c9bde6df7930752 : FAILURE in 24m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ee0cf87c4fd4759881268ddd87bd2fe : FAILURE in 31m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38d8686271724a97bf694e02a91159b7 : SUCCESS in 10m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b5afe4b57bf0455893778fbb5f2035fd : SUCCESS in 53m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b093cd8a3db44983895d393a23be54fb : SUCCESS in 1h 29m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/46bb53429c2b4ebda942073b5ea0c1f4 : SUCCESS in 1h 14m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/148cc28a2a6c468cba9dc4e993b03d4b : SUCCESS in 57m 28s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cce8927ccd0546879aaad3301299b232 : SUCCESS in 57m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a223e6784bc9412388e04f614a7be9da : SUCCESS in 1h 07m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/37460d71f99d4b53889c8305d8cdca8f : SUCCESS in 1h 14m 25s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1217a5b10b14437f8f58ac7c586ba301 : SUCCESS in 56m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/57529f056cf1465b868ac87abdb53f02 : SUCCESS in 1h 11m 06s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46a078133725402eb3798fa91eb196d8 : FAILURE in 23m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c81f22318a91302f4b10351176dfe98a6a1763a1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-10 18:01:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"997c59c69b3d65b8f042e4c8123c088ac7842164","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-10 18:08:36.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/828739/1 : SUCCESS in 1h 47m 30s","accounts_in_message":[],"_revision_number":1},{"id":"c860919a05f5c06dd859e880488e9e53cc90095b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 19:09:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7429e0596d934bcfb163fc9f26118c60 : FAILURE in 40m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/43f6faf693594572a62b8f578cc87880 : FAILURE in 49m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"27aa6af6f3d2ff06860485043cffee579780dab7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-02-15 11:08:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"48a0d784481e0ae7a42e533dccad7626ee1d4678","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-02-15 12:37:20.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2022/Arista_Tempest_Testrefs/changes/39/828739/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"921c753c8a9d12cbcb7a1986504af849819d9c90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 12:37:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/04bed134edac43fc803955ae0f0e1862 : SUCCESS in 55m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c1f58a8be89d45bab060097824848ed1 : SUCCESS in 34m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"00a4e52f4761ac36111eefaa8b8857ced63a73d5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-15 12:44:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/828739/2 : SUCCESS in 1h 34m 11s","accounts_in_message":[],"_revision_number":2},{"id":"d4651e3a86856ead2bb5924744913f52ccc4bfd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 12:47:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c56d8eb3e68f46818aee2011b804bba5 : SUCCESS in 38m 37s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4642ea0b26e2479eb19185d703c105bd : SUCCESS in 1h 06m 21s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0303899062af41ccb9fb8e6f25c61b52 : SUCCESS in 1h 05m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/37e8a6971496413faf81e4ef14cb3123 : SUCCESS in 1h 02m 57s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/98bbefb35f764d4082ac636bce9c62a2 : SUCCESS in 51m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/45248f25a8ce470ea515cb6dce2ba921 : SUCCESS in 38m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcb4f0b3c1674372ad091c89408dc766 : SUCCESS in 32m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ca490ef3b8d4e68b39403ada9cdd4fc : SUCCESS in 36m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eea842f86fc424d9b1eb42d5685bf2f : SUCCESS in 11m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a7f41c4e0664681b0eeca2ccb7d50a1 : SUCCESS in 26m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d8bc205208f4de3a3389f8340841c8b : SUCCESS in 29m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48f48edbf2a746158d2065f590ffe832 : SUCCESS in 11m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e1d3fca1c77843649f0d3666fb7dd087 : FAILURE in 48m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a428ad724c68411c8b99535607b06c05 : SUCCESS in 1h 32m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f3532e1fe746485abec6d0aef96bea84 : SUCCESS in 1h 03m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb2df9702f9b44e6be5bbb232d6599d2 : SUCCESS in 58m 53s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/85f6075095bc4de0b26ea38a848f4311 : SUCCESS in 1h 06m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d06d910b8d704432a9f1c225141e96c1 : SUCCESS in 1h 14m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/31f03bfca635408abfd1143c15349e68 : SUCCESS in 1h 18m 32s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5a2253d83124e739e7d7d769ca3b034 : SUCCESS in 57m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/64d7449759b142ae905b91cd4593638c : SUCCESS in 1h 09m 03s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4878468beefe43e4a0862067c1f51956 : SUCCESS in 26m 50s","accounts_in_message":[],"_revision_number":2},{"id":"a7a035040f2fc2481d927f93a21eef48988e2d28","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-02-23 17:39:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"df8c53bc7fe5fa50c9aa1a1c6517a46c587d1a55","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-23 17:54:04.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/828739/2 : FAILURE in 13m 51s","accounts_in_message":[],"_revision_number":2},{"id":"59503c3a77e94219df92c354d4289e0becba1a54","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 18:56:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a824d295eec4334af50b5249c923341 : SUCCESS in 1h 11m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b964859b37b4eb8918492e7b7ab3a67 : FAILURE in 1h 09m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f7a3b46619e093fdc40d87b1843551e2ddf67354","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 19:16:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c987747497aa4d79a7a2a181eee85b99 : SUCCESS in 53m 28s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/95c05510c27f43f4bf698905743de846 : SUCCESS in 1h 17m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/be7bfa008c0f4cbc9f46e9423b33ea7a : SUCCESS in 1h 23m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e37961af17bb4a68a1fe9db570ad2eb8 : SUCCESS in 46m 43s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e350ea37520046caac37add920721dd7 : SUCCESS in 1h 11m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e7f09779dd7c4b6db15d3e92df1aad35 : SUCCESS in 27m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f2774b4a3f243e39ac52aec79863231 : SUCCESS in 44m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d64a7694bc2748cbbf90034ae4ae28e9 : SUCCESS in 35m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68ba39cec9c64c019729e775d969ba4e : SUCCESS in 14m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/918b9d24dc69437ba41a46ccd0f3de5a : SUCCESS in 25m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c79ba5a9e8b437a877b1558be6e0db1 : SUCCESS in 30m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5be22d6671c041718d474721780941c9 : SUCCESS in 10m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/be601a36d9d1497f8166580a204f67ff : SUCCESS in 56m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a57dd1287f9f47d08c153f8a96b98aef : SUCCESS in 1h 31m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/963e70c703e64386b901602d811c2442 : SUCCESS in 1h 10m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fc7f30ff8cd240c78333ff9ca06713a3 : SUCCESS in 55m 57s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4a0eb80e6b314649ba05e2fbb3302344 : SUCCESS in 58m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8817b6afbb4c4a1abbc71f83d36e279e : SUCCESS in 1h 15m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c08809d0029c4a4fbe3ad2b9945ed60f : SUCCESS in 1h 16m 22s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86e09062eb5b42219f85ad74eb41dbc5 : SUCCESS in 1h 00m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/92dc8475bd1f4e829c5361c285136ddf : SUCCESS in 1h 18m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8976dcbe43a78e7f11dabccb3de176462aaca75f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-02-24 09:07:39.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"49e7237e849131092be4bbd08811e46893efb3fa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-02-24 09:08:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6f25cffa269ac97e2cf1ed3237021e713af2a42c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-02-24 09:08:49.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"efae5406f1ffb954e6337c4fad4276981f3732c1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-24 09:23:21.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/828739/4 : FAILURE in 13m 48s","accounts_in_message":[],"_revision_number":4},{"id":"c172954bd1d2cd37f166796225d10611e1b1587b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 10:02:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/47019bac3b3c452d9c2320b3e2c6cc31 : SUCCESS in 49m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f667a1c419c3496ea39c6c38920f0650 : SUCCESS in 48m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"21c30f04feb848f6ad81a637f561c172b3841c97","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-02-24 10:03:18.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2022/Arista_Tempest_Testrefs/changes/39/828739/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"963c3adabe5c2b919b66144cfbd237caa6d9de55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 10:44:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c5d2e13514464e468f92698ef61fb1ff : SUCCESS in 51m 24s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/539f6694aa8a4ce189813c4fcb1ec576 : SUCCESS in 1h 32m 49s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4ac2fa28c2cd4287b22323e5d280ef76 : SUCCESS in 1h 17m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5fecc920795242e192d78c4ae639ed8d : SUCCESS in 1h 06m 31s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d76dd0d7e326404f8e0368ee9ae2c2ad : SUCCESS in 56m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b7b467639b0649229081b9973b19a9d4 : SUCCESS in 35m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c335e1383c74769acbe35fd4e3d12ab : SUCCESS in 36m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f13ac2fe3424cf3bcbc9cf99523e91c : SUCCESS in 33m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01bfaacb67e14319b025433acc6dee46 : SUCCESS in 16m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1af477fb28ff4a2f84999aa979ab8065 : SUCCESS in 30m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab548dc1f1f14174a5988ba8af61108c : SUCCESS in 30m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/491df7cf4fd24791857a9b121b436177 : SUCCESS in 9m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0d552c9753124ea784ae832af98582fe : SUCCESS in 59m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a53b56c4879c4f10999a6fc59ee78cd5 : SUCCESS in 1h 21m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b52f5ff702f8430a8abf6d679167d42e : SUCCESS in 1h 20m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9c236eb63450495b93f808744287917c : SUCCESS in 1h 08m 01s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6eacfbb515c64a7da7730fe9606aa5cb : SUCCESS in 47m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9e5e6e6c94184b1aa44d8e2e6489e75a : SUCCESS in 1h 03m 11s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7e487b8230c84fa69f6174d2eb996049 : SUCCESS in 1h 16m 57s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1bdad88d01847bf8eee3095c15fda06 : SUCCESS in 1h 00m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f1df763d89744edba2c546cbc6e0345d : SUCCESS in 1h 09m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d513f4f16ac04cd8f261105aad4b521de6e5af0a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-02-24 10:47:48.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2022/Arista_Tempest_Testrefs/changes/39/828739/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"d94196679ebb85b5904cad14656aec418a200e9a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-24 14:58:43.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b2a211533c4050f6b687fa2787b1bf58f9b7a1cf","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-02-24 15:57:15.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"714a08be1488553b75cf7e5ec1fdf6996561bab6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 15:58:20.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"470f52beef254f52ed8599470f1c2f57f976be58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 17:41:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c8c1359bbc984140ba1df05d4c4cafbe : SUCCESS in 53m 43s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/eebf81ef6ef74712843141a6e6a41fbd : SUCCESS in 1h 17m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8ed5657015464886ac3ee28d31d0071a : SUCCESS in 1h 20m 53s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/22234db4e38f49b58298abdb47a929d1 : SUCCESS in 1h 15m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/68a09497420c4e7a929fffdc9281d599 : SUCCESS in 58m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2cbc2788441d4910a8a8c62138885e40 : SUCCESS in 40m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c95886931034177a3c6f0f39a292e57 : SUCCESS in 15m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2111f2bc5d7b4a88ad2540a4df248de3 : SUCCESS in 20m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d42cee96eb4c4859a561667eec01d078 : SUCCESS in 34m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81797454b33d4bef97d7e65b9435939b : SUCCESS in 9m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/55106fbad3cc4febbca7cbc2a422ac02 : SUCCESS in 55m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e44afed88b374b90a5aec39d30f4cfb7 : SUCCESS in 1h 31m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5de2964201214446b3072f6ac1a1ed26 : SUCCESS in 1h 38m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d9fbc0fb5cf64c248fb8fba85e18566f : SUCCESS in 1h 00m 23s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4c7d937e19d84030930dc7270256ee0a : SUCCESS in 1h 06m 56s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6dd6e2be953485088463aa4f1ae9a42 : SUCCESS in 1h 01m 30s","accounts_in_message":[],"_revision_number":4},{"id":"9a35635c9ef7bdd44fbd22ce50b34429776921e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 17:41:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"3406649fd91fe4e14e089ef251642268facc781d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 17:44:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a20e71bb56b6426c90709904b5793420 : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":4},{"id":"d6670f464e9bc2b9ad65c83c8a247dcd1c32d0f8","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-02-25 07:21:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"88778e59a2d83f985a9450f69e97a32fab710e70","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-02-25 09:11:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"982986507371e78d9c9f5e1d0112efd645c09df9","revisions":{"01c4f3b7936519ff1cd89255946216792122f207":{"kind":"REWORK","_number":1,"created":"2022-02-10 16:20:08.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/828739/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/828739/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/828739/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/828739/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/828739/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/828739/1"}}},"commit":{"parents":[{"commit":"f324ddf769b9d9b68530a191dcf2e419ff2c25d4","subject":"Merge \"Make sure \"dead vlan\" ports cannot transmit packets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f324ddf769b9d9b68530a191dcf2e419ff2c25d4"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-02-05 04:58:20.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-02-05 04:58:20.000000000","tz":0},"subject":"[DNM] Refactor session \"is_active\" handling for sqlalchemy-20","message":"[DNM] Refactor session \"is_active\" handling for sqlalchemy-20\n\nSince sqlalchemy 1.4, \"session.autocommit\" is False by default; in\nsqlalchemy this will be the only value accepted.\n\nThe ``_orm.Session`` is considered active when [1]:\n- there is a transaction and this transaction is active\n- there is no transaction [2], the class ``_orm.Session`` will\n   autobegin when it is first used.\n\nThe second one breaks the way Neutron considers a session is active:\nonly when a transaction is in place, Neutron considers a session is\nactive.\n\n[1]https://github.com/sqlalchemy/sqlalchemy/blob/rel_1_4/lib/sqlalchemy/orm/session.py#L3918-L3950\n[2]https://github.com/sqlalchemy/sqlalchemy/blob/rel_1_4/lib/sqlalchemy/orm/session.py#L3930-L3932\n\nChange-Id: Iabaee4e556afb3dc75a82d99dc4a597fe4d7dd21\nTODO: set a LP/bug, Story or BP\nTopic: sqlalchemy-20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/01c4f3b7936519ff1cd89255946216792122f207"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/01c4f3b7936519ff1cd89255946216792122f207"}]},"branch":"refs/heads/master"},"7807bd0f0706b87b131992c17adefb900b8c82d1":{"kind":"REWORK","_number":2,"created":"2022-02-15 11:08:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/828739/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/828739/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/828739/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/828739/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/828739/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/828739/2"}}},"commit":{"parents":[{"commit":"f324ddf769b9d9b68530a191dcf2e419ff2c25d4","subject":"Merge \"Make sure \"dead vlan\" ports cannot transmit packets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f324ddf769b9d9b68530a191dcf2e419ff2c25d4"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-02-05 04:58:20.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-02-06 09:41:23.000000000","tz":0},"subject":"[DNM] Refactor session \"is_active\" handling for sqlalchemy-20","message":"[DNM] Refactor session \"is_active\" handling for sqlalchemy-20\n\nSince sqlalchemy 1.4, \"session.autocommit\" is False by default; in\nsqlalchemy this will be the only value accepted.\n\nThe ``_orm.Session`` is considered active when [1]:\n- there is a transaction and this transaction is active\n- there is no transaction [2], the class ``_orm.Session`` will\n   autobegin when it is first used.\n\nThe second one breaks the way Neutron considers a session is active:\nonly when a transaction is in place, Neutron considers a session is\nactive.\n\n[1]https://github.com/sqlalchemy/sqlalchemy/blob/rel_1_4/lib/sqlalchemy/orm/session.py#L3918-L3950\n[2]https://github.com/sqlalchemy/sqlalchemy/blob/rel_1_4/lib/sqlalchemy/orm/session.py#L3930-L3932\n\nChange-Id: Iabaee4e556afb3dc75a82d99dc4a597fe4d7dd21\nTODO: set a LP/bug, Story or BP\nTopic: sqlalchemy-20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7807bd0f0706b87b131992c17adefb900b8c82d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7807bd0f0706b87b131992c17adefb900b8c82d1"}]},"branch":"refs/heads/master"},"4933cd99ecfa76a6d6cbb99af1bb3c3cec718e1a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-02-24 09:07:39.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/828739/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/828739/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/828739/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/828739/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/828739/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/828739/3"}}},"commit":{"parents":[{"commit":"f324ddf769b9d9b68530a191dcf2e419ff2c25d4","subject":"Merge \"Make sure \"dead vlan\" ports cannot transmit packets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f324ddf769b9d9b68530a191dcf2e419ff2c25d4"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-02-05 04:58:20.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-02-10 09:01:22.000000000","tz":0},"subject":"Refactor session \"is_active\" handling for sqlalchemy-20","message":"Refactor session \"is_active\" handling for sqlalchemy-20\n\nSince sqlalchemy 1.4, \"session.autocommit\" is False by default; in\nsqlalchemy this will be the only value accepted.\n\nThe ``_orm.Session`` is considered active when [1]:\n- there is a transaction and this transaction is active\n- there is no transaction [2], the class ``_orm.Session`` will\n   autobegin when it is first used.\n\nThe second one breaks the way Neutron considers a session is active:\nonly when a transaction is in place, Neutron considers a session is\nactive.\n\n[1]https://github.com/sqlalchemy/sqlalchemy/blob/rel_1_4/lib/sqlalchemy/orm/session.py#L3918-L3950\n[2]https://github.com/sqlalchemy/sqlalchemy/blob/rel_1_4/lib/sqlalchemy/orm/session.py#L3930-L3932\n\nPartial-Bug: #1962153\nTopic: sqlalchemy-20\n\nChange-Id: Iabaee4e556afb3dc75a82d99dc4a597fe4d7dd21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4933cd99ecfa76a6d6cbb99af1bb3c3cec718e1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4933cd99ecfa76a6d6cbb99af1bb3c3cec718e1a"}]},"branch":"refs/heads/master"},"982986507371e78d9c9f5e1d0112efd645c09df9":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-02-24 09:08:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/828739/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/828739/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/828739/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/828739/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/828739/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/828739/4"}}},"commit":{"parents":[{"commit":"f324ddf769b9d9b68530a191dcf2e419ff2c25d4","subject":"Merge \"Make sure \"dead vlan\" ports cannot transmit packets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f324ddf769b9d9b68530a191dcf2e419ff2c25d4"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-02-05 04:58:20.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-02-10 09:03:36.000000000","tz":0},"subject":"Refactor session \"is_active\" handling for sqlalchemy-20","message":"Refactor session \"is_active\" handling for sqlalchemy-20\n\nSince sqlalchemy 1.4, \"session.autocommit\" is False by default; in\nsqlalchemy 2.0 this will be the only value accepted.\n\nThe ``_orm.Session`` is considered active when [1]:\n- there is a transaction and this transaction is active\n- there is no transaction [2], the class ``_orm.Session`` will\n   autobegin when it is first used.\n\nThe second one breaks the way Neutron considers a session is active:\nonly when a transaction is in place, Neutron considers a session is\nactive.\n\n[1]https://github.com/sqlalchemy/sqlalchemy/blob/rel_1_4/lib/sqlalchemy/orm/session.py#L3918-L3950\n[2]https://github.com/sqlalchemy/sqlalchemy/blob/rel_1_4/lib/sqlalchemy/orm/session.py#L3930-L3932\n\nPartial-Bug: #1962153\nTopic: sqlalchemy-20\n\nChange-Id: Iabaee4e556afb3dc75a82d99dc4a597fe4d7dd21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/982986507371e78d9c9f5e1d0112efd645c09df9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/982986507371e78d9c9f5e1d0112efd645c09df9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
