)]}'
{"id":"openstack%2Fneutron~828758","triplet_id":"openstack%2Fneutron~stable%2Fxena~Ie4093dc6cd63b89e3a62363a4f805ef8287d15b9","project":"openstack/neutron","branch":"stable/xena","topic":"bug/1960006-stable/xena","attention_set":{},"removed_from_attention_set":{"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2022-03-03 13:08:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie4093dc6cd63b89e3a62363a4f805ef8287d15b9","subject":"[ovn] Prevent stale ports in the OVN database","status":"MERGED","created":"2022-02-10 21:53:42.000000000","updated":"2022-03-03 13:11:04.000000000","submitted":"2022-03-03 13:08:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828758-bug/1960006-stable/xena","cherry_pick_of_change":827834,"cherry_pick_of_patch_set":9,"meta_rev_id":"ca7c91ce04d977ce96e8205ad31ffdfc5f699278","_number":828758,"virtual_id_number":828758,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-03 13:08:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-03-03 11:22:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-03-02 07:57:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2022-03-03 11:22:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-10 21:53:42.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"CC"},{"updated":"2022-02-10 23:46:46.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-02-10 23:51:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-02 07:57:54.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-03-03 11:22:31.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"77417a705afa191e090b184412ce17bbc3b93597","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-02-10 21:53:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ab69709c0ccfeaa7e348d35c8b254c0757999ada","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-10 23:46:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/828758/1 : SUCCESS in 1h 52m 09s","accounts_in_message":[],"_revision_number":1},{"id":"6188ecf2a27e09d2bb71c550a9d9ba267bd4bc40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 23:51:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/614a4cc7d3e34921a3f9e0a178a8119a : SUCCESS in 1h 07m 10s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/7572d967c36d42348f6e872bc8ee4a23 : SUCCESS in 49m 51s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/d0b6d8e602a846778af7e0718ea36d22 : SUCCESS in 24m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5406662e1b6249309ffdd3a5cb5deaed : SUCCESS in 29m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/409b4a07937340d9875eea6672d3fc37 : SUCCESS in 13m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d772419927ee42ffa8e0d4f49336f8a4 : SUCCESS in 44m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0bc69c9478f41aaafbfb77573177859 : SUCCESS in 42m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58ee3250299842a3bdb7f79eee252bce : SUCCESS in 22m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff5f7235ccb44882946c251e05fd18be : SUCCESS in 9m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/477f4699d0d54407984f83b36203b7c8 : SUCCESS in 1h 01m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/aca5d79c9d554936947bc25761d7447d : SUCCESS in 1h 22m 53s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0460fa634584257bc8c7d4dcf2f1b14 : SUCCESS in 1h 19m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b9a95e7e943d48cb914d3208ceed0593 : SUCCESS in 1h 09m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ea972797a0d069207fcb47c528de73aaa4618fe6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 00:22:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5a6130e729a4bf0893436671feaea33 : SUCCESS in 53m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cc13c78ef1f4461491ec0167f451ed4a : SUCCESS in 53m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7bf07800fb4373d511d398841ad42fd551999ad","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-02 07:57:54.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"563b27cb57dbd61f09462a0b73ed222c6ff2a2f9","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-03-03 11:22:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ac61f9e6f5e6772f5796030cdd6979d1cf3daa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 11:26:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7b2803c593b0853ee5445d62bb1145da203e7265","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 13:08:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/2fc825e8fc3f4ded87529e2f357b4f76 : SUCCESS in 41m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62bd9ea776e7413f98e3aaafeb8b982b : SUCCESS in 14m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5e0f8902a7641c3ab06b5705ec007d7 : SUCCESS in 29m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ced4e5fd4a5437daa99c0f168a1ced4 : SUCCESS in 23m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40e65a9ae7044114abaa6c5ccd0cb9fe : SUCCESS in 11m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/85977fcb323f47be8d1a7aa7b659d5f2 : SUCCESS in 50m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/eb2bf5820ded4c469090bf0ba274503d : SUCCESS in 1h 37m 10s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3de09af5680489bb49031c931437104 : SUCCESS in 1h 01m 25s","accounts_in_message":[],"_revision_number":1},{"id":"335f776d13139c05c61164e561d55fd8256fa752","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 13:08:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ca7c91ce04d977ce96e8205ad31ffdfc5f699278","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 13:11:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18af701f71954a619d46ba51bbe5ae70 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d67b05ddea02dbe5d7a0563fa2f37f472c945e40","revisions":{"d67b05ddea02dbe5d7a0563fa2f37f472c945e40":{"kind":"REWORK","_number":1,"created":"2022-02-10 21:53:42.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/58/828758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/58/828758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/828758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/828758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/828758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/58/828758/1"}}},"commit":{"parents":[{"commit":"2796ea1f4f4d3be15dad8a24718bc319d51f9e8e","subject":"Merge \"Make sure \"dead vlan\" ports cannot transmit packets\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2796ea1f4f4d3be15dad8a24718bc319d51f9e8e"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2022-02-04 10:32:47.000000000","tz":60},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2022-02-10 21:53:42.000000000","tz":0},"subject":"[ovn] Prevent stale ports in the OVN database","message":"[ovn] Prevent stale ports in the OVN database\n\nUnder a lot of load, there can be situations where all the Neutron\nworkers have not updated their in-memory copy of the NB database\nin time before certain operations.\n\nThis scenario can lead to stale resources when a somewhat recently\ncreated port is attempted to be deleted, but the worker handling\nthis deletion doesn\u0027t know about the OVN port yet.\n\nThis patch detects this condition and allows some time (at least one\nmaintenance task cycle) before it deletes the OVN revision number.\nIf the port then shows up in the OVN database within that window, then\nit will be deleted later by the maintenance task avoiding the stale\nports. If not, the revision number row will be deleted and we won\u0027t\nstale these entries either.\n\nCloses-Bug: #1960006\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\nChange-Id: Ie4093dc6cd63b89e3a62363a4f805ef8287d15b9\n(cherry picked from commit be7331c8169c53e3900c9c1a08e12808cf5ed2ec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d67b05ddea02dbe5d7a0563fa2f37f472c945e40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d67b05ddea02dbe5d7a0563fa2f37f472c945e40"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
