)]}'
{"id":"openstack%2Fneutron~828799","triplet_id":"openstack%2Fneutron~stable%2Fussuri~Ie4093dc6cd63b89e3a62363a4f805ef8287d15b9","project":"openstack/neutron","branch":"stable/ussuri","topic":"bug/1960006-stable/ussuri","attention_set":{},"removed_from_attention_set":{"23804":{"account":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"last_update":"2022-03-04 10:08:55.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2022-03-04 10:08:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie4093dc6cd63b89e3a62363a4f805ef8287d15b9","subject":"[ovn] Prevent stale ports in the OVN database","status":"MERGED","created":"2022-02-10 21:59:29.000000000","updated":"2022-03-04 10:12:14.000000000","submitted":"2022-03-04 10:08:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828799-bug/1960006-stable/ussuri","meta_rev_id":"fcbea0dfea53f393962cd2ba59b8ccbd37d1dda8","_number":828799,"virtual_id_number":828799,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-04 10:08:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-03-03 11:22:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-03-02 07:59:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-03-04 08:16:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2022-03-03 11:22:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-10 21:59:29.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2022-02-11 01:43:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-02 07:59:03.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-03-03 11:22:59.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-03-04 08:16:06.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"e28db777cd590e759de4ffbbf7786702e9c5279e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-02-10 21:59:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0004e87a91f298315b3ef1d3be4a1876412c6f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 01:43:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/889c3fccc7d74eb2812f3cc607dbdfea : SUCCESS in 57m 14s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/786e57fa9c364771992f2edd25d4fec8 : SUCCESS in 2h 35m 35s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/9863e9203014440097add890f8cfa6bc : SUCCESS in 44m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff980590ad3b4bd7a84c89acffb27670 : SUCCESS in 45m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cdb9c22feb54f5fab9ea6588d8f79a7 : SUCCESS in 14m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57d1d8a2b9b64db8857ff9b4542f0fcf : SUCCESS in 29m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/665e17adf15a4c4ba968624ed654697b : SUCCESS in 34m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5daaa1150e7b4482a22288786e1d4d53 : SUCCESS in 24m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88b285114d994cf2b02f7beb67fdea12 : SUCCESS in 9m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3c9230ea70c24d9ba11e5ef4c3b5869a : SUCCESS in 1h 13m 36s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e53167184a894ea1afa47ae38d5a382d : SUCCESS in 1h 25m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e19788e13e064798900e54da66352e69 : SUCCESS in 1h 22m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0f3bea81afe742fdb519082419c4999c : SUCCESS in 1h 38m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb137356aef34df5b4f1bb416af218ac : SUCCESS in 1h 21m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/424d0595fea641b59519d115dc8c1887 : SUCCESS in 1h 23m 04s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/aa784f0e5b21456aa9a936d9788fd91b : SUCCESS in 1h 40m 40s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/bef29446fd4a4a379d5f7543971a8e3e : SUCCESS in 1h 17m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b1c416bc38a6483398b2783f98234356 : SUCCESS in 1h 36m 26s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2431f5d2d97f492da5fd28082bce0aa3 : SUCCESS in 1h 41m 27s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7f235cb5ca2a4adfacda57207c07ba87 : SUCCESS in 1h 26m 51s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e671f261885341eda6d98b8a89603bce : SUCCESS in 1h 21m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/20b2a4e233084f549d88273c68101e86 : SUCCESS in 1h 05m 05s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1601b34b71df4c52b6528c0449bd0da2 : SUCCESS in 1h 42m 39s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/345d4969de2a4f8992e0aecc6f7802f0 : SUCCESS in 1h 07m 37s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e4afa204638a41deb578a6d5c75a079a : SUCCESS in 1h 07m 39s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/6e432ff5a2594e6794bf3fc6a7e2b5dc : SUCCESS in 1h 04m 23s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e41172021cbb46029c98bfd0b3f5b885 : SUCCESS in 1h 23m 57s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/075e7416a0164b2080a3ddbbb2b1dad2 : SUCCESS in 1h 02m 30s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e1e6c3bf830d4df290e35f881f67a80d : SUCCESS in 3h 42m 02s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2c63519ceb9f41d4b036ab30ead1b4cf : SUCCESS in 1h 27m 04s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/065e7d3911a4442a9e6d7ae7352608eb : SUCCESS in 1h 57m 25s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/39ffc7e44ee74baabe6cb52a1015f5ab : SUCCESS in 2h 02m 16s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/05f4d4830a564026ba3c4386f5de3298 : FAILURE in 1h 26m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d14128ed17049e2698050c3b20472f42273bdc47","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-02 07:59:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a30a2e756fb72f176d6704fac92452fa44c267b4","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-03-03 11:22:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"36749f9406edad4162eb143ae484b10102d7991d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-03-04 08:16:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aaea85ce230462acde4354c632614c8af699e8d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 08:17:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4286e61f72a664607f417943d2a0046e6a3f1ff7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 10:08:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/fd96b6d7ec384a2ab96e7fb23fa0f12a : SUCCESS in 57m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d238cb4c64a4014825202c6f6d58294 : SUCCESS in 17m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc8526703f6a4a709a26b853e9ff3e5d : SUCCESS in 29m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/43ce55ff8bd744ed9ad724b36a0a7341 : SUCCESS in 30m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c5c0126aec94395a97db9fc1237437f : SUCCESS in 10m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a9a73344191149e8aebafb16816007ed : SUCCESS in 59m 00s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3c2ff0f26177482d85fe76950427e276 : SUCCESS in 1h 34m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e36263d138e2431f814b4d4234faa200 : SUCCESS in 1h 25m 37s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/647d4d4e737242b4a239d3aabbc410fa : SUCCESS in 1h 18m 26s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/75adc7bb6bd74cc9be10c0c8f00d1231 : SUCCESS in 1h 00m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d80b2b2e2064254b7a779a9dbf8b22e : SUCCESS in 1h 22m 38s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/caf482b90eb449438f2a1d6f1f0df4fc : SUCCESS in 1h 25m 30s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ab765356a7ad42c9a66aaeb88344b7de : SUCCESS in 1h 43m 44s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9082df840e3c438fa7747d08e1e8627b : SUCCESS in 1h 09m 44s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1a7a065bed4a4daf8300407240522087 : SUCCESS in 1h 24m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b23d342fccebc4a3b2172ef10ae85534051b806e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 10:08:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fcbea0dfea53f393962cd2ba59b8ccbd37d1dda8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 10:12:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab471aa80fd1468182e00235a72d468f : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8164c27d9ad1de198a5943d75ceff473d0218b3d","revisions":{"8164c27d9ad1de198a5943d75ceff473d0218b3d":{"kind":"REWORK","_number":1,"created":"2022-02-10 21:59:29.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/99/828799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/828799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/828799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/828799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/828799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/828799/1"}}},"commit":{"parents":[{"commit":"f016548d2edb17e531f4ee29fdeb75b123d31f27","subject":"[OVN] Do not delete port\u0027s revision on RowNotFound","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f016548d2edb17e531f4ee29fdeb75b123d31f27"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2022-02-04 10:32:47.000000000","tz":60},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2022-02-10 21:58:58.000000000","tz":-300},"subject":"[ovn] Prevent stale ports in the OVN database","message":"[ovn] Prevent stale ports in the OVN database\n\nUnder a lot of load, there can be situations where all the Neutron\nworkers have not updated their in-memory copy of the NB database\nin time before certain operations.\n\nThis scenario can lead to stale resources when a somewhat recently\ncreated port is attempted to be deleted, but the worker handling\nthis deletion doesn\u0027t know about the OVN port yet.\n\nThis patch detects this condition and allows some time (at least one\nmaintenance task cycle) before it deletes the OVN revision number.\nIf the port then shows up in the OVN database within that window, then\nit will be deleted later by the maintenance task avoiding the stale\nports. If not, the revision number row will be deleted and we won\u0027t\nstale these entries either.\n\nConflicts:\n\tneutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n\nCloses-Bug: #1960006\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\nChange-Id: Ie4093dc6cd63b89e3a62363a4f805ef8287d15b9\n(cherry picked from commit be7331c8169c53e3900c9c1a08e12808cf5ed2ec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8164c27d9ad1de198a5943d75ceff473d0218b3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8164c27d9ad1de198a5943d75ceff473d0218b3d"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
