)]}'
{"id":"openstack%2Fneutron~829039","triplet_id":"openstack%2Fneutron~stable%2Fxena~I5c1474b405d436d3b1e5db745d77999f1723b660","project":"openstack/neutron","branch":"stable/xena","topic":"bug-1869244-stable/xena","attention_set":{},"removed_from_attention_set":{"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2022-02-21 13:41:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5c1474b405d436d3b1e5db745d77999f1723b660","subject":"Wait longer before deleting DPDK vhu trunk bridges","status":"MERGED","created":"2022-02-15 00:03:05.000000000","updated":"2022-02-21 13:43:26.000000000","submitted":"2022-02-21 13:41:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829039-bug-1869244-stable/xena","cherry_pick_of_change":827580,"cherry_pick_of_patch_set":1,"meta_rev_id":"e2b9de8bf2d832a8cc6b780318df634ead9c95a0","_number":829039,"virtual_id_number":829039,"owner":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-21 13:41:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2022-02-21 11:51:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-02-17 16:40:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2022-02-18 07:12:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2022-02-21 11:51:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-15 01:26:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-15 01:46:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-15 01:50:23.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-02-17 16:40:54.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-02-18 07:12:54.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-02-21 11:51:52.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"b7a18e456d7b8681c4fa7bbcf8908869370c3652","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-02-15 00:03:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a9b414072a8679e7d73931a214e784766d00e44e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 01:26:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/72a6430f4b6b49fcabfd3cf15bccfa7c : SUCCESS in 53m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a3823e3b4374e35859bef79c809700e : SUCCESS in 56m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4eafcd4532840c2193c3d73c64de9cbed448cbf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 01:46:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/be1468fa4d7d4c0898510248a63cd3a9 : SUCCESS in 50m 11s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/a848f68216704511a42c832b32b9bd31 : SUCCESS in 1h 07m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/7e9a39096ecf476bb0572cfb9c82eded : SUCCESS in 44m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33579a41d57e4b38a9c893ba1022334d : SUCCESS in 44m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f1df6ef40094cd88af5fae1410e60cd : SUCCESS in 10m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33975a726ead4c9998dcaa20ff8de56b : SUCCESS in 58m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb22dfc011ce477eab01e801c6bd913a : SUCCESS in 21m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/997babb22c4b4f4db1d5e923865cedae : SUCCESS in 29m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21bbf236b7e4454fa6d64c1540aa4e6e : SUCCESS in 8m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2a28a90095e14d44bae6bb0050fecf3e : SUCCESS in 47m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/46e10e247cde41e48dfdfede9eeee0a8 : SUCCESS in 1h 40m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/91a6398923364a7a89cf46b970e40235 : SUCCESS in 1h 36m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1510fc97604f47c6b9539baa22ee2d34 : SUCCESS in 1h 01m 33s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a09bc87d0efd40a9b3848d5fde4fafca : SUCCESS in 1h 04m 00s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2f513fd950674d619381bf93162192ea : SUCCESS in 1h 18m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bcf285f8890244dea4d23b251dc326d0 : SUCCESS in 1h 13m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ed12593237c05b2836c2cc9c0a261d2a10598ad6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-15 01:50:23.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/829039/1 : SUCCESS in 1h 46m 34s","accounts_in_message":[],"_revision_number":1},{"id":"65dc3631d86b89823a53ca7a2b24db4112389621","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-17 16:40:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1a3d436f13344022e0ee04b5768af82bb12b6c7e","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-02-18 07:12:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1ab55e1b29478dcd90952fac3d3ce62e612a3948","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-02-21 11:51:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d8ab58fde955a42ebe6853db3f16b4b1bf56936c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 11:52:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bddfc684a0edcc4aa2d6e1f0d06f7e248d69d091","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 13:41:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/f31c7dffa92641f69580d411210ffda2 : SUCCESS in 40m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf789bee217241d69b7e668e21f445b1 : SUCCESS in 12m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc9dd243e3f94ce09222922a2f655183 : SUCCESS in 21m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f2dc0aad3094b92a75d7b9c2093ab83 : SUCCESS in 21m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db75dfd4633c48b285935b10eb9bccf9 : SUCCESS in 10m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fab6e1c97d164246afc9dd67a02be608 : SUCCESS in 42m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/10bcaa7e7d5a403b9901f6d6fe79922f : SUCCESS in 1h 27m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9e852b70455d40c3aca2ecffe284d34a : SUCCESS in 1h 07m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/adcd543ac40d47469ac5124042178175 : SUCCESS in 58m 26s","accounts_in_message":[],"_revision_number":1},{"id":"ae2b1ea03e12f30b5967d588f1415c1e21a556b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 13:41:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e2b9de8bf2d832a8cc6b780318df634ead9c95a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 13:43:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd36dbf0949642ebb6d0f03d71dbffb9 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac4aa7ef7e33a5ff9fbc9ce5c8249c40334e66a4","revisions":{"ac4aa7ef7e33a5ff9fbc9ce5c8249c40334e66a4":{"kind":"REWORK","_number":1,"created":"2022-02-15 00:03:05.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/39/829039/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/829039/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/829039/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/829039/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/829039/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/829039/1"}}},"commit":{"parents":[{"commit":"2796ea1f4f4d3be15dad8a24718bc319d51f9e8e","subject":"Merge \"Make sure \"dead vlan\" ports cannot transmit packets\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2796ea1f4f4d3be15dad8a24718bc319d51f9e8e"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2022-02-03 01:25:40.000000000","tz":-360},"committer":{"name":"Miguel Lavalle","email":"miguel@mlavalle.com","date":"2022-02-15 00:03:05.000000000","tz":0},"subject":"Wait longer before deleting DPDK vhu trunk bridges","message":"Wait longer before deleting DPDK vhu trunk bridges\n\nIn [1] we added a delay before deleting DPDK vhu trunk bridges to\nmitigate a race condition when instances are rebooted. As explained in\n[1], with DPDK rhu, a reboot is esentially a super fast bridge\ndelete-then-create that is prone to race conditions. We have recently\nencountered in customer deployments that the wait added in [1] is not\nlong enough. As a consequence, this change increases the wait.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/717394\n\nChange-Id: I5c1474b405d436d3b1e5db745d77999f1723b660\nPartial-Bug: #1869244\n(cherry picked from commit 140bb63665223d7cd2a7fee8c1d1494ebd2a802f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac4aa7ef7e33a5ff9fbc9ce5c8249c40334e66a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac4aa7ef7e33a5ff9fbc9ce5c8249c40334e66a4"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
