)]}'
{"id":"openstack%2Fneutron~829437","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~Ic363c5c805d02bdc93317423b95a1cd89c16876a","project":"openstack/neutron","branch":"stable/wallaby","topic":"improve-neutron-ci-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-02-24 15:23:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic363c5c805d02bdc93317423b95a1cd89c16876a","subject":"Remove run-devstack role from run_functional_job playbook","status":"MERGED","created":"2022-02-16 11:50:41.000000000","updated":"2022-02-24 15:26:58.000000000","submitted":"2022-02-24 15:23:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829437-improve-neutron-ci-stable/xena-stable/wallaby","cherry_pick_of_change":829436,"cherry_pick_of_patch_set":1,"meta_rev_id":"2fe643bf5f9554189b1a55b9b1356deece0233b1","_number":829437,"virtual_id_number":829437,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-24 15:23:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-02-24 13:53:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-02-22 14:04:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-02-24 14:47:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-02-24 13:53:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-16 12:57:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-16 13:45:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-16 14:11:40.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-02-22 14:04:12.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-02-24 13:53:33.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-02-24 14:47:29.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"8e24853a5c99cebf864f09effcf890945daeba75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-02-16 11:50:41.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"286fa18fbb6ae5266f65ff5f8fea532a5a8ee3e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 12:57:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ffbc93350d394793bc24dc10ca4a95af : TIMED_OUT in 1h 01m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b720b26f944c4c9ebcfdf2e532afc807 : TIMED_OUT in 1h 01m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0dddd79ca8e6e128cb38d3a14c0894dbd0a615f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 13:45:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/241c420317a446c6b029245fbad942fb : SUCCESS in 43m 51s\n- neutron-tempest-plugin-scenario-linuxbridge-wallaby https://zuul.opendev.org/t/openstack/build/454780ecc9a0408d9beeeb930b3489e4 : FAILURE in 1h 29m 47s\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/bfe3fc4e255e48efbf15bbc345070abf : SUCCESS in 1h 35m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-wallaby https://zuul.opendev.org/t/openstack/build/2d2e00c922164808b5931d7abad9866b : SUCCESS in 1h 12m 07s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/96fdb23795db4932ab6264ffce898f04 : SUCCESS in 1h 33m 11s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/73bc7f65e81c4975a86674dbcad70df2 : SUCCESS in 38m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82cdeb5c30d04420b56899ea86a6f5fa : SUCCESS in 33m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b71c9e33d07404fbcb11ca201102a9c : FAILURE in 12m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f9450b8fdb24ef9a298c98631b47ed9 : SUCCESS in 23m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5281f71c1b540ea9a4e1d40c2622f24 : SUCCESS in 21m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/346cb2fff92a482d848dcd4932b88c58 : SUCCESS in 23m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7a8a0b7745647668ebf72f65cd48db0 : SUCCESS in 10m 52s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/066d5b6dc82e48bda45b4345e0770583 : SUCCESS in 1h 05m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5724e5d1b5b54dc4aa11da8a488c48de : SUCCESS in 52m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5c4c8b24cda4433e8a06bf6db4568a7b : SUCCESS in 58m 07s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/50e13b46cffd4e979a11591c43e63d73 : SUCCESS in 1h 10m 17s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1993344baed5463aaf1e8add80b73c45 : SUCCESS in 1h 17m 18s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ff64a07644c0474eacf254391db5c8c3 : SUCCESS in 1h 42m 06s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3cae8daabca7423587f856c0ead534e0 : SUCCESS in 1h 10m 28s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4f06b15e7a8e49feada82d652131aa16 : SUCCESS in 59m 13s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a9502c7bb83149859eda68e2179eea49 : SUCCESS in 32m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/50eef455c827440990c4819f9e2dea9f : SUCCESS in 1h 49m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9c0a15faf275432ab5d749dc9c4b1d89 : SUCCESS in 1h 11m 49s\n\nWarning:\n  Comments left for invalid file neutron/agent/l2/extensions/fdb_population.py\n  Comments left for invalid file neutron/agent/l3/dvr_local_router.py\n  Comments left for invalid file neutron/agent/l3/router_info.py\n  Comments left for invalid file neutron/agent/linux/bridge_lib.py\n  Comments left for invalid file neutron/agent/linux/interface.py\n  Comments left for invalid file neutron/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/agent/linux/tc_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/tc_lib.py","accounts_in_message":[],"_revision_number":1},{"id":"677627f929e3d1224bb261579c046eaa265bc1ec","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-16 14:11:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/829437/1 : SUCCESS in 2h 19m 54s","accounts_in_message":[],"_revision_number":1},{"id":"e328edd80bb26fcd13e03f92fbee39ce9899b8d4","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-02-22 14:04:12.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"038b9e7c5880cb456d81ed627bdb3204bd509cf1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-22 14:18:34.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/829437/1 : FAILURE in 13m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d375d64de74d7dfe240c61b0e921ab6100c50d71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 15:07:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bb8db36be7d1443d9252395b6516f856 : SUCCESS in 53m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f77e39752bb444b8ea9e14b6b5ffb71 : SUCCESS in 53m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a16a03965ef3fce33021323633490a0c0e8ee165","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 15:53:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/c3b0411c7ed147c08058c1caa8ca6b2e : SUCCESS in 55m 04s\n- neutron-tempest-plugin-scenario-linuxbridge-wallaby https://zuul.opendev.org/t/openstack/build/30bfbcfcca4b4ff8926dd8cad6e79840 : SUCCESS in 1h 37m 09s\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/3302586564fb4a23ac0f928885e678cd : SUCCESS in 1h 23m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-wallaby https://zuul.opendev.org/t/openstack/build/1ac4b39360474491b1af1dcc19f53a5b : SUCCESS in 1h 21m 13s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/41cbe80e8394400aa7403eb450738780 : SUCCESS in 1h 28m 42s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/e414471dd39a46d197abd3b758c4787d : SUCCESS in 42m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e04c55db9f4a42e4bd5f0ec5895bebf8 : SUCCESS in 43m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5817999870a4597a9ea7420624728d7 : SUCCESS in 14m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/536cf8a314014f479f46b2f12d4cb948 : SUCCESS in 42m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bec1352ae104a3ab0893621f28650d8 : SUCCESS in 30m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43241d871c2a4ef28fce5ea812507bf2 : SUCCESS in 30m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdfc5df63c1241c985cdf243e8d55863 : SUCCESS in 11m 11s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6b0c35e1a56c4cdb8517f88debf7e173 : SUCCESS in 1h 21m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0a296f8cc66c40bd8463e13aed7d9d91 : SUCCESS in 1h 06m 58s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b9196351c74241ccbf8b8e350f0131a2 : SUCCESS in 1h 10m 01s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/305b3d9f629f4ceba0c85b01d16b24d7 : SUCCESS in 1h 19m 45s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/16361cddce4c4a4db59cdbcd44e34038 : SUCCESS in 1h 13m 07s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eb1b401c0088401486de784587f7f91e : SUCCESS in 1h 42m 41s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc6b37b06e37487d99b67ef299d66759 : SUCCESS in 1h 03m 52s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/721458a4dc64403596ac05aad6b51549 : SUCCESS in 1h 05m 55s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/deec1ff26d974b209f63eb0b3e0f6807 : SUCCESS in 37m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b79f6ca206964b3c87d7a028e2641c38 : SUCCESS in 1h 30m 38s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/41f82254799b4f1db175ef405ea95b72 : SUCCESS in 1h 21m 22s","accounts_in_message":[],"_revision_number":1},{"id":"6912bf3c8b2ce138be0b2d502d953a72cdbe816f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-02-24 13:53:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"866d77acb98e4042dd6c514f6d03da7634136eea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 13:54:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"88d80411fbbfbcdd62207f9bfe22ad4ed8d76b81","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-24 14:47:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4fcc1c801876369220560734fb89e974f37449f4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-24 14:48:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"38a78f4d9c52ff97ebe5958037a8db624162cbed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 15:23:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/94ef6dc5a12946d08b5091f5e562d795 : SUCCESS in 39m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de5eaf550275411ca1db4b017dc64125 : SUCCESS in 17m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c444762e58bc44afa168b3b35bf7bbfe : SUCCESS in 37m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/974f2cd24ef84d3c9a3fe27cbc1d3035 : SUCCESS in 26m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21287e4c08654ab981e3278104e9182d : SUCCESS in 9m 31s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2707eb34b4c34365868614d21857aa03 : SUCCESS in 1h 20m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/53301c5e328c4e5f8d46ad1da006669f : SUCCESS in 1h 03m 08s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9292d63b236f43528e898c76e7ac7b04 : SUCCESS in 1h 28m 17s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78f2aa24a3a64653875deca34369698a : SUCCESS in 52m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/14fa668009064a3aa6ecaa68cbf4e864 : SUCCESS in 58m 40s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/53d95b8da1394e948ede09007f326962 : SUCCESS in 35m 09s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5459b9981a8f4baf816fc94df8cc5663 : SUCCESS in 1h 08m 07s","accounts_in_message":[],"_revision_number":1},{"id":"9ce5bc8ef30e7165975050a3db0b08811dd4bd97","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 15:23:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2fe643bf5f9554189b1a55b9b1356deece0233b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 15:26:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02f62064a0e647c3abf33349aaf29e56 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a4f4dbcf106edfb5b43395cd6335704b4585fd3d","revisions":{"a4f4dbcf106edfb5b43395cd6335704b4585fd3d":{"kind":"REWORK","_number":1,"created":"2022-02-16 11:50:41.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/37/829437/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/829437/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/829437/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/829437/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/829437/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/829437/1"}}},"commit":{"parents":[{"commit":"8ad5beb479838cb5a9ccef513513fa06e483110c","subject":"Merge \"[Stable Only] Enforce policy for qos_policy_id attribute\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8ad5beb479838cb5a9ccef513513fa06e483110c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-11-02 13:05:14.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-02-16 11:50:41.000000000","tz":0},"subject":"Remove run-devstack role from run_functional_job playbook","message":"Remove run-devstack role from run_functional_job playbook\n\nThere is no need to run that role in the functional nor fullstack jobs\nnot running it can save few minutes in each job\u0027s run.\n\nChange-Id: Ic363c5c805d02bdc93317423b95a1cd89c16876a\n(cherry picked from commit 83f57021d6f317c86d1f2cf8cd38abd8863bd617)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4f4dbcf106edfb5b43395cd6335704b4585fd3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4f4dbcf106edfb5b43395cd6335704b4585fd3d"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
