)]}'
{"id":"openstack%2Fneutron~829475","triplet_id":"openstack%2Fneutron~master~I50d5d18056b868d03aea782bd603133872fbc8d9","project":"openstack/neutron","branch":"master","topic":"dnm","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-04-14 19:34:57.000000000","reason":"Change was abandoned"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-04-14 19:34:57.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I50d5d18056b868d03aea782bd603133872fbc8d9","subject":"[DNM] Check switch to importlib","status":"ABANDONED","created":"2022-02-16 05:37:03.000000000","updated":"2022-04-14 19:34:57.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"731bf2a86731b86d267fa9b1a47cc1caae82b74a","_number":829475,"virtual_id_number":829475,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-02-22 13:04:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-02-22 12:53:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-16 06:21:35.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-16 06:22:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-16 06:25:28.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-16 06:39:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-16 07:08:24.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-04-14 19:34:57.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"f5bd0c9ce2a1d93f6b5e3de7db608547efc21c68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-16 05:37:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93df7d53ae51548e828141636cb65e0e7767f734","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-02-16 06:21:35.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/75/829475/1/openstack-sriov-neutron-master-direct : SUCCESS\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/75/829475/1/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"13e308338fcd8373d2eba3126468d892b60119a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 06:22:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d4d64c8998414247952c1ed7fa14f34c : SUCCESS in 32m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/12818a52acf6428c97eb11e93b76e120 : SUCCESS in 43m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e6525b5e3707748e984aa934dc3ce7c964ed898","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-02-16 06:25:28.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2022/Arista_Tempest_Testrefs/changes/75/829475/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"66937e91202bc0457dd9a5092f9891c6b7926835","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 06:39:31.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84233f581a574af391ffa3a99908a4ed : SUCCESS in 56m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a063d00534904fb0b11b3b437c6c6d7a : SUCCESS in 33m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81147ecf17ac409bb43009d38fdd8a10 : FAILURE in 13m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd0ecde1afcb42c29349d6f5fb99c287 : SUCCESS in 35m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a58babfdfe1440d864cbb8084de6a92 : SUCCESS in 24m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb03134397444dd3b0a4deeb020d992b : SUCCESS in 28m 31s","accounts_in_message":[],"_revision_number":1},{"id":"b5ee5edcb11dac11af92e48513a3b061a3ad5eb7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-16 07:08:24.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/829475/1 : SUCCESS in 1h 30m 39s","accounts_in_message":[],"_revision_number":1},{"id":"902a2e545850b861de312bedd82cfc70b30fa7fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-16 12:21:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"42b83220a153337d2305d77b152e415b337122b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 13:04:37.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b84d08fdbfa47f1afbf251d624268ad : FAILURE in 36m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a0df8f852dc4582b7937c8463494fa5 : FAILURE in 39m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0549d23b6c474bad97bef934731b16a3 : FAILURE in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/744aeba203434a5e98cb024cd2904fef : FAILURE in 32m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e37ff12cbfa7410b9963de4a7be8b62a : FAILURE in 26m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e251bc76c31e450eab40bb5fc60f09d3 : FAILURE in 26m 18s","accounts_in_message":[],"_revision_number":2},{"id":"4dd151b0de5f4bfc6f2782ca75b5156cead50313","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-02-16 13:06:40.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/75/829475/2/openstack-sriov-neutron-master-direct : SUCCESS\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/75/829475/2/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"702ad4e23dd7d3cc58e54bafdd8d7df8bb3ef7dc","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-16 13:17:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"842c9c69dc882fb8d16398dadaf783b9cb3d2b1f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-02-16 13:28:29.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2022/Arista_Tempest_Testrefs/changes/75/829475/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"bff15e628c404ceb5d42dc5b09d63cc1af04cb31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 13:55:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6447097c0a474d4a8c6c76e24f67937b : FAILURE in 41m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0844bd1e5c0e464198aa4aca5e55e1e9 : FAILURE in 1h 01m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bed2b3e3176c313bffac4a7245106a2fc4c3a54d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-16 15:15:10.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/829475/2 : SUCCESS in 1h 43m 56s","accounts_in_message":[],"_revision_number":2},{"id":"4541a8b7df42271fae6e5428def8485abbaae277","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-02-22 12:13:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"be88b9438daa68179218c82e91729871155ce9d0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-22 12:28:01.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/829475/3 : FAILURE in 13m 52s","accounts_in_message":[],"_revision_number":3},{"id":"4fec654ce9b4d5ed389b90d64f907a3fa0694451","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 12:53:22.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09e1fd2e5e00497b87993acf80c80284 : SUCCESS in 28m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/323d0b7f3add4fb0bf618435020daac7 : SUCCESS in 34m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82ad6d9ab24c448aa5068b53f09bb884 : FAILURE in 8m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e192f7605f5848c5b95e7093b2bc2d73 : SUCCESS in 23m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/858fd87172194a389c656b0a1e683a9e : SUCCESS in 27m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63f07520e01146cbaa027c6f93b0ac02 : SUCCESS in 31m 36s","accounts_in_message":[],"_revision_number":3},{"id":"bebc3a9c5fad499fdd52c4ef5052674b2d797797","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-02-22 13:04:14.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2022/Arista_Tempest_Testrefs/changes/75/829475/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"96ed822c16c5a1e3ccb6f9783a03696003971866","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 13:11:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/421ad0e2b3844e32ba036dca49acffe6 : SUCCESS in 50m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75772e14247e47428dcbf60ae91a4ab7 : SUCCESS in 52m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"731bf2a86731b86d267fa9b1a47cc1caae82b74a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-14 19:34:57.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"69397a7a2ff0abaf55a97a76b08691cd6a1e2311","revisions":{"fdc0396d67e966d82faaf77b9b0224f7c7fa4255":{"kind":"REWORK","_number":1,"created":"2022-02-16 05:37:03.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/75/829475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/829475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/829475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/829475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/829475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/829475/1"}}},"commit":{"parents":[{"commit":"1df18a885e8af6a55e98d9f56d1c042dcda12db3","subject":"Merge \"Updating python testing as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1df18a885e8af6a55e98d9f56d1c042dcda12db3"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-16 05:36:39.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-16 05:36:39.000000000","tz":330},"subject":"[DNM] Check switch to importlib","message":"[DNM] Check switch to importlib\n\nChange-Id: I50d5d18056b868d03aea782bd603133872fbc8d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fdc0396d67e966d82faaf77b9b0224f7c7fa4255"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fdc0396d67e966d82faaf77b9b0224f7c7fa4255"}]},"branch":"refs/heads/master"},"fec1ed6039f56ef87dfbb8f39bd9f67cb7238dde":{"kind":"REWORK","_number":2,"created":"2022-02-16 12:21:42.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/75/829475/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/829475/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/829475/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/829475/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/829475/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/829475/2"}}},"commit":{"parents":[{"commit":"1df18a885e8af6a55e98d9f56d1c042dcda12db3","subject":"Merge \"Updating python testing as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1df18a885e8af6a55e98d9f56d1c042dcda12db3"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-16 05:36:39.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-16 12:21:23.000000000","tz":330},"subject":"[DNM] Check switch to importlib","message":"[DNM] Check switch to importlib\n\nChange-Id: I50d5d18056b868d03aea782bd603133872fbc8d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fec1ed6039f56ef87dfbb8f39bd9f67cb7238dde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fec1ed6039f56ef87dfbb8f39bd9f67cb7238dde"}]},"branch":"refs/heads/master"},"69397a7a2ff0abaf55a97a76b08691cd6a1e2311":{"kind":"REWORK","_number":3,"created":"2022-02-22 12:13:28.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/75/829475/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/829475/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/829475/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/829475/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/829475/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/829475/3"}}},"commit":{"parents":[{"commit":"1df18a885e8af6a55e98d9f56d1c042dcda12db3","subject":"Merge \"Updating python testing as per Yoga testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1df18a885e8af6a55e98d9f56d1c042dcda12db3"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-16 05:36:39.000000000","tz":330},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-02-22 12:13:18.000000000","tz":60},"subject":"[DNM] Check switch to importlib","message":"[DNM] Check switch to importlib\n\nChange-Id: I50d5d18056b868d03aea782bd603133872fbc8d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69397a7a2ff0abaf55a97a76b08691cd6a1e2311"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69397a7a2ff0abaf55a97a76b08691cd6a1e2311"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
