)]}'
{"id":"openstack%2Fneutron~830347","triplet_id":"openstack%2Fneutron~master~Id52a31b1165a966d1c2a3b3df6bb70fc2c2a42ce","project":"openstack/neutron","branch":"master","topic":"dnm","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-02-22 14:36:45.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id52a31b1165a966d1c2a3b3df6bb70fc2c2a42ce","subject":"[DNM] Test TripleO periodic c9 jobs","status":"ABANDONED","created":"2022-02-22 10:32:29.000000000","updated":"2022-02-22 14:36:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a7779a2111a0920c40f657a93b9b27456242e4ce","_number":830347,"virtual_id_number":830347,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-02-22 13:21:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 10:33:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 10:47:02.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"}],"messages":[{"id":"f84d4227892eef8008e879f5ae54ffa518dc7e81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-22 10:32:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45cf9d9d7bea02782b77f8c6358ba917cb0661c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 10:33:01.000000000","message":"Patch Set 1: Verified-1\n\nUnable to freeze job graph: Job neutron-centos-8-tripleo-standalone depends on tripleo-ci-centos-8-content-provider which was not run.","accounts_in_message":[],"_revision_number":1},{"id":"b445c60d2bd9bf2eb6a022080a944e6e66fb8261","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-22 10:47:02.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/830347/1 : FAILURE in 13m 52s","accounts_in_message":[],"_revision_number":1},{"id":"80c3b0c9a1d53c6dfd75dea9832fbdd9af815541","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-22 10:48:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"979c9bb8cca218ff60867539e076dbaabb223121","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-22 11:02:45.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/830347/2 : FAILURE in 13m 52s","accounts_in_message":[],"_revision_number":2},{"id":"e73e72c75fecb21fdbc2cde806b25f91038f1991","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 13:21:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/883cbea54a4f413bb9e81f5dc2773b49 : SUCCESS in 2h 30m 35s\n- neutron-centos-9-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8de4cf91cdee46e6a634244aa821cf55 : SUCCESS in 1h 07m 55s\n- neutron-ovn-tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/ad1284e73bc44ac389d4e075c7545c76 : SUCCESS in 1h 28m 46s","accounts_in_message":[],"_revision_number":2},{"id":"a7779a2111a0920c40f657a93b9b27456242e4ce","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-22 14:36:45.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"72cd48421437e3a3d38b1a5a7ee77ae9e83d51c0","revisions":{"c02f341923db6b7457051854fa0bb139c7f8c23d":{"kind":"REWORK","_number":1,"created":"2022-02-22 10:32:29.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/47/830347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/830347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/830347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/830347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/830347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/830347/1"}}},"commit":{"parents":[{"commit":"699869100e361ac23a8543b8d7e5e6581be144fb","subject":"Merge \"[fullstack] use noop fw driver for Local IP conntrack test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/699869100e361ac23a8543b8d7e5e6581be144fb"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-22 10:31:54.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-22 10:31:54.000000000","tz":330},"subject":"[DNM] Confirm periodic jobs issue","message":"[DNM] Confirm periodic jobs issue\n\nChange-Id: Id52a31b1165a966d1c2a3b3df6bb70fc2c2a42ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c02f341923db6b7457051854fa0bb139c7f8c23d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c02f341923db6b7457051854fa0bb139c7f8c23d"}]},"branch":"refs/heads/master"},"72cd48421437e3a3d38b1a5a7ee77ae9e83d51c0":{"kind":"REWORK","_number":2,"created":"2022-02-22 10:48:03.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/47/830347/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/830347/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/830347/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/830347/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/830347/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/830347/2"}}},"commit":{"parents":[{"commit":"e45501aee40207454c82948405e23e30856adfbf","subject":"Switch TripleO jobs to CentOS 9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e45501aee40207454c82948405e23e30856adfbf"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-22 10:45:54.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-22 10:46:03.000000000","tz":330},"subject":"[DNM] Test TripleO periodic c9 jobs","message":"[DNM] Test TripleO periodic c9 jobs\n\nChange-Id: Id52a31b1165a966d1c2a3b3df6bb70fc2c2a42ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72cd48421437e3a3d38b1a5a7ee77ae9e83d51c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72cd48421437e3a3d38b1a5a7ee77ae9e83d51c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
