)]}'
{"id":"openstack%2Fneutron~830623","triplet_id":"openstack%2Fneutron~master~If72cc4573dc52e40622dc9a458300f3f80b9e14b","project":"openstack/neutron","branch":"master","topic":"fips-compatibility","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-03-01 12:26:03.000000000","reason":"Rodolfo Alonso replied on the change"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-03-01 13:12:44.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-03-01 13:12:44.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-03-01 10:31:10.000000000","reason":"Lajos Katona replied on the change"}},"hashtags":[],"change_id":"If72cc4573dc52e40622dc9a458300f3f80b9e14b","subject":"Move FIPS jobs from the experimental to the periodic queue","status":"MERGED","created":"2022-02-23 14:37:29.000000000","updated":"2022-03-01 13:12:44.000000000","submitted":"2022-03-01 13:12:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830623-fips-compatibility","meta_rev_id":"5d785ddaf259d8d41a8fe276a4c49368abc92062","_number":830623,"virtual_id_number":830623,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-01 13:12:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-03-01 12:26:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-02-25 07:50:24.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2022-03-01 10:31:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2022-02-25 08:59:40.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-03-01 12:26:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-23 14:52:00.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-02-23 15:51:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-23 16:44:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-24 06:54:42.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-02-24 07:38:52.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2022-02-25 08:59:40.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-03-01 10:28:13.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-03-01 10:28:13.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"29503230aef88dba95378d6eaf040e33564e6251","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-02-23 14:37:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14cee145957452ddef3178aaebd417298b1f19ab","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-23 14:52:00.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/830623/1 : FAILURE in 13m 49s","accounts_in_message":[],"_revision_number":1},{"id":"7d74e52c877445d75a2f12c1d9fda14fc05514eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 15:51:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0f6f739f4274d0ba0a9db5a018b730c : SUCCESS in 1h 10m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d7f4b3fe51a548d49182a804cad83d2a : SUCCESS in 42m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"69669a1d908963aaa366b2eb9247362b932a77ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 16:44:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/04dd246a91e7485ba46a078cb11f1a32 : SUCCESS in 59m 24s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0d45a4a5fcb5453889b7674fe72be561 : SUCCESS in 1h 21m 49s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/badc514960494114a2ba634bb9c29887 : SUCCESS in 1h 10m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7b349183638e447c8efb44df44148886 : SUCCESS in 1h 11m 51s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6675484fbac14cc085b691c9f0be5ec8 : SUCCESS in 58m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7c5eb258b8e246f899023536421b9fc2 : SUCCESS in 38m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2eac4d8dea247db93d7f1411206cacb : SUCCESS in 40m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e00875158f88436692bed04d2a3b6bff : SUCCESS in 34m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61f8d34fdd4348e4af1c82de53b625c8 : SUCCESS in 14m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ece9316223614105b7d5174178d1a534 : SUCCESS in 25m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47eb9ddf54bb4719acdec2456d81785f : SUCCESS in 26m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aed6ab0c6ca4b8ab26aa63bef0187b2 : SUCCESS in 9m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/41cadb3117364e5cb1d5eb3bf79fa329 : FAILURE in 1h 08m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/366febda92bc4ad78914f7e4804cf4a6 : FAILURE in 2h 03m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/46643afb43cc444c859eea730769e000 : SUCCESS in 1h 20m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c3657e8f3d6641cd9a5d32258ba5358b : SUCCESS in 1h 03m 32s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d6c5fc0a836f42f1ac9e638477b2863a : SUCCESS in 1h 12m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dff6ae6c9151421f9c24af2a74fad9df : SUCCESS in 1h 25m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1cce6958c1184e278b75bbb1cd2e4dee : SUCCESS in 1h 22m 02s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40e798bb873d4134b301d49acba3d778 : SUCCESS in 1h 22m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1820ba0b1dec47469c8488cbb0e321d1 : SUCCESS in 1h 22m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6316f05f2b711411055cbc621d0ff959eaf6dc4d","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-24 06:54:42.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"899f59a78a6f2e454c9a88123592a5ae675d05e0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-24 07:09:08.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/830623/1 : FAILURE in 13m 53s","accounts_in_message":[],"_revision_number":1},{"id":"9722062ad8a8c80e79f5e9ccf7b34cfb50b21087","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2022-02-24 07:38:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b8ec109459fecaa4b9dd5ba1244b82d826fc7d92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 07:44:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b23671defc244cebfcc30796a66163f : SUCCESS in 46m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a60dfcebff0c4e3c948799f9acc29965 : SUCCESS in 46m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"443c423e1ca463d17b7f4808e779cbcf6cdffbc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 08:26:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/40e9389780ac46b8b2538f2a706aa2b0 : SUCCESS in 57m 35s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f8beb74514f840d091e77529eccdbbbf : SUCCESS in 1h 14m 39s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5957ed72992a40ca9209467e36125cef : SUCCESS in 1h 20m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/12e43d8657434a5fa57eb1c6ffd9dd14 : SUCCESS in 1h 13m 00s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ef8bf7082f3a4371a6d3ba488b5c06e8 : SUCCESS in 1h 05m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6ef2d80886ff4893b2e0f877201ef0da : SUCCESS in 33m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7af8aaef5a9b4483880d83b522aabc4a : SUCCESS in 36m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c89ca9a663304a8eb37a73d08db77899 : SUCCESS in 33m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1963e4db3f9a4e2c8d842e079bdbb491 : SUCCESS in 14m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3569c504e29485796b71e0a4a7e0824 : SUCCESS in 27m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb3ffe5347e54a2990fb62f32c5dd2ba : SUCCESS in 27m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0e93e2c5fa347cd93441cb1e82937ae : SUCCESS in 8m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/305864d642544fc0843434c3b612bcd3 : FAILURE in 44m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2f5f0cc70e97485a9d10820cf607cfc8 : SUCCESS in 1h 28m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bff4ef409f7e438cbca029c071805fee : SUCCESS in 1h 15m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7bc75e880fa8469b872c2cb5d0574649 : SUCCESS in 55m 05s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/69180eab766041f4b924fdba0d9c550f : SUCCESS in 54m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/603412c48b454c9aa90d580416d8695d : SUCCESS in 1h 23m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/15e3627dfd094fa68c663aa95f7ba1bd : SUCCESS in 1h 10m 17s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b59fb71f84f46c887403c28575384c0 : SUCCESS in 54m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0e9d0aa3efb94dcb93aea9a8168ce7b8 : SUCCESS in 1h 13m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a2e9e46bd35841c855710d2dc3edc8db8956e5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-25 07:49:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2a0ec3a15522ce8114cd7b5eef46e0eb77fd7894","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-25 07:50:24.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"94c3a8914e82689975a7d978e13140b9f6a79b65","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-25 08:04:19.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/830623/2 : FAILURE in 13m 46s","accounts_in_message":[],"_revision_number":2},{"id":"e996f90e808ed4dd6be6d6871f5a10474dbb2cfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 08:44:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d82ac2be5cd4a8f95fd2339006714e8 : SUCCESS in 53m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6582bc16e3aa4f6788cc9e6b9205c150 : SUCCESS in 34m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ea36a11eb934fad93b0416c61d62153 : SUCCESS in 14m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c50e00c280f8403aa6cf3bc4b25092c7 : SUCCESS in 26m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b30a10547d84cfc9d337843803449df : SUCCESS in 30m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33a5533a14cc45819ecb8b33199a0efa : SUCCESS in 9m 09s","accounts_in_message":[],"_revision_number":2},{"id":"add7f8e8f499a075861823ee2670c2d3703eef53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 08:51:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b029942a6ff4c968a54613acf1e19c9 : SUCCESS in 49m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3e2b25ffdd6c48c99a730ae383f6de1e : SUCCESS in 48m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b07ec4aa09de9aa8cf1c2968655c8cda82d83318","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-02-25 08:59:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3a25e1f660b759e719dace4e74b063e1f59a3aa3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-03-01 10:31:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c26d02fff4ee154f6efd502dd8357412051f53b1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-01 12:26:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4e1a28016a5f720a23c53ad043de4765c6f7dd88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 12:26:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e2121512db32ef95d0c7da62d1bbdf5993b80c03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 13:12:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0fa7c3552194b03b2c27e0ecea90e72 : SUCCESS in 37m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6997e85eebd940078f52dad65e8fc754 : SUCCESS in 16m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40a69c54476a4778a63aa5e0d3162556 : SUCCESS in 28m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a61a27d19b9b40deb4aeb7fe25a67530 : SUCCESS in 27m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/567de08f5f89460e83146927c8426ca3 : SUCCESS in 10m 33s","accounts_in_message":[],"_revision_number":2},{"id":"5d785ddaf259d8d41a8fe276a4c49368abc92062","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 13:12:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"210148e708ea6003a15d89ae860ba223dbf853d5","revisions":{"0c2c47630015fe5ffa4cfec08ed3599cc72372c0":{"kind":"REWORK","_number":1,"created":"2022-02-23 14:37:29.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/23/830623/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/830623/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/830623/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/830623/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/830623/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/830623/1"}}},"commit":{"parents":[{"commit":"6613d40fcb7892d13c1dc6374a0ea2db87b1053f","subject":"Merge \"Switch TripleO jobs to CentOS 9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6613d40fcb7892d13c1dc6374a0ea2db87b1053f"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-02-23 14:35:57.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-02-23 14:35:57.000000000","tz":60},"subject":"Move FIPS jobs from the experimental to the periodic queue","message":"Move FIPS jobs from the experimental to the periodic queue\n\nIt was agreed on the Neutron CI meeting [1] to move those FIPS jobs\nfrom the experimental to the periodic queue to have it run more often\nand get more data about stability of those jobs.\n\n[1] https://meetings.opendev.org/meetings/neutron_ci/2022/neutron_ci.2022-02-22-15.00.log.html#l-123\n\nChange-Id: If72cc4573dc52e40622dc9a458300f3f80b9e14b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c2c47630015fe5ffa4cfec08ed3599cc72372c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c2c47630015fe5ffa4cfec08ed3599cc72372c0"}]},"branch":"refs/heads/master"},"210148e708ea6003a15d89ae860ba223dbf853d5":{"kind":"REWORK","_number":2,"created":"2022-02-25 07:49:42.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/23/830623/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/23/830623/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/830623/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/830623/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/23/830623/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/23/830623/2"}}},"commit":{"parents":[{"commit":"22f1c99453e933ec0bcb2be2a8e258d91768ae4c","subject":"Restructure layout for periodic, experimental and tox jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/22f1c99453e933ec0bcb2be2a8e258d91768ae4c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-02-23 14:35:57.000000000","tz":60},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-25 07:48:43.000000000","tz":330},"subject":"Move FIPS jobs from the experimental to the periodic queue","message":"Move FIPS jobs from the experimental to the periodic queue\n\nIt was agreed on the Neutron CI meeting [1] to move those FIPS jobs\nfrom the experimental to the periodic queue to have it run more often\nand get more data about stability of those jobs.\n\n[1] https://meetings.opendev.org/meetings/neutron_ci/2022/neutron_ci.2022-02-22-15.00.log.html#l-123\n\nChange-Id: If72cc4573dc52e40622dc9a458300f3f80b9e14b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/210148e708ea6003a15d89ae860ba223dbf853d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/210148e708ea6003a15d89ae860ba223dbf853d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
