)]}'
{"id":"openstack%2Fneutron~830864","triplet_id":"openstack%2Fneutron~stable%2Fxena~Ic54214ca4965274cd58c1d3a278d8f3ffcda0a64","project":"openstack/neutron","branch":"stable/xena","topic":"mtu_router_removed-stable/xena","attention_set":{},"removed_from_attention_set":{"32586":{"account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"last_update":"2022-02-28 19:32:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic54214ca4965274cd58c1d3a278d8f3ffcda0a64","subject":"[OVN] Handle RouterNotFound exception in set_gateway_mtu","status":"MERGED","created":"2022-02-25 08:09:45.000000000","updated":"2022-02-28 19:34:00.000000000","submitted":"2022-02-28 19:32:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830864-mtu_router_removed-stable/xena","cherry_pick_of_change":830509,"cherry_pick_of_patch_set":3,"meta_rev_id":"75656a470c72b4278252cb5f2b028bc4ce91ca0c","_number":830864,"virtual_id_number":830864,"owner":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-28 19:32:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-02-28 18:02:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-02-28 13:44:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-02-28 18:02:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-25 08:24:12.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-02-25 09:32:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-28 13:44:10.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-02-28 18:02:13.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"9693608035cb424fee4293c597bfdde9c0d33d28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-02-25 08:09:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f5e383a940830936460c7c73511c00a8d3fa9e83","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-25 08:24:12.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/830864/1 : FAILURE in 13m 41s","accounts_in_message":[],"_revision_number":1},{"id":"0a0dbc741f664401b43b13509124cf4eb6ebfd03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 09:32:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/01095c79363942749cbc1f8263bbb570 : SUCCESS in 59m 46s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/7613f0d5d4084649ac4f57dbd8230d04 : SUCCESS in 49m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59065e83e61243b790d254227d4ddf05 : FAILURE in 48m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0617b3c9a44f45b2a08624681732de4f : SUCCESS in 11m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/363316fa2d0041399d5433529d1671c2 : FAILURE in 21m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cab2bc1085354e618c3044e83907580c : FAILURE in 38m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af578edcd506460996d636840a8dcbba : FAILURE in 21m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0390379ed63a4f988ecdd201eeb6942a : SUCCESS in 9m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e18d4295e0aa4d87af22dec68941a7e6 : SUCCESS in 59m 08s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a88260c7a9d34467b654a759c8b6e2cf : SUCCESS in 1h 15m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4f0b0146393f45bfae4a56d58451d122 : SUCCESS in 1h 00m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b064b86a9d7ab82d5b21a1bf899e687cb73543b6","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-02-25 09:43:58.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"80e75d4cdaca955bde743c73eb9281f6952c3670","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 09:45:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc10cb4429c74443a46388e83dd22e12 : FAILURE in 50m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/125758d942384987805aa9bb921c2464 : FAILURE in 49m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3cc92c29379d0d26ae59a6c8be520c0e7746caf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-02-25 09:56:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"eaab0cd4161266f7e24bcecc9e067eb6c0ff8e1f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-25 10:11:20.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/830864/2 : FAILURE in 13m 46s","accounts_in_message":[],"_revision_number":2},{"id":"b499318814d1915c3cddbf6c908fd6feecb902fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 11:10:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/89db4f67fc2a409f8fef6eb450e93036 : SUCCESS in 39m 31s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/f997a3be251e4d559534ee69561bb90c : SUCCESS in 49m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35085f9fe53e4631900a57d5db8f3171 : SUCCESS in 27m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/716eaf4ba3634642921ada04cb7119d8 : SUCCESS in 12m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/222ac49cc715411c986b16e2029ce966 : SUCCESS in 23m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b3e3969249e43a7a1684b21c1622bf3 : SUCCESS in 21m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e7a67c7a47b4a11b2706bc6ddb6cd84 : SUCCESS in 22m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3da9dd971d647138dd459ff048e5052 : SUCCESS in 9m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/52507c6844fd42a783be4d3d63af9e52 : SUCCESS in 43m 33s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c849caf760e04a3fbccbaba6b3bf47d3 : SUCCESS in 1h 03m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d1b22b4de28644eb8b6c166da49318bf : SUCCESS in 1h 01m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"93925bc12adccd66e8ff55606c6bd88432e5ac5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 11:33:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d6f2e539f29d4e668c0d89246ce6743b : SUCCESS in 54m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/52a47d6c1d054681977158f84209bf83 : SUCCESS in 53m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c1bef59110a1dd6c87d77b79e41ed958028c7ac7","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-02-28 13:44:10.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1decd83ff1efea53e542b1d09310185159a299dc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-02-28 18:02:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2884b5f5228df6cce351305081a6ccb6e5b69745","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 18:03:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f08b46d96ae3c7ef33c1ebc0e841dd74fe725875","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 19:32:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/72df83e89a764e21947b240f5af066b3 : SUCCESS in 48m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad07f741f4024bbe8b9de67ffadfdb46 : SUCCESS in 15m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29d329a8a7bc499c8936b2fb9abf74e7 : SUCCESS in 30m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6272732963b7479a92ed344edaf5acde : SUCCESS in 25m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdc8ca34d77b414ba153bc3327829cd6 : SUCCESS in 10m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9e2254c1cf2747f6aa476ea2051452f1 : SUCCESS in 51m 21s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eefa3a90eb3b4eecb147ccd215d0aa71 : SUCCESS in 55m 55s","accounts_in_message":[],"_revision_number":2},{"id":"0b73dbc46ea72543b6f44cd028792bb1369929d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 19:32:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"75656a470c72b4278252cb5f2b028bc4ce91ca0c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 19:34:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a170e6c26e248c0bbafb863dc59bee3 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c0c292efbc5a4f87041a72fbe82f3239507b1342","revisions":{"a95061cac082bcca2c42b0dd7af7ac657b3f85d6":{"kind":"REWORK","_number":1,"created":"2022-02-25 08:09:45.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/64/830864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/830864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/830864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/830864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/830864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/830864/1"}}},"commit":{"parents":[{"commit":"b647f3c1334aaef583db2c7368779d6b219efb1c","subject":"Merge \"Wait longer before deleting DPDK vhu trunk bridges\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b647f3c1334aaef583db2c7368779d6b219efb1c"}]}],"author":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2022-02-22 17:56:55.000000000","tz":60},"committer":{"name":"Elvira García Ruiz","email":"egarciar@redhat.com","date":"2022-02-25 08:09:45.000000000","tz":0},"subject":"[OVN] Handle RouterNotFound exception in set_gateway_mtu","message":"[OVN] Handle RouterNotFound exception in set_gateway_mtu\n\nWhen handling multiple router update/delete petitions, if they are\nexecuted at the same time, there is a chance that the set_gateway_mtu\ntries to update a no-longer-existing router, triggering a\nRouterNotFound error from the _get_router_ports function. Now, if there\nis not a router, the _get_router_ports function will return an empty\nlist instead of raising the error.\n\nCloses-Bug: #1961967\nSigned-off-by: Elvira García \u003cegarciar@redhat.com\u003e\nChange-Id: Ic54214ca4965274cd58c1d3a278d8f3ffcda0a64\n(cherry picked from commit b2421b01e5bc1269458b421643c13b620cdaccaf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a95061cac082bcca2c42b0dd7af7ac657b3f85d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a95061cac082bcca2c42b0dd7af7ac657b3f85d6"}]},"branch":"refs/heads/stable/xena"},"c0c292efbc5a4f87041a72fbe82f3239507b1342":{"kind":"REWORK","_number":2,"created":"2022-02-25 09:56:49.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/64/830864/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/830864/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/830864/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/830864/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/830864/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/830864/2"}}},"commit":{"parents":[{"commit":"b647f3c1334aaef583db2c7368779d6b219efb1c","subject":"Merge \"Wait longer before deleting DPDK vhu trunk bridges\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b647f3c1334aaef583db2c7368779d6b219efb1c"}]}],"author":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2022-02-22 17:56:55.000000000","tz":60},"committer":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2022-02-25 09:56:27.000000000","tz":60},"subject":"[OVN] Handle RouterNotFound exception in set_gateway_mtu","message":"[OVN] Handle RouterNotFound exception in set_gateway_mtu\n\nWhen handling multiple router update/delete petitions, if they are\nexecuted at the same time, there is a chance that the set_gateway_mtu\ntries to update a no-longer-existing router, triggering a\nRouterNotFound error from the _get_router_ports function. Now, if there\nis not a router, the _get_router_ports function will return an empty\nlist instead of raising the error.\n\nCloses-Bug: #1961967\nSigned-off-by: Elvira García \u003cegarciar@redhat.com\u003e\nChange-Id: Ic54214ca4965274cd58c1d3a278d8f3ffcda0a64\n(cherry picked from commit b2421b01e5bc1269458b421643c13b620cdaccaf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c0c292efbc5a4f87041a72fbe82f3239507b1342"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c0c292efbc5a4f87041a72fbe82f3239507b1342"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
