)]}'
{"id":"openstack%2Fneutron~830867","triplet_id":"openstack%2Fneutron~stable%2Fussuri~Ic54214ca4965274cd58c1d3a278d8f3ffcda0a64","project":"openstack/neutron","branch":"stable/ussuri","topic":"mtu_router_removed-stable/ussuri","attention_set":{},"removed_from_attention_set":{"34118":{"account":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"last_update":"2022-03-01 16:18:55.000000000","reason":"Change was submitted"},"32586":{"account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"last_update":"2022-03-01 16:18:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic54214ca4965274cd58c1d3a278d8f3ffcda0a64","subject":"[OVN] Handle RouterNotFound exception in set_gateway_mtu","status":"MERGED","created":"2022-02-25 08:10:52.000000000","updated":"2022-03-01 16:25:51.000000000","submitted":"2022-03-01 16:18:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830867-mtu_router_removed-stable/ussuri","cherry_pick_of_change":830509,"cherry_pick_of_patch_set":3,"meta_rev_id":"c148f0e2a53d79bb83ba875dfe94989a6363cc18","_number":830867,"virtual_id_number":830867,"owner":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-01 16:18:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"all":[{"value":1,"date":"2022-02-28 14:58:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":2,"date":"2022-03-01 09:28:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-03-01 13:51:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-03-01 13:51:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-25 12:20:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-28 10:26:17.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2022-02-28 14:58:03.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"REVIEWER"},{"updated":"2022-03-01 09:28:18.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-03-01 13:51:51.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"01f26325011bceda721dcf211bc0874253e70af9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-02-25 08:10:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"de8ae77ba4d6cd58d4f1c320abd03f12dede1904","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 12:20:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/855c707dd89040418d83f2afbc2f744b : SUCCESS in 48m 11s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/3ea2b26984864236bf2e96df5001acdc : SUCCESS in 1h 40m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9807c9d9a1d2424fbd9344546eab784d : FAILURE in 33m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61038e1663df4b32820dad8c29213ca1 : SUCCESS in 13m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/640c7c3839354a8ba59603ddba5de911 : FAILURE in 29m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98f3d85ac0b046ba80f057b9ae1fa1cb : FAILURE in 21m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3fb8912569a498085155f39cf4ca995 : FAILURE in 23m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51e6e247844e49d0a28afb9d60441e6a : SUCCESS in 8m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4ff6f3c4b8b74d45b639305767a1ed82 : SUCCESS in 57m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/660e504ffdeb4c72b6db77966f5085e0 : SUCCESS in 1h 26m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b57ae0f3374447d2b30f0240818b7722 : SUCCESS in 1h 18m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f53046e2f41f4bbeb698a44d985a83fe : SUCCESS in 1h 29m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d005003a0764f67ba546a91c07a0bae : SUCCESS in 1h 17m 55s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ffa03ff65d73402d959f9144aef0f822 : SUCCESS in 1h 13m 26s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/060c4754874e4983977915878571774e : SUCCESS in 1h 15m 00s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/01c7b1b22d7f4117b0d7a586ecd30bd6 : SUCCESS in 1h 13m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2cf30f400d724470bc9e74656b622ee8 : SUCCESS in 1h 04m 48s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fb85982cf74e4a959ccd3ba414acbf49 : SUCCESS in 1h 39m 42s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d19ed58b3ad2449da14b3ccecdbd4992 : SUCCESS in 1h 00m 58s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/051f1b9dc57542459ef52d83a6b51f47 : SUCCESS in 1h 19m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/975eb56df3d2454f8299cb71b778b89d : FAILURE in 1h 10m 29s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/da6860f6683e4a948ef8cbc671197fd0 : SUCCESS in 1h 39m 37s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b281d08a7f164fa69d64c9c588eddb57 : SUCCESS in 1h 02m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/05dec55a72ae425192d655b670e9f607 : SUCCESS in 1h 16m 45s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/64577f8d035143aea58b4458543a6a57 : SUCCESS in 41m 49s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8b51f4e0760a4081be9c74f2e403878c : SUCCESS in 1h 22m 39s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9120493898734ee29fc12c45cb4a082b : SUCCESS in 52m 47s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/48e1bd4c6c904e5a8be2b6cf7fc57c92 : SUCCESS in 4h 04m 00s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/758fdb6e56d4468dbea952d74e6f592e : SUCCESS in 1h 52m 31s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/14194d8125c54c25b864df1ef07f5f88 : SUCCESS in 1h 48m 35s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/497f34a7e0ad454fade01f62887d009d : SUCCESS in 2h 24m 51s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/99d55a3f9e8f435ca15918a8b2a2d82b : FAILURE in 1h 26m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"09d456965cd60a0ab1b13725056ebc00423ec5fa","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-02-28 10:26:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac2b33dfba0315c9f32494960c948427a4d191cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"date":"2022-02-28 13:53:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"66fb2c06b0446e3e2a247586164817d5650f33fd","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2022-02-28 14:58:03.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2786e0082650cdbbfc23f19a6f8c9eb0d9492a67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 17:22:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/f6670260eca64fa4b760c033e89efc76 : SUCCESS in 1h 04m 15s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/9cab5dc9f71f4aca8f3457e376113374 : SUCCESS in 2h 14m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ecbcb05a9a448e098ca9f7da20812e1 : SUCCESS in 42m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/225219671741475ebcc522ce5821aaac : SUCCESS in 16m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcedcfa9f65245ca95d2fd2c2bb78a09 : SUCCESS in 33m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da325e62520f4e368f05e79cd6cd486c : SUCCESS in 31m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/747494c03d77428183e439d4fb6a2de8 : SUCCESS in 24m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f42f688db0c45a79855af2e24de8e4a : SUCCESS in 10m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a75927c3c69e475d98d6d91fe347849c : SUCCESS in 53m 36s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c7cc55c7c8f24100b20893c6ac31f429 : SUCCESS in 1h 24m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fd1ab85672da41f6a110ec6d53be362d : SUCCESS in 1h 13m 37s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6c1a91ae64dd4e2fbc40db861c86481e : SUCCESS in 1h 51m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b4109728e1447b4832419c8614c480f : SUCCESS in 1h 05m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/669ef80925ad482987eb1f10cdfd52a8 : SUCCESS in 1h 18m 16s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/783ebdca50ce4b898bef00fb06c8e529 : SUCCESS in 1h 28m 22s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/4cfafe3f20dc43e3af206b19543e663b : SUCCESS in 1h 29m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3ece3d71d2a04673967163fd54057748 : SUCCESS in 1h 14m 37s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/52ce109f3a2e458cb723b5c7ec534620 : SUCCESS in 1h 41m 07s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ca2db4b0ad044ef9a5fcafc5947b22ee : SUCCESS in 1h 22m 31s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8cecf88c9faa47609cf2b2280951eb82 : SUCCESS in 1h 29m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/95590de002954e729b8c4672117c7e4a : FAILURE in 58m 38s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8789bf64041241838128ef52cc3560a6 : SUCCESS in 2h 25m 26s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d387f56d50ae4e93bb7596242276d2c1 : SUCCESS in 1h 20m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f303721b9ad84044aae64bbd694d87ec : SUCCESS in 1h 00m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/60dcad74331a468d805a63c8ea122221 : SUCCESS in 54m 30s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f37170f72fa84b30811dc7cd89d61d38 : SUCCESS in 1h 26m 18s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d84f0f36f70f423da61d993b17d5e7d7 : SUCCESS in 56m 31s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b76f6eb4ab6c4fdbaa4e9e0bb5a97525 : SUCCESS in 3h 26m 42s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c2ca14c219534d679e08095181e901f3 : SUCCESS in 1h 05m 18s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b0243113a6134c9490a9fd93fea7e6b7 : SUCCESS in 1h 51m 20s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/901a0bfb20264ea7ba4e07c3df40ba55 : SUCCESS in 1h 37m 22s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b37d4d1df8cd490eab686efd66ab968e : FAILURE in 1h 32m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"77a85c2e31abc4015fa5ddfa3b0b7b81bb7aca0c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-03-01 09:28:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2994d5d107ddfd6b5a16b4184735cb1ab136a72e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-01 13:51:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2cfd3400eed6169482666538129c3ddab429b74e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 13:52:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ed9efcfdfa1fbc3b6dfa598f9de9693b690a4e7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 16:18:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/a22f7f2ec33d4fd7b9bb48b6b4394a72 : SUCCESS in 52m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72acf0168a744595bd63b31ade7bd946 : SUCCESS in 17m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bc13c5eac0e450787cc42225e79b4dd : SUCCESS in 23m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97685da4e6954825ae57f18260088cba : SUCCESS in 25m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7f991481f0f414f8e14ff4baecdb06f : SUCCESS in 10m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bcb7a28a14bd454d866066ab1c21a2a2 : SUCCESS in 56m 55s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a6955b08be4d47bf82f08dc59a1964e5 : SUCCESS in 1h 14m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6d76eeebb9ca4e3e950d2482a9326e33 : SUCCESS in 1h 35m 00s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c767bc87817346d2ad2a47372627b837 : SUCCESS in 1h 12m 49s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/139771dc402a4793a76b9635fdbf7597 : SUCCESS in 55m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/78229ab9a3b445e6b504689aac84b8dc : SUCCESS in 1h 05m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a871f28b1bc64b2bbab7d8d35227701a : SUCCESS in 1h 14m 53s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/255d28783cc545fbae4acf055f2da44a : SUCCESS in 1h 21m 53s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f7666fc6d614365bbad73382472c474 : SUCCESS in 57m 56s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/bc2eee271c7b4431bddc35de14d23244 : SUCCESS in 1h 04m 40s","accounts_in_message":[],"_revision_number":2},{"id":"6b3a03b46510c81794ae549de7f8a3bfba8a3a24","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 16:18:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c148f0e2a53d79bb83ba875dfe94989a6363cc18","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 16:25:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b7fa3117b594964a8891d5a9ec54125 : SUCCESS in 2m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4e8ee06f570f5b9158e9ffefe0a67fb1e668761c","revisions":{"befd40a9d9e353134f2cbba8caf36f18d1afdf79":{"kind":"REWORK","_number":1,"created":"2022-02-25 08:10:52.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/67/830867/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/830867/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/830867/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/830867/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/830867/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/830867/1"}}},"commit":{"parents":[{"commit":"309624e189326f546efce01ba42aac66e03293d6","subject":"Merge \"Wait longer before deleting DPDK vhu trunk bridges\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/309624e189326f546efce01ba42aac66e03293d6"}]}],"author":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2022-02-22 17:56:55.000000000","tz":60},"committer":{"name":"Elvira García Ruiz","email":"egarciar@redhat.com","date":"2022-02-25 08:10:52.000000000","tz":0},"subject":"[OVN] Handle RouterNotFound exception in set_gateway_mtu","message":"[OVN] Handle RouterNotFound exception in set_gateway_mtu\n\nWhen handling multiple router update/delete petitions, if they are\nexecuted at the same time, there is a chance that the set_gateway_mtu\ntries to update a no-longer-existing router, triggering a\nRouterNotFound error from the _get_router_ports function. Now, if there\nis not a router, the _get_router_ports function will return an empty\nlist instead of raising the error.\n\nCloses-Bug: #1961967\nSigned-off-by: Elvira García \u003cegarciar@redhat.com\u003e\nChange-Id: Ic54214ca4965274cd58c1d3a278d8f3ffcda0a64\n(cherry picked from commit b2421b01e5bc1269458b421643c13b620cdaccaf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/befd40a9d9e353134f2cbba8caf36f18d1afdf79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/befd40a9d9e353134f2cbba8caf36f18d1afdf79"}]},"branch":"refs/heads/stable/ussuri"},"4e8ee06f570f5b9158e9ffefe0a67fb1e668761c":{"kind":"REWORK","_number":2,"created":"2022-02-28 13:53:27.000000000","uploader":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"ref":"refs/changes/67/830867/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/830867/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/830867/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/830867/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/830867/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/830867/2"}}},"commit":{"parents":[{"commit":"309624e189326f546efce01ba42aac66e03293d6","subject":"Merge \"Wait longer before deleting DPDK vhu trunk bridges\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/309624e189326f546efce01ba42aac66e03293d6"}]}],"author":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2022-02-22 17:56:55.000000000","tz":60},"committer":{"name":"Arnau Verdaguer","email":"averdagu@redhat.com","date":"2022-02-28 13:51:52.000000000","tz":60},"subject":"[OVN] Handle RouterNotFound exception in set_gateway_mtu","message":"[OVN] Handle RouterNotFound exception in set_gateway_mtu\n\nWhen handling multiple router update/delete petitions, if they are\nexecuted at the same time, there is a chance that the set_gateway_mtu\ntries to update a no-longer-existing router, triggering a\nRouterNotFound error from the _get_router_ports function. Now, if there\nis not a router, the _get_router_ports function will return an empty\nlist instead of raising the error.\n\nOVNL3RouterPlugin has no attribute \u0027_nb_ovn\u0027\nFakeOvsdbNbOvnIdl has no attribute \u0027lrp_set_options\u0027\nConflicts:\n    neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n\nCloses-Bug: #1961967\nSigned-off-by: Elvira García \u003cegarciar@redhat.com\u003e\nChange-Id: Ic54214ca4965274cd58c1d3a278d8f3ffcda0a64\n(cherry picked from commit b2421b01e5bc1269458b421643c13b620cdaccaf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4e8ee06f570f5b9158e9ffefe0a67fb1e668761c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4e8ee06f570f5b9158e9ffefe0a67fb1e668761c"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
