)]}'
{"id":"openstack%2Fneutron~831214","triplet_id":"openstack%2Fneutron~master~I512f7d98ac99bb0ef06fd2acba09482e3436d18d","project":"openstack/neutron","branch":"master","topic":"bug/1962306","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-03-11 20:24:42.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-03-11 20:24:42.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-03-11 15:11:05.000000000","reason":"Slawek Kaplonski replied on the change"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-03-11 10:10:39.000000000","reason":"Lajos Katona replied on the change"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2022-03-11 20:24:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I512f7d98ac99bb0ef06fd2acba09482e3436d18d","subject":"Ensure gateway is set for prefix delegated subnets","status":"MERGED","created":"2022-02-28 14:18:02.000000000","updated":"2022-03-11 20:26:41.000000000","submitted":"2022-03-11 20:24:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"831214-bug/1962306","meta_rev_id":"5a7b83bb6189c10faea9743ca66785697c5cc7fd","_number":831214,"virtual_id_number":831214,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-03-03 06:42:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-11 20:24:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2022-03-11 15:11:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-03-11 10:10:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2022-03-11 15:11:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-28 15:08:27.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-28 15:54:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-28 16:20:41.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-02-28 22:05:32.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2022-03-01 13:40:01.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-03-01 13:40:01.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-03-01 13:40:01.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-03-01 13:40:01.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"b531e4ac596d510b34412f7aee07d065e1349a80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-28 14:18:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41570cc388bb98997cd67c58a6c6d3ee586c1204","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-28 14:30:07.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d5d327a6e0582e2d3b2d7a1055001fd5f0c45763","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-02-28 15:08:27.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2022/Arista_Tempest_Testrefs/changes/14/831214/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"97dccf757670923b8433000850f10ff59b132ba3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 15:54:44.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9b041bf3c5b7417b8625976a782481bd : SUCCESS in 50m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/39fa2fa4670e4a64966b75bbdb8b17a8 : SUCCESS in 53m 39s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5a90188aff6f4c3abcf4bb7a1ace47ee : SUCCESS in 57m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e8a7a987ba114806957976441647977a : SUCCESS in 50m 32s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0ae53a5eb4b54bc797972739edb30db5 : SUCCESS in 50m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fcc02884e4a14efca7e19a1ab58bb4e9 : SUCCESS in 30m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c26934a87414578bcb2e0108e642060 : SUCCESS in 50m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7b0e63247194915826c1ed0f82bfaca : SUCCESS in 35m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6811a486cf094d18b4479ce6307056f9 : FAILURE in 8m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bec7a4caf8a448e8414fb3700fcd672 : SUCCESS in 30m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05f5554b694645649b3ca1019c8ce5de : SUCCESS in 28m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e92256dde5394fd9bfb4c49c8fe0beae : SUCCESS in 13m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a2f59b9a3e53422090e3c8f09edb3a29 : SUCCESS in 49m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1fd9534cb3f2416a8269938f9b02aec8 : FAILURE in 1h 18m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4fca35dfdbe846a69ea5d87920b2d647 : SUCCESS in 1h 15m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/eae371b5340a44289149bfa9fd887532 : SUCCESS in 55m 53s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d1a50316c6094eafb6be9a36fc272883 : SUCCESS in 1h 00m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b9336920b41c42fcb529739599dbcaaf : SUCCESS in 1h 16m 46s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c30ca7b1a3c6468f8f427fcc15e69a4f : SUCCESS in 1h 12m 52s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c99069d7217410d85b696422d28e1d3 : SUCCESS in 58m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6d9131ac2f9c417e8eba07864b0043d8 : SUCCESS in 1h 14m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d4dbffbec25a1b4707531351a0b19d5cd0c46b51","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-02-28 15:57:52.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/February-2022/Arista_Tempest_Testrefs/changes/14/831214/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"e142248a04e66882ccf8d4abc0be2699fd5ca718","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-02-28 16:20:41.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/831214/2 : SUCCESS in 1h 48m 37s","accounts_in_message":[],"_revision_number":2},{"id":"531688fd8afb9b623325c36046b2f913401786e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 16:41:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec54b20b0f8f4f6db6695bfc3321d212 : SUCCESS in 1h 07m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/966660637a454eb0b01e6be1d577c21d : SUCCESS in 1h 06m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"34109cafa0347b6f7177cee32fc52d464d4bbf4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-03-01 04:40:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4612d6116aef72d9f6ad92058606a2f3fcd23ae3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 05:25:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/179eb144116b46a3aa5d943a47cdc982 : SUCCESS in 42m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0c47f2a46f6d46e3ad27b1bd38cfe3cc : SUCCESS in 41m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0125da303927c8af3795a41ccc914e59cae27dae","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-01 05:35:26.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/14/831214/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"5caa7823200f18a2bb236632c40ab7e2b80573a0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-01 06:13:38.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/831214/3 : FAILURE in 1h 32m 30s","accounts_in_message":[],"_revision_number":3},{"id":"6fda89c7094e7368a721359538d508ebd74a39cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 06:24:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/10750ba3ce6c460d86d14259b87bb637 : SUCCESS in 1h 25m 23s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/36036a16f3244c698c8414aeadd4a461 : SUCCESS in 1h 02m 31s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/794db79ec03443b3ba28f764e9815cd1 : SUCCESS in 1h 13m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0268752597c24b259ebf3a5ac2547617 : SUCCESS in 1h 00m 22s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/78d1706e2a4e4313b9d106de3c4660e5 : SUCCESS in 1h 16m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dc1731ecb2754b3d9c2be9508e48af2a : SUCCESS in 36m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bee99809cfe4e0f9fc72ed795ac10bb : SUCCESS in 36m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/acfbee05cbe647e2bb4209f96b658562 : SUCCESS in 31m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82530281906148a789ecf9501c39b18b : SUCCESS in 13m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9570591dd7d47848a3f8964996a51ca : SUCCESS in 27m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/279f426585e541f0aa1bab93f27664d0 : SUCCESS in 22m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec3e66e718054392ba34d9932e2c04f7 : SUCCESS in 11m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/05985d94addf4cf8a294cc55fa63089c : SUCCESS in 53m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5ffd6fac082d48a0bdd44ccbc0c1a320 : SUCCESS in 1h 42m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/989ca00c8f434c308ccadc517c964896 : SUCCESS in 57m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/30ba2252458241e192eba96ba3df61c4 : SUCCESS in 58m 28s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d235d0e8aa4b4012ba37d8fc3a63e236 : SUCCESS in 59m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cff68d3969fd4ebca7669053552f37a6 : SUCCESS in 1h 07m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b134ea2c410a436db41bbf09ea7e50fe : SUCCESS in 1h 11m 42s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4590086948c84de4be66cadc336f05df : SUCCESS in 49m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b0501d02fd77494a9a31233c971b6a40 : SUCCESS in 1h 15m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0e71edfd16979731236cf3018d3db4f990d1421f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-02 09:21:38.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e1bac3f2954a4b346d2477869f01f21901595ad8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-03-02 10:07:33.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5784aa64f0652106f9f28e87b2a2c0234bb7a011","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-03-02 11:54:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d158ed408e34aa5d4a2af46301f093a23d5d7de2","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-03-02 11:54:10.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f0ccb95efd9f62764e72fca73e7009a7789b629b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 12:44:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/055ed6c66dce49d98f10ee8d12b3af28 : FAILURE in 41m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f039bc9ff0e0480e86da96d7750d9652 : FAILURE in 47m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5f81cdf8f9ab27de83db90ea936026737c596cd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 13:30:37.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ca93b96071c4420a89bbff078bcd9961 : SUCCESS in 43m 52s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b4880d14244241df8d0f6a1b9ab91326 : SUCCESS in 1h 32m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/67a237d465a847eab4fc5b3ee2329e31 : SUCCESS in 1h 01m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b6ceb81215fb42d29fd1df70e6f3a7c3 : SUCCESS in 1h 05m 50s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/427ce04c46b444388e41ff7396aed9a0 : SUCCESS in 1h 08m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0078cee502f64cbabab51e1bc9a0d437 : SUCCESS in 29m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fff74246ee8b46d4b10b8426aa53aa5a : FAILURE in 31m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d510753790644e9abd72a302e361ea60 : TIMED_OUT in 1h 02m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a66abd334ad48cd993e07ef23b80885 : SUCCESS in 13m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d527a0cd5b774f69aa403d2777bfaf14 : FAILURE in 32m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b75814f475b47af9d3a99cb1a31ce22 : FAILURE in 24m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcb5e7ce5e5845c3a4025b116ee8fdea : SUCCESS in 10m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aa9db772baee4076b091710be6fe776d : SUCCESS in 48m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/08f5778d188b408f874ecb7c5e640f86 : SUCCESS in 1h 30m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/57ffcb75c4ea434db4e8cdeda05e7b6f : SUCCESS in 1h 08m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b7f1c562be6b48fdb1cc674464ea7771 : SUCCESS in 1h 02m 57s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5ed543e241c74dd1b3a18f0ae2bf3fd6 : SUCCESS in 58m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/332d4f3dc4ae4307b850a4717101d540 : SUCCESS in 1h 17m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a34bd09a8fd34e69baf4eb0455b69219 : SUCCESS in 1h 14m 19s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fd70229536a40f687f36757eaeb0268 : SUCCESS in 55m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/aa31d3121d8648a2ada14b5072b8ba92 : SUCCESS in 1h 13m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"163c7350fa2f5428628f51d5e62debcec0f7fb3f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-02 13:31:02.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/831214/4 : SUCCESS in 1h 36m 09s","accounts_in_message":[],"_revision_number":4},{"id":"3366fa2a1335debafe037bff3f5ea825865eddd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-03-02 13:36:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"86b4ad684c51c6fb31edc3f37c05c6d2f2f004e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 14:28:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4c369509953439aaeeabeda899f07a0 : SUCCESS in 48m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/052146266fb949acb9679fe928e60268 : SUCCESS in 45m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ed2ec5b35eeb6880ec91587d95f44d86df6310c9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-02 14:55:17.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/14/831214/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"2c6496811a2e0c753c72cc340f165bbb6f6628e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 15:18:50.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/06f05d8ad4864bf8b6e1a2f33316f9b1 : SUCCESS in 47m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7cc02f04b6034ce085941eed234ccf14 : SUCCESS in 55m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/39d2f7d1fca0431285868d9ccc4a1fca : SUCCESS in 1h 05m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/12622d120d364e19b9c1f7f69935f6f1 : SUCCESS in 52m 28s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/babf8490e4c4438cb8e18c3c16ef1651 : SUCCESS in 50m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7575a4e7826a4b2fa2917207da0fe459 : SUCCESS in 27m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0999d8ba65b488dabf331122eb07604 : SUCCESS in 42m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a36bcfc258184b68af4a0cfd8303473d : SUCCESS in 34m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01fee2767b2641748366b6b67bfe8b75 : FAILURE in 9m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/429b9d1fd1a645ea93e9e60616008fbe : SUCCESS in 29m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0f1709a89ad44da976c686439898c3a : SUCCESS in 26m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae770f0012b14856b3492b7df8189d77 : SUCCESS in 10m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1343dde96557436fa7997eed1f0c584a : FAILURE in 49m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a1d463b29e57428094d069c09e1a3fa2 : SUCCESS in 1h 40m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/26b73d1f2a754631b2af483b81f0baee : SUCCESS in 1h 02m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/88620b81161042f6b5734dd55b38009a : SUCCESS in 54m 53s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/19f94940c06a443ea8c28d7af090cdea : SUCCESS in 56m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8b5cd2d910a0470d88cd276a5c3ab169 : SUCCESS in 1h 13m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3676dcaaf23d47ed85464a7bab31a547 : SUCCESS in 1h 02m 51s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29a401ba516a40e1bc7c96671015b446 : SUCCESS in 55m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e39e5bc8e9824e6598c0c0b7f8cf333c : SUCCESS in 1h 01m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ef5837be43baf35689ca2df028456daee551cec7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-02 15:27:36.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/831214/5 : FAILURE in 1h 49m 51s","accounts_in_message":[],"_revision_number":5},{"id":"957ed6e16da9ed6572b515c32cb8da2263934bd1","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-03-02 15:34:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0313b49e7bbe28df3b0280e6004885569b00c413","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-03-02 15:39:15.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"81ac4febcf53d25e4520fa59b36324e2af5d92bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 16:32:08.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/456c1dae065a4b609cdca408e07c5ce3 : SUCCESS in 48m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0a67b19a8e8e419fae59de54dc6029f6 : SUCCESS in 50m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d93574f193642eebdb46cf62c7eab18545552378","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-02 16:55:12.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/14/831214/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"82e63dee7f0e16a464fea2aab8b446ed248d1ff0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 17:10:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9b3ecdd4896d43e580e985bb3bbb3e60 : SUCCESS in 34m 58s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6cbab31e1b70417881c78520cb0f919d : SUCCESS in 1h 04m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/51740614f82a4eef8667fdf8046679d5 : SUCCESS in 1h 12m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/44135dce9d054db49bb5f066aaf05ba7 : SUCCESS in 57m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ad5b1fda3d804d69938efda8880621ca : SUCCESS in 51m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b310dd15addd4dfeafcbb038f6c0472f : SUCCESS in 32m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8ef75e3227c4e0595d1182089a0461e : SUCCESS in 37m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/342b468c8ba4484a85262adbc1d334f0 : SUCCESS in 28m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a1dcdd59ea74c3d9639a7b1892e5653 : SUCCESS in 16m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a1dca784df342ba8ee600e1854a1672 : SUCCESS in 24m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c87be7da44b74127a9a1e084b5d6ba88 : SUCCESS in 25m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5db9acbdd87644478d29eed639676a61 : SUCCESS in 11m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cdcc0e246f1941dba5e1123570838b92 : FAILURE in 45m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/93af1aa51574485db490412d9834804f : SUCCESS in 1h 28m 29s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8e3b961be16f4558b6c05a680857fb90 : SUCCESS in 1h 12m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3285b5eecd804961a65b1b162c9cf236 : SUCCESS in 53m 43s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a5faef6d33f24e4d9c5ccb9b1f46d484 : SUCCESS in 1h 00m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dad3a2d6e3064ff591b13e29a0cc1bbc : SUCCESS in 1h 15m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/11499ec9b06e4608bece56ccab278a8f : SUCCESS in 1h 13m 59s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09c8d63d4fa6472294f3fe4f8bdef510 : SUCCESS in 1h 09m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/37edf7e8738046edbce00f9eb2ea2142 : SUCCESS in 1h 15m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d7b14c41abf7d0fd0cba99824050282c81302980","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-02 17:32:12.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/831214/6 : SUCCESS in 1h 52m 12s","accounts_in_message":[],"_revision_number":6},{"id":"8ee8972563fd6894000e5213f314a46d2776ab37","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-02 18:35:33.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/14/831214/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"36cf3e67b06dcdccd2b88dfd08d8c56225d3380b","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-03-03 04:55:32.000000000","message":"Patch Set 6: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7c545720fc4ee6c34673e53b60402df9977e0ac3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-03-03 05:41:33.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"03d69d57304f8ef739966e9b77ffc01ce2478747","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-03 06:42:44.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/14/831214/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"4a2240a8f212b53c4a692cab784fb64fb00aff91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 06:57:57.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c7e4fabbd0cf47c5965add849ab53156 : SUCCESS in 56m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a7d448603dfd42a88e08f9b1ad39fffa : SUCCESS in 55m 46s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f4a717004d55fe57129279a68fb23cfa890118c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 07:10:51.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/de7dd54b33dd4732bb0c212f868f21f5 : SUCCESS in 57m 13s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b85cc124083341c3b65ad8ae2860f12f : SUCCESS in 59m 30s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f3a217decd20461981a48a8c60e4c9b5 : SUCCESS in 54m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/adc90518c44c42ef952fc8fbaf5b7abe : SUCCESS in 1h 08m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c27ec98f75ea43e8b1a5968885a471ab : SUCCESS in 56m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/81224f42a3b74b99a6c1fc3a478363b2 : SUCCESS in 26m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1eb7b69568ac4e13abeed1efe5bc80e2 : SUCCESS in 32m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90d692481e1c42ad8e69de9525fcdefb : SUCCESS in 30m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1034367bea4745d4a96ac70f5ecbbc36 : SUCCESS in 12m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdde22ecc3d5430daf05b897e1f15ff7 : SUCCESS in 23m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1062c185533c496b9e4f5dc8fceee100 : SUCCESS in 24m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4030395498c24b32916ed0cad81a1bbe : SUCCESS in 11m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/193ff5e6bbdc4aa4a64d7c398036e706 : SUCCESS in 45m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/13f4d9813660482cb878490cf3d00c68 : SUCCESS in 1h 27m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/cc8a2f0bbefd4ad68c71fded10edc58f : SUCCESS in 1h 01m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c8dd5150656481089f4b589b48e4d20 : SUCCESS in 53m 38s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d9595a8c7f034591b7c386f88eec268b : SUCCESS in 53m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/80ff10c9f388416fb4acca40c2704d73 : SUCCESS in 1h 23m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/48325370625a4822885ff67f5c758b5f : SUCCESS in 1h 09m 29s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9428e19b8d814f059aba16d9b05c720f : SUCCESS in 53m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/77b58a1025af4d6dabde1cddd4a4c74b : SUCCESS in 1h 03m 24s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fc2e7c1ac88397c5097d1fd3ae1284e184ddddd4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-03 07:14:28.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/831214/7 : SUCCESS in 1h 32m 12s","accounts_in_message":[],"_revision_number":7},{"id":"6b701792e999fd27909f3fadb975530bbc2f26ff","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-03-11 10:10:39.000000000","message":"Patch Set 7: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"5a0ab84cfefeb2f7d77e8432fd10379ca4f3fde6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-11 15:11:05.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"246f66113701dc2f36a734e866e2591f546df329","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 15:12:27.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"5792497363f016bd805aec43c49cb8ef02551ae7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 20:24:42.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9bf71c4983cb433f8ca008738eb85406 : SUCCESS in 48m 28s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e65056547ac84a71b3689ab9ab5262f9 : SUCCESS in 1h 39m 47s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4602b1e65fae41eeba57670003486e2d : SUCCESS in 1h 15m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/47d5d81dd6a74d66afed0612d35960d2 : SUCCESS in 1h 28m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b1f4f109453e44b1a85b9366902a9846 : SUCCESS in 1h 36m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d685dbc72ae4a2f88c70deb95039588 : SUCCESS in 34m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/721e6fd8381e4b56988120c69bd4e114 : SUCCESS in 13m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/126e867c745446fcb80fe3f20caee8b9 : SUCCESS in 31m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1c6360d3627490bb26dd55bae0e0ee1 : SUCCESS in 24m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47df95d62fdc41108e39db019d040285 : SUCCESS in 10m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/84bbd60a8af7488ca7c8ef6bd590de53 : SUCCESS in 44m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/50bf58e4b0fc46e8864b9f3c613a7b59 : SUCCESS in 1h 19m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/032554d7da6949f08809891e0527f8ee : SUCCESS in 1h 29m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b3cb48d40f954b398c5972987d51a72b : SUCCESS in 44m 39s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c0e6fd944ae04f6fb782ad6e4279ce06 : SUCCESS in 53m 09s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7bfe0e795524f3299f2e3fb819cb480 : SUCCESS in 46m 26s","accounts_in_message":[],"_revision_number":7},{"id":"dc6a375afbe9938d8a4cd2d1fa90bfb28752f72b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 20:24:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"5a7b83bb6189c10faea9743ca66785697c5cc7fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 20:26:41.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecb80c97f03c45038294c065c06ba371 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"820b2e26658a3fb5e00a7cedbdc2418d2196aa42","revisions":{"68017fabe35b6b8bbc0d464e763673ff64166665":{"kind":"REWORK","_number":1,"created":"2022-02-28 14:18:02.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/14/831214/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/831214/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/831214/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/831214/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/831214/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/831214/1"}}},"commit":{"parents":[{"commit":"6349d1be38b468f2ba114fc242021fbbfa6e43cd","subject":"Merge \"Add new class for Logging API methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6349d1be38b468f2ba114fc242021fbbfa6e43cd"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-28 13:24:26.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-28 14:17:05.000000000","tz":330},"subject":"Ensure gateway is set for prefix delegated subnets","message":"Ensure gateway is set for prefix delegated subnets\n\nWith [1] gateway is no longer set for subnet created\nwith prefix delegation, but when adding the subnet\nto the router it fails as it expects gateway to be\nset.\n\nThis patch ensures gateway is set temporary to the first IP\nof the subnet as it used to be just like the temporary CIDR.\nAlso need to ensure dhcp configuration is skipped to avoid the\noriginal issue[2].\n\n[1] https://review.opendev.org/c/openstack/neutron/+/699465\n[2] https://bugs.launchpad.net/neutron/+bug/1856675\n\nCloses-Bug: #1962306\nRelated-Bug: #1856675\nChange-Id: I512f7d98ac99bb0ef06fd2acba09482e3436d18d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/68017fabe35b6b8bbc0d464e763673ff64166665"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/68017fabe35b6b8bbc0d464e763673ff64166665"}]},"branch":"refs/heads/master"},"5162a49194924dbb2e9c718c6d447a1b402cb214":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-02-28 14:30:07.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/14/831214/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/831214/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/831214/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/831214/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/831214/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/831214/2"}}},"commit":{"parents":[{"commit":"0763d9284a9cf9476bb79ba4434f8e54c5061403","subject":"Merge \"Change OVS and OVN installation directories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0763d9284a9cf9476bb79ba4434f8e54c5061403"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-28 13:24:26.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2022-02-28 14:30:07.000000000","tz":0},"subject":"Ensure gateway is set for prefix delegated subnets","message":"Ensure gateway is set for prefix delegated subnets\n\nWith [1] gateway is no longer set for subnet created\nwith prefix delegation, but when adding the subnet\nto the router it fails as it expects gateway to be\nset.\n\nThis patch ensures gateway is set temporary to the first IP\nof the subnet as it used to be just like the temporary CIDR.\nAlso need to ensure dhcp configuration is skipped to avoid the\noriginal issue[2].\n\n[1] https://review.opendev.org/c/openstack/neutron/+/699465\n[2] https://bugs.launchpad.net/neutron/+bug/1856675\n\nCloses-Bug: #1962306\nRelated-Bug: #1856675\nChange-Id: I512f7d98ac99bb0ef06fd2acba09482e3436d18d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5162a49194924dbb2e9c718c6d447a1b402cb214"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5162a49194924dbb2e9c718c6d447a1b402cb214"}]},"branch":"refs/heads/master","description":"Rebase"},"a33eb266420f1695010857c2a8fe9628e075873a":{"kind":"REWORK","_number":3,"created":"2022-03-01 04:40:27.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/14/831214/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/831214/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/831214/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/831214/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/831214/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/831214/3"}}},"commit":{"parents":[{"commit":"0763d9284a9cf9476bb79ba4434f8e54c5061403","subject":"Merge \"Change OVS and OVN installation directories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0763d9284a9cf9476bb79ba4434f8e54c5061403"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-28 13:24:26.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-03-01 04:40:12.000000000","tz":330},"subject":"Ensure gateway is set for prefix delegated subnets","message":"Ensure gateway is set for prefix delegated subnets\n\nWith [1] gateway is no longer set for subnet created\nwith prefix delegation, but when adding the subnet\nto the router it fails as it expects gateway to be\nset.\n\nThis patch ensures gateway is set temporary to the first IP\nof the subnet as it used to be just like the temporary CIDR.\nAlso need to ensure dhcp configuration is skipped to avoid the\noriginal issue[2].\n\n[1] https://review.opendev.org/c/openstack/neutron/+/699465\n[2] https://bugs.launchpad.net/neutron/+bug/1856675\n\nCloses-Bug: #1962306\nRelated-Bug: #1856675\nChange-Id: I512f7d98ac99bb0ef06fd2acba09482e3436d18d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a33eb266420f1695010857c2a8fe9628e075873a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a33eb266420f1695010857c2a8fe9628e075873a"}]},"branch":"refs/heads/master"},"8c75e1b81ca17ae78d8b5797da0394ff75f98026":{"kind":"REWORK","_number":4,"created":"2022-03-02 11:54:06.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/14/831214/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/831214/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/831214/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/831214/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/831214/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/831214/4"}}},"commit":{"parents":[{"commit":"0763d9284a9cf9476bb79ba4434f8e54c5061403","subject":"Merge \"Change OVS and OVN installation directories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0763d9284a9cf9476bb79ba4434f8e54c5061403"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-28 13:24:26.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-03-02 11:52:28.000000000","tz":330},"subject":"Ensure gateway is set for prefix delegated subnets","message":"Ensure gateway is set for prefix delegated subnets\n\nWith [1] gateway is no longer set for subnet created\nwith prefix delegation, but when adding the subnet\nto the router it fails as it expects gateway to be\nset.\n\nThis patch ensures gateway is set temporary to the first IP\nof the subnet as it used to be just like the temporary CIDR.\nAlso need to ensure dhcp configuration is skipped to avoid the\noriginal issue[2].\n\n[1] https://review.opendev.org/c/openstack/neutron/+/699465\n[2] https://bugs.launchpad.net/neutron/+bug/1856675\n\nCloses-Bug: #1962306\nRelated-Bug: #1856675\nChange-Id: I512f7d98ac99bb0ef06fd2acba09482e3436d18d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c75e1b81ca17ae78d8b5797da0394ff75f98026"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c75e1b81ca17ae78d8b5797da0394ff75f98026"}]},"branch":"refs/heads/master"},"6059e67ad49baa3468841a0d13400ad207a499c2":{"kind":"REWORK","_number":5,"created":"2022-03-02 13:36:51.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/14/831214/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/831214/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/831214/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/831214/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/831214/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/831214/5"}}},"commit":{"parents":[{"commit":"0763d9284a9cf9476bb79ba4434f8e54c5061403","subject":"Merge \"Change OVS and OVN installation directories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0763d9284a9cf9476bb79ba4434f8e54c5061403"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-28 13:24:26.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-03-02 13:36:33.000000000","tz":330},"subject":"Ensure gateway is set for prefix delegated subnets","message":"Ensure gateway is set for prefix delegated subnets\n\nWith [1] gateway is no longer set for subnet created\nwith prefix delegation, but when adding the subnet\nto the router it fails as it expects gateway to be\nset.\n\nThis patch ensures gateway is set temporary to the first IP\nof the subnet as it used to be just like the temporary CIDR.\nAlso need to ensure dhcp configuration is skipped to avoid the\noriginal issue[2].\n\n[1] https://review.opendev.org/c/openstack/neutron/+/699465\n[2] https://bugs.launchpad.net/neutron/+bug/1856675\n\nCloses-Bug: #1962306\nRelated-Bug: #1856675\nChange-Id: I512f7d98ac99bb0ef06fd2acba09482e3436d18d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6059e67ad49baa3468841a0d13400ad207a499c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6059e67ad49baa3468841a0d13400ad207a499c2"}]},"branch":"refs/heads/master"},"99d1ee24232471317c095fa3a196b48b4f1ef86a":{"kind":"REWORK","_number":6,"created":"2022-03-02 15:39:15.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/14/831214/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/831214/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/831214/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/831214/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/831214/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/831214/6"}}},"commit":{"parents":[{"commit":"0763d9284a9cf9476bb79ba4434f8e54c5061403","subject":"Merge \"Change OVS and OVN installation directories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0763d9284a9cf9476bb79ba4434f8e54c5061403"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-28 13:24:26.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-03-02 15:38:52.000000000","tz":330},"subject":"Ensure gateway is set for prefix delegated subnets","message":"Ensure gateway is set for prefix delegated subnets\n\nWith [1] gateway is no longer set for subnet created\nwith prefix delegation, but when adding the subnet\nto the router it fails as it expects gateway to be\nset.\n\nThis patch ensures gateway is set temporary to the first IP\nof the subnet as it used to be just like the temporary CIDR.\nAlso need to ensure dhcp configuration is skipped to avoid the\noriginal issue[2].\n\n[1] https://review.opendev.org/c/openstack/neutron/+/699465\n[2] https://bugs.launchpad.net/neutron/+bug/1856675\n\nCloses-Bug: #1962306\nRelated-Bug: #1856675\nChange-Id: I512f7d98ac99bb0ef06fd2acba09482e3436d18d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99d1ee24232471317c095fa3a196b48b4f1ef86a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99d1ee24232471317c095fa3a196b48b4f1ef86a"}]},"branch":"refs/heads/master"},"820b2e26658a3fb5e00a7cedbdc2418d2196aa42":{"kind":"REWORK","_number":7,"created":"2022-03-03 05:41:33.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/14/831214/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/831214/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/831214/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/831214/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/831214/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/831214/7"}}},"commit":{"parents":[{"commit":"0763d9284a9cf9476bb79ba4434f8e54c5061403","subject":"Merge \"Change OVS and OVN installation directories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0763d9284a9cf9476bb79ba4434f8e54c5061403"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-02-28 13:24:26.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-03-03 05:40:54.000000000","tz":330},"subject":"Ensure gateway is set for prefix delegated subnets","message":"Ensure gateway is set for prefix delegated subnets\n\nWith [1] gateway is no longer set for subnet created\nwith prefix delegation, but when adding the subnet\nto the router it fails as it expects gateway to be\nset.\n\nThis patch ensures gateway is set temporary to the first IP\nof the subnet as it used to be just like the temporary CIDR.\nAlso need to ensure dhcp configuration is skipped to avoid the\noriginal issue[2].\n\n[1] https://review.opendev.org/c/openstack/neutron/+/699465\n[2] https://bugs.launchpad.net/neutron/+bug/1856675\n\nCloses-Bug: #1962306\nRelated-Bug: #1856675\nChange-Id: I512f7d98ac99bb0ef06fd2acba09482e3436d18d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/820b2e26658a3fb5e00a7cedbdc2418d2196aa42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/820b2e26658a3fb5e00a7cedbdc2418d2196aa42"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
