)]}'
{"id":"openstack%2Fneutron~831409","triplet_id":"openstack%2Fneutron~master~I6bca7daf4d3e46071686bb62d5988ae314af8e7e","project":"openstack/neutron","branch":"master","topic":"bug/1962594","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-03-07 16:01:04.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-03-07 14:28:04.000000000","reason":"Slawek Kaplonski replied on the change"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-03-07 16:01:04.000000000","reason":"Change was submitted"},"32586":{"account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"last_update":"2022-03-03 12:45:20.000000000","reason":"Elvira replied on the change"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2022-03-03 12:51:47.000000000","reason":"Oleg Bondarev replied on the change"}},"hashtags":[],"change_id":"I6bca7daf4d3e46071686bb62d5988ae314af8e7e","subject":"Group execution of SQL functional tests","status":"MERGED","created":"2022-03-01 17:17:45.000000000","updated":"2022-03-07 16:03:06.000000000","submitted":"2022-03-07 16:01:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"831409-bug/1962594","meta_rev_id":"1b94eeda4993036e7677e3712b1077bddc62e6ba","_number":831409,"virtual_id_number":831409,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-07 16:01:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-03-07 05:16:30.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2022-03-07 14:28:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-03-07 11:02:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2022-03-03 12:45:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":2,"date":"2022-03-03 12:51:47.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2022-03-07 14:28:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}],"CC":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-01 18:11:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-01 18:42:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-01 19:37:21.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-01 20:45:20.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-03-02 07:34:17.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-03-02 16:01:06.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2022-03-07 05:16:30.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-03-07 05:16:30.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-03-07 05:16:30.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-03-07 11:02:54.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"2a3b4ba73f9fcd20b2e06b63aaf17f87066d239b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-01 17:17:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6a64d046c001ace3a8f5788d3f179d77f144dca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 18:11:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/190acb4b746c4c65b965b4187b257a39 : SUCCESS in 46m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c08ac21e78e54ff987f17d889f45859b : SUCCESS in 45m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e3b5225950041066b55d772fbbcdf42bde1bd18c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 18:42:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9a37b106ada48a7b9030d6219d9773a : SUCCESS in 38m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e95755ae2c142708235535e0628009a : SUCCESS in 39m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca5b4991557f41e4b1aa977990848b73 : SUCCESS in 15m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e166eae3e74a40d795c6fc33e58db7a0 : SUCCESS in 24m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/675687aff4c945f9a210a2d8e82829a6 : SUCCESS in 25m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f424727a9d34d5b82e47d330c107e34 : SUCCESS in 12m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39158d731cf44896aba2ddd2160320e1 : SUCCESS in 10m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/18db40500ad4415a8d2092516d49686f : FAILURE in 41m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/253630f7887c48b08ddfac3512ada97b : SUCCESS in 1h 20m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5eda99fd9f7faa5e68196ca95bce1ed3bbb53ae7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-01 18:53:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8f2d8f31d53b46d05addedef4cfaeea793f009e9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-01 19:05:25.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"27a988e902399bae720587de6e71bd931aedeebb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-01 19:37:21.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/09/831409/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"a4b624ed1ba060c8867c3e0fe0a2753c18bf1f06","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 19:59:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/41820f9a83f44dffa03cf964943cd91f : SUCCESS in 46m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/044c868d551f4890ac715bd33044d4a5 : SUCCESS in 46m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"858820ffdda920871d40023f9dc4cf8da5bdff07","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-01 20:45:20.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/831409/2 : SUCCESS in 1h 50m 55s","accounts_in_message":[],"_revision_number":2},{"id":"00ca4ff5444b65455d6a8a30952a7b34a53518fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 20:45:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96abfa824243438c8a9dc54dcbce2be3 : SUCCESS in 30m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/515d59cf1f314ef48eeaddae55a229bb : SUCCESS in 31m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c69d6b50b70456b87f5f1cac0648a10 : SUCCESS in 14m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85002b3b0a05457ebca33c4000b81fac : SUCCESS in 22m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0dd9d2911d674d719a8cf6ad8a6e8aba : SUCCESS in 25m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46fd5172120f4e64b0d1fdac6efc82b9 : SUCCESS in 9m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/839345e1e9ab41018cf79eb44fc65573 : SUCCESS in 9m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7265c9c0c8564e4cba9cdc52e998c663 : SUCCESS in 37m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6add30ccad1249d7a1c796ca71c8f73d : SUCCESS in 1h 50m 10s","accounts_in_message":[],"_revision_number":2},{"id":"44b8b1c1ed24a4fa91a91576c9644627b143ebcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-02 08:35:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"da0d04801f3927170c2503ed7581edb56c164f56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 09:36:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c96bde255f44318bf55c23c7bc8ed83 : SUCCESS in 58m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3850f8b15c4f489b9a07f52642109e2a : SUCCESS in 58m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ae0d6b2c38b92d2289073999655b875b2ff1f027","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-02 09:49:42.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/09/831409/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"53b46148a66c4c33a45ee095c039a397f5634559","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 09:59:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6be31817606c4c22bcecf12ee309a192 : SUCCESS in 30m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c07f2e96bc0f49efa98934f7fc6c54e5 : SUCCESS in 27m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21811fb1b43549d4a91d8a1e88859851 : SUCCESS in 13m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce71869b5bc04c8591d4dbc289b9d581 : SUCCESS in 23m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4026b6a3e3946819474708ee57172be : SUCCESS in 23m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27df5fa3050c43549bc7a1241307db8f : SUCCESS in 11m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/733d4baee7724e668532a011aed2813b : SUCCESS in 10m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6663799405f54cf1a14294cb783ae8b1 : FAILURE in 44m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b59dfaf6f120404aa539d569093a08ca : SUCCESS in 1h 22m 56s","accounts_in_message":[],"_revision_number":3},{"id":"3183d657a061de7ea5bc118a43e905e77209b572","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-02 10:07:20.000000000","message":"Patch Set 3: Code-Review-2","accounts_in_message":[],"_revision_number":3},{"id":"5a916cecc3876e486b72dc41ae91297c75b5f379","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-02 10:10:23.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/831409/3 : FAILURE in 1h 33m 39s","accounts_in_message":[],"_revision_number":3},{"id":"03ebefa4f3b2a79d80f7364be99cea4c97cd4c2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-02 11:36:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5feea84482d2bd99d92841bb13ca2860ea85c622","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-02 12:13:49.000000000","message":"Patch Set 4: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"691688a933395dbc663dc63d5257fa3c722e7354","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-02 12:16:26.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c1fe7ab2af6f7c90863d4748e0ebfeaf80cc5499","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 13:03:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/255f83cc6a7a45809cbbce33a42ef7e0 : SUCCESS in 44m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ccef9e2115a14042b1e03e0dfa3d84da : SUCCESS in 43m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8b52f40cb3b2780f404edd8c56b27845d1f976ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-02 13:55:09.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"665f1a1e6bed27c4f76a15869c4c733e2bec97ed","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-02 13:55:19.000000000","message":"Removed Code-Review-2 by Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":6},{"id":"7ebb605db442af7a8f3f0866786063330ce8d060","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-02 14:06:20.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/09/831409/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"5ff0a3760c76372cb2edcfe00ea45390819cb5d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 14:41:22.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb4829c104d741a7ad8bdac653bcb6c7 : SUCCESS in 42m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/179099e94fae42149c3fa1bde8d9a8b6 : SUCCESS in 42m 11s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6cf4d419c4ee4d9df0ee1ff1b05b23ea603517f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 15:28:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84288a2f2654458c9ac028a8cb5943cd : SUCCESS in 37m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30d7acfdb91c43d4855b27c69689a425 : SUCCESS in 35m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6833c08571643878721e9c6ab1dca26 : SUCCESS in 13m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1f6fb8b426e4ddcbf72d929ac6c2646 : SUCCESS in 29m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03906d03e8ca4eb197c2510358f0fd69 : SUCCESS in 25m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/097c5724bdb043d29e098ddd83257c41 : SUCCESS in 10m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb1e92d05c70427ba4a735da9f4b2f89 : SUCCESS in 12m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/015119705c994fabbb3e2acf3e6f3605 : SUCCESS in 52m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/28023e4d462d4e64918ffc3da348df98 : SUCCESS in 1h 29m 07s","accounts_in_message":[],"_revision_number":6},{"id":"4e4278af7564779ba785845740e06b331873c64a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-02 15:49:50.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/831409/6 : SUCCESS in 1h 53m 55s","accounts_in_message":[],"_revision_number":6},{"id":"409957d56976583a8598d065b00451c22f443ff0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-02 16:00:18.000000000","message":"Patch Set 5:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/09/831409/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"575f0787e32edaf72f168c2695bc8d616afe2091","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-03-02 16:01:06.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"7d065a8bad8138e3350f804c9141748cb6f20660","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-03-02 16:07:30.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"364354cc81391cd8256647c882980bff3b7d23e3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-02 17:45:15.000000000","message":"Patch Set 6:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/09/831409/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"9fa7eb50cb73521a2bea3b38e2ed806f12fb1bb0","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-03-03 07:47:35.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0ff6b1e6a26756fe7130d52a0564faaf0725a790","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-03 12:09:54.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7825c332d2713e9064fa5f1ec9abc09768a8634c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-03 12:10:56.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"ea441f61e9b35913235615d1303cb674477bf650","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-03-03 12:45:20.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"ab29811167142f11067634ac573ec4201d5a2163","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-03-03 12:51:47.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"9da1bf91f6344ba9edbecfa43601d46171a89a8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 12:56:54.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/52ad44ffc518421c9ec7de4da44e7f00 : SUCCESS in 43m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/28d9d75f00b44252b7d8fe3088c73418 : SUCCESS in 42m 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"59c006ccb4600de2efba1ba2df8bd21825515602","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-03 13:10:22.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/09/831409/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"d98b6de8237cceeeb92cfd9001aaa0e6f39abac9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 13:45:39.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dae78588073b4a28907ecf836d0d374d : SUCCESS in 37m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c834c6e90f44d8da4846e08e35a3830 : SUCCESS in 30m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b3a98c1e85f404db8a40f0d433e1a70 : SUCCESS in 14m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f36ba3ec2be415c8cd67483e1fe73bb : SUCCESS in 25m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57073b4957f84b97a9dfd47d8cbc13a6 : SUCCESS in 25m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c7cc08315a948aeb25ec643f3997b8b : SUCCESS in 11m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2013dad13e344235877ec5f444db8f9c : SUCCESS in 13m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ab858f10260d4d7fb0175c93733326ac : FAILURE in 40m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d5b3de45598c4887a3c6d549816086ce : SUCCESS in 1h 29m 29s","accounts_in_message":[],"_revision_number":7},{"id":"d0e83bbd911849af7fadd86868d2abcac0dce947","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-03 13:51:11.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/831409/7 : SUCCESS in 1h 39m 32s","accounts_in_message":[],"_revision_number":7},{"id":"31c868a2125e111bc5f6969701d07d155b1bdc78","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-04 16:54:37.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9d8fd21579cdb1609149f0091ecb67d2cdd30e8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 18:02:03.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/08fa0d0e01974f1d927b249e80cc1ba3 : SUCCESS in 1h 03m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9193b21bda2c4779ad9e0e1adf84bb3d : SUCCESS in 1h 05m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"852d73bb1b3b29d7f492f350b5f0658c402ae105","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-04 18:43:38.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/831409/7 : SUCCESS in 1h 48m 28s","accounts_in_message":[],"_revision_number":7},{"id":"aa039e83b1402cb32f17ac7f50ba42d6de1313d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 19:04:13.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a3d95d71f0e4c38b066c5a2f996518f : SUCCESS in 42m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb48340054d044cd8f2f9c3f535f6f9f : SUCCESS in 35m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/448d9cde430a49a6946d6b498315d7fc : SUCCESS in 15m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c1bd143dc8e45a0821a405dab38cad8 : SUCCESS in 30m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acba1646377041b29a2237b334460424 : SUCCESS in 29m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a729fe4a676c4b44b4e12f47db1203d9 : SUCCESS in 10m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1677c5cb0f94987baa0bbfd2239cce0 : SUCCESS in 12m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b7d307189f294b6e84cfb85305b6bab5 : SUCCESS in 1h 04m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3043c069d9bd4aa6bbe72cf01cf695ba : SUCCESS in 1h 59m 07s","accounts_in_message":[],"_revision_number":7},{"id":"1ab9d3886a402c434a1f347595c72195687db5d2","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-03-07 05:16:30.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"fb440c1da4ee15585e512b7265f7d06b366bb5c1","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-03-07 11:02:54.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"5c10017006908fa40334ec0fcad71a2203487b7b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-07 14:28:04.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"79ba80e381124a0f58cbc88256b5a42e156c20a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 14:28:35.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"87ca65c0022c235ceba709678ecc3470ae886a7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 16:01:03.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5d805af5c3a4c6caf649cb1a1623d2c : SUCCESS in 36m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a129c98ac094e91ae27bbc2826a40d7 : SUCCESS in 15m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a893b39419ee4a84b55b8d778ebfd173 : SUCCESS in 32m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2343a0184724434ca070a1b9eecdc1cc : SUCCESS in 22m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86bcbbc3de0c4c1a8908f6a12cb8f39a : SUCCESS in 9m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23c79583d2ed437ebd109d93280fef29 : SUCCESS in 11m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/23dfe96d32b84e948d3aede880d877cf : SUCCESS in 44m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b1b023e6bf364bf6b868a6fd3fe10b16 : SUCCESS in 1h 27m 08s","accounts_in_message":[],"_revision_number":7},{"id":"ae5263986748f4aec540955614963441923a27b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 16:01:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"1b94eeda4993036e7677e3712b1077bddc62e6ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 16:03:06.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d67f1b3e9c049f0815d70799fb69997 : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/faee12a3e483458b8b2c337a2ff28b12 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"4a66adf94c4e01ca92dfe12cf8646d11acfc7176","revisions":{"d3d7fc388296f60a629a7c0878d63e2c51e19e24":{"kind":"REWORK","_number":1,"created":"2022-03-01 17:17:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/09/831409/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/831409/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/831409/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/831409/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/831409/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/831409/1"}}},"commit":{"parents":[{"commit":"ed267b01518fd8a2c0292cdcf4d543c05b959d62","subject":"Merge \"Fix typo in the third instruction paragraph of \"Controller nodes\" block\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed267b01518fd8a2c0292cdcf4d543c05b959d62"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-02-11 17:43:34.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-02-11 17:47:03.000000000","tz":0},"subject":"Group execution of SQL functional tests","message":"Group execution of SQL functional tests\n\nFunctional SQL tests (MySQL and PostgreSQL) use the DB installed\nbackend. To avoid overloading the DB engine, the test cases are\ngrouped [1] together and executed by the same worker process.\nThat will ensure those tests are not executed in parallel,\ninterfering each other.\n\nClasses that are grouped:\n* TestModelsMigrationsMysql (patch [1])\n* TestModelsMigrationsPsql (patch [1])\n* TestIsSessionActiveMySQL\n* TestIsSessionActivePostgreSQL\n\n[1]https://review.opendev.org/c/openstack/neutron/+/785568\n\nCloses-Bug: #1962594\nRelated-Bug: #1687027\nChange-Id: I6bca7daf4d3e46071686bb62d5988ae314af8e7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d3d7fc388296f60a629a7c0878d63e2c51e19e24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d3d7fc388296f60a629a7c0878d63e2c51e19e24"}]},"branch":"refs/heads/master"},"0038c067de5a28b9d919da64b760e98d7f9f4938":{"kind":"REWORK","_number":2,"created":"2022-03-01 18:53:30.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/09/831409/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/831409/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/831409/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/831409/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/831409/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/831409/2"}}},"commit":{"parents":[{"commit":"ed267b01518fd8a2c0292cdcf4d543c05b959d62","subject":"Merge \"Fix typo in the third instruction paragraph of \"Controller nodes\" block\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed267b01518fd8a2c0292cdcf4d543c05b959d62"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-02-11 17:43:34.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-02-11 19:30:18.000000000","tz":0},"subject":"Group execution of SQL functional tests","message":"Group execution of SQL functional tests\n\nFunctional SQL tests (MySQL and PostgreSQL) use the DB installed\nbackend. To avoid overloading the DB engine, the test cases are\ngrouped [1] together and executed by the same worker process.\nThat will ensure those tests are not executed in parallel,\ninterfering each other.\n\nAll classes ended in \"MySQL\" or \"PostgreSQL\" are now grouped.\n\nCloses-Bug: #1962594\nRelated-Bug: #1687027\nChange-Id: I6bca7daf4d3e46071686bb62d5988ae314af8e7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0038c067de5a28b9d919da64b760e98d7f9f4938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0038c067de5a28b9d919da64b760e98d7f9f4938"}]},"branch":"refs/heads/master"},"6a0551895e88ca7a58e81bb1a120fc13b9bb74bd":{"kind":"REWORK","_number":3,"created":"2022-03-02 08:35:23.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/09/831409/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/831409/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/831409/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/831409/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/831409/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/831409/3"}}},"commit":{"parents":[{"commit":"ed267b01518fd8a2c0292cdcf4d543c05b959d62","subject":"Merge \"Fix typo in the third instruction paragraph of \"Controller nodes\" block\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed267b01518fd8a2c0292cdcf4d543c05b959d62"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-02-11 17:43:34.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-02-11 20:10:11.000000000","tz":0},"subject":"Group execution of SQL functional tests","message":"Group execution of SQL functional tests\n\nFunctional SQL tests (MySQL and PostgreSQL) use the DB installed\nbackend. To avoid overloading the DB engine, the test cases are\ngrouped [1] together and executed by the same worker process.\nThat will ensure those tests are not executed in parallel,\ninterfering each other.\n\nAll classes ended in \"MySQL\" or \"PostgreSQL\" are now grouped.\n\nThis patch also discards the DB testing resource cleanup. Once a\nDB test finishes, the ``OpportunisticSqlFixture`` class adds a\n``addCleanup`` method to tear down the resources. This is when\nthe timeouts usually happen. Skipping the resource removal saves\nexecution time too. For example, the execution of: all five tests\nof ``TestModelsMigrationsPostgreSQL``\n\n                                  No cleanup     With cleanup\n  test_branches:                  7.6            8.4\n  test_forbid_offline...:         0.6            4.2\n  test_models_sync:               4.7            9.7\n  test_upgrade_contract_branch:   3.8            4.8\n  test_upgrade_expand_branch:     1.4            5.4\n\nCloses-Bug: #1962594\nRelated-Bug: #1687027\nChange-Id: I6bca7daf4d3e46071686bb62d5988ae314af8e7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a0551895e88ca7a58e81bb1a120fc13b9bb74bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a0551895e88ca7a58e81bb1a120fc13b9bb74bd"}]},"branch":"refs/heads/master"},"50ab4e73fe0a5a17acbf331d5be5c00ad1478a20":{"kind":"REWORK","_number":4,"created":"2022-03-02 11:36:38.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/09/831409/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/831409/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/831409/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/831409/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/831409/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/831409/4"}}},"commit":{"parents":[{"commit":"6a571f8068e863a90929358b4fb13b02ebc4f3ce","subject":"Merge \"[OVN][Cosmetic] Improve gen_router_port_options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a571f8068e863a90929358b4fb13b02ebc4f3ce"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-02-11 17:43:34.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-02 11:36:10.000000000","tz":0},"subject":"[WIP] Group execution of SQL functional tests","message":"[WIP] Group execution of SQL functional tests\n\nFunctional SQL tests (MySQL and PostgreSQL) use the DB installed\nbackend. To avoid overloading the DB engine, the test cases are\ngrouped [1] together and executed by the same worker process.\nThat will ensure those tests are not executed in parallel,\ninterfering each other.\n\nAll classes ended in \"MySQL\" or \"PostgreSQL\" are now grouped.\n\nCloses-Bug: #1962594\nRelated-Bug: #1687027\nChange-Id: I6bca7daf4d3e46071686bb62d5988ae314af8e7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/50ab4e73fe0a5a17acbf331d5be5c00ad1478a20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/50ab4e73fe0a5a17acbf331d5be5c00ad1478a20"}]},"branch":"refs/heads/master"},"dd7afe34e9278a5ae708ced40c5eaf27fe839cb5":{"kind":"REWORK","_number":5,"created":"2022-03-02 12:16:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/09/831409/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/831409/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/831409/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/831409/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/831409/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/831409/5"}}},"commit":{"parents":[{"commit":"6a571f8068e863a90929358b4fb13b02ebc4f3ce","subject":"Merge \"[OVN][Cosmetic] Improve gen_router_port_options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a571f8068e863a90929358b4fb13b02ebc4f3ce"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-02-11 17:43:34.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-02 12:16:06.000000000","tz":0},"subject":"[WIP] Group execution of SQL functional tests","message":"[WIP] Group execution of SQL functional tests\n\nFunctional SQL tests (MySQL and PostgreSQL) use the DB installed\nbackend. To avoid overloading the DB engine, the test cases are\ngrouped [1] together and executed by the same worker process.\nThat will ensure those tests are not executed in parallel,\ninterfering each other.\n\nAll classes ended in \"MySQL\" or \"PostgreSQL\" are now grouped.\n\nCloses-Bug: #1962594\nRelated-Bug: #1687027\nChange-Id: I6bca7daf4d3e46071686bb62d5988ae314af8e7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd7afe34e9278a5ae708ced40c5eaf27fe839cb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd7afe34e9278a5ae708ced40c5eaf27fe839cb5"}]},"branch":"refs/heads/master"},"8d86061729beab71c9c3f6b9c9ddc7e52a5df680":{"kind":"REWORK","_number":6,"created":"2022-03-02 13:55:09.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/09/831409/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/831409/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/831409/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/831409/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/831409/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/831409/6"}}},"commit":{"parents":[{"commit":"6a571f8068e863a90929358b4fb13b02ebc4f3ce","subject":"Merge \"[OVN][Cosmetic] Improve gen_router_port_options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a571f8068e863a90929358b4fb13b02ebc4f3ce"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-02-11 17:43:34.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-02 13:54:59.000000000","tz":0},"subject":"Group execution of SQL functional tests","message":"Group execution of SQL functional tests\n\nFunctional SQL tests (MySQL and PostgreSQL) use the DB installed\nbackend. To avoid overloading the DB engine, the test cases are\ngrouped [1] together and executed by the same worker process.\nThat will ensure those tests are not executed in parallel,\ninterfering each other.\n\nAll test classes ended in \"MySQL\" or \"PostgreSQL\" are now grouped.\nThat means the tests implemented inside the same class are\nexecuted in the same worker, not in parallel. **However**, that\ndoesn\u0027t guarantee all test cases using the same database backend\nwill run in the same worker serially. The FT job can still execute\ntest cases from different classes, using the same DB backend,\nin two workers at the same time.\n\nThis patch also fixes the ``TestReservation*`` classes to inherit\nfrom the parent classes in the correct order. That sets the\n\"DRIVER\" value to the expected one \"mysql\" or \"postgresql\".\n\nCloses-Bug: #1962594\nRelated-Bug: #1687027\nChange-Id: I6bca7daf4d3e46071686bb62d5988ae314af8e7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d86061729beab71c9c3f6b9c9ddc7e52a5df680"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d86061729beab71c9c3f6b9c9ddc7e52a5df680"}]},"branch":"refs/heads/master"},"4a66adf94c4e01ca92dfe12cf8646d11acfc7176":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2022-03-03 12:10:56.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/09/831409/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/831409/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/831409/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/831409/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/831409/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/831409/7"}}},"commit":{"parents":[{"commit":"6a571f8068e863a90929358b4fb13b02ebc4f3ce","subject":"Merge \"[OVN][Cosmetic] Improve gen_router_port_options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a571f8068e863a90929358b4fb13b02ebc4f3ce"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-02-11 17:43:34.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2022-03-03 12:10:56.000000000","tz":0},"subject":"Group execution of SQL functional tests","message":"Group execution of SQL functional tests\n\nFunctional SQL tests (MySQL and PostgreSQL) use the DB installed\nbackend. To avoid overloading the DB engine, the test cases are\ngrouped [1] together and executed by the same worker process.\nThat will ensure those tests are not executed in parallel,\ninterfering each other.\n\nAll test classes ended in \"MySQL\" or \"PostgreSQL\" are now grouped.\nThat means the tests implemented inside the same class are\nexecuted in the same worker, not in parallel. **HOWEVER**, that\ndoesn\u0027t guarantee all test cases using the same database backend\nwill run in the same worker serially. The FT job can still execute\ntest cases from different classes, using the same DB backend,\nin two workers at the same time.\n\nThis patch also fixes the ``TestReservation*`` classes to inherit\nfrom the parent classes in the correct order. That sets the\n\"DRIVER\" value to the expected one \"mysql\" or \"postgresql\".\n\n[1]https://review.opendev.org/c/openstack/neutron/+/785568\n\nCloses-Bug: #1962594\nRelated-Bug: #1687027\nChange-Id: I6bca7daf4d3e46071686bb62d5988ae314af8e7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4a66adf94c4e01ca92dfe12cf8646d11acfc7176"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4a66adf94c4e01ca92dfe12cf8646d11acfc7176"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
