)]}'
{"id":"openstack%2Fneutron~831595","triplet_id":"openstack%2Fneutron~stable%2Fxena~If51d7f25375768f8c60492c84d84e91d91886025","project":"openstack/neutron","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"34118":{"account":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"last_update":"2022-03-04 13:10:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If51d7f25375768f8c60492c84d84e91d91886025","subject":"[ OVN ][Migration] Reload systemctl daemon after removal","status":"MERGED","created":"2022-03-02 18:45:26.000000000","updated":"2022-03-04 13:12:06.000000000","submitted":"2022-03-04 13:10:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831595","meta_rev_id":"8bbab1590b3f8ce5e366e7a6f1ee5b0606ddb5f1","_number":831595,"virtual_id_number":831595,"owner":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-04 13:10:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-03-04 10:53:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-03-03 11:18:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-03-03 11:15:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-03-04 10:53:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-02 19:38:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-02 20:37:33.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-03-03 11:15:02.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2022-03-03 11:18:54.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-03-04 10:53:21.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"1f1a397eafe0c88122d6fae31724b1de45dd0415","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"date":"2022-03-02 18:45:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0910b3aa8894e8c150b6133d0ca3db2c6b44d79f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"date":"2022-03-02 19:03:15.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"25a7123e2497ff9bc6c809046721c3ba4dd8526c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 19:38:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a65fd73e8eee412282aef0e63846f436 : SUCCESS in 17m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c76038d3e1ef4ff8879a20a71bae861c : SUCCESS in 29m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d698c0293fc4439c989fa7970517a4be : SUCCESS in 10m 29s","accounts_in_message":[],"_revision_number":2},{"id":"86a4cb6709d9853caa6c2b9d5dd11b50d8b9f4b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 19:48:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/62f39a19fd194e03b8566094b83dccb3 : SUCCESS in 43m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a52fca62ade4b4e91bf0faaea2cf032 : SUCCESS in 43m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f419680bdb2789ee7eb6af03f145ed1bafa0e2d6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-02 20:37:33.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/831595/2 : SUCCESS in 1h 33m 33s","accounts_in_message":[],"_revision_number":2},{"id":"76b00eaf5ea4949917b6f2dc2dcdbde0d3da48c8","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-03-03 11:15:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4e9a93f0351f7f60ffbe225baaa2b0355794bd1d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-03-03 11:18:54.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a75f352e66a295d2e205e43e3ef073760f1f761f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-04 10:53:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fa7a0f5926c6f41932b65c5e90809f60ef9073d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 10:54:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"be8eb2f0c84f78669e08ad5cfe6ac14b927cfa55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 13:10:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55b87b08e3794c1cbc2aef87a0460d0e : SUCCESS in 21m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b36c88cb5680434899665987b3f72d18 : SUCCESS in 34m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e36c609ad05447dda59c436c968a3f48 : SUCCESS in 10m 46s","accounts_in_message":[],"_revision_number":2},{"id":"97f2f5abb9536d926aa246cd6c2aafcf021b160e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 13:10:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8bbab1590b3f8ce5e366e7a6f1ee5b0606ddb5f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 13:12:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38c9c660776b4460b3ed4d056902d445 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c9de2e89df7c9c296fcdf8770289d41537b0a376","revisions":{"99cdb83ef3fd1f3cb36202c276fe7ed1b735d215":{"kind":"REWORK","_number":1,"created":"2022-03-02 18:45:26.000000000","uploader":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"ref":"refs/changes/95/831595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/831595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/831595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/831595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/831595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/831595/1"}}},"commit":{"parents":[{"commit":"b0b17dad0aee54031efc767cc58614491550ba8b","subject":"Restructure layout for periodic, experimental and tox jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b0b17dad0aee54031efc767cc58614491550ba8b"}]}],"author":{"name":"Arnau Verdaguer","email":"averdagu@redhat.com","date":"2022-03-02 11:23:37.000000000","tz":60},"committer":{"name":"Arnau Verdaguer","email":"averdagu@redhat.com","date":"2022-03-02 18:43:56.000000000","tz":60},"subject":"[ OVN ][Migration] Reload systemctl daemon after removal","message":"[ OVN ][Migration] Reload systemctl daemon after removal\n\nAfter stopping and deleting the services if this role is runned\nagain it could fail bc systemd has still some ovs services loaded\n(eventhough the service is stopped) this will cause that ansible\nwill try to delete again and fail while trying to disable and\ndelete the service.\n\nChange-Id: If51d7f25375768f8c60492c84d84e91d91886025\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99cdb83ef3fd1f3cb36202c276fe7ed1b735d215"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99cdb83ef3fd1f3cb36202c276fe7ed1b735d215"}]},"branch":"refs/heads/stable/xena"},"c9de2e89df7c9c296fcdf8770289d41537b0a376":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-03-02 19:03:15.000000000","uploader":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"ref":"refs/changes/95/831595/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/831595/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/831595/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/831595/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/831595/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/831595/2"}}},"commit":{"parents":[{"commit":"b0b17dad0aee54031efc767cc58614491550ba8b","subject":"Restructure layout for periodic, experimental and tox jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b0b17dad0aee54031efc767cc58614491550ba8b"}]}],"author":{"name":"Arnau Verdaguer","email":"averdagu@redhat.com","date":"2022-03-02 11:23:37.000000000","tz":60},"committer":{"name":"Arnau Verdaguer","email":"averdagu@redhat.com","date":"2022-03-02 18:59:32.000000000","tz":60},"subject":"[ OVN ][Migration] Reload systemctl daemon after removal","message":"[ OVN ][Migration] Reload systemctl daemon after removal\n\nAfter stopping and deleting the services if this role is runned\nagain it could fail bc systemd has still some ovs services loaded\n(eventhough the service is stopped) this will cause that ansible\nwill try to delete again and fail while trying to disable and\ndelete the service.\n\nChange-Id: If51d7f25375768f8c60492c84d84e91d91886025\n(cherry picked from commit 5ee94a5b9b5235ae290304e6648bd3b027a7b90f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c9de2e89df7c9c296fcdf8770289d41537b0a376"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c9de2e89df7c9c296fcdf8770289d41537b0a376"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
