)]}'
{"id":"openstack%2Fneutron~833247","triplet_id":"openstack%2Fneutron~master~I3da37fee205b8d67d10673075b9130147d9eab5f","project":"openstack/neutron","branch":"master","topic":"bug/1964575","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-04-08 13:08:32.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2022-04-08 13:08:32.000000000","reason":"Change was submitted"},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2022-04-08 13:08:32.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-04-08 09:14:26.000000000","reason":"Slawek Kaplonski replied on the change"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-04-08 13:08:32.000000000","reason":"Change was submitted"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2022-04-08 11:10:45.000000000","reason":"Oleg Bondarev replied on the change"}},"hashtags":[],"change_id":"I3da37fee205b8d67d10673075b9130147d9eab5f","subject":"Add the corresponding DB context to all SQL transactions","status":"MERGED","created":"2022-03-11 10:13:06.000000000","updated":"2022-04-08 13:10:25.000000000","submitted":"2022-04-08 13:08:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":27,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"833247-bug/1964575","meta_rev_id":"4deb785815ffbb05df0a3b90b9913f68f22c6e77","_number":833247,"virtual_id_number":833247,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-04-08 10:18:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-08 13:08:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-04-08 09:14:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-04-08 11:23:52.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-04-08 11:23:52.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-11 10:48:36.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-11 11:30:55.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-11 11:37:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-11 12:17:03.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-11 12:30:28.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-03-22 08:41:29.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-04-01 07:38:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-04-01 07:38:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2022-04-01 07:38:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-04-01 14:00:32.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"0bbca3037928f27bdb339220ad09f5f1c4a37329","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-11 10:13:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09f5d825ca0c5625997b2b86b908d043ea9ce1cc","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-11 10:13:30.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"61e5c1c6a8d93193ccecdd32c5f1bece2a6039da","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-11 10:47:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0feaee2d9139dd8333f380a32c1d324cdce9aeee","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-03-11 10:48:36.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/47/833247/1/openstack-sriov-neutron-master-direct : ABORTED\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/47/833247/1/openstack-sriov-neutron-master-macvtap : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"967c48bc9f4d4bc6b60ff2502ae66139fa27aeb6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-03-11 11:23:41.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/47/833247/2/openstack-sriov-neutron-master-direct : SUCCESS\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/47/833247/2/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"3a2d584bca77e74812985972f88fda4da23a256d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-11 11:30:55.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/47/833247/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"3e479f686896f8f98944b039633e2d24c4b68ca4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 11:37:39.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"1a11b5c2e459be10affdf6f1108fa58a1f62bf25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 11:39:18.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"4384b85742851373fdb15ca32f446c98498250f0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-11 11:39:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b9ef97b5278a2054c9b93a8ae8ec1df15dfc276a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-11 11:40:48.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"add03515d2b26ddc556d989dda6157e3261c8c06","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-11 12:17:03.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/47/833247/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"24243dc066aa206ca0be363867f69c09faf3d681","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-11 12:30:28.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/833247/2 : FAILURE in 1h 42m 01s","accounts_in_message":[],"_revision_number":2},{"id":"a01d9d30e24f8e240cb1e22cee5ecbbdce7b0ab8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-03-11 12:31:26.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/47/833247/2/openstack-sriov-neutron-master-direct : ABORTED\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/47/833247/2/openstack-sriov-neutron-master-macvtap : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"aca5b24137cab3b6250a00bd031536ce5f6d4933","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 13:22:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1dfa65041eab4189aea04d8967ec0742 : SUCCESS in 43m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/38575e82d2c341f992be01ce21a20db1 : SUCCESS in 1h 10m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"38dd6d666c5d3dcfe0bec0f1291cdeb30152d9ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 13:29:16.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b451d594107f4522bb523da9f0201a7a : SUCCESS in 41m 06s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3ac6f1d8f4cf470fabeec2362f9e6592 : SUCCESS in 1h 20m 24s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/63d10d260508431296f3bc9e5e68db72 : SUCCESS in 1h 06m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e7e85e4b29124fe0a414d45013bfe6c1 : SUCCESS in 58m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ead03540f6be48d7b81f34d3824b3bf9 : SUCCESS in 1h 18m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6fcd36d297e54c568acd1e22e4b4f2a1 : SUCCESS in 40m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/614f53155e974675b761f111e6f8f424 : SUCCESS in 30m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dec90bd0086647f181e643a581fc27bd : SUCCESS in 38m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37450e4272404dfdb8e5ba63bd374c1b : FAILURE in 9m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/adb2e708707e49b0b88af2deda30eabe : SUCCESS in 30m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/479538b55e3841b18faac7c95e115448 : SUCCESS in 31m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4611c89747554048ae76c51f00a81d57 : SUCCESS in 11m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c7c315862d34aed9f1b88a6105403cb : SUCCESS in 52m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b37402481286497395189b5ee99220ef : SUCCESS in 1h 27m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/dbc23077e4974b129256841039123dec : SUCCESS in 1h 00m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2cc21d0560174b30b9f7db73ce85cbf8 : SUCCESS in 52m 35s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/407ebc673da348faa3ad963e07cf27f1 : SUCCESS in 1h 00m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fa70a0fdc75147f19c98d938fdd087c8 : SUCCESS in 1h 26m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/472dc634bb5a48a69219180f4cfedfdf : SUCCESS in 1h 12m 28s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f20856c4f714478ab7c12deb782520f6 : SUCCESS in 1h 06m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/38859a909d6e434bb336cb09e7914e9a : SUCCESS in 1h 14m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1dee93709ad39fa24ec1775b3b33f44b021d3acb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-03-11 13:49:28.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/47/833247/2/openstack-sriov-neutron-master-direct : SUCCESS\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/47/833247/2/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"ea65fe965f0ef3a2d5d8bb997a522dc11a30645e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-11 14:01:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fad9df4e9ba81d2d8087d6cc7589dd19c52eb20e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-03-11 14:53:09.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/47/833247/3/openstack-sriov-neutron-master-direct : SUCCESS\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/47/833247/3/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"b21691eafd8f178f939993f109a9be29f06116ab","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-11 15:02:00.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/47/833247/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"34271b9591f1d78c11ea5ece63df5977ef480a7e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-11 15:44:28.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/833247/3 : SUCCESS in 1h 36m 23s","accounts_in_message":[],"_revision_number":3},{"id":"63e62879a231ff9dea4898dc8207f9f0054ceddd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 15:51:24.000000000","message":"Patch Set 3: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f2d09730457f4670a47579cea575eebf : SUCCESS in 55m 27s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/52bbc01b33564230b5cb00d31dbe390a : SUCCESS in 1h 05m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a2bdd812491d4e478150f8553d20373e : SUCCESS in 56m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/16aa8b583b314298abfad38207866e45 : SUCCESS in 48m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0590ba7401944f39974261b840857f1a : SUCCESS in 50m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2d902fa3cdff436a9d818834d2207e74 : SUCCESS in 28m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/749a4ad5e52448ed96b9f1e2deb16ae6 : SUCCESS in 49m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a65645448794df5a77e4f088900e00c : SUCCESS in 45m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eda22a0ceb2475490e3abead4c41bbc : FAILURE in 8m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5cb47f0358a460eb926889a93e46092 : SUCCESS in 37m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb25504a78e64c95898e2cc61c685e74 : SUCCESS in 29m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9df78b547414c53b0c3d70e974778d4 : SUCCESS in 10m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ed089689c19548a49ee092b6281a1778 : SUCCESS in 1h 03m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ee90c744f945436099bec4604c5426c1 : SUCCESS in 1h 40m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2cb4a9ff9efe4ab4af57b64049b28749 : SUCCESS in 1h 31m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5cf909f2aafc46e1a7e0134ba6292fc7 : SUCCESS in 1h 03m 15s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/75304951d986446c8132dd52a6880662 : SUCCESS in 1h 02m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/07e632b0ccac4d31afbc84eec30d4523 : SUCCESS in 1h 28m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0af78ce9c4ce4272a8ab3963f293a868 : SUCCESS in 1h 29m 21s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59d2ba2f9af8429e9ee0795812233c3c : SUCCESS in 1h 07m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/19ee36fc094b478bbec116b71bfe89d9 : SUCCESS in 1h 11m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fcfc41e7799aae0cda0122af7c0546d3ca76e4d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-11 16:02:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7ac0f338c1f21f1f48f7efac8994622a8d0f5765","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-03-11 16:39:57.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/47/833247/4/openstack-sriov-neutron-master-direct : SUCCESS\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/47/833247/4/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"90b6c6b98781c1d0faf4c56cc4af491299941881","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 17:37:36.000000000","message":"Patch Set 4: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/02c3b5583ba048798c45930d05d85c40 : FAILURE in 27m 23s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3b96ad4a54fb457d9303dbd471a54b1a : FAILURE in 20m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/93ddfa97b1224c35a01238dd905c63dc : FAILURE in 18m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dcf58681560240da84afd0849324b12c : FAILURE in 18m 38s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/cb6665b7b2ec48f49aebecb8767af681 : FAILURE in 19m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1226a1ae2bb34df09ce6d2c9d6d35df1 : FAILURE in 20m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b4ff466c0034bb4b234146f0a834803 : FAILURE in 36m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08783f86b36743c2af605b4f296f92ba : FAILURE in 35m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0ea07d1acfe4e6799464b4555e1ac04 : FAILURE in 9m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/387a737cb0e24a04aef0d7faf9291779 : FAILURE in 33m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/345fd159b88a4ffdbc322bfebe775a4f : FAILURE in 32m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff1fdbf49c05487f9c34da4cdf53094b : SUCCESS in 12m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bf3bd8e5f6a447828ac24b67ce0defb1 : FAILURE in 49m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/44333fb451bf45c99c77f4531bb63c78 : POST_FAILURE in 51m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/34304f90a8ab497785017481ac9a48ec : FAILURE in 26m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d88e946bb784f68b639b902c2c22a63 : SUCCESS in 56m 11s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8cafeddf05f94fbcb3afa15b03b553d5 : SUCCESS in 53m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fb77ba54d4b1459ba21780702e6bfa65 : FAILURE in 32m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ad18c834fb9c4d72bb44e33778db26f1 : FAILURE in 34m 55s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eaab1e19b71945b6a3602c98975f91f2 : FAILURE in 32m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/28e3bbb3efc2477496c1dd79eb9dec1e : FAILURE in 30m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"07ed2034fc58d1ca5f262b3709df25993ffd0e52","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-11 17:40:16.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/47/833247/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"7343a7f3a546ff87f5d33fa76a9644f3aa8dc371","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-11 18:23:35.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/833247/4 : FAILURE in 1h 43m 45s","accounts_in_message":[],"_revision_number":4},{"id":"cda6826c414930ee2c0eda6a443d8c826d1f186a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-11 18:25:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"06c3e31419d070026625592b1512444af3cf1306","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-03-11 18:25:33.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/47/833247/4/openstack-sriov-neutron-master-direct : ABORTED\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/47/833247/4/openstack-sriov-neutron-master-macvtap : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"7c15699507102119b9e4ef7f3cf869f5af89dd0a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-03-11 20:08:08.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/47/833247/5/openstack-sriov-neutron-master-direct : SUCCESS\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/47/833247/5/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"d2a36ccfb7bda913935754008002a9a4b051da84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 20:14:19.000000000","message":"Patch Set 5: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ca1e1cc850ca4cf08de15b2b8401cc42 : SUCCESS in 51m 52s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7f42ea83971545468244ce5a3bfdd0d6 : SUCCESS in 1h 18m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a320cc0018f841a2a7ad9cccb93c3f77 : SUCCESS in 1h 00m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/48e810ff60324fcf86520e4f2a45d169 : SUCCESS in 1h 26m 16s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/29b8dc9507fc4e579958fb4cea63d58e : SUCCESS in 50m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cc171e1670cc4627b0f49307e71bca5e : FAILURE in 34m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/842eb9a54bb843e09cf9d4386f05a2dc : SUCCESS in 49m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91811373adc4471da9cf7c559b8a0b15 : SUCCESS in 47m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63db330005de4e03b3c890b0d5e69f3b : FAILURE in 8m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b51ec7f4a8d41acbdb507efdf6882d4 : SUCCESS in 29m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80e8d98f9b0741d79160f4908d3f7c27 : SUCCESS in 32m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad82110dfc79433e8521d05a6c816bac : SUCCESS in 11m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f5d739d3bb7d42b996de02f2e3a61529 : SUCCESS in 50m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/552412c5779f48cb88638920f234da55 : SUCCESS in 1h 39m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2a549febf1f64870a62ee7dd1bc9e154 : SUCCESS in 1h 14m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e0009fadff7c480e95241eecd9bb0dfa : SUCCESS in 1h 05m 59s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bc89905131e1462897cc4cdf9fcf69cf : SUCCESS in 1h 06m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/77166ae78d044f2a937eb762c9faf363 : SUCCESS in 1h 28m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ea89efbe23494565819fbcbc0c2d9cb8 : SUCCESS in 1h 16m 47s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c23cc7fd1f064aa3ae909e2aec4fcb69 : SUCCESS in 1h 05m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/04d73d5702d94480b39d36954758139a : SUCCESS in 1h 05m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d80ed9328ee13bfece78d86ea5b3f96ec86668b2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-11 20:19:41.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/47/833247/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"b686a7a563d9c38a88e99599164fdf79cce17553","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-11 20:20:08.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/833247/5 : FAILURE in 1h 54m 10s","accounts_in_message":[],"_revision_number":5},{"id":"1d8b48d09d16bd48987f77b91f6a8114a0ce3423","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-03-11 21:22:45.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/47/833247/5/openstack-sriov-neutron-master-direct : SUCCESS\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/47/833247/5/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"a9d98c340c16863ca99e13958ed9c2658a1041af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 21:26:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cf1fc00ed5814d34b56b25a33d58bcf6 : FAILURE in 1h 20m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/abc48dfea03c4c84b6fd6c72fb7daae8 : FAILURE in 1h 20m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"83e518b9b9dd5b4621c240102a0c7c9c1587dd53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-12 11:59:46.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"785044d4c58cce5d17aeb97e436a5b205140c0b4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-03-12 12:35:59.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/47/833247/6/openstack-sriov-neutron-master-direct : SUCCESS\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/47/833247/6/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"6ff21543f10463bac57c2f8bff3554b75c521dbf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-12 12:55:42.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/47/833247/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"828e660059a2e4f0a8e42fdb2171744d8a838d22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-12 13:24:25.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8da3ac6c12f4a1fb383b53836a79667 : TIMED_OUT in 1h 22m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6fd6bf7c73ec43a588f5885b91693ec5 : TIMED_OUT in 1h 22m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"86c673c50e410cc74c47b6a3d4630fe318b8fc7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-12 13:45:41.000000000","message":"Patch Set 6: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0d27d96d398b4028ac6b3e7b8a08e4de : FAILURE in 37m 27s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/15bcbba94a6f4adf9587a435209177be : SUCCESS in 1h 10m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7fc0bd801035440da9684dceed695f6e : SUCCESS in 1h 00m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/716ba5a22a6243aeb2064d65f6386534 : SUCCESS in 54m 30s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c57979612ab943d98bd38cde5cc25ef2 : SUCCESS in 53m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/353c77e5c6cd4fdaa71e27041c01771b : SUCCESS in 30m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01bbdba598524cf38a31b32ea179198a : FAILURE in 27m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a53e933aa9054a248e9ce66666b45200 : FAILURE in 27m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32fb010708b84d7fa4f79aab7d70d072 : FAILURE in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb92fef5729f4e4d8b05718f0c000b25 : FAILURE in 22m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9baec1513f05440f843c2b0ca65202b7 : FAILURE in 31m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6863799340b64708b09749e3696393ae : RETRY_LIMIT in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/606bd527902a40b6a47181bb55d5386f : SUCCESS in 9m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61ad20dcfab74a619cf1f95171fd9c1e : FAILURE in 25m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eac81104a173456b85721d46a06a3374 : SUCCESS in 46m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/13a858afc75446f8b4d6e0131911319b : SUCCESS in 1h 44m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/529a3d936726449badb0b5253d721ce0 : SUCCESS in 1h 21m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6b30134faf23468b9e89adeb4b87b0fb : SUCCESS in 59m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/475969ac079c45b19a2a4b0d925aa15b : SUCCESS in 1h 04m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9f00a10a409d40709b657fd8bd04f333 : SUCCESS in 1h 15m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a1140fa8f9fa43f79927afb429d766e2 : SUCCESS in 1h 19m 56s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a7ce483aa664963994b70f3ce8cce59 : SUCCESS in 53m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/351a57dfbc8e4b4dba12633ea1731794 : SUCCESS in 1h 03m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3265dc12c2fbfdc9bcfe322cfc502f829e46f8a6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-12 13:50:56.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/833247/6 : SUCCESS in 1h 50m 26s","accounts_in_message":[],"_revision_number":6},{"id":"b47cacbf5cad79b63e42d43c470d8eef1535843c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-14 17:50:39.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5b1ac5410cfc8767a468e015f52c77f752d9b6c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 19:18:41.000000000","message":"Patch Set 7: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cd9e311be39b4c83a7fc703dc8531603 : FAILURE in 39m 36s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d318df930cb54ce0bf5f28291b3b5eee : SUCCESS in 1h 18m 17s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b00d5a10d6964f8788cfc3837d8951ba : SUCCESS in 1h 06m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/67ff796b55e14d94a47aaeb430310307 : SUCCESS in 58m 47s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/1953df074ae6404596d897558ad22782 : SUCCESS in 49m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5179594f4f6a41a7bb62ba1900fe3af5 : SUCCESS in 34m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/544d76cf6d604577a697c04fb0d9a110 : FAILURE in 1h 01m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa261001e4804677a8dd9df7bdd931b6 : FAILURE in 28m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53711782164b4079a5432d703c6c322d : FAILURE in 8m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f8bc3e0d26546ccb0887fb59f6775a6 : FAILURE in 26m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/408ab50267b44bd3996db429feb2ca63 : FAILURE in 23m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4637db8fd90240a18ce0f7ac177264b0 : RETRY_LIMIT in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96525ca443bb44369cbce794a2ea5f8b : SUCCESS in 11m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8c4dface1e945cf8ad70d41bb8de7e3 : FAILURE in 32m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/28370f38ec17439a884a46127cac67ea : FAILURE in 55m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c267d94202e5482c8717b07ab3c7fdad : SUCCESS in 1h 18m 29s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d2b6e80d474b47a58dc42c034f9ba2ae : SUCCESS in 1h 16m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b552607c8b9f4bef955c59b36317c0fd : SUCCESS in 53m 18s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5644eb71607b421499afc40eb9adcdd7 : SUCCESS in 47m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d0fbaa15506a4b2eaae8848ebcb3fd9a : SUCCESS in 1h 19m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/16345a2bb3274728a691e288d04e6dfb : SUCCESS in 1h 19m 06s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1236b31d00374844b0d0957fac723b26 : SUCCESS in 45m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7a523a1451084cc095d2042e1eb06eaf : SUCCESS in 1h 05m 02s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b8533b041e61b6547d833170a2b67a7c6e36341d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-03-14 20:19:09.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/47/833247/7/openstack-sriov-neutron-master-direct : SUCCESS\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/47/833247/7/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"2bc277d12ee41c7709cb1029d5993c628c159e23","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-14 20:20:27.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/833247/7 : FAILURE in 2h 28m 51s","accounts_in_message":[],"_revision_number":7},{"id":"4ee813d5389537aba7c6838b9b8034cc26c8bb03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 21:00:29.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ebae92bae5647c1954fb04928f08928 : TIMED_OUT in 1h 23m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/78b33317ca2f4015a94c9b1ecafb9ee7 : TIMED_OUT in 1h 23m 02s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"166e516ca525f4a95dbd734b9d99cb43043aea92","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-03-14 21:44:15.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/47/833247/7/openstack-sriov-neutron-master-direct : SUCCESS\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/47/833247/7/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"31688d6cfd979c3e5da52bf76154dbdd21d81ccd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-15 04:39:53.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/47/833247/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"ff33fa0b71fb463d15eb352f1788c8121792d7a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-15 11:09:11.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"cbad45557ebd42f7e3d4579d8d223d7b36b6f309","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 12:35:05.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d9899ee98674c5facf74de2adbab0fc : TIMED_OUT in 1h 23m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2b4149c1bee54efa88c69ea1a9079eb9 : TIMED_OUT in 1h 23m 34s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e927061d56688c57d50a4ce502e88863b7f43d8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 12:52:16.000000000","message":"Patch Set 8: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/aabac918583540168e09853330cf80c9 : SUCCESS in 39m 22s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6414d836c5c247b08996de37cf66c585 : SUCCESS in 1h 04m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1c88efe98ba740fb9fbbeba2145f4a8a : SUCCESS in 1h 00m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/98d443095a144efea70423ee02f7477d : SUCCESS in 1h 13m 15s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/01ad60ef55384788ae6489246f3abd57 : SUCCESS in 1h 11m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c3be551735404d53975c65fa98ee7606 : SUCCESS in 33m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63f3dae5c3cd4fc48a85b2fddff6c57d : POST_FAILURE in 37m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f432954b0414797995156ccf043243b : FAILURE in 36m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1952aa426fff4fe487414457f2917286 : FAILURE in 8m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99ca0cce6e6146d6b9127cd4b612db23 : FAILURE in 27m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fd85aad96704bec90db4e653158cf55 : FAILURE in 26m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/856d509d82a24a7590699924d90284de : FAILURE in 31m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eda6ef8c2614851bfea349a39d8db9c : SUCCESS in 11m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3474f035e6a3414d8786eaf051ef7d92 : FAILURE in 30m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bb03508613664e66a73f40c28e527abf : FAILURE in 54m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bfe655b22f0d47f3aa18421057364a21 : SUCCESS in 1h 37m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7e8f8cca06194a6a822bbd51eb9fe654 : SUCCESS in 1h 11m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b0bbc35806ac436f80c8f77b7bf7eafc : SUCCESS in 57m 15s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c91e6eba67554e8498232f890f2f0a6c : FAILURE in 1h 00m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/52032e61ddc34920b6e992f3ac93f01d : FAILURE in 1h 00m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8c5f174b3c5a4204a9596ea6de86e941 : POST_FAILURE in 1h 00m 22s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb4aa555749842d1870c19934c08da9d : SUCCESS in 1h 00m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/769d2f6e15e04e5c9911934c69d9625e : SUCCESS in 1h 09m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fd2d0e05771a451279eca59acf1dcbbeb4105d59","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-15 13:29:39.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/833247/8 : FAILURE in 2h 19m 33s","accounts_in_message":[],"_revision_number":8},{"id":"5719e882d7b48a0e3bcf8ecf27c35d89aa975a0c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-15 16:37:33.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/47/833247/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"9cb37ac11775e91558f79caa4f35639750e85ec4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-16 11:21:20.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"55f643958e1ef38a7e95ee4c16338fe2195b0054","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-16 11:55:54.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/47/833247/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"d6adaf87f402633c797fc3b0031b0549d9f36651","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 12:22:28.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/291156a1fd304d50a8ceea8af2300ab8 : SUCCESS in 58m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ea0d16996834cdf8d73ed66bd9cd1ae : SUCCESS in 59m 00s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"64e175a839fbaed9c04205cff261b43384412150","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-16 13:11:31.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/833247/9 : SUCCESS in 1h 49m 25s","accounts_in_message":[],"_revision_number":9},{"id":"26c9e73337babe5516a0039eaf1022f1aa870c0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 13:13:11.000000000","message":"Patch Set 9: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bee06cfa1005448eabde5d169d1c47dd : SUCCESS in 40m 42s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/54ed8866775b40bfa1ffaf2113288bcd : SUCCESS in 1h 11m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/760deb2457734f04a86dfdbe2e6e5b82 : SUCCESS in 1h 13m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/043e0aa70b374d969e131a89ef4bb02e : SUCCESS in 1h 05m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ae287914454c4bb2a018bb05f51d233c : SUCCESS in 49m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9bb1f919e88f44959dafeb712e3e4b66 : SUCCESS in 38m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5608d618d054447db145338b453219ba : SUCCESS in 46m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e39332eebe1740cfb079ec9f7600886d : FAILURE in 36m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58eeb999cfa4480492252372bf7a49d2 : FAILURE in 8m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd219d1d268a455daee594953a90b6cc : SUCCESS in 27m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f41ffc05bfc4d84b5845e1ba8193c2f : SUCCESS in 28m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b3d5feea19e942ae9b28aa8e1b258898 : SUCCESS in 32m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d50bc3619fb64ee6af2f14753a6ba53a : SUCCESS in 10m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6eeac793a9904bdf8d1799fb44b84dc7 : SUCCESS in 1h 03m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c968d355faf6454f9fdda49e17cdc66d : SUCCESS in 1h 41m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/86d5484835d349ffbad6fcd6bf9cc8ff : SUCCESS in 1h 28m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e7c4e52336a4a1ca2d721ed430f3b89 : SUCCESS in 59m 39s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ed6df34b1fd34603842b80958b2856d0 : SUCCESS in 1h 06m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/32e88926dbb646689ee609eccebaa285 : SUCCESS in 1h 22m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/784e57a48c724a4aabf0c58f3a4ef63f : SUCCESS in 1h 30m 26s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fef1f886bfba46d5b581f0f1b6d6e205 : SUCCESS in 1h 05m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1474f6db968c454c943a1c8d27feae97 : SUCCESS in 1h 21m 37s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e91b57479da4e463b9e28af298e68a7a5d737b25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-16 14:00:15.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"f843f73f8126a67fc706e4588e2e8aaf4f0801d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 15:15:12.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc926a55a5284e4b9e43266e67c879ff : SUCCESS in 43m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9df4cec2b2a541599000064cd6f241e0 : SUCCESS in 1h 08m 45s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5f2ef0e4dde77683254559f564974e917021e56d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 15:29:14.000000000","message":"Patch Set 10: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e034830d24d046ab83289eff592461e1 : SUCCESS in 53m 03s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/87741b5665f34c73b5606ca2676f645b : SUCCESS in 1h 02m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/66be6dd9360c44688de7b052bba4429b : SUCCESS in 1h 00m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2c0770347c0f4d3fb7088fc268da7eaf : SUCCESS in 52m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/59a7c3752c9c49129be70584a74e91c6 : SUCCESS in 48m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cd3625fbaba7477a82d26a1ea320f691 : SUCCESS in 29m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6181a8ff5a804d6b94ceec94596bb5c3 : SUCCESS in 44m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/314b905d5b154caaa01b6e57a0514403 : FAILURE in 36m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f494e68f57234ec2b12a4e90a9392faf : FAILURE in 8m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/365d2ed8006640b0af9a37e578b83977 : SUCCESS in 26m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c911ab6c7e947fca7da66b945364719 : SUCCESS in 27m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5848560d0fa4efaa018a6bfd0c4d5b5 : SUCCESS in 35m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86f47f2a4f234232b0ceff38a91ca533 : SUCCESS in 10m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/758ff2d3e1274e0d9f4362d325cb03c8 : SUCCESS in 55m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0c47234631724016b0aaacc8fe4870e6 : SUCCESS in 1h 25m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/47e7e5d5ee9748c480209c675f43d0d3 : SUCCESS in 1h 02m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/598e653131d44a0096ccc94953f29252 : SUCCESS in 1h 04m 20s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9673b5766f9741e8a62b1e71afa17f1d : SUCCESS in 53m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4c9840a2b3ee4f42a0a7d359bd336120 : SUCCESS in 1h 11m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9cbbf361b77542eb93714c5065717996 : SUCCESS in 1h 04m 22s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3f905dc294c4abd81865cdd96bf78c0 : SUCCESS in 55m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b2716631c2464e44acc55c044e9b64be : SUCCESS in 1h 07m 19s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"716e07460c573de991c6c72cf80b47349a4dd235","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-16 15:35:36.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"c29a418ffca365daa77a29f829c7bf94bd5f0a8b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-16 16:16:14.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/47/833247/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"c667f1b8a61c6dc3665acab033535e2c359cfb35","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-16 17:07:48.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/47/833247/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"0720336ac1824d0c58e2b4c702a7e984dad4e36c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 17:12:05.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/24e58fedfae84898825e448ce420b572 : SUCCESS in 1h 02m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e1f872d245cf4027883be67d00cae04c : SUCCESS in 1h 04m 40s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c26a0b03eca0aafc4d1ac7cb395f029c679b8a63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 17:15:03.000000000","message":"Patch Set 11: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/49598e515de8473398cd73c541a32978 : SUCCESS in 44m 10s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/515c439c3d7e4229980d90a642446884 : FAILURE in 26m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5d7398f633fb47e3a38ea8a9ce454448 : FAILURE in 40m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/91be0947f62f4f3eb1ca0b3c3590e94f : FAILURE in 38m 29s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b58b0e42a8d54a39b9f118b382d6bd22 : SUCCESS in 1h 06m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/434aafe382a04128bfc354cb923368ec : SUCCESS in 51m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/276938e9d74749e08ce42ee5afad3c46 : SUCCESS in 34m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28b3afce155640ec98c1f458aae5de45 : FAILURE in 28m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/378cc77050764348aca1247592e3ba6c : FAILURE in 8m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b6cb6215dd644b0bcc03d0662991349 : SUCCESS in 24m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58590327b21a4bb0a4472cc9b3bd5b86 : SUCCESS in 25m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f522417fd9fa495ca258b06c25b949bc : SUCCESS in 34m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28016dd8079443c0b9c708a06f1754f4 : SUCCESS in 11m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/286ae2aa2cc24867ac177def7b7ab28b : SUCCESS in 52m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0838ce086f874152bbb5ad0176ef70be : SUCCESS in 1h 36m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d7687d7655a742178c9060ea3bb5b27b : FAILURE in 25m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6597d946c5904848bfddbc8b677b657b : SUCCESS in 52m 23s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ff1789d395e148d1b95a7a99161a3e94 : SUCCESS in 57m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9d0bab5d1ad443f3851fbeed1220f8ea : FAILURE in 24m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0a72f952faaa4fb8a6c4c2351c468d9d : FAILURE in 27m 30s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2b146e106184faabbefcfc382f59451 : SUCCESS in 55m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4cb627cad13b40beaa5088618b75f025 : SUCCESS in 1h 08m 32s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"704a466d5962f8a4d10f7d41b9bd14cdd7dcc6a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-16 17:44:46.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"b24b45d1b3f95d191ade869d1c58b533fcba539f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 19:18:01.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/643b845f688e482a9138e7ee9c33767e : SUCCESS in 59m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75a209fb6aa54e99a671be493e360efb : SUCCESS in 1h 01m 19s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"c130d47abcb7ac859f7c54b5202299f5ef485bdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 19:31:29.000000000","message":"Patch Set 12: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9bd7de0428cc4e1194f561e8bd298e88 : SUCCESS in 51m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2e9e577ffa564d93b7c9cad960005184 : SUCCESS in 1h 05m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3b865d7cfca14bd5993b99b94c060b51 : SUCCESS in 1h 22m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f787b9dee96440b2a850c9d5be7f8ec6 : SUCCESS in 46m 09s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8b22314148b74661a1a4a2f8beea4dfe : SUCCESS in 1h 15m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2441bf885f4944b8a64f172b6b064ede : SUCCESS in 39m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/daab8eb3d6c74d11b8cbdbe361a37cc6 : SUCCESS in 40m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6443e373565429889c97ab16abecc3a : FAILURE in 1h 02m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb307f78f2624a5ab45fdd8cfc2529e8 : FAILURE in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2295a3d5fec4478893170f55876d909 : SUCCESS in 28m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97156fe5bde24be39b3d4be5b114d27d : SUCCESS in 29m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c45dc9e20f64b4b837bf634d07cd8b1 : SUCCESS in 36m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a583dfaec6a94871945f3434e21600df : SUCCESS in 11m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4695e51e02dc4564aaa663b6d9fc018e : SUCCESS in 52m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fe619acd35864b929b45f10ee92e0838 : SUCCESS in 1h 39m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7ee961affb1242189b62305b5d602fc1 : SUCCESS in 1h 11m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/77dbfaf0e4cc404cbd487c7ec2dc411d : SUCCESS in 1h 03m 14s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fccc0ae298ac42b7b27ae07565ae065f : SUCCESS in 57m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4141311bb1604ba5a468172fa37991b0 : SUCCESS in 1h 20m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ed78455246ed41a0adf794567bf0e9d9 : SUCCESS in 1h 14m 40s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5514a97ae87845c39dbc4e77fb3f348c : SUCCESS in 58m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bb65c49fca994d6abfffc12f34965fba : SUCCESS in 1h 21m 22s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"7b6788a7b45ed1b9b074e06aafe402b6e054e6e7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-16 19:46:04.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/47/833247/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"896e1ad041ed012f72ddf64faa112a153546bd4f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-16 21:01:30.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/833247/12 : FAILURE in 3h 15m 47s","accounts_in_message":[],"_revision_number":12},{"id":"e198e15b448c5edf6eee916d7097c1b18b1091b5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-22 08:41:29.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"7a8863a8dba93057677923a2773de50b94bcb6bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-22 12:33:22.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"3a0d7a9812ee2908cc08aa0720d8ec148c86e073","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-22 12:34:19.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"bfed9cd73973c5ca5ad8a2bbbf300385b1aba7f1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-22 13:17:52.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/47/833247/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"1def72bc3ebf5d0c203352e951bfe7d6699caef6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 14:06:21.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c04512ec79614d5188d62222a525b57e : SUCCESS in 1h 02m 10s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/598c064686e24b6daecf5f70b12d2bc8 : SUCCESS in 1h 26m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6d609d95c88e4dedadb6b972e6e38660 : SUCCESS in 55m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bb675f4f23bb43f9ba4e4de15faf109e : SUCCESS in 49m 04s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/52f3b6e7a922483990046eb610d46e9c : SUCCESS in 58m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e35338984ff94694b1e80faa0ed1bd5d : SUCCESS in 27m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d92d987789f4ef78f456e945aece55e : SUCCESS in 39m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba1f8482a20e42d29971b680a5dd2a6d : SUCCESS in 36m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9be99ac04c074e52a7facaaf5e8cfd69 : SUCCESS in 20m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b13aff856ff45a3aa30580cf8f65469 : SUCCESS in 26m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/520c80c2bea54adbb474be9e352216f0 : SUCCESS in 28m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06935d8f228a4c76ad81abc89a47a303 : SUCCESS in 28m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fef57ba897a844ab978a03b946278161 : SUCCESS in 11m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/704fea8d1b6d4cf6b2b611f9739cad2d : FAILURE in 1h 02m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/09ea3bcbfac24b79894346a8201e2ec1 : SUCCESS in 1h 30m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8ce449b4e3ee43259b8ccaa52b7f6b5f : SUCCESS in 1h 08m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/10c3f9b842ff4c049d12495830dbc442 : SUCCESS in 1h 00m 32s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6a2b36837f16413b8814de119f904fcb : SUCCESS in 58m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d3d0a040fb8944ec81c212fa97e3cf20 : SUCCESS in 1h 13m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7dec98c328ae4b83b1092910db48f4bf : SUCCESS in 1h 20m 48s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6519c54bd23b434e826046a3a0f157e8 : SUCCESS in 56m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/87dbe0e79b4d43f9a195ed6abc1743b8 : SUCCESS in 1h 12m 26s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"063653a6eed337a9aa3cd12ea1d4f7d97a37b9b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 14:20:48.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c85be46f48644b5789dfe320925a0d43 : TIMED_OUT in 1h 23m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/17ada8b0ac284f04895ced887edc79b7 : TIMED_OUT in 1h 23m 01s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"bde3e51edaed486e7dfefcc3152f4e005800f0d5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-22 14:34:36.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/833247/13 : SUCCESS in 2h 00m 29s","accounts_in_message":[],"_revision_number":13},{"id":"7d80ad1b28ead69b33ea6a05aab0ce694d873074","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-22 15:37:28.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"4205d4f309ff0f0bc1b0d1dc90ca40c7a473f2af","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-22 16:23:02.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/47/833247/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"06f9d4bca3b2a4234b9778b604df1c006e03f812","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 16:58:18.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce816d0645a74bdeae5d08f9d821cc9d : SUCCESS in 48m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f4762ce2df9f4b90a62e6970bc225bfb : FAILURE in 1h 16m 37s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"c5e79e046fed28cbd310a6435cdd4bf29f68be7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 17:19:34.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/869255c6897d4fcabb528c7a06387ce2 : SUCCESS in 53m 03s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/dcf6acda07104b28a08a5f44a5d460fd : SUCCESS in 58m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/42bd249b3edf46c6aed1f82f1e21bd2f : SUCCESS in 1h 14m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0d6a6dcd32d54e5f957a1ebc9ab054ea : SUCCESS in 50m 36s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/de4e591c5bbb4d6aa413962e64e9d040 : SUCCESS in 48m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1b4e334802b24ced866f415f0b0fc683 : SUCCESS in 29m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b29ec53555f40f69a7ee57aebed7166 : SUCCESS in 40m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b070c300e21c423999d13109fb56ae56 : SUCCESS in 33m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f763f90b628495d8f59850982be8beb : SUCCESS in 16m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3540b982df0c4678a98544e857efeb51 : SUCCESS in 30m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90c7245584e141ebab8d994798b45040 : SUCCESS in 25m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93165c5c01d14d499c21555ff915b003 : SUCCESS in 31m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bef7ef47eed4d52a39d1f5c31a5ed2a : SUCCESS in 11m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/89838a1327ee4473b7ef0d76486f3d3c : SUCCESS in 48m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d7b7f54c5ad74e75878868b526522668 : SUCCESS in 1h 38m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/daaeb8a0bc57457f8a7497e1ba534740 : SUCCESS in 1h 07m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f893b4944e741fc9b7cf77343eb8a5d : SUCCESS in 55m 08s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/41b4ab286ff74c8a9e9d7c928230e036 : SUCCESS in 53m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8b5974930ac7437ca00773b6508f638d : SUCCESS in 1h 13m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/95e13acb664c45c9a124705f89dd04e8 : SUCCESS in 1h 13m 39s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df6d910ee7d44726af31bfbb05173c9d : SUCCESS in 57m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0b51b3d660a34551b060461c1f16eb8a : SUCCESS in 1h 07m 49s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"f3630300fce850f1f72c29a5d49d266503c1bd37","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-22 17:46:31.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/833247/14 : SUCCESS in 2h 08m 15s","accounts_in_message":[],"_revision_number":14},{"id":"589dcd069851e3b0c02ea70f762285631e943ed6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-22 17:58:16.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"e90dc1b227f7d7c3bb03285829a74636553baaa7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-22 18:48:04.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/47/833247/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"97a9ccab2b78f0c488629fc5698c9ff4b5031949","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 19:27:59.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c677365d16e429f90d16823a3bfee21 : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a7bad93730b47e38aefad097e2eb23e : SUCCESS in 50m 33s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"13c7ea42d118b621445e94d21123d7e18f55b662","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 20:19:20.000000000","message":"Patch Set 15: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/87074ba993ad4c559f48bf01b6804a37 : SUCCESS in 52m 27s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/003562c20c254731b1b596454368d70f : SUCCESS in 1h 08m 17s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/af30a68cddfd4f0e994ef10f754ded92 : SUCCESS in 2h 13m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e697a774f41642418d93f4f9276e307e : SUCCESS in 1h 09m 49s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/848b31d4801a453fb46c1d4c1d8f907d : SUCCESS in 1h 54m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ccfaf32863394260896aec0352062345 : SUCCESS in 39m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b81b1369b0e45b3ab65b444a94407a4 : SUCCESS in 39m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6f8b8c148514bc78951d144202b7757 : POST_FAILURE in 34m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80c312a51b5f435293d0e4bd51624e0b : FAILURE in 9m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e7b73855bad429a826f4304afa19f6d : SUCCESS in 29m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97d0102fe59545e798fbb111c6a445bb : SUCCESS in 24m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/421b21b501d84f82ae890294a1eb5175 : SUCCESS in 30m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6ff2df9d4994d9fabd91d8e895c3c2d : SUCCESS in 10m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/79c881454f3d4a75a8fb4c58cabcf1e1 : SUCCESS in 56m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/584d8cac7b39431c85d8d946096cd691 : SUCCESS in 1h 31m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1f6c5d8a5a8648388d997892c46f3f80 : SUCCESS in 1h 05m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fdf92e44d8af42829765adc1076f97fc : SUCCESS in 1h 11m 36s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1f0ceee9a7b4475a8ab8941ef3988977 : SUCCESS in 58m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/386f865b1a11402db937039e8e11a89e : SUCCESS in 1h 30m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f2b9c019b7054a10b12d2181702ed44e : SUCCESS in 1h 15m 07s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fdb77df8b13a425eac2c986b3a2ecd68 : SUCCESS in 55m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fa992b39268d4755ba5bed71b51cd25f : SUCCESS in 1h 31m 05s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"30ef2351deb1a33301035b0395b1a8a6514b8d4a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-22 20:24:55.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/833247/15 : SUCCESS in 2h 25m 50s","accounts_in_message":[],"_revision_number":15},{"id":"18683fcfcf67bc7825f431edf53267979b1fed17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-23 09:15:09.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"4b50c2f4269efa1b5d99b88d55d625dda4ca2298","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 10:18:48.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ab3f92c68094ca292e44b504753842d : SUCCESS in 1h 01m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ab14229d8bcf438cb523b3853e7479ef : SUCCESS in 59m 10s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"1eed3e58029bd9fe4bdcad5454b0eae8017ebde1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-23 10:52:52.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/833247/16 : SUCCESS in 1h 36m 54s","accounts_in_message":[],"_revision_number":16},{"id":"bbd0cf0b832e1c5c0a9f6b9873712939bb1d26de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 10:56:34.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/737bc82f9fa841a295254d43c86c62ef : SUCCESS in 48m 41s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d727e7c7cd7040e0b2b55f8d819a45b8 : FAILURE in 1h 15m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f98d853ac1624bb4b2d0c78f279414f5 : SUCCESS in 1h 29m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ba09b435b3724198bee66140d61693bb : SUCCESS in 1h 21m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/42cb3ffc22c74b93969a4cd2689f4a99 : SUCCESS in 1h 07m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/25dc4d5858ce42db9faf418bf8281c06 : SUCCESS in 42m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/077b94253c654ccda5cce340d7963e90 : SUCCESS in 36m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d39d99522df744cd867504b1ba972a9a : SUCCESS in 38m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30205e52feb34ff39fe2f1ea877e9295 : SUCCESS in 15m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a55f081a3524426a0868a77e6fa5aed : SUCCESS in 33m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0d52e42891a4a408a592530d3a0f914 : SUCCESS in 27m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/124a0085bf6a4acd957ff07c201e134b : SUCCESS in 32m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b76aa7eaf3544dfd8dc2465ba7e7d4cb : SUCCESS in 10m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e1aa9acc5b8e4ba9a05682df24d6d235 : SUCCESS in 59m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/522135b47c234dc4a3aaa3b619512a0e : SUCCESS in 1h 35m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f0fb39207cc0420a892327423a7da7be : SUCCESS in 1h 11m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/633dec7be1484a128ac9400ba3667e78 : FAILURE in 49m 09s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2efff392ff8b44459abd169cc1126bc7 : SUCCESS in 58m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/163f4751b8184ee4bd9c6c332ca72ec2 : SUCCESS in 1h 14m 46s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f431ca9d5c344d25acd018dffbdb1549 : SUCCESS in 1h 14m 35s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5561840e0d36439993cb2a72f5c3daca : SUCCESS in 50m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ac96055c9f56461a8f43d1ebe98d8bbd : SUCCESS in 1h 16m 22s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"dbe0fbd645eb36ec7cfd04c1540e5c9fedcb90ab","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-23 10:57:44.000000000","message":"Patch Set 16: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/47/833247/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"8f273147929fc2c65269e344281bfffe0de5016f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-23 15:29:44.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"4194871f2d72b1c084ea5412362724a282beff97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 16:38:36.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2edab536d0c4d8aab27a94737d4a5eb : SUCCESS in 1h 01m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4a995effa30e4ef6a9b51049042dcd76 : SUCCESS in 1h 01m 22s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"0605819cf9a69089d5e496399ff2cf229efe6232","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 17:07:26.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/824daaf3947b4f7f8f03a24a7de84345 : SUCCESS in 48m 37s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/785c9c6a32674a6a8e8e1aac438ad0d5 : SUCCESS in 1h 00m 35s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/93fb381af31047ba8781a02f82555447 : SUCCESS in 1h 00m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/00ea4bd559504a578c9d12d2f9aa413b : SUCCESS in 48m 47s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f3b58744b23747d0bf9e2d0ecdb75012 : SUCCESS in 50m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/654fd62760a9474c9ef5d74fe6f1659b : SUCCESS in 31m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00daa41b93d744699177e43610e084f8 : SUCCESS in 40m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc1c68e437574eac850d9b2212fb8fb7 : SUCCESS in 37m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a429b210b8444e7b633b5e8d52857ce : SUCCESS in 17m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c96631caac94392a5b15adc7ca19cac : SUCCESS in 25m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46f5c1ef62df4a7098a8633cd623e2f4 : SUCCESS in 27m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/552edde6e9f54b068207c0e8f932d46b : SUCCESS in 30m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2be500d4e3c34afa8c1ecad834c81720 : SUCCESS in 10m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e89ba84bbbbd4f2a80da97a151dd36f0 : POST_FAILURE in 52m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ad4d7cc04d8a45df96eccb083fbb4003 : SUCCESS in 1h 31m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a31f9a276217419784cdb15dba12fa8a : SUCCESS in 1h 10m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/be6921cccbae447787eb25ddf6b38558 : SUCCESS in 58m 27s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b1268d0c943d4a668c691735202095b4 : SUCCESS in 1h 00m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8be9685429fc43fe85b6acb734674f17 : SUCCESS in 1h 17m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/98e962fe44894a7f90760296e0dd8022 : SUCCESS in 1h 15m 08s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae4c54d6ac5d4869959f0212873a6e85 : SUCCESS in 1h 02m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f37c07b6e4a94f0e937a8852141e40d2 : SUCCESS in 1h 13m 25s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"21adcb29829364412c1813e10be7ba341b53bedd","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-23 17:37:19.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/833247/16 : SUCCESS in 2h 06m 59s","accounts_in_message":[],"_revision_number":16},{"id":"25c762bc9680e5b4cf8b5e2ef0189e7f20b26c7c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-24 09:10:59.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"ad8708d1da98f6b9a8f184d864384c70dd276301","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 10:03:08.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a802325720a4a2496a54691d59cffab : SUCCESS in 50m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6fb75b180b0540e5bf8a399dbdc15b3b : SUCCESS in 42m 37s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"1d6a17b7034ec1cbac785e9779475ced16092636","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-24 10:44:12.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/833247/16 : SUCCESS in 1h 32m 31s","accounts_in_message":[],"_revision_number":16},{"id":"bafc796fa4b548d13e84a4034274f3a1d11e7382","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 10:55:40.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/816eae575a1140b2acc71e2f444c6e9e : SUCCESS in 54m 34s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/83b131d2dbff4db992ba938cb919564f : SUCCESS in 58m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5fbd3fbd8dd74cecbaefc888e99f585f : SUCCESS in 59m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dbb6c9303c404f8995d5c90e9f7152e9 : SUCCESS in 51m 45s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ec437ac9fcd94428a9d225b91b2297f8 : SUCCESS in 58m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/02a7b8f32f72437bb73086b74d3fb83f : SUCCESS in 29m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a92111cd71264eab854be0e17c88a068 : SUCCESS in 36m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/169bfe5c256547c8bdd86a580117f624 : SUCCESS in 40m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f60b8914882945798877feb4ff04dba3 : SUCCESS in 15m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acd7fa8ba1004e32bd4671d02ed7f8e9 : SUCCESS in 28m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a021d3f7a8c45d083acaa291206faff : SUCCESS in 33m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e456151d2c024632b252a8cef041aa2f : SUCCESS in 29m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa97590f0539433384bea7311d9a1ac8 : SUCCESS in 11m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ecf4af49124c43b6b869c8053050ee30 : SUCCESS in 1h 01m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef08310a1a494f80a669c61f2acdf702 : SUCCESS in 1h 32m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d4e7b1be0541453aa57d226c2cde958c : SUCCESS in 1h 18m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb23a48545014d3195a83bcdf9e98f4c : SUCCESS in 1h 01m 43s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/66ea3f6b7c054b08afeed5aa11c26c90 : SUCCESS in 1h 02m 59s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6a88a749848843fab4c23595354ba450 : SUCCESS in 1h 31m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/38a8becad40c4bb69fa7858cade26d23 : SUCCESS in 1h 14m 27s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c88399d9af244f08d43756b6479c7ad : SUCCESS in 1h 14m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/76a045105bee46bf99cfa91cef36a90c : SUCCESS in 1h 36m 47s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"cbe6abd36193af1a93d14ef9348bd536b8f0dcdf","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-04-01 14:00:32.000000000","message":"Patch Set 16: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":16},{"id":"909e83644fa5960614b2f5f4cb17976d7fe31a05","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-04 10:44:13.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"dce1676f90ee677c88537e9ac6354c1863c5f390","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-05 10:00:35.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"132f8f408704da5789ea9f40371df67aeb162058","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-08 07:17:46.000000000","message":"Patch Set 16: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":16},{"id":"2ae0375cbdd184a8948d482a2f8aa5e2a6ac6549","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-08 08:57:49.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"242db191146b5e6dc2f1defc191620fd55d7fb8c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-08 09:05:42.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"e0e41ad0570e20c1a82ad4f98d971bca2ec666aa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-08 09:07:58.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"1f6789a997e689291693d162494f8892d06ebeca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-08 09:10:12.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"1733ced3d3dd8f9bb0f5aa719ebfa176b8812e7d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-08 09:14:26.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"f40766c493ff6a82df1c85c1f99e9e78962b2ee7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-08 09:14:59.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"a55ad0dc1c8917fd131918831ea6b9100f8e28de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 10:03:54.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c0c6a862fef44db8bc3aabe05ae2751a : SUCCESS in 51m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1d3d2bff202146228f5c94c4b8b2f5df : SUCCESS in 51m 36s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"13dda09b3c8ae711ab147b86197c25a079597848","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-04-08 10:18:35.000000000","message":"Patch Set 17: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2022/Arista_Tempest_Testrefs/changes/47/833247/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"75266c6646a5162afb0ee96698c498ac09cab16f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 10:49:51.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d517397df5bc4fc39c5d1e798075fde9 : SUCCESS in 40m 27s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d773454a600a486aaccf63112987ad19 : SUCCESS in 1h 16m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a7f719806eeb4e8983e6fa32a9678d02 : SUCCESS in 1h 23m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1e1f2eb593b749ac93c70a092a03e1b7 : SUCCESS in 51m 13s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/f9b27f6290ff4f4eaab0bcc44c17010b : SUCCESS in 1h 22m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/21d37db431d747019fbca5380c4faffd : SUCCESS in 29m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10200f42bf564577b11c6c3282d8c4cd : SUCCESS in 39m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/254f07c49d684506bce2fbde13c9f5eb : SUCCESS in 29m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cec1c8ed2f574b808599a772d94c91d4 : SUCCESS in 13m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bd890781251474fa136be533745513b : SUCCESS in 21m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/770e2d38e5444b4db6f921ebf15db733 : SUCCESS in 21m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb0412c474394045831ad909b42fb760 : SUCCESS in 22m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7277170a9d5f43f483bf2066d4e7ba06 : SUCCESS in 10m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4676498a03da4c1fb5287457b16f95ff : SUCCESS in 51m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a4d252b319e44a208e286719c434b510 : SUCCESS in 1h 37m 05s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e4ddd495e3034ef595fa82cb6d2c283d : SUCCESS in 1h 04m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/78b642ece36044688b465c7d2c991044 : SUCCESS in 1h 04m 40s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fd1f82457ec642edb3d926a1135fb932 : SUCCESS in 56m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5bdd7d17d89b4426b7d5ba5149e24f4b : SUCCESS in 1h 08m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1a6bbab7338c4e0086e45d25b97e1ef4 : SUCCESS in 1h 10m 16s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0eee6b791f094bf1b915df0561d8f932 : SUCCESS in 54m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4b232d7265a745369ebca5b2ed077fcc : SUCCESS in 1h 06m 07s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"527a63c2e51ebb7d1d1af124f829831936d92995","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-08 11:10:27.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/833247/17 : SUCCESS in 1h 59m 21s","accounts_in_message":[],"_revision_number":17},{"id":"910081250d1d513d1fb15607321c7ca62cf66810","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-04-08 11:10:45.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"b967870e25f5e229c23dd8dcccebde56d70cc2ea","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-04-08 11:15:38.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"a0cdf3bed6fcf687bb58994d22935c043153f2c7","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-04-08 11:23:52.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"8ea04a665987facc336daa5111d9884d971f272b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 11:24:39.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"8a1fbec28d33a50c5292bc2398c5f50810f7087d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 13:08:32.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7c3f3261902440238d0c433b036b1b15 : SUCCESS in 53m 41s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1dded784e8d34c988ffd8d7631198e8c : SUCCESS in 1h 15m 46s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2468d084f60a4d239006877b50f5b670 : SUCCESS in 1h 12m 28s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d08a8ab7ac684502916f7d2517206187 : SUCCESS in 1h 05m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e019f95c299c4fd5ba31b8361485e162 : SUCCESS in 1h 03m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6437bd140b9c44c38f22d8bb65e44a2f : SUCCESS in 31m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac61f9dbeeff48b1bbdf0ca0b3abb56b : SUCCESS in 15m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/990dafbadb6b44f79b87fac48b32b06c : SUCCESS in 28m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e28025db668948f9a34785ce539a8985 : SUCCESS in 37m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8901323ae1d4927a9d9ff334f332a38 : SUCCESS in 9m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1828d37edc1d40c6a453c52f5b5e4297 : SUCCESS in 59m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9cb6230df502464b9702ad41d9b11cfa : SUCCESS in 1h 29m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dc73788bc2c54bdb958d1e92615a6a05 : SUCCESS in 1h 22m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c9af80daacdd4e4f9c480ad70175d792 : SUCCESS in 1h 04m 09s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/43de92fdd19f4d9fb240ce91b3b3fdde : SUCCESS in 46m 22s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abb03b7361bf48bfb2b81a4292722d05 : SUCCESS in 47m 08s","accounts_in_message":[],"_revision_number":17},{"id":"dbf7847e70106d45708064fca0e197f8856fa173","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 13:08:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17},{"id":"4deb785815ffbb05df0a3b90b9913f68f22c6e77","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 13:10:25.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89188e6f54dc476db3ed56f2f464d3a2 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"eeb918e1b9c27d3b5f0747cb856a4a8483f636ad","revisions":{"24d1c5225c9bbc621b134e25ebf69a6bde951f2f":{"kind":"REWORK","_number":1,"created":"2022-03-11 10:13:06.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/833247/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/833247/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/833247/1"}}},"commit":{"parents":[{"commit":"452a3093f62b314d0508bc92eee3e7912f12ecf1","subject":"Merge \"Local IP: use LOCAL_IP_TABLE for back flows if no OVS fw\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/452a3093f62b314d0508bc92eee3e7912f12ecf1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 14:34:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 14:34:40.000000000","tz":0},"subject":"[DNM] [WIP] Refactor the DB contexts","message":"[DNM] [WIP] Refactor the DB contexts\n\nRelated-Bug: #1964575\nChange-Id: I3da37fee205b8d67d10673075b9130147d9eab5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24d1c5225c9bbc621b134e25ebf69a6bde951f2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24d1c5225c9bbc621b134e25ebf69a6bde951f2f"}]},"branch":"refs/heads/master"},"99a88299b707f6d026993fa7666562555b3f8a39":{"kind":"REWORK","_number":2,"created":"2022-03-11 10:47:24.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/833247/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/833247/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/833247/2"}}},"commit":{"parents":[{"commit":"452a3093f62b314d0508bc92eee3e7912f12ecf1","subject":"Merge \"Local IP: use LOCAL_IP_TABLE for back flows if no OVS fw\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/452a3093f62b314d0508bc92eee3e7912f12ecf1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 14:34:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 15:12:08.000000000","tz":0},"subject":"[DNM] [WIP] Refactor the DB contexts","message":"[DNM] [WIP] Refactor the DB contexts\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/833103\n\nRelated-Bug: #1964575\nChange-Id: I3da37fee205b8d67d10673075b9130147d9eab5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99a88299b707f6d026993fa7666562555b3f8a39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99a88299b707f6d026993fa7666562555b3f8a39"}]},"branch":"refs/heads/master"},"c8b63022a2361fceed61355114dc8c09014f28f3":{"kind":"REWORK","_number":3,"created":"2022-03-11 14:01:41.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/833247/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/833247/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/833247/3"}}},"commit":{"parents":[{"commit":"452a3093f62b314d0508bc92eee3e7912f12ecf1","subject":"Merge \"Local IP: use LOCAL_IP_TABLE for back flows if no OVS fw\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/452a3093f62b314d0508bc92eee3e7912f12ecf1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 14:34:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 18:42:42.000000000","tz":0},"subject":"[DNM] [WIP] Refactor the DB contexts","message":"[DNM] [WIP] Refactor the DB contexts\n\nneutron-lib patch:\n  https://review.opendev.org/c/openstack/neutron-lib/+/833103\n\nRelated-Bug: #1964575\nChange-Id: I3da37fee205b8d67d10673075b9130147d9eab5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c8b63022a2361fceed61355114dc8c09014f28f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c8b63022a2361fceed61355114dc8c09014f28f3"}]},"branch":"refs/heads/master"},"8e45bc471f0374524d32e8198a46a96c1d0ff404":{"kind":"REWORK","_number":4,"created":"2022-03-11 16:02:58.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/833247/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/833247/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/833247/4"}}},"commit":{"parents":[{"commit":"452a3093f62b314d0508bc92eee3e7912f12ecf1","subject":"Merge \"Local IP: use LOCAL_IP_TABLE for back flows if no OVS fw\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/452a3093f62b314d0508bc92eee3e7912f12ecf1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 14:34:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 20:54:19.000000000","tz":0},"subject":"[DNM] [WIP] Refactor the DB contexts","message":"[DNM] [WIP] Refactor the DB contexts\n\nneutron-lib patch:\n  https://review.opendev.org/c/openstack/neutron-lib/+/833103\n\nRelated-Bug: #1964575\nChange-Id: I3da37fee205b8d67d10673075b9130147d9eab5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e45bc471f0374524d32e8198a46a96c1d0ff404"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e45bc471f0374524d32e8198a46a96c1d0ff404"}]},"branch":"refs/heads/master"},"5ca8c19fd9c92dd4e0b5ff085e09977f93f47063":{"kind":"REWORK","_number":5,"created":"2022-03-11 18:25:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/833247/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/833247/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/833247/5"}}},"commit":{"parents":[{"commit":"452a3093f62b314d0508bc92eee3e7912f12ecf1","subject":"Merge \"Local IP: use LOCAL_IP_TABLE for back flows if no OVS fw\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/452a3093f62b314d0508bc92eee3e7912f12ecf1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 14:34:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 23:28:30.000000000","tz":0},"subject":"[DNM] [WIP] Refactor the DB contexts","message":"[DNM] [WIP] Refactor the DB contexts\n\nneutron-lib patch:\n  https://review.opendev.org/c/openstack/neutron-lib/+/833103\n\nRelated-Bug: #1964575\nChange-Id: I3da37fee205b8d67d10673075b9130147d9eab5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ca8c19fd9c92dd4e0b5ff085e09977f93f47063"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ca8c19fd9c92dd4e0b5ff085e09977f93f47063"}]},"branch":"refs/heads/master"},"b3a5ffd137d3cae94818538c50d5e87a6218eff9":{"kind":"REWORK","_number":6,"created":"2022-03-12 11:59:46.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/833247/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/833247/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/833247/6"}}},"commit":{"parents":[{"commit":"452a3093f62b314d0508bc92eee3e7912f12ecf1","subject":"Merge \"Local IP: use LOCAL_IP_TABLE for back flows if no OVS fw\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/452a3093f62b314d0508bc92eee3e7912f12ecf1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 14:34:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-06 01:27:12.000000000","tz":0},"subject":"[DNM] [WIP] Refactor the DB contexts","message":"[DNM] [WIP] Refactor the DB contexts\n\nneutron-lib patch:\n  https://review.opendev.org/c/openstack/neutron-lib/+/833103\n\nRelated-Bug: #1964575\nChange-Id: I3da37fee205b8d67d10673075b9130147d9eab5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3a5ffd137d3cae94818538c50d5e87a6218eff9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3a5ffd137d3cae94818538c50d5e87a6218eff9"}]},"branch":"refs/heads/master"},"a919611cb1af33506c609032d300402d55578b9f":{"kind":"REWORK","_number":7,"created":"2022-03-14 17:50:39.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/833247/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/833247/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/833247/7"}}},"commit":{"parents":[{"commit":"452a3093f62b314d0508bc92eee3e7912f12ecf1","subject":"Merge \"Local IP: use LOCAL_IP_TABLE for back flows if no OVS fw\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/452a3093f62b314d0508bc92eee3e7912f12ecf1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 14:34:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-06 12:42:03.000000000","tz":0},"subject":"[DNM] [WIP] Refactor the DB contexts","message":"[DNM] [WIP] Refactor the DB contexts\n\nneutron-lib patch:\n  https://review.opendev.org/c/openstack/neutron-lib/+/833103\n\nRelated-Bug: #1964575\nChange-Id: I3da37fee205b8d67d10673075b9130147d9eab5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a919611cb1af33506c609032d300402d55578b9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a919611cb1af33506c609032d300402d55578b9f"}]},"branch":"refs/heads/master"},"e971d645495dd1730a9bd807b439d7fb0066622d":{"kind":"REWORK","_number":8,"created":"2022-03-15 11:09:11.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/833247/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/833247/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/833247/8"}}},"commit":{"parents":[{"commit":"452a3093f62b314d0508bc92eee3e7912f12ecf1","subject":"Merge \"Local IP: use LOCAL_IP_TABLE for back flows if no OVS fw\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/452a3093f62b314d0508bc92eee3e7912f12ecf1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 14:34:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-15 11:08:46.000000000","tz":0},"subject":"[DNM] [WIP] Refactor the DB contexts","message":"[DNM] [WIP] Refactor the DB contexts\n\nneutron-lib patch:\n  https://review.opendev.org/c/openstack/neutron-lib/+/833103\n\nRelated-Bug: #1964575\nChange-Id: I3da37fee205b8d67d10673075b9130147d9eab5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e971d645495dd1730a9bd807b439d7fb0066622d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e971d645495dd1730a9bd807b439d7fb0066622d"}]},"branch":"refs/heads/master"},"2fa681da139535f9962a7c78575153e8cebff96e":{"kind":"REWORK","_number":9,"created":"2022-03-16 11:21:20.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/833247/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/833247/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/833247/9"}}},"commit":{"parents":[{"commit":"452a3093f62b314d0508bc92eee3e7912f12ecf1","subject":"Merge \"Local IP: use LOCAL_IP_TABLE for back flows if no OVS fw\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/452a3093f62b314d0508bc92eee3e7912f12ecf1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 14:34:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-15 21:45:51.000000000","tz":0},"subject":"[DNM] [WIP] Refactor the DB contexts","message":"[DNM] [WIP] Refactor the DB contexts\n\nneutron-lib patch:\n  https://review.opendev.org/c/openstack/neutron-lib/+/833103\n\nRelated-Bug: #1964575\nChange-Id: I3da37fee205b8d67d10673075b9130147d9eab5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2fa681da139535f9962a7c78575153e8cebff96e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2fa681da139535f9962a7c78575153e8cebff96e"}]},"branch":"refs/heads/master"},"89b083391e1ee2bbab6c818f014143227f751ee3":{"kind":"REWORK","_number":10,"created":"2022-03-16 14:00:15.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/833247/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/833247/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/833247/10"}}},"commit":{"parents":[{"commit":"452a3093f62b314d0508bc92eee3e7912f12ecf1","subject":"Merge \"Local IP: use LOCAL_IP_TABLE for back flows if no OVS fw\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/452a3093f62b314d0508bc92eee3e7912f12ecf1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 14:34:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-16 00:37:51.000000000","tz":0},"subject":"[DNM] [WIP] Refactor the DB contexts","message":"[DNM] [WIP] Refactor the DB contexts\n\nneutron-lib patch:\n  https://review.opendev.org/c/openstack/neutron-lib/+/833103\n\nRelated-Bug: #1964575\nChange-Id: I3da37fee205b8d67d10673075b9130147d9eab5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/89b083391e1ee2bbab6c818f014143227f751ee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/89b083391e1ee2bbab6c818f014143227f751ee3"}]},"branch":"refs/heads/master"},"934f02805181eec02e003f37e09bd3dcf6ce6a01":{"kind":"REWORK","_number":11,"created":"2022-03-16 15:35:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/833247/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/833247/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/833247/11"}}},"commit":{"parents":[{"commit":"452a3093f62b314d0508bc92eee3e7912f12ecf1","subject":"Merge \"Local IP: use LOCAL_IP_TABLE for back flows if no OVS fw\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/452a3093f62b314d0508bc92eee3e7912f12ecf1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 14:34:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-16 02:21:13.000000000","tz":0},"subject":"Add the corresponding DB context to all SQL transactions","message":"Add the corresponding DB context to all SQL transactions\n\nThe goal of this patch is to make the Neutron code compliant\nwith SQLAlchemy 2.0.\n\nAll SQL transactions must be executed inside an explicit\nwriter/reader context. SQLAlchemy no longer will create an\nimplicit transaction if the session has no active transaction.\n\nA warning message, only available in debug mode, is added. When\nan ORM session calls \"do_orm_execute\", if there is no active\ntransaction, a warning message with a traceback will be logged\nto help to debug the regression introduced.\n\nRelated-Bug: #1964575\n\nChange-Id: I3da37fee205b8d67d10673075b9130147d9eab5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/934f02805181eec02e003f37e09bd3dcf6ce6a01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/934f02805181eec02e003f37e09bd3dcf6ce6a01"}]},"branch":"refs/heads/master"},"ba73bd865eee77496d8f4ded346037d67a757cac":{"kind":"REWORK","_number":12,"created":"2022-03-16 17:44:46.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/833247/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/833247/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/833247/12"}}},"commit":{"parents":[{"commit":"452a3093f62b314d0508bc92eee3e7912f12ecf1","subject":"Merge \"Local IP: use LOCAL_IP_TABLE for back flows if no OVS fw\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/452a3093f62b314d0508bc92eee3e7912f12ecf1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 14:34:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-16 04:41:11.000000000","tz":0},"subject":"Add the corresponding DB context to all SQL transactions","message":"Add the corresponding DB context to all SQL transactions\n\nThe goal of this patch is to make the Neutron code compliant\nwith SQLAlchemy 2.0.\n\nAll SQL transactions must be executed inside an explicit\nwriter/reader context. SQLAlchemy no longer will create an\nimplicit transaction if the session has no active transaction.\n\nA warning message, only available in debug mode, is added. When\nan ORM session calls \"do_orm_execute\", if there is no active\ntransaction, a warning message with a traceback will be logged\nto help to debug the regression introduced.\n\nRelated-Bug: #1964575\n\nChange-Id: I3da37fee205b8d67d10673075b9130147d9eab5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba73bd865eee77496d8f4ded346037d67a757cac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba73bd865eee77496d8f4ded346037d67a757cac"}]},"branch":"refs/heads/master"},"5d85aab9ada0277782fa6d8ab7ae0698b1548691":{"kind":"REWORK","_number":13,"created":"2022-03-22 12:33:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/833247/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/833247/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/833247/13"}}},"commit":{"parents":[{"commit":"452a3093f62b314d0508bc92eee3e7912f12ecf1","subject":"Merge \"Local IP: use LOCAL_IP_TABLE for back flows if no OVS fw\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/452a3093f62b314d0508bc92eee3e7912f12ecf1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 14:34:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-16 21:56:32.000000000","tz":0},"subject":"Add the corresponding DB context to all SQL transactions","message":"Add the corresponding DB context to all SQL transactions\n\nThe goal of this patch is to make the Neutron code compliant\nwith SQLAlchemy 2.0.\n\nAll SQL transactions must be executed inside an explicit\nwriter/reader context. SQLAlchemy no longer will create an\nimplicit transaction if the session has no active transaction.\n\nA warning message, only available in debug mode, is added. When\nan ORM session calls \"do_orm_execute\", if there is no active\ntransaction, a warning message with a traceback will be logged\nto help to debug the regression introduced.\n\nRelated-Bug: #1964575\n\nChange-Id: I3da37fee205b8d67d10673075b9130147d9eab5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5d85aab9ada0277782fa6d8ab7ae0698b1548691"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5d85aab9ada0277782fa6d8ab7ae0698b1548691"}]},"branch":"refs/heads/master"},"00d9e4e7474d874f1f089cabb00436b257b22355":{"kind":"REWORK","_number":14,"created":"2022-03-22 15:37:28.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/833247/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/833247/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/833247/14"}}},"commit":{"parents":[{"commit":"452a3093f62b314d0508bc92eee3e7912f12ecf1","subject":"Merge \"Local IP: use LOCAL_IP_TABLE for back flows if no OVS fw\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/452a3093f62b314d0508bc92eee3e7912f12ecf1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 14:34:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-17 01:15:53.000000000","tz":0},"subject":"Add the corresponding DB context to all SQL transactions","message":"Add the corresponding DB context to all SQL transactions\n\nThe goal of this patch is to make the Neutron code compliant\nwith SQLAlchemy 2.0.\n\nAll SQL transactions must be executed inside an explicit\nwriter/reader context. SQLAlchemy no longer will create an\nimplicit transaction if the session has no active transaction.\n\nA warning message, only available in debug mode, is added. When\nan ORM session calls \"do_orm_execute\", if there is no active\ntransaction, a warning message with a traceback will be logged\nto help to debug the regression introduced.\n\nRelated-Bug: #1964575\n\nChange-Id: I3da37fee205b8d67d10673075b9130147d9eab5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00d9e4e7474d874f1f089cabb00436b257b22355"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00d9e4e7474d874f1f089cabb00436b257b22355"}]},"branch":"refs/heads/master"},"d9a84b98686064dd54d53f148e98f96f96093091":{"kind":"REWORK","_number":15,"created":"2022-03-22 17:58:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/833247/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/833247/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/833247/15"}}},"commit":{"parents":[{"commit":"452a3093f62b314d0508bc92eee3e7912f12ecf1","subject":"Merge \"Local IP: use LOCAL_IP_TABLE for back flows if no OVS fw\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/452a3093f62b314d0508bc92eee3e7912f12ecf1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 14:34:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-17 03:48:30.000000000","tz":0},"subject":"Add the corresponding DB context to all SQL transactions","message":"Add the corresponding DB context to all SQL transactions\n\nThe goal of this patch is to make the Neutron code compliant\nwith SQLAlchemy 2.0.\n\nAll SQL transactions must be executed inside an explicit\nwriter/reader context. SQLAlchemy no longer will create an\nimplicit transaction if the session has no active transaction.\n\nA warning message, only available in debug mode, is added. When\nan ORM session calls \"do_orm_execute\", if there is no active\ntransaction, a warning message with a traceback will be logged\nto help to debug the regression introduced.\n\nRelated-Bug: #1964575\n\nChange-Id: I3da37fee205b8d67d10673075b9130147d9eab5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d9a84b98686064dd54d53f148e98f96f96093091"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d9a84b98686064dd54d53f148e98f96f96093091"}]},"branch":"refs/heads/master"},"e4086862de1cadcb1347b65543ddaab159c81d9d":{"kind":"REWORK","_number":16,"created":"2022-03-23 09:15:09.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/833247/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/833247/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/833247/16"}}},"commit":{"parents":[{"commit":"452a3093f62b314d0508bc92eee3e7912f12ecf1","subject":"Merge \"Local IP: use LOCAL_IP_TABLE for back flows if no OVS fw\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/452a3093f62b314d0508bc92eee3e7912f12ecf1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 14:34:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-17 06:11:25.000000000","tz":0},"subject":"Add the corresponding DB context to all SQL transactions","message":"Add the corresponding DB context to all SQL transactions\n\nThe goal of this patch is to make the Neutron code compliant\nwith SQLAlchemy 2.0.\n\nAll SQL transactions must be executed inside an explicit\nwriter/reader context. SQLAlchemy no longer will create an\nimplicit transaction if the session has no active transaction.\n\nA warning message, only available in debug mode, is added. When\nan ORM session calls \"do_orm_execute\", if there is no active\ntransaction, a warning message with a traceback will be logged\nto help to debug the regression introduced.\n\nRelated-Bug: #1964575\n\nChange-Id: I3da37fee205b8d67d10673075b9130147d9eab5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e4086862de1cadcb1347b65543ddaab159c81d9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e4086862de1cadcb1347b65543ddaab159c81d9d"}]},"branch":"refs/heads/master"},"eeb918e1b9c27d3b5f0747cb856a4a8483f636ad":{"kind":"REWORK","_number":17,"created":"2022-04-08 09:10:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/833247/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/833247/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/833247/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/833247/17"}}},"commit":{"parents":[{"commit":"452a3093f62b314d0508bc92eee3e7912f12ecf1","subject":"Merge \"Local IP: use LOCAL_IP_TABLE for back flows if no OVS fw\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/452a3093f62b314d0508bc92eee3e7912f12ecf1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-05 14:34:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-04-08 09:09:54.000000000","tz":0},"subject":"Add the corresponding DB context to all SQL transactions","message":"Add the corresponding DB context to all SQL transactions\n\nThe goal of this patch is to make the Neutron code compliant\nwith SQLAlchemy 2.0.\n\nAll SQL transactions must be executed inside an explicit\nwriter/reader context. SQLAlchemy no longer will create an\nimplicit transaction if the session has no active transaction.\n\nA warning message, only available in debug mode, is added. When\nan ORM session calls \"do_orm_execute\", if there is no active\ntransaction, a warning message with a traceback will be logged\nto help to debug the regression introduced.\n\nRelated-Bug: #1964575\n\nChange-Id: I3da37fee205b8d67d10673075b9130147d9eab5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eeb918e1b9c27d3b5f0747cb856a4a8483f636ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eeb918e1b9c27d3b5f0747cb856a4a8483f636ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
