)]}'
{"id":"openstack%2Fneutron~833469","triplet_id":"openstack%2Fneutron~master~I912c3145383e9a18dfd46b8e4dbb0cf927ca76e4","project":"openstack/neutron","branch":"master","topic":"bug/1964640","attention_set":{},"removed_from_attention_set":{"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2022-03-14 21:34:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I912c3145383e9a18dfd46b8e4dbb0cf927ca76e4","subject":"ovn: Retry port binding on deadlock when migrating to ovn","status":"MERGED","created":"2022-03-11 21:33:34.000000000","updated":"2022-03-14 21:36:21.000000000","submitted":"2022-03-14 21:34:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"833469-bug/1964640","meta_rev_id":"44ee7c50af9baab1e2b223da91cd60d4952b29a4","_number":833469,"virtual_id_number":833469,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-03-11 23:12:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-14 21:34:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-03-12 12:06:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-14 13:24:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-14 13:24:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-11 22:24:25.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-11 22:28:10.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-11 22:43:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-11 22:48:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-11 23:15:03.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-03-12 12:06:52.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-03-14 13:24:14.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"925b88492dfbf4de04de6e3b761693e744c8a8b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-03-11 21:33:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0716fd84b7988975c82b8ef0f00457976a227df6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-03-11 21:33:54.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c8ed8fd6d6bff24026004fa9da842abcac128f80","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-03-11 22:24:25.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/69/833469/1/openstack-sriov-neutron-master-direct : ABORTED\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/69/833469/1/openstack-sriov-neutron-master-macvtap : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"26c50df5b42e17e825842ee2a3a5cf0c998b08f5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-11 22:28:10.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/69/833469/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"ca9a54865099d7734195b6133adef9babf7f2bf7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-03-11 22:32:42.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/69/833469/2/openstack-sriov-neutron-master-direct : SUCCESS\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/69/833469/2/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"141ca701d09e5ee1340a83a68526086a65e56243","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 22:43:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/21df9dd782f54bdf8da774225fa33961 : SUCCESS in 43m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c5daa64d9b84162baed45f5c1ceaa1b : SUCCESS in 50m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ace82635ab3dde0eeb3a24a4e20a089188688ad8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 22:48:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c06c80203d9746e0937ecd20878280d4 : SUCCESS in 44m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9ae76aabddea41dbbd529dc8cc8dc9e9 : SUCCESS in 1h 11m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e541b52355245618345b3bb900df2e8 : SUCCESS in 34m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c42930ecaa2f4f8eb78695e1fa55f038 : SUCCESS in 35m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7c8c945496b46779ce6c18a68f7bfd4 : SUCCESS in 14m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5479b0080ea84e30857bc2d64a44ae97 : SUCCESS in 22m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1aea54498da94ca8a218c5e574142c11 : SUCCESS in 22m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/374102bb5c5f43baa897237a92e1d360 : SUCCESS in 10m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1510088ea7cd41599bc4cabf9b5c128a : FAILURE in 39m 19s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55a1cad4b6b0444e916cd0e09208c327 : SUCCESS in 57m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/22e2f1ff7aa749e89630b09bc6df0c4c : SUCCESS in 1h 01m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1798e14885986fc2c9f7cd84485783ed1d24cf85","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-11 23:12:41.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/69/833469/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"f3eec6dab4c0b170ee63c422f6a51b90c9be5b67","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-11 23:15:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/833469/2 : SUCCESS in 1h 40m 15s","accounts_in_message":[],"_revision_number":2},{"id":"9041c5fb323700aee47986cc426e5f8911c0123b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-12 12:06:52.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9ef1fe70546264999e3c8d3ee74d58826d0a37ea","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-03-12 12:57:50.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/69/833469/2/openstack-sriov-neutron-master-direct : SUCCESS\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/69/833469/2/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"30c460db16ccf078b275f8e4e51695f248c7fb6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-12 13:32:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5365530b150c4393af1f1026629a9670 : TIMED_OUT in 1h 22m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c5518508c454a8e9395a4b686572ea4 : TIMED_OUT in 1h 22m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c6477a5a2f15e361ac6cc83b147c3a4d9ca4927f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-12 13:35:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/648840d4bed647218a1c7e87527af5e0 : SUCCESS in 56m 49s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/722db5282970401cb425145441da1ffa : SUCCESS in 1h 04m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b188d9e7e5b43d4b7618b9db115f62f : SUCCESS in 44m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80c6bce638824cd5acd5984f9f3a1893 : SUCCESS in 42m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43e55f0257bd4659bd727b02482ae785 : SUCCESS in 15m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c9503db33a14ace9829827bd8cd1edf : SUCCESS in 29m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e8f33e6f1c941bcbee0a3fa977b0107 : SUCCESS in 24m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/791f6c20fcec4b5c87a84d1461a9cb46 : RETRY_LIMIT in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cc5da457b7d4c74994f1f3ba886d8b4 : SUCCESS in 9m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f733c57160d643c9a1aee329626a57e5 : SUCCESS in 32m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/45f8ab2763344ab0be99fa8fca9cb68c : SUCCESS in 55m 47s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d005abee18b4a4aba5d4f3d8db67596 : SUCCESS in 1h 03m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/908a534b2e7d487282230ba2aa7e0fa1 : SUCCESS in 1h 23m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"34653d9375801d661a270dcd61a1a9f76bafc0c0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-12 14:28:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/833469/2 : SUCCESS in 2h 21m 17s","accounts_in_message":[],"_revision_number":2},{"id":"4e4428479984ce65f315ae2c83ad20def7b7e2cd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-14 13:24:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"23f0a25f27e135c1538a5d07f5653b60da1343f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 13:25:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a4d522360c31ef7efa34ec46d2f21b772ae7429e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 14:29:43.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e3875aa5f7c94453a1e1a27a4321ac2e : SUCCESS in 43m 11s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8fb95b4806d540cd91b6b8a3bede3668 : SUCCESS in 46m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c4961eae3204e5aa037ac03e451d9ee : SUCCESS in 30m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9f4124e0940443fb2542a44e0a952ce : SUCCESS in 14m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c60ec0434ac4998952e34bb2a1c6abf : SUCCESS in 24m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3865cc6effe94c7ba53b8ef24ed97f6d : SUCCESS in 22m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba69c151b0fc4589aa538f18ff422548 : SUCCESS in 11m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bdc2273162e40f7aba24a443b8c2b46 : SUCCESS in 29m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1fea75072da14f889dacd4e346cf149f : FAILURE in 47m 57s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f79f41d9fbdc4fc4a3c8b3e2af1b15b6 : SUCCESS in 1h 00m 23s","accounts_in_message":[],"_revision_number":2},{"id":"3c96fb7bf7e58ddf9791abfb41f1fd8823e366eb","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-03-14 19:04:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eba40885510ec4e542aa6ae226a757ae528d430b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 20:19:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1ad94db4afe148a4bff2d665c3622149 : SUCCESS in 40m 26s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/032108bcf8154618abdec8624ea9f5d1 : SUCCESS in 1h 13m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/898f3297a9894da5ac31b8560035a256 : SUCCESS in 33m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b24c988033d4eca91630fcfba4bb184 : SUCCESS in 29m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7860977c227f47c98a1245af8c06663f : SUCCESS in 12m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a76c065ac514b7c83ca2a631428efdb : SUCCESS in 25m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/596f76b5cdb84014b24a39ab52bb66fa : SUCCESS in 24m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df54b096b53a431788bd5b66d5a98c3e : RETRY_LIMIT in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c7d6ae2bf64471eaedf08c6c8ed5cc4 : SUCCESS in 12m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47c8835b42bf4460a8c2e0549b4bcac8 : SUCCESS in 28m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e542e203352646eb8ab413fe7db5c256 : SUCCESS in 53m 10s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/971317665c524b37a7ee4632bb12ff19 : SUCCESS in 50m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/66898ba97f944ae2b3e372686834ec14 : SUCCESS in 59m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2b6642fdb2b2a01da3a4e4e3a83b976e920ea58b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 20:20:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9ba8d070ad358984df7fb0d189cff70c6cafe7eb","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-14 20:58:07.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/833469/2 : SUCCESS in 1h 53m 20s","accounts_in_message":[],"_revision_number":2},{"id":"3c9335f5d8e37378e140692eddaa91f0c2932965","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 21:06:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b092f6e433f243058e2f5eba73511f33 : FAILURE in 1h 21m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ede4be088294404ea80921eeb3101aac : FAILURE in 1h 17m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"94a7bd4869c73011e1c73c3ca5ac45683623b580","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-03-14 21:11:08.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/69/833469/2/openstack-sriov-neutron-master-direct : SUCCESS\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/69/833469/2/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"ce28ecfb8cca38cc7b24b9b4c0e260bd49b53c58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 21:34:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a8009dee059748f29aea78fde564eed1 : SUCCESS in 1h 07m 32s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a3753f5e453d4da888a913e28c4482c6 : SUCCESS in 1h 02m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/348ce653da9f4699ab328c3d7906e0e1 : SUCCESS in 35m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0a30d49e744414193e9ab35bfd30cd1 : SUCCESS in 15m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c214640551514f758b4678d9bcbb760b : SUCCESS in 25m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61bf934d6b714cb0acb6461051635708 : SUCCESS in 28m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99d8fb0a4ae74379827fed1ee23f30fc : SUCCESS in 10m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32ad7f07e4e64b46ae1a9f40925be688 : SUCCESS in 29m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4825cd85342b460bb54084a23d38ab0d : SUCCESS in 52m 55s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f79a1a80d88e4c3a84f425860d193023 : SUCCESS in 1h 05m 27s","accounts_in_message":[],"_revision_number":2},{"id":"f0d7d3ad5f8867205702bf3f7aa2e8206a87e2ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 21:34:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"44ee7c50af9baab1e2b223da91cd60d4952b29a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 21:36:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2911f5377b58443bb41ec7a35480df0c : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0ec523aac11c31d9d8d2f0e7169332b876a61391","revisions":{"52975681dcb99f5bbc69c26ea30e3f5fd2fa496b":{"kind":"REWORK","_number":1,"created":"2022-03-11 21:33:34.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/69/833469/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/69/833469/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/833469/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/833469/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/833469/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/69/833469/1"}}},"commit":{"parents":[{"commit":"99345b714f04804f1887584508adcb7c795ecccf","subject":"Merge \"Ensure gateway is set for prefix delegated subnets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99345b714f04804f1887584508adcb7c795ecccf"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2022-03-11 21:32:18.000000000","tz":-300},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2022-03-11 21:33:27.000000000","tz":-300},"subject":"ovn: Re-try port binding on deadlock when migrating to ovn","message":"ovn: Re-try port binding on deadlock when migrating to ovn\n\nIf the port binding is being changed while Neutron keeps changing\nthe port binding because ovn controller starts claiming the port\nbindings then the db sync script may fail.\n\nThe patch adds DBDeadLock to the re-try mechanism when running db sync\nscript in migration mode.\n\nChange-Id: I912c3145383e9a18dfd46b8e4dbb0cf927ca76e4\nCloses-bug: #1964640\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/52975681dcb99f5bbc69c26ea30e3f5fd2fa496b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/52975681dcb99f5bbc69c26ea30e3f5fd2fa496b"}]},"branch":"refs/heads/master"},"0ec523aac11c31d9d8d2f0e7169332b876a61391":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-03-11 21:33:54.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/69/833469/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/69/833469/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/833469/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/833469/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/833469/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/69/833469/2"}}},"commit":{"parents":[{"commit":"99345b714f04804f1887584508adcb7c795ecccf","subject":"Merge \"Ensure gateway is set for prefix delegated subnets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99345b714f04804f1887584508adcb7c795ecccf"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2022-03-11 21:32:18.000000000","tz":-300},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2022-03-11 21:33:46.000000000","tz":-300},"subject":"ovn: Retry port binding on deadlock when migrating to ovn","message":"ovn: Retry port binding on deadlock when migrating to ovn\n\nIf the port binding is being changed while Neutron keeps changing\nthe port binding because ovn controller starts claiming the port\nbindings then the db sync script may fail.\n\nThe patch adds DBDeadLock to the re-try mechanism when running db sync\nscript in migration mode.\n\nChange-Id: I912c3145383e9a18dfd46b8e4dbb0cf927ca76e4\nCloses-bug: #1964640\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ec523aac11c31d9d8d2f0e7169332b876a61391"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ec523aac11c31d9d8d2f0e7169332b876a61391"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
