)]}'
{"id":"openstack%2Fneutron~833730","triplet_id":"openstack%2Fneutron~stable%2Fyoga~Ifc2e5b2cc64c0b3beafd0a2e6eb9b9f248970cc5","project":"openstack/neutron","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2022-03-16 10:19:06.000000000","reason":"Change was submitted"},"23567":{"account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"last_update":"2022-03-16 10:19:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifc2e5b2cc64c0b3beafd0a2e6eb9b9f248970cc5","subject":"Ensure no GARPs are sent for Load Balancer VIPs on tenant networks","status":"MERGED","created":"2022-03-15 08:05:50.000000000","updated":"2022-03-16 10:20:59.000000000","submitted":"2022-03-16 10:19:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"833730","cherry_pick_of_change":832594,"cherry_pick_of_patch_set":4,"meta_rev_id":"791af04590ab88bde43d5af3c37021b2e98c78f9","_number":833730,"virtual_id_number":833730,"owner":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-16 10:19:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2022-03-16 05:15:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2022-03-16 08:46:42.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2022-03-15 17:01:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2022-03-16 08:46:42.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-15 08:41:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-15 15:23:47.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-03-16 05:15:41.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-03-16 08:46:42.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"}],"messages":[{"id":"6656e92151122f0872e323899a87abe27ba9f99f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2022-03-15 08:05:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"467fd5090c55d696c4abc52a738acad70dfed086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 08:41:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6edd1f5245044df9911efb22b3625c36 : SUCCESS in 34m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb262b910a414a4c834f130a8c41a45b : SUCCESS in 26m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fed56fa8baf4991ad82326b0b9bdcae : SUCCESS in 14m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8b4f11e41f9449aa505197d5f4a16d7 : SUCCESS in 27m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4ac27b81cf34b29a1a94970dbab219d : SUCCESS in 28m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aeca2d2adaf4cec92ed632c1d05cdd6 : SUCCESS in 12m 22s","accounts_in_message":[],"_revision_number":1},{"id":"861d1e66c65042c272435fe931c0d6ea72b4759c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 08:59:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/315054770e7645f5a5ce71873f6b81de : SUCCESS in 45m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2465a59e43b54c618864c13e9b475069 : SUCCESS in 50m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85ee3630a9a11ef63657b77cccad73040865d980","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-03-15 15:23:47.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c487dfb9643093964297a3bc00408e2f957ed76a","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2022-03-15 15:46:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4c0372bbf848359971eb49e5e21d89b98a232b52","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-03-15 17:01:32.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"71d631fd476fd2ca5da2dd9b7b721d54f2a549e2","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-03-16 05:15:41.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5946d43615a4cda6d6af42d535294517f61e16e8","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-03-16 08:46:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"22895b3131a9da283c9267879d838bae7c98a929","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 08:47:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fa6a33118bd85bf5f14db1363ab5aeb9a2c6e238","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 10:19:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a36b0d9fc9f47ff8e573cb8822e5b19 : SUCCESS in 47m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e9049f863264ffe8442574d58b7d646 : SUCCESS in 17m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e869d73f154141aebb0b960b68c6c3c0 : SUCCESS in 31m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/622214084beb442fb688a5d83ac3f757 : SUCCESS in 30m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa9d1760ba064192a4b74a0ae7fe65a1 : SUCCESS in 11m 21s","accounts_in_message":[],"_revision_number":1},{"id":"c7488a833da9ce8a8778002f26d6bfc8fe6663ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 10:19:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"791af04590ab88bde43d5af3c37021b2e98c78f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 10:20:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b1a15cd80f542a2874b473930b770b6 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e61b117b3319188089e0e38727d6eb49c2ad5fec","revisions":{"e61b117b3319188089e0e38727d6eb49c2ad5fec":{"kind":"REWORK","_number":1,"created":"2022-03-15 08:05:50.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/30/833730/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/833730/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/833730/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/833730/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/833730/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/833730/1"}}},"commit":{"parents":[{"commit":"6799e85e5df52af2dc61448d9773f0be95501369","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6799e85e5df52af2dc61448d9773f0be95501369"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2022-03-08 12:04:58.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2022-03-15 08:05:50.000000000","tz":0},"subject":"Ensure no GARPs are sent for Load Balancer VIPs on tenant networks","message":"Ensure no GARPs are sent for Load Balancer VIPs on tenant networks\n\nWhen a loadbalancer is created in tenant network (VIP and members),\nand that tenant network is connected to a router, which is connected\nto the provider network, the ovn loadbalancer gets\nassociated to the ovn logical router. If the \"router\" option is used for\n\"nat-addresses\" (as currently done), ovn-controller will send GARPs for\nthat VIP.\n\nAs there is nothing blocking different tenants in openstack to create a\nsubnet with the same CIDR and then a loadbalancer with the same VIP,\nthere may be several ovn-controllers generating GARPs on the provider\nnetwork for the same IP, each one with the MAC of the logical router\nport belonging to each tenant.\nThis could be a problem for the physical network infrastructure.\n\nThis patch fixes it by setting a new option added in OVN[1], named\n\"exclude-lb-vips-from-garp\", on the router gateway port, ensuring\nno GARPs are sent for the load balancer VIPs.\n\n[1] https://patchwork.ozlabs.org/project/ovn/patch/183edfc446633c5c38d7d7361089d34432c527dd.1645793899.git.lorenzo.bianconi@redhat.com/\n\nCloses-Bug: #1964901\n\nChange-Id: Ifc2e5b2cc64c0b3beafd0a2e6eb9b9f248970cc5\n(cherry picked from commit 9d4e6edd5b4c92c88fbe8d06f6a176fae1b35f25)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e61b117b3319188089e0e38727d6eb49c2ad5fec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e61b117b3319188089e0e38727d6eb49c2ad5fec"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
