)]}'
{"id":"openstack%2Fneutron~834536","triplet_id":"openstack%2Fneutron~master~I5b96b3a91aacffe118ddbb91a75c4892818ba97a","project":"openstack/neutron","branch":"master","topic":"bug/1965807","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-03-22 12:46:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5b96b3a91aacffe118ddbb91a75c4892818ba97a","subject":"Remove exception ``IpAddressAllocationNotFound``","status":"MERGED","created":"2022-03-21 18:49:23.000000000","updated":"2022-03-22 12:48:06.000000000","submitted":"2022-03-22 12:46:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834536-bug/1965807","meta_rev_id":"71b58493c8ab3a1cc662881ed2c0cfdc545a7574","_number":834536,"virtual_id_number":834536,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-03-21 20:27:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-22 12:46:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-22 11:07:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-03-22 10:00:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2022-03-22 05:57:30.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-22 11:07:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-21 19:31:07.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-21 19:36:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-21 20:47:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-21 21:14:34.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-03-22 05:57:30.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-03-22 10:00:39.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-03-22 11:07:43.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"7b3a7e370c29c4b6f607067615ecd83ce13c7932","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-21 18:49:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c20c8e3018ce86709ae2cf2f343a6ae0003666d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-21 18:50:42.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4acb45e4d5fde8d4033be1e1639b7e4c1839cfb2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-21 19:31:07.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/36/834536/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"d00a199b420dca59343a3f0e969b773957bfee34","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 19:36:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a403fa461fd84649a733db7a269de698 : SUCCESS in 43m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/81dfb7b5cdd84d0497cf8ef574964ae7 : SUCCESS in 42m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4bb6d28a0cdc55c76e9aa15c83ba5ed8cd4f0ce0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-21 20:27:23.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/36/834536/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"2e6eca4baf542ea92297e343120feca78070139a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 20:47:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9437df3678e5475b892413a3bd61bbea : SUCCESS in 56m 05s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/31d2b02c3e2b40c7babaca7d09344589 : SUCCESS in 1h 17m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/67c812b74ad54030a314f14226513dcc : SUCCESS in 1h 24m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8fa77072b2744275a0c4e2f09d7e43dd : SUCCESS in 1h 06m 34s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d98ec48a3544448ca7d1c71bbddf17cf : SUCCESS in 1h 15m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6fcff4587e7046439f3e6f10474559de : SUCCESS in 37m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6afe8a32cc414fce919b443a6c41f210 : SUCCESS in 48m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5d15e469186413ab70e9541d1e720c5 : SUCCESS in 41m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dedc84fde57f44ec858b9b029b8e09d0 : SUCCESS in 16m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/540e36b8ebf746489371ea58c4d88b68 : SUCCESS in 30m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31b9f34fa4c945e98db2b076d3fa40d4 : SUCCESS in 32m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a73a3d8d82249839cff4172b06c247f : SUCCESS in 33m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e23d253b2f134b338472c6282d4217c9 : SUCCESS in 11m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cdccd1718c4b4c24b41c1d1ba48e0d9e : SUCCESS in 52m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/82cbee789e1e4fbebc341636fe78fa48 : SUCCESS in 1h 30m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/27beda2208d14861a7960145ed56747b : SUCCESS in 1h 09m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d5f70b0e7c6e4e0ebaafe78f3a413b32 : SUCCESS in 59m 34s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1bd85d10e1ab43dc97cf386b76d44a6a : SUCCESS in 1h 03m 02s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/74e928bf050c47988f6af3e8a1526808 : SUCCESS in 1h 26m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6eff96c3d6e7402a8cec84bc594ede76 : SUCCESS in 1h 49m 28s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3923ec9f47e74ef794d07f3bd07f509f : SUCCESS in 1h 00m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8172a3bc0b054cbc92ce25809e692582 : SUCCESS in 1h 05m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"361d5fd0e882c7614635fd8df219c57c33f762c3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-21 21:14:34.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/834536/2 : FAILURE in 2h 23m 10s","accounts_in_message":[],"_revision_number":2},{"id":"e367842e1afa3f0f3f49f7423d6173211fc55620","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-03-22 05:57:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e414198f768740e3aa1dcc0c4ad4f5f10546d777","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-03-22 10:00:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2239007aa32f436625f491a37f422fc1b0c0fc3f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-22 11:07:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"568df4f78d13d630ca07a52b0ac726690e36f59a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 11:09:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a57d466a5380267d186f54648e12e119ebd8f783","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 12:46:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/06428ece490b4c58875258f801983c21 : SUCCESS in 52m 04s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/51c0466269114a0faeeca466779c89fe : SUCCESS in 1h 13m 22s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/96a1d59cb6f540f89f074f301e3ac063 : SUCCESS in 58m 21s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/71a6fc6aa775417ca43f4c82575156d2 : SUCCESS in 1h 06m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d6b26004fa3449149ba2570e9c30c794 : SUCCESS in 48m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fae014580a9c46aaa8eba99c84ac8bca : SUCCESS in 38m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ae23fae623746838c768530f5da65d4 : SUCCESS in 15m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/568cfac9666748c0a0a796ad4d1e6487 : SUCCESS in 25m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d68d6d3c58f443f992c7491f06a52db3 : SUCCESS in 28m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0004ee77d504ef184324388e989a1c1 : SUCCESS in 11m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5ef866b9da06441bb39a8e53eb049ab5 : SUCCESS in 51m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/34c0d22c73284256849ee9ee1efe35b5 : SUCCESS in 1h 31m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8e42d8d7b0cc434da9b1875ae251e8dd : SUCCESS in 1h 20m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/780479f562e44a91be2e20b88e13f9d8 : SUCCESS in 55m 30s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1b6de445676140698a50413b45e22478 : SUCCESS in 56m 46s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dca482300ec64ae48778346e84421989 : SUCCESS in 50m 53s","accounts_in_message":[],"_revision_number":2},{"id":"19593f9034452b51f1611fe9a8b74e12d04ceea5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 12:46:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"71b58493c8ab3a1cc662881ed2c0cfdc545a7574","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 12:48:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e27972d15f3452a9aec0fa1a6e6e6a4 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"83b6ce9e9ec2f0261fed630e18cbe9cfa363a0dd","revisions":{"4170e2ddc6ca3ab28d511124430d767f045bec98":{"kind":"REWORK","_number":1,"created":"2022-03-21 18:49:23.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/36/834536/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/834536/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/834536/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/834536/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/834536/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/834536/1"}}},"commit":{"parents":[{"commit":"28a941ce8c242f64a8cbc4fc1c8e17fffff2fa10","subject":"Merge \"[OVN] Remove maintenance \"check_metadata_ports\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/28a941ce8c242f64a8cbc4fc1c8e17fffff2fa10"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-16 16:32:31.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-16 16:46:36.000000000","tz":0},"subject":"Remove exception ``IpAddressAllocationNotFound``","message":"Remove exception ``IpAddressAllocationNotFound``\n\nThis patch removes the ``IpAddressAllocationNotFound`` exception. This\nexception was raised when a IPAM register was called to be deleted\nbut not found.\n\nAs reported in the LP bug, this IPAM register deletion can be called\nseveral times if a port fails during the creation. The IPAM register\ndeletion calls the DB deletion but doesn\u0027t raise any exception if the\nregister does not exist. The code ensures the IPAM register is\ndeleted and there is no need to fail if it is not present anymore.\n\nThis patch also removes the exception catch and try in \"update_port\",\nthat was added in [0] as a fix for [1]. That was added because the\nsubnet deletion code involved a port update call [2] during the\nIP allocation deletion, if any port was still present in the subnet.\nSince [3], this code is not needed because the subnet deletion does\ncall a port update anymore.\n\n[0]https://review.opendev.org/c/openstack/neutron/+/373536\n[1]https://bugs.launchpad.net/neutron/+bug/1622616\n[2]https://github.com/openstack/neutron/blob/pike-em/neutron/db/db_base_plugin_v2.py#L1017-L1018\n[3]https://review.opendev.org/c/openstack/neutron/+/713045\n\nCloses-Bug: #1965807\nRelated-Bug: #1954763\nRelated-Bug: #1622616\n\nChange-Id: I5b96b3a91aacffe118ddbb91a75c4892818ba97a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4170e2ddc6ca3ab28d511124430d767f045bec98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4170e2ddc6ca3ab28d511124430d767f045bec98"}]},"branch":"refs/heads/master"},"83b6ce9e9ec2f0261fed630e18cbe9cfa363a0dd":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-03-21 18:50:42.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/36/834536/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/36/834536/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/834536/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/834536/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/36/834536/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/36/834536/2"}}},"commit":{"parents":[{"commit":"28a941ce8c242f64a8cbc4fc1c8e17fffff2fa10","subject":"Merge \"[OVN] Remove maintenance \"check_metadata_ports\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/28a941ce8c242f64a8cbc4fc1c8e17fffff2fa10"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-16 16:32:31.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-16 16:48:06.000000000","tz":0},"subject":"Remove exception ``IpAddressAllocationNotFound``","message":"Remove exception ``IpAddressAllocationNotFound``\n\nThis patch removes the ``IpAddressAllocationNotFound`` exception. This\nexception was raised when a IPAM register was called to be deleted\nbut not found.\n\nAs reported in the LP bug, this IPAM register deletion can be called\nseveral times if a port fails during the creation. The IPAM register\ndeletion calls the DB deletion but doesn\u0027t raise any exception if the\nregister does not exist. The code ensures the IPAM register is\ndeleted and there is no need to fail if it is not present anymore.\n\nThis patch also removes the exception catch and try in \"update_port\",\nthat was added in [0] as a fix for [1]. That was added because the\nsubnet deletion code involved a port update call [2] during the\nIP allocation deletion, if any port was still present in the subnet.\nSince [3], this code is not needed because the subnet deletion does\nnot call a port update anymore.\n\n[0]https://review.opendev.org/c/openstack/neutron/+/373536\n[1]https://bugs.launchpad.net/neutron/+bug/1622616\n[2]https://github.com/openstack/neutron/blob/pike-em/neutron/db/db_base_plugin_v2.py#L1017-L1018\n[3]https://review.opendev.org/c/openstack/neutron/+/713045\n\nCloses-Bug: #1965807\nRelated-Bug: #1954763\nRelated-Bug: #1622616\n\nChange-Id: I5b96b3a91aacffe118ddbb91a75c4892818ba97a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/83b6ce9e9ec2f0261fed630e18cbe9cfa363a0dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/83b6ce9e9ec2f0261fed630e18cbe9cfa363a0dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
